)]}'
{"id":"openstack%2Ftempest~930092","triplet_id":"openstack%2Ftempest~master~I20c205c8ac8679f7f202ef854a406973f8f22136","project":"openstack/tempest","branch":"master","topic":"qa-2024-2-release","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-09-24 07:16:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I20c205c8ac8679f7f202ef854a406973f8f22136","subject":"Switch to the master constraint in tox.ini","status":"MERGED","created":"2024-09-20 21:49:06.000000000","updated":"2024-09-24 07:18:53.000000000","submitted":"2024-09-24 07:16:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930092-qa-2024-2-release","meta_rev_id":"4185ca4922beb3d07198df44cb4bfb5ce0553432","_number":930092,"virtual_id_number":930092,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-24 07:16:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2024-09-23 20:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2024-09-23 20:11:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-21 00:22:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-23 12:07:43.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"3440242040daa78b706c901e622f77ca6d0aae22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-20 21:49:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90a3dca9ba50bfb98bdada5933de2adf9ae229ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 00:22:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f0f01e066c445e4b9ae3b7b72ceee6a\n\n- grenade https://zuul.opendev.org/t/openstack/build/774ba1400afb4920982f0e4d4527f7c6 : SUCCESS in 40m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b8e90d16a5f46e18379ca9773ff6955 : SUCCESS in 1h 19m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7f0e12ded7f4c80807270beae194ab3 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409fccfb6441459892409907feb06213 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98d992485bfe490992ab0c8fb6fad365 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b6ab1414554f8f81831db86efcac03 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fd2127eb57a42a0b7b7331390920428 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86920d61e07448409f7945e74b86fa1b : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5815c8f68d2d40df8e1432313d66e25b : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e00e87e57f1f4dedb59b8d3d14692a9d : SUCCESS in 3m 48s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5c4cbe9b39934ae5bf71faaeec73487a : SUCCESS in 1h 00m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/88fb4b0887894c85b1aae074d0b648c3 : SUCCESS in 2h 25m 42s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/beb277b168da4facae2d44298ea8e047 : SUCCESS in 1h 19m 14s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/c9a5907eec1c46d580eaad5362f32da7 : SUCCESS in 1h 08m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fad091d6148d4373b05dcf267c6201a5 : SUCCESS in 51m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5688e83390c14710a903ab00c34a2b64 : SUCCESS in 14m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45c8765407b8456bb109bd1c7479fe4f : SUCCESS in 38m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8fc4dcb508a045388e988cffe9c6c635 : SUCCESS in 1h 42m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bf22d79d8b834f4aa477905625199b16 : SUCCESS in 52m 00s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1346de10baf446a587daf58c5b0d8dca : SUCCESS in 1h 21m 05s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/58b702c996294037aa4a1fce3ad0fc9a : SUCCESS in 1h 15m 46s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/174b597a4482419289f5c19b8d6d9b64 : SUCCESS in 41m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7cf748851f80410a8422bda97301f3ed : SUCCESS in 51m 44s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/80f3533ac520432989e9458fdd567f7c : SUCCESS in 1h 12m 13s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d4b85b707c0647abbf86255fee8c6d3f : SUCCESS in 4m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/add7e30ae72343b4b551a70f17c6a56b : FAILURE in 43m 34s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/98ba0c931f0d4697b781e2f50db77342 : SUCCESS in 47m 41s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/77b7191ff2ed4fcb8a890187a98b721e : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4ded8c02e5fa016a9ebd8b967603dd8474fb69db","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-09-23 12:07:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"770727664f81d9df80be2f24723f55f96677e9b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-23 20:00:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f987317805ecb05833ad2adf07dba5a260113830","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-09-23 20:11:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"31a9c8055b7ed34d19b8ba6c769bc097c5b04498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 22:22:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd974c38941240a7a4a5c0229d3fe4ae\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5de497cc1104f5281f285ea8142abd8 : SUCCESS in 53m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76b3a9d9e76141f795568a888c8e0812 : SUCCESS in 1h 33m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c57579f2140a40a6983072b75e76a7c9 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f2315cfc2c454ba765e21e68224a60 : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f25f860d41a45aabf6c8bde7e6b6c0b : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85c34d6c1acd4c3294dcf242cdfee0f4 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90dd0f931a4241c1b7fdb142ed7d237b : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39feddcb232549719dd2df015bba9034 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b42dc9cf388469f9847930c96f2ed4c : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/121d79b5eb584a06b128c6e3dd98b5c9 : SUCCESS in 4m 27s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/a582a6f3a25840c5abf7159da0c07150 : SUCCESS in 37m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/52d818922b2744a4be7fd963f45fb4b5 : SUCCESS in 2h 15m 03s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/1fc884a41c6d49f9b9d6f601314f3159 : SUCCESS in 1h 13m 09s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/27fcda8bc7124f108a35803fb4de7d6d : SUCCESS in 1h 11m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4e0580726c9a425498a35b56bac8c518 : SUCCESS in 58m 22s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/fa82ba0f8185487c9234519f1f0e76ee : SUCCESS in 26m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0da601cf8214400f968a3c4610fb5968 : SUCCESS in 44m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8fe3285ded6f46728ca7b6675448f875 : SUCCESS in 59m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/605e3304ed6041fb813cf005f60cd246 : SUCCESS in 45m 54s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/76bfe56eca3b4e579030fedf2efa3fb3 : SUCCESS in 1h 53m 56s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/473f0d672e2b44d4a16fdaac033e9eb2 : SUCCESS in 2h 01m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2faee30848154ab6b110fd6c8f512ab7 : SUCCESS in 1h 11m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd621bcb4c524befbc21f2dba7f46c2d : SUCCESS in 1h 13m 38s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/3653baa3827947e6b1de14416c58dcee : SUCCESS in 1h 17m 56s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/cc5ff27bc1e54139a712954955732e96 : SUCCESS in 4m 15s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/49da284d7b9240e7a590e152db359475 : FAILURE in 54m 37s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c4f1d2991eb440ffa1bf6ae0dd2595e5 : SUCCESS in 39m 51s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d2f5b9ba0e0741a88fa86c65c8a20c06 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"344448c376ff04ae3381f735a3c6fb8c27105bb9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-24 00:25:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99e7390150667aaee64bd6d29161016b53e6a64f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 02:40:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd07395b9cf44cabb16eb4fe5bcb4d67\n\n- grenade https://zuul.opendev.org/t/openstack/build/04d082ca87a44b1aa26d78d51177b300 : SUCCESS in 1h 14m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50ce8eb484b04c73ab708972ec4b5f63 : SUCCESS in 1h 28m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b9c5eac3b4b4dd09b87f91c6dd9a0a3 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d195b14d1a4ad69cd83124f010c3ab : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0740a882c80421888fa332b0d660752 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc3a11dc668408c839c423cc65de852 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c0c264cf1e64e86b2073b9fc0165931 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30183541e7f745ed86810066080d03d7 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eda32960dd154097971ed8da987233aa : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1489db1865a442e88341e8ddf20152d2 : SUCCESS in 3m 39s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/d5f540e9c3014f0eb2ac349b86f490b9 : SUCCESS in 33m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/68f79911d55b4c95be7caf6644847d59 : SUCCESS in 2h 12m 02s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/610251d2692f4ae893d4c1a449f71bac : SUCCESS in 1h 37m 13s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/2eefe7ae55294113b03b467c2e6375dc : SUCCESS in 1h 26m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6f7f2b259a4946aaa9551da7285bf254 : SUCCESS in 1h 01m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/396f688f81194a579a1b2f45dab0a57a : SUCCESS in 26m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b81c9fa3c24a44089866d3ceaa29702f : SUCCESS in 58m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a50ce7bb90245758332a63b14d80c6d : SUCCESS in 1h 36m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d6ef912c91e1437582c9046df713589a : SUCCESS in 49m 30s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7a64c37e50c2417ca1c35fc43eee3ad4 : SUCCESS in 1h 46m 35s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/be3b2c2975a346469325e24ed3372dfe : SUCCESS in 1h 45m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc4ac384bcc645648a3138a50d8609e6 : SUCCESS in 52m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/17affc3253e24a308427010b382d9734 : SUCCESS in 51m 31s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/67751b27c02147329df936d5dcfcbc79 : SUCCESS in 1h 13m 43s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/663e803646a94f28a68275792acff4b6 : SUCCESS in 4m 16s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8e0b6df327e84450bb657099eddf3f04 : FAILURE in 1h 11m 33s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/7e159bc38d894c738fce73379b05cab9 : SUCCESS in 50m 58s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1a175b43cce745b69aa2d607a20a9b07 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"27788e8ab89dc33a60d69dc611a8f2ab83bda047","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-24 05:03:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13254824b3de0332c8e92a296addc7ba833ab890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 05:19:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d8738248ad4f53bdd48254de6dc50e8de9934677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 07:04:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cdfca76b1f34b03863ca06228d1daec\n\n- grenade https://zuul.opendev.org/t/openstack/build/7d59eb857e4344f7898ba30194525bff : SUCCESS in 51m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2b99b9be1b947618187ac0ab24e6ad3 : SUCCESS in 1h 26m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72347affec834ce49116ae4c227b3718 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31cc0258482f4b0c9f31c76ac681e43f : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c700f3814b94871b5abfd0edbbea77b : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d8aede28f04d75bf00ce23298788a3 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e83beaebde1454c96b495626b6fb3e8 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d84f67db6d36420dabe8a8db9471a709 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/577977af27dc4fe4891bd6afb13fc523 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31f483eae2f747ce852e8f439a644f6f : SUCCESS in 3m 19s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/a3a1aee52ad84005be6a9c87fb76a796 : SUCCESS in 41m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7a7bad1d705447499d2768212a44baec : FAILURE in 1h 23m 00s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/2727937df50741208e1aefcfa4806e30 : SUCCESS in 1h 31m 20s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/79d61755d5084771911b115da0131472 : SUCCESS in 1h 18m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/130815d000eb4711975faae6630178cf : SUCCESS in 1h 17m 30s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/63fb6400db894ff1a1e4205d742bf52a : SUCCESS in 26m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/580fe8ffffab4e1998b64736f28256e2 : SUCCESS in 39m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cf14f7636e86457aba975bb205744e2f : SUCCESS in 1h 52m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/55467af0b3644f99ba96252a872be06c : POST_FAILURE in 24m 20s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c155dd85798c42a69211d5dd06f44854 : SUCCESS in 1h 42m 53s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c4c714205a624e6789e6b4f309ed2802 : SUCCESS in 1h 18m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/79687e18c81642cebaf94e43c1a9ed1b : SUCCESS in 46m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad3d4a4d043b443b84656ab859385c87 : SUCCESS in 1h 07m 29s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/9c2e89e9e61d4c1c90367ab689e035a6 : SUCCESS in 1h 13m 38s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/31747dc72a0e4d07a141cb556e07c1e6 : SUCCESS in 3m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/098295fe60214c03adbb31531c714384 : FAILURE in 1h 24m 39s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/86dd05a55293485ca6611f5a39fdaf18 : SUCCESS in 36m 58s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c1baca056303454e9aeb4bf326bbb240 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":2},{"id":"c58f58f17f7edcd1b5584f6760a13903e56d55a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 07:16:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d6646df21344c42b6dd4fc888514b73\n\n- grenade https://zuul.opendev.org/t/openstack/build/94e5c8b0db5e40c5ab35eccab9112840 : SUCCESS in 39m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88cb9a0aa35c4424a90e5af9a742366f : SUCCESS in 1h 52m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342e0d4a09834dd7910bcb78c1a334f7 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0653342776f94b44b89dcb248be75ddd : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbc9601c9f9544278ee975bbdbf70230 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99a79989a24644ca8a54ddc5d95897f6 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b778043263348c18fe7b6a72471db58 : SUCCESS in 7m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b678c03f977420b9501b84c527da730 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6e9f47d0221443fbdb9c89894e857fd : SUCCESS in 4m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/16e5fb52af794a809f7d26900ae5a686 : SUCCESS in 1h 30m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2162421d1604e61bf4b1b85a3566281 : SUCCESS in 59m 27s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/444240335b8a4a10b0ed790e23fcfd59 : SUCCESS in 55m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ed381b5935149e58b02602be3aae294 : SUCCESS in 55m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/960769b6a32b411188adcbc778f2299e : SUCCESS in 50m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dc8a10bb0ba84866a92ae9b902ff5135 : SUCCESS in 1h 02m 10s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6f71dc36535343b3a0fdad12d585dbd8 : SUCCESS in 1h 40m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/26def4fa66a94b60ba2950b97db462f8 : SUCCESS in 36m 34s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4042745621204aba97d62055cc9d49c8 : SUCCESS in 48m 42s","accounts_in_message":[],"_revision_number":2},{"id":"978a834352f834c12cffa1bdb71715bae139fb09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 07:16:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4185ca4922beb3d07198df44cb4bfb5ce0553432","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 07:18:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0f88e51a77843bfa1900052d6ab3f06\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6b08db1a4c4dffbb5095ca31a0434a : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa792b2e019a4fc29d82a97d3cde7717 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6c389b211823ecca179d456aa48598773a15a5c3","revisions":{"0f5f60c59a704a6bb1d8c7f1e0d1dc945046890c":{"kind":"REWORK","_number":1,"created":"2024-09-20 21:49:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/92/930092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/92/930092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/92/930092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/92/930092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/92/930092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/92/930092/1"}}},"commit":{"parents":[{"commit":"a008e1da6fb41fd1c89af25c66ab2c5b16b8fbb4","subject":"Use stable constraint in tox to release new tag for 2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a008e1da6fb41fd1c89af25c66ab2c5b16b8fbb4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-20 21:42:39.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-20 21:42:39.000000000","tz":-420},"subject":"Switch to the master constraint in tox.ini","message":"Switch to the master constraint in tox.ini\n\nTo releas a new tag for Tempest 2024.2, we modified the\ncompatible constraint in tox.ini\n- https://review.opendev.org/c/openstack/tempest/+/930091\n\nAs new tag is released now so we can switch back to the master\nconstraint in tox.ini.\n\nChange-Id: I20c205c8ac8679f7f202ef854a406973f8f22136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0f5f60c59a704a6bb1d8c7f1e0d1dc945046890c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0f5f60c59a704a6bb1d8c7f1e0d1dc945046890c"}]},"branch":"refs/heads/master"},"6c389b211823ecca179d456aa48598773a15a5c3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-23 20:00:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/92/930092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/92/930092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/92/930092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/92/930092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/92/930092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/92/930092/2"}}},"commit":{"parents":[{"commit":"7e247a7b09a523368e289956b2f75a061573a35e","subject":"Use stable constraint in tox to release new tag for 2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7e247a7b09a523368e289956b2f75a061573a35e"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-09-20 21:42:39.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-09-23 20:00:54.000000000","tz":0},"subject":"Switch to the master constraint in tox.ini","message":"Switch to the master constraint in tox.ini\n\nTo releas a new tag for Tempest 2024.2, we modified the\ncompatible constraint in tox.ini\n- https://review.opendev.org/c/openstack/tempest/+/930091\n\nAs new tag is released now so we can switch back to the master\nconstraint in tox.ini.\n\nChange-Id: I20c205c8ac8679f7f202ef854a406973f8f22136\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6c389b211823ecca179d456aa48598773a15a5c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6c389b211823ecca179d456aa48598773a15a5c3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
