)]}'
{"id":"openstack%2Ftempest~931358","triplet_id":"openstack%2Ftempest~master~I7b1e32b7ac3622029336f1919fb5cc2731050a2d","project":"openstack/tempest","branch":"master","topic":"compute-response-schema-servers-2.96","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-10-04 19:01:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7b1e32b7ac3622029336f1919fb5cc2731050a2d","subject":"Add compute response schema for microversion 2.96","status":"MERGED","created":"2024-10-03 23:41:49.000000000","updated":"2024-10-04 19:02:46.000000000","submitted":"2024-10-04 19:01:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931358-compute-response-schema-servers-2.96","meta_rev_id":"4ff72a2ebc399b65093cd8ae30ed273ae6c1726c","_number":931358,"virtual_id_number":931358,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-04 19:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2024-10-04 17:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2024-10-04 17:18:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-04 01:51:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 17:18:55.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"d34295ec85f972cbb471c095dc42342b73e17dd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-10-03 23:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"388536ba1cf27927f6acc6a2cc6860015fd9a24d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 01:51:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607060bbad38461c9b09862dd5509a60\n\n- grenade https://zuul.opendev.org/t/openstack/build/11a5760bc14a4b4ba05b2ccc2d752c47 : SUCCESS in 43m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f36632b658e4c96ac55932c94fe1afa : SUCCESS in 1h 12m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bb199616c514cdc96aabe64ebf32280 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0abbde9d41d4c728853c384a9d55349 : SUCCESS in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6bb737168554e258539420c5fff1347 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3ec887a67ea4b1d83bdfc0e72ec1a90 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ebd25485224498bb439e7fdfec8672b : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78bb336cb2ae4647aca4756c7d7c657b : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/923fc06a9c644062a197b29bab4298ef : SUCCESS in 4m 20s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/a5c97d02c1c44f2ea33013824104fdfe : SUCCESS in 59m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1c23e23ce3b4d429ed25b91d26e8acb : FAILURE in 2h 01m 55s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/e8232f0644f04c30b8440c424ebadcdc : SUCCESS in 1h 40m 07s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/5cc8ec64b5c84af199a78385c085f0ef : SUCCESS in 1h 37m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7bcce6aefebf49d48301361bf276a696 : SUCCESS in 1h 05m 51s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/004be27d499b4de49caaf534e426c043 : SUCCESS in 28m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36440fbffbf2499c829bc0bac72423ef : SUCCESS in 1h 05m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e406c22dc9274ac3a3e73440b533eae7 : SUCCESS in 1h 36m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/311caa81a70846ab8411d1a1d0b0e96d : SUCCESS in 54m 33s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/468f3c3ba24243a9b4ef1f6cd4047bc1 : SUCCESS in 1h 38m 50s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6fe5e69ac9ba4541b772807ecc1010fc : SUCCESS in 1h 45m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd1a5695aa7746c28a56e40221ed05db : SUCCESS in 1h 00m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d27ed1543fae45029f8a5640fbdf9595 : SUCCESS in 53m 52s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/7ced823bbe224f2b875198804e8ebaf2 : SUCCESS in 56m 08s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0e4eb84a2c174d3493d076569afb9b23 : SUCCESS in 2m 53s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8dd2b27d44ba4ed28f3a7e537041ed28 : FAILURE in 1h 03m 45s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8878568f2d9b4831b7481e5e4d3906b2 : SUCCESS in 44m 52s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ef94d1653b4b4ceeb5ddf1e8785167ba : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"cd71961ef6dc2f0d1bd42df4c124be7ec7649cff","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-04 17:18:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"263d7c3cd1ddb1966b39d7ffc3ecfb89e6d7220e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 17:19:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19e45a5d79e0561df4268e1f991bee99a0992116","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 19:01:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b470a6c938334ae4a2885deee90baef2\n\n- grenade https://zuul.opendev.org/t/openstack/build/2d7371e651a143c5ae344914ae804f63 : SUCCESS in 50m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3881725fdae469d97c6e8cbf7857460 : SUCCESS in 1h 35m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb081ad2f444613ad83b9c1bb257aad : SUCCESS in 11m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c164867452245b0a4df2e4a66c68bdd : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e65fb2490a9f4f0b91418353992f2307 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f534b26cc64f4fe588c8ccbf8e877798 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1114292e9d7b43038c5008f3f537e71a : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3dc8a5c640364723a88d3c1c7133ab4d : SUCCESS in 3m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dfaa8ccc5ac4479ba275d8d3d6a8f976 : SUCCESS in 59m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/14c324b63dfc4300a559680896a34299 : SUCCESS in 1h 04m 08s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/9c3aa04a029e47b182a71823bed1dffb : SUCCESS in 55m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/344f94d270044474a8638b9ed3aa048f : SUCCESS in 49m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9769819d0f0a4e52946fe1d4aea54685 : SUCCESS in 1h 00m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ff5a8f97729341e0bdd9e2f3271d4ae6 : SUCCESS in 56m 56s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/486fdd3ed33c4001b4d402c7db239f08 : SUCCESS in 1h 37m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7125f892e0694cd9a5fe4bc30b7c1188 : SUCCESS in 50m 47s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/200a9dd64d1a4d39911a57af164b68c8 : SUCCESS in 47m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7a0aabfb8c67a26ec4528cb55cd784159c156715","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 19:01:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4ff72a2ebc399b65093cd8ae30ed273ae6c1726c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 19:02:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/278e95a72c8a46c4aef0e48d35bcedc1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80551e32ee14f429275bf0be3dd8d26 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8e48949565d40acd7e8cddd32abff9eed5f962d","revisions":{"d8e48949565d40acd7e8cddd32abff9eed5f962d":{"kind":"REWORK","_number":1,"created":"2024-10-03 23:41:49.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/58/931358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/58/931358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/931358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/931358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/931358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/58/931358/1"}}},"commit":{"parents":[{"commit":"da14e2972243e4890e3ef6889bf5a85ee8cffe39","subject":"Replace deprecated datetime.utcnow()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/da14e2972243e4890e3ef6889bf5a85ee8cffe39"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 23:21:53.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-10-03 23:38:57.000000000","tz":0},"subject":"Add compute response schema for microversion 2.96","message":"Add compute response schema for microversion 2.96\n\nCompute microversion 2.96 added the pinned_availability_zone field to\nthe GET /servers/detail and GET /servers/{server_id} responses.\n\nhttps://docs.openstack.org/nova/latest/reference/api-microversion-history.html#maximum-in-2024-1-caracal-and-2024-2-dalmatian\n\nChange-Id: I7b1e32b7ac3622029336f1919fb5cc2731050a2d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d8e48949565d40acd7e8cddd32abff9eed5f962d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d8e48949565d40acd7e8cddd32abff9eed5f962d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
