)]}'
{"id":"openstack%2Ftempest~933716","triplet_id":"openstack%2Ftempest~master~I8d2ff33384aec93820aa27ad27f1ff9bd7e87098","project":"openstack/tempest","branch":"master","topic":"bug/2085886","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-10-31 01:41:00.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-10-30 20:03:58.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-10-30 21:15:58.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I8d2ff33384aec93820aa27ad27f1ff9bd7e87098","subject":"Use stable branch upper-constraints till 2023.1","status":"MERGED","created":"2024-10-30 10:36:03.000000000","updated":"2024-10-31 01:42:03.000000000","submitted":"2024-10-31 01:41:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"933716-bug/2085886","meta_rev_id":"218bf67396821d236af026605edeabe49cbfeb90","_number":933716,"virtual_id_number":933716,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-31 01:41:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2024-10-30 20:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2024-10-30 21:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2024-10-30 21:16:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-30 12:33:08.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2024-10-30 12:33:08.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-10-30 12:46:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4f3ce3796b7fbab9a27c51a62392ebd1d7f10a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-10-30 10:36:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fad7454dceab73569c7a7915257ee9e2fdb493d7","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-10-30 12:33:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a91f129618528ec41bd2a42e2ad677b2e5d7f570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 12:46:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ad7f8def16c46f2819e4a4e02aab783\n\n- grenade https://zuul.opendev.org/t/openstack/build/145ad869842f4cc6a5eb01301d705666 : SUCCESS in 42m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22196aa77d994e678785036f9d86cd62 : SUCCESS in 1h 14m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9f10131ac55446188d77a4d38a4a1b0 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/181f12d363764a9dbeef5e70a40bd43c : SUCCESS in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b314ed5de14c1dba9a98ac73c435a5 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8df7dc2988d04d1a941ad4ccef3ca83b : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33fc0429600b4fde9758782eeb06c29d : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/92fccbbb938d4b01992fff1983484c46 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59220446f2cf461faa86a9281ad51c6d : SUCCESS in 4m 35s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/426ed827eaa9492588fe0f6088aa52f0 : SUCCESS in 51m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/86417a33d6254e148fcd529f644b30f6 : SUCCESS in 1h 57m 21s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/6109ff18955b4f40aba463bd8ab21781 : SUCCESS in 1h 43m 58s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/59f52185d73b494b8638b20bae0215b2 : FAILURE in 40m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/019012c501aa4891a4b477e62d89c571 : SUCCESS in 1h 08m 45s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/7c55cd60d69d4444b5059a762fbe961a : SUCCESS in 25m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff3553869e0348ecbf1ed8b7b305714d : SUCCESS in 47m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/562dde19c456447a97297875ccdb1faf : SUCCESS in 1h 22m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a13b74e8578d422f81b7feb1bf35cc09 : SUCCESS in 46m 43s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0471a910bbe443d9b49cbb55840a8f94 : SUCCESS in 1h 42m 10s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a3aab91698e94b6e9b8c9e5214e8752d : SUCCESS in 1h 49m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c076df7ebf8c4edda87077969e35c30a : SUCCESS in 59m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee61392f42644ed09daba86033431351 : SUCCESS in 1h 03m 33s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/eab3c4062a0641ba90f8d2ee0a464164 : SUCCESS in 1h 17m 48s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/dedb92754fc64dae9d288c9ea23e1355 : SUCCESS in 4m 45s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/797eec7c684a4ce5b48a019a4ae3a1bf : FAILURE in 1h 25m 16s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f44e46d115994896960aa86449492993 : SUCCESS in 1h 08m 15s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/96ccb922160f463899f1733e454e36fb : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2c9978bbdf2372b94cb7f643db31e91aba32e11d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-30 17:55:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cf6b826232900759b892b25afa43d0b1de89335","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-30 17:59:18.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d70b79189634a16c47b88dc6a72c7b916f89e3c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-10-30 18:03:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"47919cff51dfda630c2a09b5665884cd79e9f84a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-10-30 18:05:29.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b5f6439a6707c9b373bd51ca7abfcac6766b5927","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-30 18:22:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4431a4ed256671e7e31f3143c934d18a4cd218f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-30 18:26:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca550713c3c69852020913ba9426e26034461b08","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-10-30 18:35:01.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"433391a1daf7273d7441465837b0265c7cdb0a44","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-30 18:41:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57a37b5c264deb200faf5c4da213912051b23166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-30 18:46:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8bd1a4a3476f8487a5c2a91d4a6138d0c2b32254","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-10-30 20:03:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e59f8ab9bd4848cbe31304cedc1d13927e1e67ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 20:56:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa63d9397fcb4ad0bb4410167cc3f48f\n\n- grenade https://zuul.opendev.org/t/openstack/build/a2e01fd60ccc4dfcbf23d0a2d058b1c0 : SUCCESS in 1h 04m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf69f0cfc22d496e97235151e4070dfb : SUCCESS in 1h 40m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f6bc2c17b794deab856856c5f9018a4 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1862813696c47cba9915c2ff0d83a20 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617a4e943a3941cf83694b844d17fa8b : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47fbaf9aff434d06bfcb330704236b1d : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f040be2d8d4748509cb78f6ff6337c41 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b864985960ef4965af8675aa6693ab6c : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d74d5d915524a07b0445f0aefd8e55a : SUCCESS in 3m 31s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/c05a43c1ca6e474eae080ba5946ab86c : SUCCESS in 50m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c87c3e02bac247da92ba9985c9a52b05 : SUCCESS in 2h 09m 13s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/7df47190e60b42669d050fb0e1abfcf1 : FAILURE in 1h 21m 36s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/6e37e626e6124a91a51f0e452a3b1779 : SUCCESS in 1h 26m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7a864dc9557a4308bbe14d2bbe29e797 : SUCCESS in 1h 04m 26s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b63e681fcd774f9a88c62243c6b336db : SUCCESS in 25m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69c4280e85d9421080ddbfd2a6924d88 : SUCCESS in 57m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e58c97a8711b4e0c925fa66ee1a03ed5 : SUCCESS in 1h 33m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/af4859841ac146f1bde8b217f3ffe7e1 : SUCCESS in 50m 26s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fd566a5b6cb842dc959f2360434df873 : SUCCESS in 1h 25m 26s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dad6517365154169993f985d57b6f49e : SUCCESS in 1h 52m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/adbd07b2e12344809005d7ad1c20cde4 : SUCCESS in 57m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dadf3f1d320045eaa9707d3aa952104c : SUCCESS in 55m 41s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/856c6e29c7874ee5b23382a0ee773744 : SUCCESS in 1h 07m 41s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d712e98b9733413ab1864eaadb28d507 : SUCCESS in 4m 15s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ebf1578815e3428caed884175b8e3201 : FAILURE in 1h 10m 27s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/9e088123e636433abbd61057f2c202e3 : SUCCESS in 57m 29s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ee93a3b1e2a249bdbaa40db611ef7908 : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":3},{"id":"246949f2e3e486661c56357b59796db5903e9ee0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-30 21:15:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"77519278db73eb08a7d7be4eac0b4eed6ed4b47d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-30 21:16:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"36e1189432c4118dca11782a02404d3b84092099","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-10-30 21:16:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e6ad119951300678d9a4617b6da71bc9873ff702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 23:46:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a824782a683c41cfbbb01e01ae303df8\n\n- grenade https://zuul.opendev.org/t/openstack/build/911ea27533294fa88da9a72c6f6e4041 : SUCCESS in 42m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6abd8b9c92ad4941a7cdfa85608ed050 : SUCCESS in 1h 21m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cebe5af31894795832aa6fc355a950c : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12600ae673342c2b1f9788d05bc44ea : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05a287265f694af6a7b5c258c886b1f7 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54dd12f754d641108af72ee2431c1031 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/991354b01b8f4a2c96e2f430daf2999b : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b88102d892a4ab4b43cb1a03e02adc4 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/245d47bfc75249058a026c19a6fe2b06 : SUCCESS in 4m 11s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/bef0e3a6b58244ea942039b37b1b179f : SUCCESS in 45m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/111738daaebc4a1da05f8f9051e1c039 : SUCCESS in 2h 23m 29s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/ceb22dba5ed841e88920e6e46f508bd7 : SUCCESS in 1h 10m 00s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/57c21e8a0b754c62b1ae1128ba387b03 : SUCCESS in 1h 36m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b4f9ed29ce9146378974c088d08e7fba : SUCCESS in 1h 04m 43s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/76740371296842a792f97fee218ee84a : SUCCESS in 19m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c763014523d041c4baff3c4e1542da59 : SUCCESS in 1h 01m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/062222fdf69d4507b7811b083c1d323f : SUCCESS in 1h 19m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6d4965bf29c34e9eb2c9fe895359b6ee : SUCCESS in 52m 02s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/04b66595bda249f1a18fa66adb9cb304 : SUCCESS in 1h 33m 56s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2c9bdd34c8d14dedbed0e2c134820183 : SUCCESS in 1h 31m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c4dc5c6c761498fb908c482551417df : SUCCESS in 50m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6461d6645dc4533af36402704a9013f : SUCCESS in 52m 10s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/3ef79df4eb2c461280018382c4c2e77d : SUCCESS in 48m 24s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/27935607e5ac4952bd5f487ecbb87420 : SUCCESS in 3m 06s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a8377238e122483cafe14c5e15f45cb2 : FAILURE in 55m 45s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/d267914108be43a39cb90c1d16495f5d : SUCCESS in 36m 22s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/61bc9cb56f7641b5affad1c7022cc538 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":3},{"id":"c71118204ac1ff1af8a612d6d43229463147613f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-30 23:46:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8b6b5ba34545aea4654635831c0d2fb8570f8279","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 01:41:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2245da7c3e0840e3a4a62361e0865a1e\n\n- grenade https://zuul.opendev.org/t/openstack/build/3a5419fa46484b3b91910eee9e7f3922 : SUCCESS in 1h 00m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cea4f3cb2a1142fdaba39f3da3e1c50d : SUCCESS in 1h 39m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84301f3b3f445c194395111dc7bddb1 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28cdc1791cea4ce781b879c33823ce3f : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c60c645c8f224b97b834f9612e07a4a7 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff797c658e6a474f8c30d1a82ebf87be : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/363a1aa5165e4958b896be6ee3fbab15 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3485005760c245b9b4f4788cbbd34a8c : SUCCESS in 4m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7cdb21c0cc50490a9f3fcd45df74fe87 : SUCCESS in 1h 48m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5268cd70741f4ee792c321d3fb2f55a0 : SUCCESS in 1h 00m 29s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/03406659e9ef4836b5a3eb7b557e076b : SUCCESS in 32m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bdb38a7d18bd4eccb915620706a06f87 : SUCCESS in 41m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4544e686a4104251a7c08779b337f086 : SUCCESS in 1h 00m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fa0e168e066e442499a165561cb23ea3 : SUCCESS in 40m 00s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cd354d6bb1aa40dfbd2f36507c76cefc : SUCCESS in 1h 14m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2de75baf2fdf416097553845f0dbd89d : SUCCESS in 30m 35s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/b8a0525b04e64fbb9e44537d9fe196e4 : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":3},{"id":"aded68ee97e0f094eeeb3db231f2673ad0302f26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 01:41:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"218bf67396821d236af026605edeabe49cbfeb90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 01:42:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2457402a85a9496bb4fa401894f3b10f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4da0095458444c8eb0c49a493af35e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"98f2393641439cdde4c6873ca30ef2a4b0679a13","revisions":{"ad29e8b016572d8654fce0291355a8c121b5c774":{"kind":"REWORK","_number":1,"created":"2024-10-30 10:36:03.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/16/933716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/16/933716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/933716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/933716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/933716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/16/933716/1"}}},"commit":{"parents":[{"commit":"b530c1ffb70110ab3d9112c90cd55c48db0285a0","subject":"Measure metadata downtime during live-migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b530c1ffb70110ab3d9112c90cd55c48db0285a0"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-10-30 10:31:28.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-10-30 10:34:58.000000000","tz":330},"subject":"Use stable branch upper-constraints till 2023.1","message":"Use stable branch upper-constraints till 2023.1\n\nStable/2023.1 and Unmaintained xena/yoga/zed runs ubuntu-focal\njobs and since with [1] py3.8 pins are dropped from master\nupper-constraints, so these branches can no longer use master\nconstraints with envs running with python3.8.\nWith this patch making these branches to use stable constraints.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/925201\n[2] https://lists.openstack.org/archives/list/openstack-discuss@\\\n    lists.openstack.org/thread/FOWV4UQZTH4DPDA67QDEROAESYU5Z3LE/\n\nRelated-Bug: #2085886\nChange-Id: I8d2ff33384aec93820aa27ad27f1ff9bd7e87098\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad29e8b016572d8654fce0291355a8c121b5c774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad29e8b016572d8654fce0291355a8c121b5c774"}]},"branch":"refs/heads/master"},"c57d5a09cfe472dd36f0b6f470cc3c4042771c8e":{"kind":"REWORK","_number":2,"created":"2024-10-30 18:03:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/16/933716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/16/933716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/933716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/933716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/933716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/16/933716/2"}}},"commit":{"parents":[{"commit":"b530c1ffb70110ab3d9112c90cd55c48db0285a0","subject":"Measure metadata downtime during live-migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b530c1ffb70110ab3d9112c90cd55c48db0285a0"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-10-30 10:31:28.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-10-30 18:00:18.000000000","tz":330},"subject":"Use stable branch upper-constraints till 2023.1","message":"Use stable branch upper-constraints till 2023.1\n\nStable/2023.1 and Unmaintained xena/yoga/zed runs ubuntu-focal\njobs and since with [1] py3.8 pins are dropped from master\nupper-constraints, so these branches can no longer use master\nconstraints with envs running with python3.8.\nWith this patch making these branches to use stable constraints.\n\nAlso dropped tempest-full-2023-1 job from check queue as that\u0027s\nblocking this change to get merged, there is dependency on\nthe Needed by patch below to get it green.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/925201\n[2] https://lists.openstack.org/archives/list/openstack-discuss@\\\n    lists.openstack.org/thread/FOWV4UQZTH4DPDA67QDEROAESYU5Z3LE/\n\nNeeded-By: https://review.opendev.org/c/openstack/devstack/+/933704\nRelated-Bug: #2085886\nChange-Id: I8d2ff33384aec93820aa27ad27f1ff9bd7e87098\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c57d5a09cfe472dd36f0b6f470cc3c4042771c8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c57d5a09cfe472dd36f0b6f470cc3c4042771c8e"}]},"branch":"refs/heads/master"},"98f2393641439cdde4c6873ca30ef2a4b0679a13":{"kind":"REWORK","_number":3,"created":"2024-10-30 18:46:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/933716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/16/933716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/933716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/933716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/16/933716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/16/933716/3"}}},"commit":{"parents":[{"commit":"3e5592f58c4018f7aa32f53c5821ee341f387ab5","subject":"End support for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3e5592f58c4018f7aa32f53c5821ee341f387ab5"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-10-30 18:44:11.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-10-30 18:45:55.000000000","tz":-420},"subject":"Use stable branch upper-constraints till 2023.1","message":"Use stable branch upper-constraints till 2023.1\n\nStable/2023.1 and Unmaintained xena/yoga/zed runs ubuntu-focal\njobs and since with [1] py3.8 pins are dropped from master\nupper-constraints, so these branches can no longer use master\nconstraints with envs running with python3.8.\nWith this patch making these branches to use stable constraints.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/925201\n[2] https://lists.openstack.org/archives/list/openstack-discuss@\\\n    lists.openstack.org/thread/FOWV4UQZTH4DPDA67QDEROAESYU5Z3LE/\n\nNeeded-By: https://review.opendev.org/c/openstack/devstack/+/933704\nRelated-Bug: #2085886\nChange-Id: I8d2ff33384aec93820aa27ad27f1ff9bd7e87098\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/98f2393641439cdde4c6873ca30ef2a4b0679a13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/98f2393641439cdde4c6873ca30ef2a4b0679a13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
