)]}'
{"id":"openstack%2Ftempest~934168","triplet_id":"openstack%2Ftempest~master~I86637ae2fd4e00ce5571d7fac8ae41418637cf29","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-11-06 00:37:08.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-06 00:37:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I86637ae2fd4e00ce5571d7fac8ae41418637cf29","subject":"Check for glance service for glance tests","status":"MERGED","created":"2024-11-05 18:12:19.000000000","updated":"2024-11-06 00:38:22.000000000","submitted":"2024-11-06 00:37:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934168","meta_rev_id":"5e8cf89322b8b152d788bf2d71c01ad23c10cadb","_number":934168,"virtual_id_number":934168,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-06 00:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2024-11-05 22:42:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2024-11-05 22:42:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-05 18:30:53.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2024-11-05 20:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-05 22:42:24.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"0d2fe58a4d0112feb9c93020305e6998b05933ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-05 18:12:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"777bae58dd5c200fb8e236e10fe08ca86dafaff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 20:22:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b645793aaaf743b99668d7247cf0e5d7\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8fd82c7523c4194a5cdf737f0225cb1 : SUCCESS in 1h 08m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1d9f47d2a434b74abba0a832586b922 : SUCCESS in 1h 38m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/740d7ee486224338ad56f1b5cc59a9ef : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2dcaa53c13b47a6962c1bf1f51fb6d2 : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2030b04fa9a43daa7fd13f8c1ee6eb0 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea57a79b9190447d88e6e22b3d30c513 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50855f73b57148aa8d7abbb1533333ba : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a6d138823b24f1e8eee1587c40cbb89 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64611c22baa347cfa4835674482f72c6 : SUCCESS in 4m 25s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/05a09862bbb4460bb517cf37ae8f6dc1 : SUCCESS in 33m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9c92e29af0414bd8859fb12c1f05f5e6 : SUCCESS in 2h 03m 54s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/9e856102983a40be84da30a519246b79 : SUCCESS in 1h 32m 36s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/1fc7014c890c4a5292a5deb7cde623f1 : SUCCESS in 1h 10m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0d44ecf6ac0143da8ee10e453cc52cd6 : SUCCESS in 44m 53s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/156a39e743414405a4fb669ed2aa509a : SUCCESS in 27m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17ff6def4513417dbd09132bdd5dbed3 : SUCCESS in 43m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d9dfd7f99c7549078d16d5ca68fa14d2 : SUCCESS in 1h 22m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fa2d48b2e25d4860b4b58c646e0f62d7 : SUCCESS in 40m 02s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c722d699f909422fa1ff3162e69ef0d6 : SUCCESS in 1h 37m 35s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2db5956ab91e485cb717650caa660bd4 : SUCCESS in 2h 02m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c66fbddb79b466db6b72629946f0f2a : SUCCESS in 1h 08m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/70a01ce2b93746939606072e23f00ff3 : SUCCESS in 1h 07m 59s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1fa52566aef341ccb21b958fa804d839 : SUCCESS in 1h 13m 06s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d0d9e9cfb9f442a1a9adf3e1e5d01994 : SUCCESS in 4m 55s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/fdb96d5be95e491c8b8e9ec36be2fae5 : FAILURE in 1h 11m 58s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/3613e607a98a4e6fb26407604a386aef : SUCCESS in 1h 04m 03s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/62d55dff735b4382b60cc324107118f0 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1},{"id":"008f1c6ef647c03c567ca30bd14604021ad30fa9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-11-05 22:42:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a2b8d6cfc261e8fe7c019f9e5acb31c964f6066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 22:43:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be6d7c6516eb9a61e742786c0047cf14942d61da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 00:37:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d265fe763a4c21a42acfc6a20ba797\n\n- grenade https://zuul.opendev.org/t/openstack/build/dda8fdf716d0430d9f48c324f592d1d0 : SUCCESS in 1h 00m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/886dd52646334fc6b0d0e98486fde757 : SUCCESS in 1h 43m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec5c5054cd94b20b5bcf0ec79c7c719 : SUCCESS in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9600166b5f9423692e825728ac5ecf0 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c292627493ea4e08b1888a26560996ab : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f6e69f8cb364c03995889de57e4f7bc : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2fb55d9a880443ba1a9840dca36d724 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/552b1e8cf3194e1f9e616f73b34de914 : SUCCESS in 4m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bbaca5d1a4db41b6bec30bd7176c1cd1 : SUCCESS in 1h 41m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb1d139bdaba4a489c1a57343e221363 : SUCCESS in 58m 20s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/3b21250fcc26461a892e910bc4159e4f : SUCCESS in 54m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44193923992d4cb2b9fcc15570d328df : SUCCESS in 59m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1030d12ca8844aaa29d41b8a74e0554 : SUCCESS in 41m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/88ec37d419e34a01ac7b382e7f65ec68 : SUCCESS in 52m 26s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b88342c76d7149f5a2cf13ffdaae7537 : SUCCESS in 1h 18m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f7f8db84360b4180ae6ad3ce43401e91 : SUCCESS in 50m 24s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4af8cf44dd374fa8b9e3a68b1f534b0b : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ff0127475b59cde3f544c78ddbd6a76bd680f0ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 00:37:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5e8cf89322b8b152d788bf2d71c01ad23c10cadb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 00:38:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e13bafe14bd47d8aaaf3f7e12d0e3b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c248d896a3f4f558e1a3a060eeac4bb : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"906b6b85a871288722cf0d670fdda0e7e55ba8c1","revisions":{"906b6b85a871288722cf0d670fdda0e7e55ba8c1":{"kind":"REWORK","_number":1,"created":"2024-11-05 18:12:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/68/934168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/68/934168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/934168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/934168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/934168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/68/934168/1"}}},"commit":{"parents":[{"commit":"2731314d4423b6b21cac95d7c736527b62aa49f8","subject":"Merge \"Check only ports owned by the dhcp service in the dhcp scheduler tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2731314d4423b6b21cac95d7c736527b62aa49f8"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-05 18:08:49.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-05 18:12:18.000000000","tz":-480},"subject":"Check for glance service for glance tests","message":"Check for glance service for glance tests\n\ntest_unified_limits scenario test the image service\nlimits and we should check the glance service is enable\nor not. Adding it in skip_checks method which run very\nfirst thing (before resource_setup()) to start the test.\n\nChange-Id: I86637ae2fd4e00ce5571d7fac8ae41418637cf29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/906b6b85a871288722cf0d670fdda0e7e55ba8c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/906b6b85a871288722cf0d670fdda0e7e55ba8c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
