)]}'
{"id":"openstack%2Ftempest~934194","triplet_id":"openstack%2Ftempest~master~Ib295ea1c6ff521d880180143a8ed40fbc73168c9","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-11-06 20:24:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Ib295ea1c6ff521d880180143a8ed40fbc73168c9","subject":"Fix \"pause_teardown\" help text for docutils","status":"MERGED","created":"2024-11-06 00:14:44.000000000","updated":"2024-11-07 00:34:44.000000000","submitted":"2024-11-07 00:33:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934194","meta_rev_id":"3119e6a6e448f27db5e57e76f31546d04b9c5bdd","_number":934194,"virtual_id_number":934194,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-07 00:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2024-11-06 02:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2024-11-06 02:27:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-06 02:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-06 02:27:47.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"0fc9e6e2b619f24b511b0258358790a235c64195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-06 00:14:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"715cca879af52c17468c235532b79ce454551f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 02:09:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3c0a7ac79c2464780a0f2f75e3cd711\n\n- grenade https://zuul.opendev.org/t/openstack/build/94d92df85a0a45218c7e7b8562f05de6 : SUCCESS in 49m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fdfcfe5e4a94f4b86dfa20e765c5934 : SUCCESS in 1h 26m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f086d50128d4762a3127b72b730d51c : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2bd6cd8f97c478fb1848c452c03809a : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc49239c9794a369199208d4e62def2 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f2c87ba489b4e40a807d49167619444 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3a79e76f5e84628b3222e3bf70e87b8 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c292a35cea814aa1af6c6ab7b4d3a0d1 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4897874bfb5476693f788bf10144165 : SUCCESS in 4m 39s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/71b2c668723148b48102cf4b43ac5e75 : SUCCESS in 52m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f7b3158edc834a2b95aca74a535cfd31 : SUCCESS in 1h 52m 59s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/6b3767823a4c4bd3996ffacc13db41e7 : SUCCESS in 1h 17m 47s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/7548b3af3b2949b6a6f2f379ae859722 : SUCCESS in 1h 30m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0bc27adac73d4c149bc04f90662898f4 : SUCCESS in 1h 15m 05s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6e675188369b459e885fbbce50a5b068 : SUCCESS in 25m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84f0c06a631e49849eef4ec312773647 : SUCCESS in 37m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/53342fcfc4594d03862ad9db95892895 : SUCCESS in 1h 45m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8aba2530d3ee4518b34a859d916749ce : SUCCESS in 52m 19s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8f17431305234321844d0335ecfddbdb : SUCCESS in 1h 13m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/df96402b518444c7a1a8e3d54ea014b4 : SUCCESS in 1h 04m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3abac4d37f5457a9e9a62d6da93f383 : SUCCESS in 58m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/969b8837779a4a369577562ce50b08f9 : SUCCESS in 59m 33s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2a75538a1a9d48ffabf4aac75824b3ba : SUCCESS in 52m 14s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/615db62dee674e13b67067eaf86a44f5 : SUCCESS in 5m 32s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/56975c3e6d3d4532ae276307e4ace7c1 : FAILURE in 55m 15s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/56ddbec7ccfb48808e983f25b722d10d : SUCCESS in 1h 00m 48s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/96992ff0cf1b490c888c92b03be0d951 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e10547b23a5f09a78ea5833656c10255b4c39a7c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-06 02:27:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3977ad65614e1012937cfa825447303ad155ad4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 02:28:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e379f97b671be654db40f80bea543ac439d70e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 04:29:36.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e68a690521c04a989f3d6da807c81bb1\n\n- grenade https://zuul.opendev.org/t/openstack/build/0bc06551953b458ebcd83d02a3dad9db : SUCCESS in 1h 08m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe289594354e46cbb0e1f79da1620cc7 : FAILURE in 1h 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba75eba443b649a5b2a4b51f93052a1d : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2634b839918641f5ba6dde11db9795d3 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc087874d4bd4d4f9c1af9c39542a0dd : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/406f4dd326ff42f8a475299a446b8c02 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ef8b0c940f543189f9a65e3003e48f2 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b419e7188195422fb8f8e6455367264c : SUCCESS in 5m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f0895140bfb49068d71f6aeb7c9052d : SUCCESS in 1h 43m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/713fa94205c54380a605ad6a9d94d75c : SUCCESS in 1h 04m 22s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/c4b55f08f1414e57a31286ec553ffc9e : SUCCESS in 1h 14m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/95860827263c46c48206954318a3b745 : SUCCESS in 1h 05m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f8f5cf590f4413d9b76c21c943e23ef : SUCCESS in 1h 03m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7f78eb1925fa4e8eb1cc5343c229a60d : SUCCESS in 1h 12m 23s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bbead79425934f95b12d5f580eee3e0f : SUCCESS in 1h 32m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/10ff4f5d3bc84dbc9fd63833be2204fb : SUCCESS in 39m 03s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c609b8f8c3b244929c4b361ba5c9b5c9 : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":1},{"id":"556f00c77b260cfb41179626801d2d1188a88309","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-06 16:58:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f012092c3972fdb6fbc32b9552a31c8ceb252bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 19:13:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b7d011777f44d6fb6bb916c2fa9eddd\n\n- grenade https://zuul.opendev.org/t/openstack/build/a443dbe8ef4f45c285ce1ed15c6a6bc0 : SUCCESS in 52m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d5c320855b84c9fa77a9b3d8d3b082c : SUCCESS in 1h 16m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5d2fda391804ae691a7ba73380ccd47 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ac090a5d49341e0a77055e37efe22bb : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96a6b655a1c6444cbd0b7a3989473af5 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0994526ecb28447ba4c8212370dc5c31 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd0d64f6d4a949a6aada72ed0d41e95e : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/942e6972b86f4be9a0fb85447c2394ec : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/662454c666cc40ef9bdd2c1e15fe7f75 : SUCCESS in 4m 30s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/b93b9f3fa120478aa75212325fb029da : SUCCESS in 43m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/55fc47f5eede4027857366c518e9b3d7 : SUCCESS in 2h 14m 11s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/9a9b3c334e714bb4a136ba05f61b9a2d : SUCCESS in 1h 37m 05s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/8989a2f99aec4dc2a77385d8cd036ec8 : SUCCESS in 1h 04m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a1e3a722c8654115886ea626ff38bf11 : FAILURE in 1h 14m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/bd003d7d1c064a1996815c37dbb0ce56 : SUCCESS in 26m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7db1dba77c649f3869d4fc813591421 : SUCCESS in 58m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29f32f11512546f2b3694d3a8c18dd50 : FAILURE in 1h 43m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/99628f870b9147f68724209f3c51106a : SUCCESS in 50m 51s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8728f002ede34872a1e725526a2ce96b : SUCCESS in 1h 44m 16s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9008c34d9fe24ec394540cfb211e7b5e : SUCCESS in 1h 51m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a3b9e31eb9b4145a9f88d23d9946f5b : SUCCESS in 58m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5136e4c9919a45588e8f395a90dd8998 : SUCCESS in 58m 36s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2ce265c4bbed424e884e8d0951fb1234 : SUCCESS in 53m 08s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/bff4c6bd580b4e82bb7fb7f43074061d : SUCCESS in 6m 54s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/dd332961bc974b1f89f90d65936b5010 : FAILURE in 51m 04s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/0ec2cb122a7c4a4596f9f5e7230aa817 : SUCCESS in 58m 50s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0adb4b996239474b84e565cb8c6fe940 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"896f9305839842b8508ef71e91e9076f99acd757","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-06 20:24:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7cab0506c049936fbe0401a04a8c566f2bbc5f14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 22:39:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ce74ca803b74cfd854a8489382b8feb\n\n- grenade https://zuul.opendev.org/t/openstack/build/f5987e165e234143a278d9536d1f522d : SUCCESS in 49m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d10a0833cd204baba1e554c2a73d088d : SUCCESS in 1h 32m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18fbc293ab90454199a83bb918d00d23 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf22799fd124abb8682e14199da929b : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c77607342546b2b2203d4188feb7a7 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e5cf2fef2394e9a9dfce6c0a95f09cf : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35e338d6fa9a47f5900476fa1e4615a4 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/670f701bcae94bd0bb61df5de3ac6ab4 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/364edc1fa95a4978b5862d9050c0c128 : SUCCESS in 4m 38s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/dd68ea455d594c0aa772d97746da1bd3 : SUCCESS in 50m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/31567f5818b04bddad7c74d023a7b455 : FAILURE in 2h 07m 45s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/d9e75220d640461a972b240993e0b3cf : SUCCESS in 1h 27m 05s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/20892e90c55449d8b7151712533a6345 : SUCCESS in 1h 20m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/823d2a97d39a474eb909a2e25b792831 : SUCCESS in 1h 10m 52s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2592a2b826b045c69009fab7fd9e05e7 : SUCCESS in 23m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e03af6feff774c8599c92d2b194c2594 : SUCCESS in 50m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67e9b6be685845d9bd37744ad16dd125 : SUCCESS in 1h 20m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4310413e16064370a1bb0eb24e6706d2 : SUCCESS in 45m 37s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fb1513fdf56348b8b37f3a4433a07ce9 : SUCCESS in 1h 26m 13s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c427fe7f9a704aecbf793b5cce6a22f9 : SUCCESS in 1h 27m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/54a86a48a9094b87a8cee7684bb16649 : SUCCESS in 58m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8fe2aa447c5a4f52b0efae6ad65a8cef : SUCCESS in 42m 23s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1bbd9136d921456aa88bfd22961ddd98 : SUCCESS in 1h 09m 55s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/66a68d9278424d559c27c4f3a60f3877 : SUCCESS in 4m 45s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ac4d38df596a412ab3b19c1454612ceb : FAILURE in 1h 27m 20s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/9819956eb6684dc5aab726995425a554 : SUCCESS in 42m 42s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d2e26782e0f34400867607ce7576f466 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9a3a6e414efbaa603b819db554763dac45125a1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 22:39:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2201a972c8b8b9efaaf6d74c0201b82f256d4d04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 00:33:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db4d4b65b7f4468d9afb6311093d3577\n\n- grenade https://zuul.opendev.org/t/openstack/build/992cd894593a4f83b7404a503361a80c : SUCCESS in 1h 00m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0465a73dfb7474fb197453c6fea7b26 : SUCCESS in 1h 48m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2039becc42164acba431eab06226b29b : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab31c05d8104295b7f65c7c5eff427c : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9484d1cf39624678bb931964af4bd2ff : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d86031b042c4b8887566f0dfdaf13b0 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44605eb2abc9400eaca6c4dce47d6754 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0020d5cf86aa4b618af01f18269085b9 : SUCCESS in 3m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/03b3edad99e94ae18799a1118c580a29 : SUCCESS in 1h 35m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/10bd6be7838d48d491381bb5f8d2497c : SUCCESS in 51m 31s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/39bffb71e5374c9b889d3c608764fb13 : SUCCESS in 46m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9dfd01d0c5c44fba73952c587531133 : SUCCESS in 33m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/278e16e576a84498a889fa005fdd8e12 : SUCCESS in 48m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/16dbbf46128b4d39980dd0e8e2e082ef : SUCCESS in 57m 48s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5b915cf8319247918c8758c57b117410 : SUCCESS in 57m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dac789aa2c494129840937b1ccda9cd2 : SUCCESS in 43m 09s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/82a9247b765c44d0982751afe4cb5d91 : SUCCESS in 59m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d9fa2806b58e1f9a82bcb4eb3c28071d9ea55ee4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 00:33:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3119e6a6e448f27db5e57e76f31546d04b9c5bdd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 00:34:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55cc962422d4471b91536ff8f6ecf472\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7babcf0ec16446648de7ebbf0416cda1 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"afe134a401c42b0eec309d67778eaaf95e0ee445","revisions":{"afe134a401c42b0eec309d67778eaaf95e0ee445":{"kind":"REWORK","_number":1,"created":"2024-11-06 00:14:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/934194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/934194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/934194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/934194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/934194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/934194/1"}}},"commit":{"parents":[{"commit":"2731314d4423b6b21cac95d7c736527b62aa49f8","subject":"Merge \"Check only ports owned by the dhcp service in the dhcp scheduler tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2731314d4423b6b21cac95d7c736527b62aa49f8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-06 00:05:58.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-11-06 00:14:43.000000000","tz":0},"subject":"Fix \"pause_teardown\" help text for docutils","message":"Fix \"pause_teardown\" help text for docutils\n\nThis patch removes the space in front of two lines in the help text of\nthe \"pause_teardown\" configuration option to resolve issues with modern\nversions of docutils which will raise ERROR: Unexpected indentation. [docutils]\n\nCloses-Bug: 2086747\nChange-Id: Ib295ea1c6ff521d880180143a8ed40fbc73168c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/afe134a401c42b0eec309d67778eaaf95e0ee445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/afe134a401c42b0eec309d67778eaaf95e0ee445"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
