)]}'
{"id":"openstack%2Ftempest~934923","triplet_id":"openstack%2Ftempest~master~I5e82704ae81e5c0ef016fac02d153d6ed003d12b","project":"openstack/tempest","branch":"master","topic":"bug/2085886","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-11-13 20:46:38.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-11-13 16:20:45.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-11-13 15:07:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-11-13 14:48:49.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-13 18:56:54.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I5e82704ae81e5c0ef016fac02d153d6ed003d12b","subject":"[Xena Jobs] use \u0027full\u0027 tox env","status":"MERGED","created":"2024-11-13 04:55:23.000000000","updated":"2024-11-14 04:44:39.000000000","submitted":"2024-11-13 20:46:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"934923-bug/2085886","meta_rev_id":"5147425beb7a6cd071651b511e65e45f79322060","_number":934923,"virtual_id_number":934923,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-13 20:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2024-11-13 16:20:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-11-13 14:48:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-11-13 15:07:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2024-11-13 18:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2024-11-13 18:56:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-13 06:43:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-13 10:33:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2024-11-13 10:33:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-11-13 14:48:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-11-13 14:48:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1de90a8db06d5a333380f90b6435e5e37cb98d1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-13 04:55:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"baa4235cb034f1a98383799c5f5297d0fdc83af6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 06:43:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4e76270493c4244a102bb451053321c\n\n- grenade https://zuul.opendev.org/t/openstack/build/aaa14ef993654688a1599a9081020e74 : SUCCESS in 1h 07m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afac964210d8454d8ba55fcac1298bcb : SUCCESS in 1h 29m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea69a6d774784b989007daa389ea7b96 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635f2b00e622454f95308f5dca669bfd : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7419557a66c94ff8b4fd4279c37a286e : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e9fc584672e401ead6940e98f4fdb34 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20fd491dcb71409da42e81a4a99bae0f : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26c2bc7865114cd48488360f8eedc806 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4cd7e667969e49d78ecc951b0aa689bf : SUCCESS in 4m 25s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/354b9ce2c6524563a071ef7427473d5c : SUCCESS in 39m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5d1adc97a8b9427b8707c705cb615ad8 : SUCCESS in 1h 38m 20s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/39118310947d4735a16d238d170ff577 : SUCCESS in 1h 47m 28s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/d15e1d49f5ff41408f670ca297720de5 : SUCCESS in 56m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7145796c6fda4e0a9cdef83a32fec4bd : FAILURE in 58m 01s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6f69330bb69c4681a67a6dc6797c8d21 : SUCCESS in 20m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a551d6ebd41f4451acde934d5c0236a7 : SUCCESS in 37m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2ecff6f2669e4b92bcec89a78e752e0c : SUCCESS in 1h 17m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4fcc43db1f4f45e29d9d4750f774739d : SUCCESS in 33m 17s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b2545e0c976747b0acec274c954c0830 : SUCCESS in 1h 03m 38s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3127cc090eca4164b3589e7f3b6f55cb : SUCCESS in 1h 11m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d29dd727a6c44c6c9ee52ea2bcded5c2 : SUCCESS in 42m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ebae106d9f644f7eb134250a1c8f4bb2 : SUCCESS in 1h 09m 22s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f5e6c2838c244b18b097924ad587ba02 : SUCCESS in 1h 14m 42s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/516203de327248acb0dc6789073d61ea : SUCCESS in 4m 17s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/32a17675953c408eaed1e8aa63e1ce12 : FAILURE in 1h 17m 40s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/6e5528ddaac74aa8ab14757ecb296542 : SUCCESS in 58m 10s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a7f976fa063a441584fd262e42cdc7a3 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0cb6cb2aaa852913112aacd73a4010c6b830a112","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-13 07:21:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8cc2dd112a135137bc961ab0cbcbcb92cdc9c037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 09:04:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71dc3133ea45407184c3b434e1a5d721\n\n- grenade https://zuul.opendev.org/t/openstack/build/c3cf3babbd3549048bb1905dca67d5a5 : SUCCESS in 52m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd825328c3354f3c82b9596d8b0c832a : SUCCESS in 1h 20m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37d7576f28944b6fb9419ce7cc60ac76 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d3f42b800b4d3d8253dd9dfbe4dc92 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b03c02252374b738927c3ae4e1d0507 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82aa6485849847dc900ab74934328709 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a017de6e56345fabc7edaadad6b0d06 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64ef5bdc18d04fb7868f141eec42b246 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1adc87105c44d4f920dd14e5add3a84 : SUCCESS in 3m 57s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/e9ec968cfaa645c5b8213b5cfb3480aa : SUCCESS in 53m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c9627db20dc496d825685d270184a8e : SUCCESS in 1h 35m 05s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/c60468bbbc7c49589f2212951335be53 : SUCCESS in 1h 36m 19s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/a9068d5b2aab4f77bce110448140be7c : SUCCESS in 1h 29m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2c77e9d663be4c85a1e47e195f0c564c : SUCCESS in 1h 10m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b56897624b7c4a5e8288d072916c70d2 : SUCCESS in 13m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40a7e4497f964831a38174dea995a863 : SUCCESS in 37m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d17ab43117574fb8a0a2e66290bc9e00 : SUCCESS in 1h 35m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8821052566644e2b8278852432ebc168 : SUCCESS in 55m 36s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e4c917458bf54638a42e2f3db642e6e6 : SUCCESS in 58m 27s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/94174e5f93084bebb25d87fcf437810b : SUCCESS in 1h 09m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/718fc5e71ae3493a8129c9d182512bce : SUCCESS in 1h 07m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f3f292f1fa64c198257a17300cb61e1 : SUCCESS in 1h 03m 14s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/9746ef5c6b934af8af9b2fc8ec21d28a : SUCCESS in 47m 23s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7ea2abebc7c848819b2b57565b6603e1 : SUCCESS in 5m 12s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6c81d948824d4af6b0d833ed1e382939 : FAILURE in 48m 22s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/fd0e9d16904e40c895e383e84b4df654 : SUCCESS in 59m 27s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7044cea8bb014f3c9acd94542e6eb43a : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":2},{"id":"479da52fe99f31c0574063c4556cc84dd7495cf4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-13 14:48:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6427566658cc00a08c9d3cf7be1beac673cb6c7","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-11-13 15:07:54.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3458a0344d0135c8236c5d093c39dd2c7095083e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-11-13 16:20:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bee9088413c9a3b0f48a759e98f659d81608e07a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-13 18:56:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d521e027220293d3dfe8beda94aa49062b4676b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 19:01:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d34c24d6371c5a90077a2ff2992b7b315d73d95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 20:46:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41682dc97b0741878ab635c7caa6a63c\n\n- grenade https://zuul.opendev.org/t/openstack/build/e1ff11e451ea4f2493ed2daef62fd164 : SUCCESS in 55m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00367ac59d664c58b24eba11572ee694 : SUCCESS in 1h 33m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a006ba0faced4783be0f3cf42e012be3 : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f93fb2ed14df4d388c71fe65ed7d1f71 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ea037f6016d4610a7534113789f8956 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb73acf878fc4de9a569b537125a43f3 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07a848a21e2447f6bc3d39f4180dc2e4 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c2fe047bf98433d8cd222903696d1e8 : SUCCESS in 5m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4ce2b0724d9b4e8ab899068dbfd2c1d5 : SUCCESS in 1h 28m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/46d64d6bf21f47fd8324875263c2eacd : SUCCESS in 58m 58s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/e8d530be9fb74b9d8286ae72e29163b6 : SUCCESS in 41m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57f541a35d8c4153941c86b7f5c848a2 : SUCCESS in 46m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/858fefbdc52541e8807491910485a5a6 : SUCCESS in 41m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2ce0395fcfd440f4ac227d60b96da7d5 : SUCCESS in 1h 15m 21s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/15faabe1bcf94207a46c5ba4b859855e : SUCCESS in 1h 44m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/63ba6953a73d4f0989e391314c579723 : SUCCESS in 57m 39s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/3ebb2b80816544edab1fe1024a4b5797 : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f8b30db451a443f1996a1976c75459dfccea0045","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 20:46:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"92c7ab1aaeddac26b0fcd352fa78d2f8c40bb120","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 20:49:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e027908e1ee245d2b265efc649501aaa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722c6052bcc441549c9cf859e5693d0d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2},{"id":"5147425beb7a6cd071651b511e65e45f79322060","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-14 04:44:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22022ccc58b6e4bd4977bd1255d6f0ca3b5ea116","revisions":{"7bf23168c82cd52042d01bd9bc9ac75195f8cf01":{"kind":"REWORK","_number":1,"created":"2024-11-13 04:55:23.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/23/934923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/23/934923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/934923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/934923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/934923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/23/934923/1"}}},"commit":{"parents":[{"commit":"e60b31706240d989ba6d0244acbbb97154c47a13","subject":"Added statuscode 406 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e60b31706240d989ba6d0244acbbb97154c47a13"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-13 04:50:49.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-13 04:50:59.000000000","tz":330},"subject":"[Xena Jobs] use \u0027full\u0027 tox env","message":"[Xena Jobs] use \u0027full\u0027 tox env\n\nWith [1] Xena is being pinned to 33.0.0 and that version\ndon\u0027t have \u0027multinode\u0027 and \u0027integrated-full\u0027 tox envs,\nso using \u0027full\u0027. xena-last tag which have both these envs\ncan\u0027t be used due to[2][3].\n\n[1] https://review.opendev.org/c/openstack/devstack/+/933741\n[2] https://review.opendev.org/c/openstack/tempest/+/821732\n[3] https://review.opendev.org/c/openstack/tempest/+/873442\n\nNeeded-By: https://review.opendev.org/c/openstack/devstack/+/933741\nChange-Id: I5e82704ae81e5c0ef016fac02d153d6ed003d12b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7bf23168c82cd52042d01bd9bc9ac75195f8cf01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7bf23168c82cd52042d01bd9bc9ac75195f8cf01"}]},"branch":"refs/heads/master"},"22022ccc58b6e4bd4977bd1255d6f0ca3b5ea116":{"kind":"REWORK","_number":2,"created":"2024-11-13 07:21:22.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/23/934923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/23/934923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/934923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/934923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/934923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/23/934923/2"}}},"commit":{"parents":[{"commit":"a350e320b162abc42c4b5236b258e1e8742048d0","subject":"Merge \"Wait for VM ACTIVE state is added for fixed network case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a350e320b162abc42c4b5236b258e1e8742048d0"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-13 04:50:49.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-13 07:19:48.000000000","tz":330},"subject":"[Xena Jobs] use \u0027full\u0027 tox env","message":"[Xena Jobs] use \u0027full\u0027 tox env\n\nWith [1] Xena is being pinned to 33.0.0 and that version\ndon\u0027t have \u0027multinode\u0027 and \u0027integrated-full\u0027 tox envs,\nso using \u0027full\u0027. xena-last tag which have both these envs\ncan\u0027t be used due to[2][3].\n\nAlso exclude DHCPAgentSchedulersTestJSON and Multiattach tests as\n[4][5] missing in tempest-33.0.0.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/933741\n[2] https://review.opendev.org/c/openstack/tempest/+/821732\n[3] https://review.opendev.org/c/openstack/tempest/+/873442\n[4] https://review.opendev.org/c/openstack/tempest/+/898090\n[5] https://review.opendev.org/c/openstack/tempest/+/875372\n\nNeeded-By: https://review.opendev.org/c/openstack/devstack/+/933741\nChange-Id: I5e82704ae81e5c0ef016fac02d153d6ed003d12b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/22022ccc58b6e4bd4977bd1255d6f0ca3b5ea116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/22022ccc58b6e4bd4977bd1255d6f0ca3b5ea116"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
