)]}'
{"id":"openstack%2Ftempest~938368","triplet_id":"openstack%2Ftempest~master~I36f6e7fefa2f6cca98befb0ef4338e40332df6ae","project":"openstack/tempest","branch":"master","topic":"boot-windows-vms","attention_set":{"35840":{"account":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"last_update":"2025-01-24 18:31:47.000000000","reason":"A robot voted negatively on a label"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2025-01-24 15:12:24.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2025-01-24 15:12:24.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2025-01-24 17:18:32.000000000","reason":"\u003cGERRIT_ACCOUNT_10459\u003e replied on the change","reason_account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}}},"hashtags":[],"change_id":"I36f6e7fefa2f6cca98befb0ef4338e40332df6ae","subject":"Add tests to boot WindowsVM","status":"NEW","created":"2025-01-02 16:30:29.000000000","updated":"2025-01-24 18:31:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":24,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"6f5f11dba14087dbdabed4c83927ffd9686016ac","_number":938368,"virtual_id_number":938368,"owner":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-24 18:31:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-02 18:09:13.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-01-02 18:55:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-05 10:19:30.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2025-01-10 13:54:15.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"}],"messages":[{"id":"12ed5f8224a665a3a62ebce1150e440d7c1a14f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"date":"2025-01-02 16:30:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb5f0e842c7f998c535fc539db0b4a3cbe7781bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"date":"2025-01-02 16:37:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1690464f2370e4b99ee70c4b8824b23f2ca225b8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-02 18:09:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"50a41b1c7b08aab4a0b7f88953e5af03e02490a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 18:55:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45ea1c2606144f0db03f4107d0f17b26\n\n- grenade https://zuul.opendev.org/t/openstack/build/22f4f39b00934be68c6127a86e5f4fc0 : SUCCESS in 55m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21caf86880884c059ff5aa558c8ab83c : SUCCESS in 1h 41m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04ef96acfc8d4743aeac4efc4858ae5a : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7eec345a6984cacab9c1d7780f1d4bd : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb39cce15e0402998fa75e431719b66 : FAILURE in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e1604a4100d4f049860774565aaa8c4 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38ba9027ac8e4683ac51a88f8044e573 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17e33cf226524284897bc5bbe767ce8e : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bfd4cdd1dab4a6e887f8380e1fc030a : SUCCESS in 4m 22s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/986ac1ee4a3545b990793c5f9519d460 : SUCCESS in 56m 01s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d6a75d1f50647a1abeea835eacb5ac8 : SUCCESS in 1h 22m 59s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0d987cd00a54efc9b455deab4b36bfa : SUCCESS in 1h 09m 39s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/bb8aeb0cc1414f2893c0839f9fd10923 : SUCCESS in 44m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cffeb381646b487184635d649a70d818 : SUCCESS in 2h 11m 59s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/56ab299829a84301bfba22a7ac6464aa : SUCCESS in 1h 34m 45s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/6b5a9cac50424a46b7e5ce7eced1c753 : SUCCESS in 1h 22m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0723bc9bd6b9472cbb16d10bf5c4c0b4 : SUCCESS in 1h 15m 36s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/83c0d89b6ffd4a4cbec544b018ec09af : SUCCESS in 27m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daa4fb76d62641ed955d57eca193b097 : SUCCESS in 44m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c2721df1c3b5415ca2e7209f5fa9bd31 : SUCCESS in 1h 25m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1e2f27298f91410882ffd125c9311c2d : SUCCESS in 1h 01m 42s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/708ed5f00bd948f48e37e815a59bb7ae : SUCCESS in 1h 27m 36s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/010454bbbd224ac8b496814fd8a341a7 : SUCCESS in 1h 38m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84e275ce619a48389538514b5dc0518f : SUCCESS in 1h 08m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec59ae60212e4ef0bdb076a0029dfe0b : SUCCESS in 52m 21s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a974295c57134d4d824501c8aa416620 : SUCCESS in 1h 01m 27s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8538e1adbced41f9ac7e1487d91c1d4f : SUCCESS in 5m 02s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e63a483c56ec4b93b88ba4fb53ce0ecb : FAILURE in 1h 21m 38s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/63e54aad1d0e4a5caf44de6e6855de0c : SUCCESS in 54m 50s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7fb0d9db834e49d6beab811b918f4fd5 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"43325d9d2f785b3b3740a5645c85f411a1f0f010","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-05 10:19:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1957e22030803e7261de1e6d3fc5aa7904095147","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-10 13:54:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ff616bb49aa19d41c517e8b4e39422b44635de6","author":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"date":"2025-01-13 16:49:40.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3dbb312314500a87522f9d0f9f925f221bd354d8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-01-13 16:55:41.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5572353b1162150c2ad8246f271db55d363cb2f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"date":"2025-01-24 15:12:24.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"03557e03698cacb345b1683db9b53d15b2d2242f","author":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"date":"2025-01-24 15:14:35.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f9fef27cb2da9bbef391e0960d62b18f3d3469dd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-24 17:18:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f5f11dba14087dbdabed4c83927ffd9686016ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 18:31:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fa5845df2ad41b9a90fd857ab884872\n\n- grenade https://zuul.opendev.org/t/openstack/build/4709fbf652a146069e6b667e6db33d78 : SUCCESS in 55m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c317c65d7ad474bb84ec7180b5f8f8c : SUCCESS in 1h 38m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2382a2d56d1472893d0319e709b3e14 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73a3ce9d1a6c4258b1038864d26f70c3 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59af0b2716b6423681cb8481c5cb356e : FAILURE in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52bea3c43fc04200bb86765faa7baea4 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f2cddad08e34280a386379ca15b6538 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1288732360ac45758dcbebcae5413533 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a88adfcdf0a4b58a0b05b4e4394d59e : SUCCESS in 4m 58s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9aa00a9e5d3744889b430ebb479f906b : SUCCESS in 1h 38m 49s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac6aaae97afe42fd928a10b168adfb5f : SUCCESS in 1h 13m 01s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ca8b5e6592d4c559ebd8822e74e6ac1 : FAILURE in 1h 31m 13s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/325f75d12d1c4e1caefb8d0a96217df4 : FAILURE in 1h 20m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c00adc83edf249c5b2c8e2e3843e0938 : FAILURE in 2h 52m 08s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/94ba8070f2264d24aa2adf6ddc8fbdb9 : SUCCESS in 1h 37m 35s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/46f93308471c4f4194a033da2bdcf837 : SUCCESS in 1h 46m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2a8a9176778a4a8389826fe136c9031f : SUCCESS in 1h 24m 30s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/20eb1597d6b84db6b800192f8756cf4b : SUCCESS in 45m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e7cb6307e424b598dc443f0a7dbbd45 : SUCCESS in 52m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d24e9f196f249f8bf084e03b69a0891 : SUCCESS in 1h 37m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/998d9513884c446f9d686b952f35c1d5 : SUCCESS in 54m 49s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/19fd8e703e9f4c63bae99866faaee767 : SUCCESS in 1h 51m 30s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d60289246da54595b5c8bffff6972f7e : FAILURE in 1h 46m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0428d169214b45e2bef8515455984fbc : SUCCESS in 55m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c62fdc85694c434bb98605b891b83b4b : SUCCESS in 1h 08m 39s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2fbcbbcf3b1f4b6c8b7a8a76e132bc6e : FAILURE in 48m 55s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8dfba66d54914f39b78cf6dcfab92ac3 : SUCCESS in 4m 44s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/27287887a48d41b9a62263f5650abf97 : FAILURE in 1h 34m 40s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/517e1b8d0c444d629509e103362c6a80 : SUCCESS in 1h 27m 56s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8b5bc261977d48e3994328a5d1de8d9d : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1cb3b8144006637f7274bd6be11e8e22f20df437","revisions":{"49f71e8ef1f45eb0593df143cedb2a9650a529af":{"kind":"REWORK","_number":1,"created":"2025-01-02 16:30:29.000000000","uploader":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"ref":"refs/changes/68/938368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/68/938368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/938368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/938368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/938368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/68/938368/1"}}},"commit":{"parents":[{"commit":"b23e9fcc280a51bf84fae68e7719bd05c0447806","subject":"Merge \"Wait for the router port activation before deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b23e9fcc280a51bf84fae68e7719bd05c0447806"}]}],"author":{"name":"rbubyr","email":"rbubyr@mirantis.com","date":"2025-01-02 16:22:44.000000000","tz":60},"committer":{"name":"rbubyr","email":"rbubyr@mirantis.com","date":"2025-01-02 16:22:52.000000000","tz":60},"subject":"Add tests to boot WindowsVM","message":"Add tests to boot WindowsVM\n\nTests:\n      - WindowsServers10Test.test_create_server\n      - WindowsServers11Test.test_create_server\n      - WindowsServers10Test.test_create_server_from_volume\n      - WindowsServers11Test.test_create_server_from_volume\n\nChange-Id: I36f6e7fefa2f6cca98befb0ef4338e40332df6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/49f71e8ef1f45eb0593df143cedb2a9650a529af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/49f71e8ef1f45eb0593df143cedb2a9650a529af"}]},"branch":"refs/heads/master"},"913b3ff2cc66f4b0d3856bc620e8ba0e8d59ccea":{"kind":"REWORK","_number":2,"created":"2025-01-02 16:37:25.000000000","uploader":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"ref":"refs/changes/68/938368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/68/938368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/938368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/938368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/938368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/68/938368/2"}}},"commit":{"parents":[{"commit":"b23e9fcc280a51bf84fae68e7719bd05c0447806","subject":"Merge \"Wait for the router port activation before deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b23e9fcc280a51bf84fae68e7719bd05c0447806"}]}],"author":{"name":"rbubyr","email":"rbubyr@mirantis.com","date":"2025-01-02 16:22:44.000000000","tz":60},"committer":{"name":"rbubyr","email":"rbubyr@mirantis.com","date":"2025-01-02 16:37:15.000000000","tz":60},"subject":"Add tests to boot WindowsVM","message":"Add tests to boot WindowsVM\n\nTests:\n      - WindowsServers10Test.test_create_server\n      - WindowsServers11Test.test_create_server\n      - WindowsServers10Test.test_create_server_from_volume\n      - WindowsServers11Test.test_create_server_from_volume\n\nChange-Id: I36f6e7fefa2f6cca98befb0ef4338e40332df6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/913b3ff2cc66f4b0d3856bc620e8ba0e8d59ccea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/913b3ff2cc66f4b0d3856bc620e8ba0e8d59ccea"}]},"branch":"refs/heads/master"},"1cb3b8144006637f7274bd6be11e8e22f20df437":{"kind":"REWORK","_number":3,"created":"2025-01-24 15:12:24.000000000","uploader":{"_account_id":35840,"name":"Roman Bubyr","email":"rbubyr@mirantis.com","username":"rbubyr","status":"Mirantis"},"ref":"refs/changes/68/938368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/68/938368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/938368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/938368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/68/938368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/68/938368/3"}}},"commit":{"parents":[{"commit":"b23e9fcc280a51bf84fae68e7719bd05c0447806","subject":"Merge \"Wait for the router port activation before deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b23e9fcc280a51bf84fae68e7719bd05c0447806"}]}],"author":{"name":"rbubyr","email":"rbubyr@mirantis.com","date":"2025-01-02 16:22:44.000000000","tz":60},"committer":{"name":"rbubyr","email":"rbubyr@mirantis.com","date":"2025-01-24 15:12:10.000000000","tz":60},"subject":"Add tests to boot WindowsVM","message":"Add tests to boot WindowsVM\n\nTests:\n      - WindowsServers10Test.test_create_server\n      - WindowsServers11Test.test_create_server\n      - WindowsServers10Test.test_create_server_from_volume\n      - WindowsServers11Test.test_create_server_from_volume\n\nChange-Id: I36f6e7fefa2f6cca98befb0ef4338e40332df6ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1cb3b8144006637f7274bd6be11e8e22f20df437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1cb3b8144006637f7274bd6be11e8e22f20df437"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
