)]}'
{"id":"openstack%2Ftempest~938766","triplet_id":"openstack%2Ftempest~master~I07d13a61053519b6ae0ac847b2c6eeeab75801a4","project":"openstack/tempest","branch":"master","topic":"add_401_case_for_attatchment","attention_set":{"25613":{"account":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"last_update":"2026-05-25 15:05:21.000000000","reason":"\u003cGERRIT_ACCOUNT_34860\u003e replied on the change","reason_account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}}},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-05-21 00:48:10.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-05-21 00:48:10.000000000","reason":"Change was submitted"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-05-21 00:48:10.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-05-20 20:28:25.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I07d13a61053519b6ae0ac847b2c6eeeab75801a4","subject":"Add Unauthorized exception for operations.","status":"MERGED","created":"2025-01-09 02:50:48.000000000","updated":"2026-05-26 16:56:17.000000000","submitted":"2026-05-21 00:48:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"938766-add_401_case_for_attatchment","meta_rev_id":"56c3659aa7b3e748608eaa768df90a97c041c306","_number":938766,"virtual_id_number":938766,"owner":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-21 00:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-05-20 20:24:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-20 20:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-20 20:19:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-20 20:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-09 05:09:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-15 19:06:29.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-03-07 23:20:40.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-05-20 18:32:20.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-05-20 20:19:07.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-05-25 15:05:21.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"CC"}],"messages":[{"id":"39e21f6896c235ad24b98f002dbe1667e5474609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2025-01-09 02:50:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"701edf47b1ec17e9ff3b593f479d1c6071467996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 05:09:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c6aff3c48524b8d82bf3d116eec705a\n\n- grenade https://zuul.opendev.org/t/openstack/build/5320c98f91ac4d658c34af3379a0f7b1 : SUCCESS in 37m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/004953480f874c0db9040cea632d003c : SUCCESS in 1h 01m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c02347b4a1144e2827792aadf0e811d : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21d4eedce6ea41e59933dc3fb2a3c17e : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a651a60f8143129e98d6f79501ff8e : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a762f157fc01403d85110629098a47be : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad3b627c0caf411181e6e6fdb5f79093 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43e4b81931dd4e47869483de7ce4885f : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/afdd4d00a6234172a56bce2e438b81d9 : SUCCESS in 5m 51s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a12f8b12c6d04866a8c3f1c1cd100351 : SUCCESS in 1h 37m 55s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0205696f63984f8ea12d28ce2df60c1c : SUCCESS in 1h 03m 37s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/873a50af082442af84496a928fc6a332 : SUCCESS in 1h 38m 00s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/2f1d429cbea14034bdaca3886e90f13a : SUCCESS in 1h 16m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6cc18cf746f64cde8810e00f0be28f0c : SUCCESS in 2h 16m 58s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/2226b5a41e514b1cb27ef982019bb047 : SUCCESS in 1h 30m 42s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/3fa59f78e5624657bdf4d1238f8db577 : SUCCESS in 1h 32m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b9f533e01a2e40e89e7151eee6fa94ec : SUCCESS in 1h 07m 03s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/364bdfd6a9184e7f9c4ab53bb50a3760 : SUCCESS in 33m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d65db3ddba3a4ee2a366020a92f9e8c6 : SUCCESS in 53m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7fbd3fa0dce944859eba46e5cb5bbf3e : SUCCESS in 1h 22m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/532fef8ecbb545989b2c9ede941b2263 : SUCCESS in 51m 47s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/062b31ab6024411da0c796e0994f39eb : SUCCESS in 1h 42m 54s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/27d2e74b9c1f47aebcf96c49c01cac90 : SUCCESS in 1h 58m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fae627a0a30740278e99337f78f6e158 : SUCCESS in 1h 01m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c54396aa62a240e393fef19a44a37874 : SUCCESS in 1h 05m 51s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/e19e187678e54daf9281345c54bcbbd6 : SUCCESS in 1h 15m 12s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/486fd49c7403409594d3aa4fbfb64ccd : SUCCESS in 5m 12s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/dc6b7ec8a1b64858a947d3e1d2e712ea : FAILURE in 1h 26m 59s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ef273bd87b9a42ce86db0f73bbd88ddf : SUCCESS in 1h 01m 23s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a7c31f026a994c0db16598dd6fe0bb6a : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"cac45d39e05a991e4a5083e7301eb1382e2f8da9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-15 19:06:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9ccbadf88de0ff68e5effd493c442e9696c6172","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2025-01-17 00:44:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be0aac79a1191daf206442e7d3299c34ac7607d7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-17 02:09:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2599669d159b3dfb9ccd8a7d157570c305663536","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2025-01-17 02:43:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55bde76d7c8009ec9c8e47b0aad5d9d2efe193fc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-24 03:14:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"671476e4e80adcc3a13d1f25120617daf4573e3c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-03-07 23:20:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5f7f4344e453f71051d29cae833499b61881cb16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"date":"2025-09-24 05:43:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5e91ac4e4200d9fb12c0585dccb7e8069573020c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 08:10:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3fedede29004cc4a6397addf5ea40dc\n\n- grenade https://zuul.opendev.org/t/openstack/build/39c7b92bff8e462c9748368a5efdb375 : SUCCESS in 1h 01m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb7c9f74488647048880a99f63ddd098 : SUCCESS in 1h 24m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89a1f46f875d4dce89bd61011e922f72 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/841407aec2e341e4b0c4f374d826e6d5 : SUCCESS in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ad955341be4f31a7c939be3c3a5853 : FAILURE in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30d6b1be01084bd284cc710df1f8d1a9 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90056d35525e48a69c53fe94eb090c0d : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f1ab5022e7b495aa28e1735dccb11ea : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2e41e4b4b674751888971ef82b8a6ef : SUCCESS in 5m 41s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/810b21faff4e463bbff9d6aa3b6c732a : SUCCESS in 56m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8be70aaad21649ac862f4a102a723b84 : SUCCESS in 2h 21m 58s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/204961865cde4bf1b5cfd7eb4297f0af : SUCCESS in 1h 19m 08s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/2804b42c13964afa9afa88734794a6b6 : SUCCESS in 1h 24m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bce9961253df4840b6e72d4763deeab3 : SUCCESS in 1h 21m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/148264cef4bb485cb845e5190d353188 : SUCCESS in 35m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfc8fab1c52f44208e523babee41bd19 : SUCCESS in 29m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/99a22dbc0b854ab89eac7ed59eba5c96 : SUCCESS in 1h 34m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/995cc7c13a734a78aaa72b1d1b5bcc4a : SUCCESS in 32m 02s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1723a9ae36ab48928cdd5effd69c8a3a : SUCCESS in 57m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/120febbc6e21468bb8920411eb239c04 : SUCCESS in 1h 46m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d911785706a434ba98d2899fe3fc8d1 : SUCCESS in 46m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92b020f284444d94bd1e6a8c6ddaf7f8 : SUCCESS in 32m 10s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/e9b4715c21d04ecb9a67224518529334 : SUCCESS in 43m 09s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8d844435bdcf44cc80e39512ebdd4270 : SUCCESS in 4m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4dad961362bd41a48795780bb1d8679f : FAILURE in 57m 46s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ef5e2b215df648559a9d7c1081eb384d : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4812364c725b9d6d7b95da8d447c0ed9b7bfab39","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-20 17:49:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"44492a8ee0d5cc840fbd819205d5bc3e8bc312c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"real_author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-20 17:49:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_25613\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"}],"_revision_number":3},{"id":"054c940d81d790bb3c042d1b0614d7a1b2259931","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-20 18:32:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8ced4ea6469cd68e38a5bee7a724e95d48415182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 20:06:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00e907a3f777457eab9ae1311a1a3a01\n\n- grenade https://zuul.opendev.org/t/openstack/build/df8268a81eb94cdbae1bf30d0ab5c9fc : SUCCESS in 1h 07m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26bea56a8d724fc6af1ee5fbc84e70d5 : SUCCESS in 50m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09772b9e75f8440c996118a53c5cfb1c : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3cbc7e675814acb96d6fc2861f96679 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d9de9924bd6407bbe551a7eca6dac6b : SUCCESS in 28m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf10d22868f74fa0a41f23783e3249d6 : FAILURE in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81f0b32c65ec4447b2fc2e8d27aecf75 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54d0f71bf0e342698be7695aef9b0ddb : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/324a45fd73764e21bfe36acd9d2382c0 : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08ce4d721e60483cac34cc26aa90830b : SUCCESS in 4m 29s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/9f4838fa1fc345e8ab1757ed3adc3fe4 : SUCCESS in 1h 12m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/71c9d3eab6b840b0a277a76146c5459b : FAILURE in 2h 00m 18s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/9f45d72af51d49c78a5c0c27c17b259d : SUCCESS in 2h 08m 06s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/356baa9b51c540b1989a70f0e8e42faa : SUCCESS in 1h 39m 50s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/2e393eaf9d0e4784abec434e266693e6 : SUCCESS in 1h 32m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bf07838526c6486fb79a176a048feebf : SUCCESS in 39m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/94c616b230f74ec9a8f0b889020e18d9 : SUCCESS in 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9d649a6801242dc949b9fc13a9cc310 : SUCCESS in 50m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3ed8d17f65d84b8ea629ca2b6d23678f : SUCCESS in 1h 47m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6043af23974441038905ad6e5378422e : SUCCESS in 1h 06m 58s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d0191918e9604324a57f2af895f55db9 : SUCCESS in 56m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1e8bac10845148b08a9232f512231fe7 : SUCCESS in 58m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e74d9b19c36485bb51c309ad95d8e48 : SUCCESS in 1h 12m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e5664504f6ad4c62a034cdc9bb3b37b3 : SUCCESS in 1h 11m 08s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/fc53aace6b0e4a6ca22eeb16a54945ee : SUCCESS in 1h 18m 21s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/be29415ed014494c861c69fd1cf527b0 : SUCCESS in 1h 44m 10s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/788762c06b4a44518cdccee8a97bc518 : SUCCESS in 1h 02m 48s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bfd55971b83b4de28144e2c088adf883 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3},{"id":"0b9983b5a4d67fff8873f5ef9aa8115aae2a4235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-20 20:08:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6ca0a11f0e0f3740dd4e051d9ac0ef7c61705cb4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-20 20:19:00.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9df39c934514a85da0b037919b12bf7386fbb843","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-05-20 20:19:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8fc2e30488b792b37259f95f9823bc01f0cc358a","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-05-20 20:24:24.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"beebcc3e39ede1523caf4dc9616c546098571a61","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-20 20:28:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6a80138fd417281a73ea251f8a9504983780ee90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 22:43:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50f6067cb43a4a1194cf204c08f1c00e\n\n- grenade https://zuul.opendev.org/t/openstack/build/ddb585e6c30b48c488da9a13062db94d : SUCCESS in 56m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0300c843e0aa4a338ddd93b78562eaa0 : SUCCESS in 1h 46m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be266b5a6d4d4eebaf2709949cea0462 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09abe56af6c04f9c9a145351d6975442 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d32901322add44f6a8a35f805d5a3254 : SUCCESS in 29m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d582f8b30b314ee0853ab1ecca2740d1 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7a8f9769e11493189fe013d2d1341dc : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4225de597f6416d845a8cecdd048e5f : SUCCESS in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9726503dd4e4271ab22b9cb5f4991dc : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/448f610395d4470bafcf63ad2690d0e5 : SUCCESS in 4m 01s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/dcd2ea323c6c423ebe88771949684c30 : SUCCESS in 1h 16m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/18d262cba8104a2bb51f36354aabad88 : FAILURE in 2h 27m 36s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/34ed1dacac124992a3ef40c54dd173d4 : SUCCESS in 1h 48m 26s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/9d4efe5c3a23495f9b3a171dd8f46e94 : SUCCESS in 1h 27m 59s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/0bcd23a13a5b4482ba5fa870dcf3d891 : SUCCESS in 1h 28m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ecaadd920cf3451bbf7e8f250c5c4c4d : SUCCESS in 1h 05m 06s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e2975e7d280c4ca383144a0b9c31911a : SUCCESS in 17m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a640f6ee361246e49c71dc0b4737bc22 : SUCCESS in 1h 06m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b448ba70ce204feb8fa37c1023559a2f : SUCCESS in 1h 30m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/87c549b29d0d4e76ab6b26050a344348 : SUCCESS in 57m 54s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0ff3dc6f6eea49bba6d081b95e49a3d0 : SUCCESS in 1h 27m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a1ffd37252d24af096663c592c398aa4 : SUCCESS in 1h 38m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed37c6d258f44a4193dec3326310934d : SUCCESS in 1h 05m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5a03dea05b8743a19fc4103b552f83d1 : SUCCESS in 1h 05m 31s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/37faf93884a94ea6a40d7c05a30fe394 : SUCCESS in 1h 20m 46s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/223b2fb04ccb4140b57a8796cf4f1df4 : SUCCESS in 1h 43m 54s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/04d6ec665b944742bc700023e46a7e46 : SUCCESS in 1h 05m 31s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2ad527f336044e868d4fbc51c434e210 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":4},{"id":"eb64c8022d2c139596d164ddae2e31ab69e2f45f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 22:44:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"36617a858630253cdfdb136247cc1a166bed5442","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 00:48:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdd2c52c057a49fb94ae73efa93067ad\n\n- grenade https://zuul.opendev.org/t/openstack/build/cc93c9f7263e415aa26156cc36e48e9d : SUCCESS in 1h 05m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff78b39e85034e529b78fb9aea7c11b5 : SUCCESS in 1h 58m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc873dec07c84d1895f63cd1cf384793 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e860c4eb6f56496e8715928aff3ca6ad : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd7b77d20b043c28b96db8813cdf933 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca201732f2b547ab96700185b962ae16 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b9166589da84192aaa63e0b041eaaa9 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7aff5057d2114defa048c3064d026487 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/333d06c2c5614a6187f5673e555803af : SUCCESS in 4m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/200d4a016ad144d69129c450679e499a : SUCCESS in 1h 07m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd97f40655954296a79fd3cf155191f6 : SUCCESS in 1h 12m 53s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/fa1fc6bb2d224befbec92876742876f5 : SUCCESS in 1h 55m 27s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/4f5265a29f8a44bcad0077ebdc56509c : SUCCESS in 1h 04m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83f9fe5ba46846aa8582066f9cb4cbb1 : SUCCESS in 1h 04m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e215acdda4bb4bbb8dd8e067855c32b9 : SUCCESS in 1h 02m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f9ec9def30b44acd823c03b04affa21a : SUCCESS in 1h 03m 44s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f5058eddc31144279f03b0cdf7fd2e2a : SUCCESS in 1h 23m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1468252c5a2c4cdab3f3ef3dea6a0f06 : SUCCESS in 1h 00m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f752bd5842ce4791a47514346048c320 : SUCCESS in 51m 17s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/79ab2e2081084948bd1c1f9fb08ce912 : SUCCESS in 27m 20s","accounts_in_message":[],"_revision_number":4},{"id":"c1c192040d630c55f24695966137e8d4cac81ac1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 00:48:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"5215772f7e4b80b7085400c1a0f8a98457f8ab44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 00:49:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e88b5f8b7523432b9bb6d3edff9b257e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9423feae40b64c95a71c53f2ffb3a00f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4},{"id":"d93aa60631ef820878fce75c5efb96522279cb9e","tag":"autogenerated:gerrit:revert","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-21 10:10:23.000000000","message":"Created a revert of this change as Ie6332a8ac50344bb5fb112986caec18728f82bdb","accounts_in_message":[],"_revision_number":4},{"id":"809ac57a230fc8d0898eda7d6a8cbada2f7d480f","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-25 15:05:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c1d59d23ad5f57c9a884ea762569408f11019221","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-25 17:34:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ab5d93db14e324d04d34852d59528eabc110211","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-05-26 05:38:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"56c3659aa7b3e748608eaa768df90a97c041c306","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-26 16:56:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"65f0c1ae4ca5b3e599291d309b068976b9bdc9e1","revisions":{"6431d71220d3fe753a5d6f0c3595e1aeef8bc250":{"kind":"REWORK","_number":1,"created":"2025-01-09 02:50:48.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/66/938766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/66/938766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/66/938766/1"}}},"commit":{"parents":[{"commit":"b23e9fcc280a51bf84fae68e7719bd05c0447806","subject":"Merge \"Wait for the router port activation before deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b23e9fcc280a51bf84fae68e7719bd05c0447806"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2025-01-09 02:44:59.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2025-01-09 02:50:10.000000000","tz":540},"subject":"Remove Forbidden exception for operations.","message":"Remove Forbidden exception for operations.\n\nIf Cinder is configured with `service_token_roles_required \u003d true`,\nit returns 401 Unauthorized to non-service user\u0027s request.\nBut the former implementation uses Forbidden to it.\nTo align with the Cinder\u0027s implementation, this change replaces\nForbidden with Unauthorized for cinder operation.\n\nChange-Id: I07d13a61053519b6ae0ac847b2c6eeeab75801a4\nSigned-off-by: Keigo Noha \u003cknoha@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6431d71220d3fe753a5d6f0c3595e1aeef8bc250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6431d71220d3fe753a5d6f0c3595e1aeef8bc250"}]},"branch":"refs/heads/master"},"ff03e308684aede5591d82a8e6f69920942945fd":{"kind":"REWORK","_number":2,"created":"2025-09-24 05:43:04.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"ref":"refs/changes/66/938766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/66/938766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/66/938766/2"}}},"commit":{"parents":[{"commit":"b23e9fcc280a51bf84fae68e7719bd05c0447806","subject":"Merge \"Wait for the router port activation before deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b23e9fcc280a51bf84fae68e7719bd05c0447806"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2025-01-09 02:44:59.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2025-09-24 05:40:02.000000000","tz":540},"subject":"Add Unauthorized exception for operations.","message":"Add Unauthorized exception for operations.\n\nIf Cinder is configured with `service_token_roles_required \u003d true`\nand has fixes for OSSA-2023-003, it returns 401 Unauthorized to\nnon-service user\u0027s request.\n\nChange-Id: I07d13a61053519b6ae0ac847b2c6eeeab75801a4\nSigned-off-by: Keigo Noha \u003cknoha@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff03e308684aede5591d82a8e6f69920942945fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ff03e308684aede5591d82a8e6f69920942945fd"}]},"branch":"refs/heads/master"},"ebaad6d710271423de6fee919dd6baf1f95cf7f1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-05-20 17:49:39.000000000","uploader":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"real_uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/938766/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/66/938766/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/66/938766/3"}}},"commit":{"parents":[{"commit":"ecf86b6dff3ae35a405ffd24b951b071137f58a7","subject":"Merge \"Add 46.3.0 release notes page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ecf86b6dff3ae35a405ffd24b951b071137f58a7"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2025-01-09 02:44:59.000000000","tz":540},"committer":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2026-05-20 17:49:39.000000000","tz":0},"subject":"Add Unauthorized exception for operations.","message":"Add Unauthorized exception for operations.\n\nIf Cinder is configured with `service_token_roles_required \u003d true`\nand has fixes for OSSA-2023-003, it returns 401 Unauthorized to\nnon-service user\u0027s request.\n\nChange-Id: I07d13a61053519b6ae0ac847b2c6eeeab75801a4\nSigned-off-by: Keigo Noha \u003cknoha@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ebaad6d710271423de6fee919dd6baf1f95cf7f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ebaad6d710271423de6fee919dd6baf1f95cf7f1"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"ff03e308684aede5591d82a8e6f69920942945fd","theirs":"ecf86b6dff3ae35a405ffd24b951b071137f58a7","contains_conflicts":false}},"65f0c1ae4ca5b3e599291d309b068976b9bdc9e1":{"kind":"REWORK","_number":4,"created":"2026-05-20 20:08:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/66/938766/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/66/938766/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/66/938766/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/66/938766/4"}}},"commit":{"parents":[{"commit":"ecf86b6dff3ae35a405ffd24b951b071137f58a7","subject":"Merge \"Add 46.3.0 release notes page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ecf86b6dff3ae35a405ffd24b951b071137f58a7"}]}],"author":{"name":"Keigo Noha","email":"knoha@redhat.com","date":"2025-01-09 02:44:59.000000000","tz":540},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-05-20 20:06:38.000000000","tz":0},"subject":"Add Unauthorized exception for operations.","message":"Add Unauthorized exception for operations.\n\nIf Cinder is configured with `service_token_roles_required \u003d true`\nand has fixes for OSSA-2023-003, it returns 401 Unauthorized to\nnon-service user\u0027s request.\n\nNow keystonemiddleware changed the default of service_token_roles_required\nto True which is what our CI has now. Fix the tempest test to handle this\ncase.\n\nCloses-Bug: #2153382\nChange-Id: I07d13a61053519b6ae0ac847b2c6eeeab75801a4\nSigned-off-by: Keigo Noha \u003cknoha@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/65f0c1ae4ca5b3e599291d309b068976b9bdc9e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/65f0c1ae4ca5b3e599291d309b068976b9bdc9e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
