)]}'
{"id":"openstack%2Ftempest~939329","triplet_id":"openstack%2Ftempest~master~I2e09b5b0fc6b49cb64cfcbe4394121741e0df3ed","project":"openstack/tempest","branch":"master","topic":"volume_snapshot_stop_start","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-01-23 22:59:01.000000000","reason":"Change was submitted"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2025-01-23 22:59:01.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-01-17 02:04:14.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2025-01-23 22:59:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e09b5b0fc6b49cb64cfcbe4394121741e0df3ed","subject":"Add test to verify instance bootability after volume snapshot","status":"MERGED","created":"2025-01-15 10:52:04.000000000","updated":"2025-03-14 05:25:57.000000000","submitted":"2025-01-23 22:59:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939329-volume_snapshot_stop_start","meta_rev_id":"f152ce560794bdc39e56bd859f8da66add90f10d","_number":939329,"virtual_id_number":939329,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-23 22:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2025-01-23 03:24:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-23 19:50:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-23 19:50:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"}],"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-15 12:06:04.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2025-01-15 13:06:27.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"state":"CC"},{"updated":"2025-01-15 13:06:27.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2025-01-15 13:06:27.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"CC"},{"updated":"2025-01-15 13:06:27.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-01-15 15:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-16 00:09:17.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2025-01-16 23:06:49.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2025-01-23 03:24:22.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2025-01-23 19:50:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"1947198cfdbb78eff758badec4fff600831dee99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-15 10:52:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ceb4aa903fbc5884d01ede8c04825ec34d8796fa","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-15 12:06:04.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4bfc303cd0d1c7339f870e99c8b61310ebea9000","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-15 12:06:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bb24ab2ae6f6b676398d497043f0128353ba433","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-15 12:59:35.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"eb1592e02a25fd282c094a3c9014c2e9262fc6a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-15 13:01:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"84f9401ab2290dd906b35eda73e3e4b048f6b3dd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-15 13:48:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90370b06ca80ccbf8b37b6af9eed0605e88ecdc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 15:11:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65443cb3962045f8af3f6dcae74a7813\n\n- grenade https://zuul.opendev.org/t/openstack/build/bc3503a8876843c3948de86d28d0da02 : SUCCESS in 1h 00m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/beaace7753274080868fea3f0796515d : SUCCESS in 1h 19m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdff2d3c0f7d499fa2503e569ce3f9f0 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a11dd49ac0e4d10ba32a7314e4ee054 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a0f4d52d15c47838f4f07ec4814d57f : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ed661fa9b0f4afcadca36f68539f85b : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7af69a2623664e458bbb5da62bcd51cc : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21d39ea7e1184790b55db329846636a0 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9113e197b394a438aa6d5605f125457 : SUCCESS in 4m 18s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8490b50ec94148dcaf522c02a8306912 : SUCCESS in 1h 21m 24s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c87d1cf1732d4dda91a0f409a5ad58e9 : SUCCESS in 55m 16s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46642925457a4613a9388c3e0e4293f9 : SUCCESS in 1h 06m 54s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/56e3d898c3d04740b3e48c19cf98d97d : SUCCESS in 1h 33m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5dd2153fc6d3412aac81feb80c2cef05 : SUCCESS in 2h 09m 13s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/08a0bcdc181146e8bfa7f3a86de17548 : SUCCESS in 1h 07m 58s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/adf8fff58a14439ba3314ecdc942c837 : SUCCESS in 1h 44m 17s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/68b3fe903fdf468dafac8d18fa098414 : SUCCESS in 1h 00m 52s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/cd7f92ec5796420d820c75fef5140fde : SUCCESS in 41m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5a33122d55642b08ce19abc6dcdd053 : SUCCESS in 1h 02m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/37aba3d53b5342e796efab44d1cc2e24 : SUCCESS in 1h 05m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d81161c2a03d40b58afd97de38566bdc : SUCCESS in 45m 49s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e89beb1401f9402e8654fc0e2a7592d9 : FAILURE in 1h 05m 29s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1fe9daccfa774d73a4e002d7025c9858 : SUCCESS in 2h 01m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b06dddcdf4444ddaa1fa386860035714 : SUCCESS in 53m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d4c755315ba4f08ba15638215bd75b7 : SUCCESS in 50m 56s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2da6c8d0d4b5452489997b160ea9ed11 : SUCCESS in 1h 26m 22s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/3008d1a4f8a943458e566aac429b1d89 : SUCCESS in 5m 12s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6385f851473f4dfe9a4fb7141f25f01a : FAILURE in 59m 25s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/08bf7d58d0dc423d918f23e965ecc44e : SUCCESS in 58m 09s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a6df44f612714efb84de0bb1df4ff164 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2},{"id":"1ff3d3d5fa26d228220953d0f50cfc7bdf890ba8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2025-01-15 15:29:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43a6617a38db06186566882972363d187073cd00","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-15 20:41:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5a28baedb23753bf4bc7eb3c675990655fabfe83","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-16 00:09:17.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1427fd4aecbf5340801ad9d05875a864175f21de","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-16 10:17:14.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cb97536e5cacf7123f27767a2964e60c6e2f759d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-16 20:58:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65077166e49e854ff5a8244f6a77aec185090088","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-16 22:00:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7478440e1aa2b0b984955d29a40da8a2141f1ced","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-16 22:00:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3400fb9d44d72c8d9f6bb511cf077b13672d714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-16 22:01:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"23f4bdbb8db4d9ad8e27410f2df96e4db3682c68","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-16 23:06:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61d59a5376532bfd3279b1e91a3447a666c5fb47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 00:30:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b07f7779d9234bc185c222cf4daa66e5\n\n- grenade https://zuul.opendev.org/t/openstack/build/a2acd283c3c54a4098e1e1642564e559 : SUCCESS in 48m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f785850b44f4b95b4dbfa5e840562d8 : SUCCESS in 1h 28m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16aa0dce2bf54d2c8c833f9e3167f23e : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c9dcee96db4c77af2d003bf4cf0a7a : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eb8723822b1406f89ed18cae76f8a58 : FAILURE in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40eb95da65f3473aba3d70a9cf16a943 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89e8181e78a34776a878bee19946f133 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd6bcc80b7154d649e2acaea854e409e : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94d4fc69ac874ca09ff088ff5cf4b8cb : SUCCESS in 4m 12s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4404e3485bc2401c8e3feaa216b0ff97 : SUCCESS in 1h 29m 16s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3e7b6123ad74463bff693ffdfa02652 : SUCCESS in 1h 15m 55s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5afa3f9ed82249978ba2550232521a9e : SUCCESS in 1h 02m 14s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/6d24ff0ff5d2467ca359727fc9d41061 : SUCCESS in 1h 12m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7e5072c5dcfa4acf935856839f72db54 : SUCCESS in 2h 21m 31s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/e1d1c6e402364ae1b5ef95d331742ce6 : SUCCESS in 1h 24m 34s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/d464ab2524864a838db3ff13cd4e6cb7 : SUCCESS in 2h 05m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ae74be70da9a4621810d3db6777da2b1 : SUCCESS in 1h 02m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d6ef90149b0340ed9cbb1bd9dd56b730 : SUCCESS in 39m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c830eeb6407c457e9482f6aee2366579 : SUCCESS in 39m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ecac1470584b4ae8b98ece37b19d764c : SUCCESS in 1h 30m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/edb30fdeec484dc28474b5963be108a8 : SUCCESS in 43m 36s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/da8b437a5d4c4dd58020d7b28cc45434 : SUCCESS in 1h 38m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fc637843f6ff4bf296e5f32fe89253be : SUCCESS in 1h 48m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/da885086b7554fe6b69168b8d3beabf2 : SUCCESS in 48m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a43027f8dd64acbba100d53749a2ea9 : SUCCESS in 1h 07m 14s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/b85961e9f1b3480b80b314171beab3f9 : FAILURE in 1h 01m 10s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8132e22a42ec42a8bef7e63dfddd6899 : SUCCESS in 4m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/852daadfa706486b94ffe80d7621c899 : FAILURE in 1h 02m 49s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8e897e14747840c2931b05c7454cc492 : SUCCESS in 47m 07s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/097ab770de7042caa1f057575010c32a : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":4},{"id":"3622b9c9a8a1d65a0524fdb3810544e4d12890a7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-17 02:04:14.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"70cc5a72c1369063750d253018791450d96dd157","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-20 19:06:36.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c9121a3fe234d555962656bd39df866a40c4e207","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-20 19:13:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2665b06694708c839bb0ec49e4ae89d30b2937b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 21:37:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e28a5828cf9417e9e4a62363c26f6f2\n\n- grenade https://zuul.opendev.org/t/openstack/build/fe87317deb7d42c89f139290e6b9cc75 : SUCCESS in 55m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86bdd30b9bb34638bb1736b344aaffae : SUCCESS in 1h 35m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebd75f7c3ee74072a86ced66fad44c7b : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19bd039f18d4b69999922963ebf682b : SUCCESS in 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83d129031edc4453a7c9e00ca7833f49 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c90014ec5be4c5b803e5782893cc2ab : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fb23575bc55471dba2f3827db0d8749 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/92b56002c34741bcb245ea86d0d3f8ae : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/237dbc75f2fa41829e6cc09fa410af90 : SUCCESS in 4m 06s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0728bd9871b4bd6b630c622bc60b65c : SUCCESS in 1h 37m 38s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ee2b8670771456fb0b0494a549c011f : SUCCESS in 1h 06m 15s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce5e1e1108cd4e969bc1b56068e1cf92 : SUCCESS in 49m 42s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/f70b733374d043c7ace41d6584e74974 : SUCCESS in 50m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aae0570a9c354cfebda52024d39fb9d1 : SUCCESS in 2h 14m 53s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/223f72f9306b4846ad94942b5ab04962 : SUCCESS in 1h 35m 10s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/e5646ef0f17f4b4094c29b03c29c0849 : SUCCESS in 1h 33m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f0d0e449a5cf409a8985e1beaefb2e1d : SUCCESS in 1h 07m 51s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e4fd8df487d14f3a8207b42cbf43c558 : SUCCESS in 40m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6ce02ef80b54d49a89c8e427ae37855 : SUCCESS in 53m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6a1a1c48fa734de5ab09a657bc4951bf : SUCCESS in 1h 42m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f8c98f8ea90d4934802d343340822920 : SUCCESS in 43m 56s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/65c481ecd3ef4aa7a06edf634e0140f9 : SUCCESS in 1h 35m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/27e5cec1e5bd4343a5cee6059fb22aac : SUCCESS in 1h 39m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d54d061166b6445bbd5bfc05fae4a507 : SUCCESS in 53m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4f47b5b7c6704c08994e35fec0fc78e8 : SUCCESS in 56m 12s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/07f6d75f2dfd46bdb7d1a3e6b9267511 : SUCCESS in 1h 14m 55s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d2efcec887244e2c8065b364afa19433 : SUCCESS in 6m 37s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f3bb4f66fa2c42baa821aa4f15b775e2 : FAILURE in 1h 12m 31s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/260fa4e19d884a1cb3272e3ca339911e : SUCCESS in 48m 33s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/82fb4d87b8ec4140b0f1374e476ea92b : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":5},{"id":"18045f0cdc54c073d39c7fbb5cd0b3bb20a9ab32","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-21 03:08:55.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"65ac97ac148cefa416b394d9e3bbf1bc4ac9a1cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-21 05:56:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"7b40539bc594b715b2f5ea9defc750509f823108","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-01-21 05:59:07.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"db73fe884beb8a6016e3ab244a42f72f1c729a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 08:33:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c78351b9624745589bad71a2c7cd151e\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce04ffdfa59c41a9a5652d00d08352be : SUCCESS in 47m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89387aa17890450fb0fd279f3d3ef309 : SUCCESS in 1h 36m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/360b8f993a5349b8b70bfa9bbbcf421e : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db259dbfbbc441989e47fd0e2eb7de9f : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82626e71f714e0abb940399c8671250 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68bb22329d8342f3921cb742f2cd726b : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc8b5a9121b34e15b40445b85b0763a6 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c3cdd66a19d4643a90cf7dae82e3d5c : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5e9ae23e6c7471094839a6a4569f1f3 : SUCCESS in 4m 56s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39d7df8b090e4fb6868cfaffa9659057 : SUCCESS in 1h 14m 36s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/463d7769a9b04c09889799920d210259 : SUCCESS in 1h 14m 46s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9594064026444f8cad65b0dca2b8024a : SUCCESS in 1h 12m 06s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/0cd1cf17ed194dba943023a01e034dc0 : SUCCESS in 1h 36m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8696993a6ae744baab9e6f2225e0ce1f : SUCCESS in 2h 30m 28s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/27815c38b6c6420c875965e18c400e5e : SUCCESS in 1h 38m 18s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/00f91817edb541d9a369f95bdd0f2807 : SUCCESS in 1h 18m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7bc6ad4190e747cca435751f616f3e15 : SUCCESS in 52m 43s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/13773bf61086473992933243b6d041ed : SUCCESS in 45m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d984bb0e77b4ea690a456ec0df888db : SUCCESS in 41m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/319c5a1333b1472d85523ae879862db6 : SUCCESS in 1h 40m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2e489b38a6504217afb72b340b7cf883 : SUCCESS in 1h 00m 02s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/18eeaadab6424cc6951d25097eca6bac : SUCCESS in 1h 44m 35s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/66503fd700ef4e93849c55e1adcfecf3 : SUCCESS in 1h 58m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3bc465fdc41438aa93f392820d31aed : SUCCESS in 46m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6a0d3eb271c438686454b025df61c84 : SUCCESS in 55m 49s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f4e27f64c3f04ec3b84c432dbfa03518 : SUCCESS in 1h 29m 46s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5452e8751a0943b1b439dc3bfc56e046 : SUCCESS in 4m 04s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/01ff2bb154d74dde9a99a095835ca0be : FAILURE in 1h 04m 58s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/22c399a512a84c2cb5a87107c48afd8d : SUCCESS in 41m 58s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/de7215e9c9c8422cbb3e00d158c95ae4 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":6},{"id":"51dd85d5496233e19c6e6864d47ba3640d0f8877","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-01-23 03:24:22.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"59f39b330a8c6e809c1d9dc9cadc28148a2fb737","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-23 19:50:05.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8d9130649d376775a6d94b907ea706ee9272c9cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 20:00:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"54bdf75133c1690909d83debf83f481aa6d80ee3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 22:59:01.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c3657539dd44d38825269da2996b9fb\n\n- grenade https://zuul.opendev.org/t/openstack/build/0ad19a1ee91a47e68a5f64cde3f5d2e8 : SUCCESS in 44m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c56e05bff8e84d36a453067fe771e92f : SUCCESS in 1h 30m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac651c830b14ab28823fa4110409eff : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd356625d504071a6dac27dc515e5ad : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0da071eeb8c45529928766067a58111 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7af510d6c0b1447c9601c125953c41a1 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15104e46079841a392ff42ee8be6c8b8 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/615e08adbaad4146b863d7e3b87df3c7 : SUCCESS in 5m 56s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/70f8cbd441d7473baacfc1f1047075d0 : SUCCESS in 1h 41m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c240de2403e4211b38dc4a31c2abf7b : SUCCESS in 1h 04m 00s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/46f5e8506dd04e43a063a6ab8dcd06f7 : SUCCESS in 1h 26m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/06d80b6af09d4a58a83f203a9ac48682 : SUCCESS in 58m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fb10e1a8fb44060bd998f0f712642f5 : SUCCESS in 37m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d034d77b241943fa910dca8be87fd1af : SUCCESS in 1h 19m 17s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a032b25cbc0743428a3b6df86beec837 : SUCCESS in 1h 06m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/39585311b75f4e318e9750902675aa36 : SUCCESS in 42m 03s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c8f5d930cd5a4947b39f7ddd133c56b5 : SUCCESS in 59m 48s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7079a41a0eda4f7c8d631c8cc59e1ba4 : SUCCESS in 1h 07m 25s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01034dd1bd374634a1189f24703eb459 : SUCCESS in 1h 23m 58s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed9ff7d6ef734a1da036535be32dc3cc : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":6},{"id":"916dca6afb11ab9408c58590a32cff526c7bfe22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 22:59:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"67e84096f961416f310872902e810b9c32b3e82a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 23:00:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f855495a247c4d33a64543b8b05be1d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/332f59140f1a47798c89430e0e426fe7 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":6},{"id":"f152ce560794bdc39e56bd859f8da66add90f10d","tag":"autogenerated:gerrit:revert","author":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"date":"2025-03-14 05:25:57.000000000","message":"Created a revert of this change as I6caab436ef0fba373efe146b8c879c6c01ab3d0a","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"eabbd35510240b58dfe2b728ece633b603ce681a","revisions":{"ac056fda22f5694159ab36b6f09cd2041f8c55bc":{"kind":"REWORK","_number":1,"created":"2025-01-15 10:52:04.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/29/939329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/939329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/939329/1"}}},"commit":{"parents":[{"commit":"a110fc009a832dc74f304cad0faa613a938ff829","subject":"Merge \"Add tests for creating servers with (anti-)affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a110fc009a832dc74f304cad0faa613a938ff829"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-15 10:40:16.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-15 10:40:16.000000000","tz":120},"subject":"Add test to verify instance bootability after volume snapshot","message":"Add test to verify instance bootability after volume snapshot\n\nThis patch adds a test to ensure an instance can boot\nfrom a bootable volume after a snapshot is created while\nthe instance is in a stopped state. The issue being\nvalidated impacts volumes that use an NFS backend,\ncausing boot failures post-snapshot creation.\n\nChange-Id: I2e09b5b0fc6b49cb64cfcbe4394121741e0df3ed\nRelated-bug: #1989514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ac056fda22f5694159ab36b6f09cd2041f8c55bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ac056fda22f5694159ab36b6f09cd2041f8c55bc"}]},"branch":"refs/heads/master"},"c21b286c99ce7c02d2ba4a8bb1ed5d38176324ce":{"kind":"REWORK","_number":2,"created":"2025-01-15 13:01:18.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/29/939329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/939329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/939329/2"}}},"commit":{"parents":[{"commit":"a110fc009a832dc74f304cad0faa613a938ff829","subject":"Merge \"Add tests for creating servers with (anti-)affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a110fc009a832dc74f304cad0faa613a938ff829"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-15 10:40:16.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-15 13:00:31.000000000","tz":120},"subject":"Add test to verify instance bootability after volume snapshot","message":"Add test to verify instance bootability after volume snapshot\n\nThis patch adds a test to ensure an instance can boot\nfrom a bootable volume after a snapshot is created while\nthe instance is in a stopped state. The issue being\nvalidated impacts volumes that use an NFS backend,\ncausing boot failures post-snapshot creation.\n\nRelated-bug: #1989514\nChange-Id: I2e09b5b0fc6b49cb64cfcbe4394121741e0df3ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c21b286c99ce7c02d2ba4a8bb1ed5d38176324ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c21b286c99ce7c02d2ba4a8bb1ed5d38176324ce"}]},"branch":"refs/heads/master"},"2c928066532ed8e1f513dd62edc7331eef74ead3":{"kind":"REWORK","_number":3,"created":"2025-01-16 22:00:12.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/29/939329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/939329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/939329/3"}}},"commit":{"parents":[{"commit":"a110fc009a832dc74f304cad0faa613a938ff829","subject":"Merge \"Add tests for creating servers with (anti-)affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a110fc009a832dc74f304cad0faa613a938ff829"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-15 10:40:16.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-16 21:43:44.000000000","tz":120},"subject":"Test instance bootability after snapshot and volume extension","message":"Test instance bootability after snapshot and volume extension\n\nThis patch adds a test to ensure an instance can boot from a\nbootable volume after a snapshot is created while the instance\nis in a stopped state. Additionally, the test extends the volume\nsize and verifies that the instance remains functional after the\nextension.\nThe issue being validated impacts volumes that use an NFS backend,\ncausing boot failures post-snapshot creation.\n\nRelated-bug: #1989514\nChange-Id: I2e09b5b0fc6b49cb64cfcbe4394121741e0df3ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2c928066532ed8e1f513dd62edc7331eef74ead3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2c928066532ed8e1f513dd62edc7331eef74ead3"}]},"branch":"refs/heads/master"},"f2337f25578c78d82ae7a7851ba3cf203579a8c0":{"kind":"REWORK","_number":4,"created":"2025-01-16 22:01:45.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/29/939329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/939329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/939329/4"}}},"commit":{"parents":[{"commit":"a110fc009a832dc74f304cad0faa613a938ff829","subject":"Merge \"Add tests for creating servers with (anti-)affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a110fc009a832dc74f304cad0faa613a938ff829"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-15 10:40:16.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-16 22:01:37.000000000","tz":120},"subject":"Test instance bootability after snapshot and volume extension","message":"Test instance bootability after snapshot and volume extension\n\nThis patch adds a test to ensure an instance can boot from a\nbootable volume after a snapshot is created while the instance\nis in a stopped state. Additionally, the test extends the volume\nsize and verifies that the instance remains functional after the\nextension.\nThe issue being validated impacts volumes that use an NFS backend,\ncausing boot failures post-snapshot creation.\n\nRelated-bug: #1989514\nChange-Id: I2e09b5b0fc6b49cb64cfcbe4394121741e0df3ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2337f25578c78d82ae7a7851ba3cf203579a8c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2337f25578c78d82ae7a7851ba3cf203579a8c0"}]},"branch":"refs/heads/master"},"30375ce7e5018d5f363147c81e88152fb7f3d14f":{"kind":"REWORK","_number":5,"created":"2025-01-20 19:06:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/29/939329/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/939329/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/939329/5"}}},"commit":{"parents":[{"commit":"a110fc009a832dc74f304cad0faa613a938ff829","subject":"Merge \"Add tests for creating servers with (anti-)affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a110fc009a832dc74f304cad0faa613a938ff829"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-15 10:40:16.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-20 19:06:11.000000000","tz":120},"subject":"Add test to verify instance bootability after volume snapshot","message":"Add test to verify instance bootability after volume snapshot\n\nThis patch adds a test to ensure an instance can boot\nfrom a bootable volume after a snapshot is created while\nthe instance is in a stopped state. The issue being\nvalidated impacts volumes that use an NFS backend,\ncausing boot failures post-snapshot creation.\n\nRelated-bug: #1989514\nChange-Id: I2e09b5b0fc6b49cb64cfcbe4394121741e0df3ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/30375ce7e5018d5f363147c81e88152fb7f3d14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/30375ce7e5018d5f363147c81e88152fb7f3d14f"}]},"branch":"refs/heads/master"},"eabbd35510240b58dfe2b728ece633b603ce681a":{"kind":"REWORK","_number":6,"created":"2025-01-21 05:56:17.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/29/939329/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/29/939329/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/29/939329/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/29/939329/6"}}},"commit":{"parents":[{"commit":"a110fc009a832dc74f304cad0faa613a938ff829","subject":"Merge \"Add tests for creating servers with (anti-)affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a110fc009a832dc74f304cad0faa613a938ff829"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-15 10:40:16.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2025-01-21 05:55:59.000000000","tz":120},"subject":"Add test to verify instance bootability after volume snapshot","message":"Add test to verify instance bootability after volume snapshot\n\nThis patch adds a test to ensure an instance can boot\nfrom a bootable volume after a snapshot is created while\nthe instance is in a stopped state. The issue being\nvalidated impacts volumes that use an NFS backend,\ncausing boot failures post-snapshot creation.\n\nRelated-bug: #1989514\nChange-Id: I2e09b5b0fc6b49cb64cfcbe4394121741e0df3ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eabbd35510240b58dfe2b728ece633b603ce681a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eabbd35510240b58dfe2b728ece633b603ce681a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
