)]}'
{"id":"openstack%2Ftempest~942723","triplet_id":"openstack%2Ftempest~master~I580aafa4a58c7bb38de53cc97d71e927602a6bcc","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-03-10 08:29:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I580aafa4a58c7bb38de53cc97d71e927602a6bcc","subject":"DNM: Test RBD attached encrypted extend test","status":"ABANDONED","created":"2025-02-25 19:00:10.000000000","updated":"2025-03-10 08:29:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f9250a8460e7a0b6475c583cd02e1e82784c69ee","_number":942723,"virtual_id_number":942723,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-02-25 20:57:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-25 20:57:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3eeffda1181dc6fae0efe4cc258dc0f0d21eec71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-02-25 19:00:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23edc2ac7a73b69297a291755c08cfaa158b94f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 20:57:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/099ac85387e3498c863b706140e39214\n\n- grenade https://zuul.opendev.org/t/openstack/build/d302cc4c40f448ef9940e95c0a9fd262 : SUCCESS in 57m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6fcc4b4fc78416ebd404230bd3e6480 : SUCCESS in 1h 37m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/044a87a1a5774580a957a52c8c81fa39 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59935178af6b4a9d92aef5034381d85f : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c87e2f31a3744fdea696524c5de14e02 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef441e139c894290b69254ff085ec4ed : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e14e553ad2ee4f42bcaff76133d599b3 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d64b9a4da9c42839d42987bb54629d7 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9e62de1b73045ae82930b1a6f00b4fa : SUCCESS in 4m 43s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcbcc7deb22d441aba444d62e76083c9 : SUCCESS in 1h 47m 20s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9affb2f1e9794e7a9665ecf60574f622 : SUCCESS in 1h 04m 08s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d74db49435fe44a88dec01ae11e7c7b2 : SUCCESS in 1h 21m 24s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/aff517f0fc3046a1b54a8f846e1c07f3 : SUCCESS in 54m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a1950b87440b4f99b6f4e02467714367 : FAILURE in 1h 45m 15s (non-voting)\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/8315feae13e546f79bf81f5ac55534da : SUCCESS in 1h 31m 37s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/512ec13ec66c403ea5fd26ab24f667b8 : SUCCESS in 1h 23m 41s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4f9987a1b6bd45299e965fca9b04e7cb : FAILURE in 1h 18m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/8a68bd26f014457ebdf002375c6effdb : SUCCESS in 30m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a8b7f3962e24714a8c753d1a6a3c888 : SUCCESS in 46m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/83a96d5a3811489dae4ab73763f5e0da : SUCCESS in 1h 39m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/71afb80ddf384b9988ebe22cf762466b : SUCCESS in 47m 05s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4ce219166ece45d2ae9224a8eff846e4 : SUCCESS in 1h 26m 13s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9b119da096ad4881a6a632f100a228ec : SUCCESS in 1h 54m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1298237bed354b87802f4f0085df7552 : SUCCESS in 52m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/827b9417f68e4fa294ac9a790560e26f : SUCCESS in 49m 16s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/3ca00954af9c47e0ba2eb81842e634d7 : SUCCESS in 1h 14m 19s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ed50b9049a6f450ab0131f1b57214ee1 : SUCCESS in 4m 22s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/41e92794b0134dbeba762af03782f6fb : FAILURE in 1h 14m 39s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/75a6291a110b4215b7771076e40d5173 : SUCCESS in 54m 26s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bcdfeec7b74f4e5eb61b50a8acaeaf21 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f9250a8460e7a0b6475c583cd02e1e82784c69ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-03-10 08:29:50.000000000","message":"Abandoned\n\ntesting is done, not needed anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cee944599dd44f2e321099ddbcb057c6b909e9f","revisions":{"3cee944599dd44f2e321099ddbcb057c6b909e9f":{"kind":"REWORK","_number":1,"created":"2025-02-25 19:00:10.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/23/942723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/23/942723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/942723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/942723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/942723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/23/942723/1"}}},"commit":{"parents":[{"commit":"cb03598a65f47c51406fc86c9c1503fe42424848","subject":"Add compute response schema for microversion 2.98","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cb03598a65f47c51406fc86c9c1503fe42424848"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-02-25 18:59:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-02-25 18:59:03.000000000","tz":330},"subject":"DNM: Test RBD attached encrypted extend test","message":"DNM: Test RBD attached encrypted extend test\n\nTEST: test_extend_attached_encrypted_volume_luksv1\nChange-Id: I580aafa4a58c7bb38de53cc97d71e927602a6bcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3cee944599dd44f2e321099ddbcb057c6b909e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3cee944599dd44f2e321099ddbcb057c6b909e9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
