)]}'
{"id":"openstack%2Ftempest~945160","triplet_id":"openstack%2Ftempest~master~I40596dd46444f463512ef9b198c841af6e9956d1","project":"openstack/tempest","branch":"master","topic":"qa-2025-1-release","hashtags":[],"change_id":"I40596dd46444f463512ef9b198c841af6e9956d1","subject":"Add Tempest gate job for stable/2025.1","status":"MERGED","created":"2025-03-20 19:29:31.000000000","updated":"2025-03-21 23:17:40.000000000","submitted":"2025-03-21 23:15:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945160-qa-2025-1-release","meta_rev_id":"621eabdafdf4f0499e8322238e71560d997f1141","_number":945160,"virtual_id_number":945160,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 23:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2025-03-21 21:27:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2025-03-21 21:27:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 22:43:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"001742155c7a93cd72b01325c45bc52dddf24eef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-03-20 19:29:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e916a90043e7d835d79618b2d4d61ad2fb1b6ed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 22:43:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34c71588985c4c98ac0ad6e99e67799a\n\n- grenade https://zuul.opendev.org/t/openstack/build/530f04e0c6ab44a09d4fe52bee8c96c2 : SUCCESS in 53m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54e7d6d560b74d68aa07ffd97e8b9ba9 : SUCCESS in 1h 05m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0d8a7e7f3184dd7b4c395ad1d3bd06c : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a774ee7da574fb791e61861a8f6bf29 : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e23ed512e31b4d21bd6d46e95a3be6b1 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a80461025dd642cb82f904f82fdaa82d : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a8cf99de9794bfca49be854fc909a06 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7595efd9297f49ecba5055b52ec0b0b3 : SUCCESS in 6m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/24b85d8dfdf64e8d9e61355e5a2e3476 : SUCCESS in 4m 34s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcba9cfb6c9e486da7aed3a0e8c8094e : SUCCESS in 2h 12m 29s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/436739989909466c821a85f0767fe42e : SUCCESS in 48m 16s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63c7b4a229924f0bae9b4e7c4d9f3def : SUCCESS in 1h 21m 52s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/1888e9bc0d774fe09f4482a1afcc51a2 : SUCCESS in 2h 02m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/88c9580899604261b7873b77c84b4a7b : TIMED_OUT in 3h 07m 59s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/8e515aec3c8145b4ad5c6a6bb756ede0 : SUCCESS in 1h 00m 09s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/29895d6b71de48e3972bccbc12a3e971 : SUCCESS in 1h 53m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dc07871719f848cba5a5ce0edf67492a : SUCCESS in 1h 07m 49s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/da89c0549a3d41adaaa3965a2de77e13 : SUCCESS in 46m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac071bf7f7ab4964ba9d3ede45eacf49 : SUCCESS in 33m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/daed9939a557465cb31b67fc9fcdf478 : SUCCESS in 3h 00m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/47bf18b9cca2440e819a836d014a8b58 : SUCCESS in 48m 33s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f694797bc5f84528baae9a9f6f3ace69 : SUCCESS in 1h 37m 05s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0869d4e64b2047c3992104a983d5df1f : SUCCESS in 2h 33m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f250079c5b34418f9eb883b577ab5e75 : SUCCESS in 58m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51f2c51018204d89a0304ab8dc9879ee : SUCCESS in 54m 51s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/05c5fa1a8d9b43839d8ed1ae61414c51 : SUCCESS in 1h 21m 00s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8dc764b9659d40279e779c3bf2a7a997 : SUCCESS in 4m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9d2bda2a682d45ee9b31ffe2896a48df : FAILURE in 1h 05m 03s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/85eba73bdb37464bb180e6644aa4483f : SUCCESS in 54m 32s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f49363b8eb334114b99ea8e0f3d96316 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"df3d9232ebdedf07e1c9f55aa277806e85d2f016","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-03-21 21:27:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d1c03d87fa6fd0a4f0f945873121edc5e482ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 21:27:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7368f06357cf2ea97b1140cb31d71b0d123f5e68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 23:15:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67f67c4ad68e405393b8db01289e2a61\n\n- grenade https://zuul.opendev.org/t/openstack/build/f09a7f17d3f54dc38f3e95413a99edd1 : SUCCESS in 44m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57be98f8cbfd41b6a6268e97ec92c32b : SUCCESS in 1h 28m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c785f59fb5174302b8829bfc032c4b31 : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/618c596739fd48e59d9b45ce81110762 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5d8c29b13ed47faaf21c8656c9515e7 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd4fa9ab69e14bb48b546a91f26e35cf : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b5d50837c4b4fd19d254e2dcd8832c7 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/92ff73e61fd44414b1181ec80ccc1e3e : SUCCESS in 4m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dfb58fb6e55744c682f890d59f4c4fd2 : SUCCESS in 1h 36m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b25bccb8f7f343a19873183a94e4ecc2 : SUCCESS in 27m 48s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/fc1cecde20cf4729ba6df9143e2b6974 : SUCCESS in 1h 09m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9471e8ac8a4d497b8c12730cacc41b33 : SUCCESS in 55m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac80cb770d374b73991aee154d2ce335 : SUCCESS in 47m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8111f8ad0cad44edb538f177525f907f : SUCCESS in 32m 25s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f484839ad9474fcf8f25dc128acc5bdf : SUCCESS in 1h 43m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3b70d7fc92fd4cd693b9989beec4e6ee : SUCCESS in 53m 01s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4d290418934940d7800b1554222d6f31 : SUCCESS in 46m 30s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef0556cda4d544c8837467af8e1399f3 : SUCCESS in 1h 15m 57s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5060042405e74af993f84aa5708d24ad : SUCCESS in 59m 19s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2dde574505f4a28bfb7bca345da69b7 : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c71af467a71f63bce059c5ece93fd89a498c288e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 23:15:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"621eabdafdf4f0499e8322238e71560d997f1141","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 23:17:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a0589db8efb4df79bcb6baa29e9e778\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf089b1be554f1e82f459d3069ea678 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4a8698b97a5433b8e64df408aace2fb14c2cb22","revisions":{"f4a8698b97a5433b8e64df408aace2fb14c2cb22":{"kind":"REWORK","_number":1,"created":"2025-03-20 19:29:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/60/945160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/60/945160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/60/945160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/60/945160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/60/945160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/60/945160/1"}}},"commit":{"parents":[{"commit":"da8a25bfce83bc8e49f15186259c4cbc9819b569","subject":"Create releasenotes page for tempest 42.0.0/43.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/da8a25bfce83bc8e49f15186259c4cbc9819b569"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-03-20 19:28:33.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2025-03-20 19:28:33.000000000","tz":-420},"subject":"Add Tempest gate job for stable/2025.1","message":"Add Tempest gate job for stable/2025.1\n\nNow we have stable/2025.1 branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Tempest testing for stable/2025.1\nby adding new jobs running on stable/2025.1 version of\nopenstack.\n\nChange-Id: I40596dd46444f463512ef9b198c841af6e9956d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f4a8698b97a5433b8e64df408aace2fb14c2cb22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f4a8698b97a5433b8e64df408aace2fb14c2cb22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
