)]}'
{"id":"openstack%2Ftempest~947242","triplet_id":"openstack%2Ftempest~master~Ic6672846f4dd3ccccdcb417a0b50b1d542e1813b","project":"openstack/tempest","branch":"master","topic":"multiattach-detach","attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2025-04-15 15:11:29.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic6672846f4dd3ccccdcb417a0b50b1d542e1813b","subject":"Wait for actual volume detach in multiattach tests","status":"NEW","created":"2025-04-15 13:15:40.000000000","updated":"2025-10-27 16:33:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3672cd99e647681f1afa21ea45ec56b07daa20d7","_number":947242,"virtual_id_number":947242,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-04-15 15:11:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 15:11:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-27 14:43:29.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"CC"}],"messages":[{"id":"9a080cf0de1b4f8fba194e2cbcb2819c1464ad46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-04-15 13:15:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e882ead08ba5415da1b08a5e8cb7f303c46d21df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 15:11:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6425403243444d3ebb6b9da4efa6930f\n\n- grenade https://zuul.opendev.org/t/openstack/build/f89b1b9638964a8fa2e33204e192de69 : FAILURE in 1h 14m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ca6c57f22bc84caa91c222dc9b8dd90d : SUCCESS in 1h 39m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bee60d433a53458dacc83f9f51e69edf : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f2c28550304ee889374853da01493e : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c62080c0b7d4ae9a7b86ae1b544a325 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7e0bb745daa40b09e2365e719121659 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a799d38743c4b03b76cd5619745f558 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aff20079341343c8b6419ac2c710eb53 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/844e8f7d138542eb9749e04433e04ca1 : SUCCESS in 4m 47s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/504eaf25e7a948e2907802d9e1bf1c37 : SUCCESS in 54m 13s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24156ea5918a47f39a656e3ed98c0ffe : SUCCESS in 42m 34s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/375ea89e8e2147e59f3deed26bfc7336 : SUCCESS in 1h 07m 30s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/d8e3c12d13264869b374c6c2f5a57d7d : SUCCESS in 1h 17m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c94cdf117e654cf2bb5c645c885d93c2 : SUCCESS in 1h 53m 30s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/7bd31ca6eff7448b98de63a8ffaee3ee : SUCCESS in 1h 35m 39s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/afd5db144a6b44ed8700caa2e0149347 : SUCCESS in 38m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3f5a4e83c2dc49cc98bc7ee07177d152 : SUCCESS in 41m 19s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4820039a5c614ae7944decfe58b2c636 : SUCCESS in 30m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2734bd5d83ff4bb4b2a470a6c95a8a45 : SUCCESS in 26m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/856a099975b741d4aab94bbd911cb220 : SUCCESS in 1h 16m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c479f3a5cf0a42699f340bdba26032b4 : SUCCESS in 39m 29s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a93ec9072b084e24a3013df6642ffa56 : SUCCESS in 1h 01m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/94707ec8905e45f685ceac1e9a679a63 : SUCCESS in 1h 51m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d0c7e0c290943d295a245daec6044a8 : SUCCESS in 26m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d4f606b0c83b438eb68b09fe250eaf1b : SUCCESS in 43m 52s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ef835774065145b38f6f1d4d548f846e : SUCCESS in 1h 07m 41s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f5ee89a56d924e2a8a0a779303d5ea02 : SUCCESS in 2m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3d3a089d95e54ff582a696be324bc9da : FAILURE in 1h 03m 01s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/21ec6094a4004e6b9d77a52ec3c05082 : SUCCESS in 1h 03m 02s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/602bba5d3b2b432d9d76b5318f26f367 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d7909418a5839d4fbe3ccacd7f79d70ec177ccea","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-10-27 14:43:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3672cd99e647681f1afa21ea45ec56b07daa20d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 16:33:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/049678ab6f1e45539936cf622f534a76\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce3283f607c54eb4b9700fda4aaac10b : SUCCESS in 42m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/771de769aa744246a361f0f3161cab6d : SUCCESS in 1h 42m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f11df625d8e4a9f9d12dc260c7bc6cd : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a9930c956d46eba30e9c3061a3290d : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1bcf31402f4119a0c311f042c5d7e1 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c28d9d3c14d94703b3c4f503de2f3844 : FAILURE in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ada340fa57274c9a97c28ae0872ce755 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/380d5c054a7245bbbe26fedc1e48ce14 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32fb455f201f49eda971510bce40dcee : SUCCESS in 2m 49s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/51bc75b2adc94b8e81423aaf9db84a50 : SUCCESS in 44m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c5b9a59dcac4edf86c4f8b6bdb6066c : SUCCESS in 1h 17m 26s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/f1146dfd8e964ded9fc0aabf4e78a898 : SUCCESS in 1h 44m 45s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/4bb8252d241a4cf7a6fb0429d1602dd9 : FAILURE in 11m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dd515725bc104898a0b18cfa32ebaa93 : SUCCESS in 40m 17s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/12bb3845c8d24429a5dddefc2599a20b : SUCCESS in 23m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6efd24937764b6f8adea85bfb066b9d : SUCCESS in 29m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1f8a49c56eef42f2bbb8e1337e274d83 : SUCCESS in 1h 26m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/76a911fd91e1407aa8a2adcbd1fe8f95 : SUCCESS in 57m 33s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6ee243cf376b4e5fbec3e646ad85e1cc : SUCCESS in 1h 39m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a5136445b39b416d8b4db5602dcf7bd5 : SUCCESS in 1h 41m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/320f9aa47fe84414b2905357d837d94f : SUCCESS in 1h 09m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fc6e130fd9b54920b6abe8b9d27fbc39 : SUCCESS in 1h 04m 12s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ad57a4e63f994dde84f41aa5ec1ec7f4 : SUCCESS in 1h 05m 51s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f0db2e8821aa4485b352f26d65253236 : FAILURE in 1h 05m 52s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/415f24395b54443e93df62b488084ff1 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"263c37e9b7e937c920f161d1cf6c2f880e4ef640","revisions":{"263c37e9b7e937c920f161d1cf6c2f880e4ef640":{"kind":"REWORK","_number":1,"created":"2025-04-15 13:15:40.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/42/947242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/42/947242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/947242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/947242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/947242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/42/947242/1"}}},"commit":{"parents":[{"commit":"60364b39f174749bd5bd255016fc6584ded1f66c","subject":"Revert \"Adds tests for resize server for swap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/60364b39f174749bd5bd255016fc6584ded1f66c"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-04-11 14:15:11.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-04-15 13:14:25.000000000","tz":0},"subject":"Wait for actual volume detach in multiattach tests","message":"Wait for actual volume detach in multiattach tests\n\ncatching \"in-use\" volume transitioning back to \"in-use\" is error prone,\nthe volume might\u0027ve been not even started to be detached,\nbut the test logic already may consider it as \u0027detached\u0027 and proceed.\nAs a result, such \"in-use\" volume might fail to be deleted later.\n\nWith this patch, when detaching multiattach-enabled volumes,\ntests will ensure that both nova and cinder actually have detached\nthe volume from the server by inspecting the \u0027attachments\u0027 field,\nand only then proceed further.\n\nChange-Id: Ic6672846f4dd3ccccdcb417a0b50b1d542e1813b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/263c37e9b7e937c920f161d1cf6c2f880e4ef640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/263c37e9b7e937c920f161d1cf6c2f880e4ef640"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
