)]}'
{"id":"openstack%2Ftempest~948578","triplet_id":"openstack%2Ftempest~master~If7a4d8dca1c10c71d806a8c6b732a3d457553944","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"If7a4d8dca1c10c71d806a8c6b732a3d457553944","subject":"Skip running c9s jobs","status":"ABANDONED","created":"2025-04-30 19:35:26.000000000","updated":"2025-05-12 20:14:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"04e6e00a5aa76860ba9ef51e65d390f59fd45971","_number":948578,"virtual_id_number":948578,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-30 22:03:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 22:03:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2cdb421fe58a6dd98bd10202720ea462533c86ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-04-30 19:35:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e43893347947b31aaa0e3b9394914b7a1aaab23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 22:03:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bbed65e985e42bab4ef6e6814898e07\n\n- grenade https://zuul.opendev.org/t/openstack/build/762e1945efb54fc0acb612f3130c1738 : SUCCESS in 1h 12m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a45ed51031542c5be17a9ae3c535fcd : SUCCESS in 1h 42m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4439f17ec4f44064a37d3666848503d7 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc9d269149984138828a63959fd5c9b2 : SUCCESS in 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e32fdb861ddc4ba7a6012b4b9a30293b : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3301d0afaa0a4dd98ef122b8f17c883a : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3342e245993466f978225bfae7ec644 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4586977989664a879626db774e4ea990 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d35c539b01f14beb90ad6b4034fd5545 : SUCCESS in 4m 06s\n- tempest-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c2896680bdd480f9f5bdfb490084b75 : SUCCESS in 1h 22m 41s\n- tempest-multinode-full-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dbc2a7ea4e64d95a51e40bc38a3b3b8 : SUCCESS in 57m 56s\n- tempest-extra-tests-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc023a8235af4f3c80bed734e9b35bf1 : SUCCESS in 1h 34m 11s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/3c3bb9faa53142798f1d77357b805fdc : SUCCESS in 1h 01m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fb5ccee37e0046418910af9f42ca4f8c : SUCCESS in 2h 14m 49s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/d6912833216945d48cc357d597c75f74 : SUCCESS in 1h 15m 07s\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/093a28eab2d2457b9375052c182785f4 : SUCCESS in 1h 23m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2490d6526a2f4ad8aae4c5ae02727c25 : SUCCESS in 1h 26m 00s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b40e12b7244647ae986d1fba61d65770 : SUCCESS in 36m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89563c42dc9743cf94f4324aa17427a5 : SUCCESS in 58m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35a63258efd1477997097b3d6e12cf25 : SUCCESS in 1h 58m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6a2ec71bb748406b8fada7bc6b4cf1cc : SUCCESS in 50m 59s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d377e1a070af4d81bb8f020dbbcd5893 : SUCCESS in 1h 55m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2719138563644294af0ae43527ca709c : SUCCESS in 2h 05m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d8dbab805e24fc29ca8a9f04f49565c : SUCCESS in 28m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b76c84fab2e441beb7e4fdecea809b9f : SUCCESS in 46m 33s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/089f06272eb944f6a98492ad4149a1b9 : SUCCESS in 1h 28m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f014f9f3b26842eb9a5f4251f0c52b0b : SUCCESS in 4m 40s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2695be6c77df4bd6ae5d051b25e4bb98 : FAILURE in 1h 12m 46s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/3dbf343a3e534511a0999461e9f56e83 : SUCCESS in 46m 07s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/852580e4d75349c08e4cb1944c007d4b : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1},{"id":"04e6e00a5aa76860ba9ef51e65d390f59fd45971","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-05-12 20:14:08.000000000","message":"Abandoned\n\njobs are fixed in https://review.opendev.org/c/openstack/devstack/+/948511","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8490329630fc015b6519a817704db44be4867299","revisions":{"8490329630fc015b6519a817704db44be4867299":{"kind":"REWORK","_number":1,"created":"2025-04-30 19:35:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/78/948578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/78/948578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/948578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/948578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/78/948578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/78/948578/1"}}},"commit":{"parents":[{"commit":"60364b39f174749bd5bd255016fc6584ded1f66c","subject":"Revert \"Adds tests for resize server for swap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/60364b39f174749bd5bd255016fc6584ded1f66c"}]}],"author":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-04-30 19:34:56.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-04-30 19:34:56.000000000","tz":-420},"subject":"Skip running c9s jobs","message":"Skip running c9s jobs\n\nAs master dropped the python 3.9, c9s jobs are\nfailing 100%. Let\u0027s not run those failing\njobs until those are fixed to run with newer python.\n\n- https://zuul.openstack.org/builds?job_name\u003dtempest-full-centos-9-stream\u0026job_name\u003ddevstack-platform-centos-9-stream\u0026job_name\u003dtempe\u003e\n\nChange-Id: If7a4d8dca1c10c71d806a8c6b732a3d457553944\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8490329630fc015b6519a817704db44be4867299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8490329630fc015b6519a817704db44be4867299"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
