)]}'
{"id":"openstack%2Ftempest~955569","triplet_id":"openstack%2Ftempest~master~Ib6f33a240ba78b39a28cb5aad6366cb29c26a568","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-07-22 16:58:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6f33a240ba78b39a28cb5aad6366cb29c26a568","subject":"Use next cidr to the configured one in the allowed_address_pair tests","status":"MERGED","created":"2025-07-22 08:53:33.000000000","updated":"2025-07-22 16:59:29.000000000","submitted":"2025-07-22 16:58:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955569","meta_rev_id":"992f4fb16dbdbe032f9e26fef971cbd52ee2398b","_number":955569,"virtual_id_number":955569,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-22 16:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2025-07-22 15:30:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2025-07-22 15:30:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-22 10:00:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-07-22 11:09:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-22 15:30:27.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"bb357f453927186cc551bc695b920a7e00b5a56d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2025-07-22 08:53:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3e10eeb365b815ef8782da6bc20bc2d2f750891","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 11:09:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b8db6ba2eba4ad8a9ef793c2206c830\n\n- grenade https://zuul.opendev.org/t/openstack/build/408cc776276046efb4ceffcfe03e4960 : SUCCESS in 59m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/200c0ae62811446fae73e817c375edf5 : SUCCESS in 1h 42m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e262a6175dc4a30bbafe3f637170d36 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96952931d57b410cad01c53501d469bc : SUCCESS in 22m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a09bedc355d402aa90eb404405dc996 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ff419ba94674537b257ebd24f1f7ab0 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5038afac5f3346a59970c3c8d5930108 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27f3df25bc2d4af6b0d21b5d1978fc7f : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7aac4c43f066417c93e8e4d59f3dcaef : SUCCESS in 4m 09s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/da7fe193c11841cfa816f21ff486980d : SUCCESS in 1h 02m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/01ddad15e491415ba6a68af750695002 : SUCCESS in 2h 13m 59s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/67744a9d49814887ac2d9d071f70207a : SUCCESS in 1h 29m 29s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/aefb6bd7ba674f7da53660a83c292474 : SUCCESS in 1h 27m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/50703423b075486196acba9d1608d55e : SUCCESS in 1h 07m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/31f60ba4de454717a9b3947c558beddc : SUCCESS in 39m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0044d12230e447e99546d2b53384bf8a : SUCCESS in 52m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/618e3b8a5dc9486f97236f18fed8c8ad : SUCCESS in 1h 44m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dbeb6416b60742429776e7b792d249e2 : SUCCESS in 35m 54s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/acc2b4ff5c8b438d8184075d2f643ac8 : SUCCESS in 1h 16m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/afb4057d531a40a38c736e88d7b421c9 : SUCCESS in 1h 49m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c6ed5ebb8af43bab249805d713e6779 : SUCCESS in 47m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/046d76ab03cb4a5b85495b7a2b685d7c : SUCCESS in 51m 45s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1a1d9112c84249178346ade09a243007 : SUCCESS in 1h 36m 59s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1b9b4f23cb8142f8be455b3edbf14b59 : SUCCESS in 4m 34s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9cadea082a944da6afe5b0cf6d7c7cda : FAILURE in 1h 26m 43s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/45f933d3dc4b463e975f3541eee21ff8 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"dedb8a3e2c29e7a137c5ad2b349c26ae644d1ddd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-07-22 15:30:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e715a464a7f36875791ad2678feff906120959e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 15:31:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e41fe8c96a0eebc78f71e0e0a48832439ebda08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 16:58:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b35b23194b248edacd545ef139eae45\n\n- grenade https://zuul.opendev.org/t/openstack/build/c2bea20f51bf412b9c67f9dad68c1acc : SUCCESS in 30m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b56e0a5a3f284d5cbc064ebba979bc17 : SUCCESS in 1h 15m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b272c508a244f27a05f03063eca287d : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa795e0df13c453d900e0c1bc155292a : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7443fe9961f74cbc98420c01640f8682 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cd04adfb2de427da2ddd5ed28251bbd : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/518651a4ddb34982b68dd643213570e1 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/815ecce8fd7e483392f5b5d4234e0eaf : SUCCESS in 2m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f230599571c04d708915b0eed9ed68e2 : SUCCESS in 56m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/429cd43c84da479a904068c2f34d7cb1 : SUCCESS in 58m 18s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/cfa9426e52ca41428a82f93eab69c8f1 : SUCCESS in 48m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f070737aa364453836e26b2dbac08a8 : SUCCESS in 33m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55c908ac71344b6d96ca358b3421413e : SUCCESS in 32m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1b8e8637e57d4c4ebeadcffb9b468af6 : SUCCESS in 35m 37s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7770e26089fb4e51a357cc174bf3da58 : SUCCESS in 1h 16m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c7d7c89e50eb42b5a5c94e1508a35fec : SUCCESS in 1h 26m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/093437e549154e5a9024e59616e5eb07 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8f28d36fb0e66803fd66c4ac40fd158ccaee517b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 16:58:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"992f4fb16dbdbe032f9e26fef971cbd52ee2398b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 16:59:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d9481d9d5124aeeb44bc4cd770cc0e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d210ecec769f468d9a7cbf03a1d31f57 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d9888abea78865849d471637c0df9e5bcd3188e","revisions":{"7d9888abea78865849d471637c0df9e5bcd3188e":{"kind":"REWORK","_number":1,"created":"2025-07-22 08:53:33.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/69/955569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/69/955569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/955569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/955569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/955569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/69/955569/1"}}},"commit":{"parents":[{"commit":"905cb14d8d6d7225356d88a1db7f03afcc54a435","subject":"Switch back the tox constraint to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/905cb14d8d6d7225356d88a1db7f03afcc54a435"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-07-22 08:48:35.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2025-07-22 08:53:29.000000000","tz":120},"subject":"Use next cidr to the configured one in the allowed_address_pair tests","message":"Use next cidr to the configured one in the allowed_address_pair tests\n\nIn the\nAllowedAddressPairTestJSON.test_update_port_with_cidr_address_pair API\ntest \"self.cidr\" was used as cidr to be set as allowed_address_pair for\nthe port. Since [1] it is forbidden by the ML2/OVN backend to set IP\naddress assigned to the metadata port (owner \"network:distributed\") as\nallowed_address_pair for any port in the same network. This constraint\nalso applies to the CIDRs set as allowed_address_pair if metadata port\u0027s\nIP address is included in that CIDR.\nTo avoid the issue in the tempest test, it now uses subnet which is next\nto the one set as self.cidr - that way it will for sure not overlap with\nthe IP of the metadata port.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/954570\n\nRelated-bug: #2116249\n\nChange-Id: Ib6f33a240ba78b39a28cb5aad6366cb29c26a568\nSigned-off-by: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7d9888abea78865849d471637c0df9e5bcd3188e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7d9888abea78865849d471637c0df9e5bcd3188e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
