)]}'
{"id":"openstack%2Ftempest~957612","triplet_id":"openstack%2Ftempest~master~I3a593b4acf1322ef954661dc026548d79134cffd","project":"openstack/tempest","branch":"master","topic":"glance-image-cache-clear-tests","attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-08-06 09:55:17.000000000","reason":"Change was marked ready for review"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-08-06 09:55:17.000000000","reason":"Change was marked ready for review"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2026-08-06 09:55:17.000000000","reason":"Change was marked ready for review"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-06 09:55:17.000000000","reason":"Change was marked ready for review"},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2026-08-06 09:55:17.000000000","reason":"Change was marked ready for review"}},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-05-31 19:20:02.000000000","reason":"Change was marked work in progress"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-07-15 14:55:13.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I3a593b4acf1322ef954661dc026548d79134cffd","subject":"Add test with delete specific image from cache","status":"NEW","created":"2025-08-17 19:12:20.000000000","updated":"2026-08-06 09:55:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":35,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"957d60310dd7b5567845391b9e54d2479b0e261d","_number":957612,"virtual_id_number":957612,"owner":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-05 14:14:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2026-08-05 12:23:41.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-17 21:20:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-22 10:42:00.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2025-08-22 10:42:00.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-08-22 10:42:00.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-08-22 11:10:25.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2025-09-11 12:18:13.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2026-01-21 04:32:47.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2026-02-26 14:15:28.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"},{"updated":"2026-05-14 18:41:59.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"}],"messages":[{"id":"59cfddcc3cb49062ca0fa1c950def71ca0d81d45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-17 19:12:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04ad8d3342925d7b192cd1a887259173e6b4213f","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-17 19:14:44.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0752c4dddf771e9623f16e34f79996e9a5bb12ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-17 21:20:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d9b66d9e6654a44ab52bc5c57c47c41\n\n- grenade https://zuul.opendev.org/t/openstack/build/3107f0dc44c44fb98c5d881f76064b6d : SUCCESS in 58m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b39dc67c806749f9af312cab38f2b204 : SUCCESS in 1h 38m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e73d78ae785442dca7059238e3fe32ae : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e433e69d25b845c386c1d3987671cc2f : SUCCESS in 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90dfa28fd13e499bb9d3fc55f382738a : FAILURE in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f0edc6e2fa4471c8f21c356048f6700 : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cd8fb440d0f409bb1dbdd9f7a6e7908 : SUCCESS in 7m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f4a5a8c9176425db9907acd14559bfd : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc86fdbf2af74dd5819844b3619bca0c : SUCCESS in 7m 00s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/4f8772a881cd4b719606f659cae7e0b0 : SUCCESS in 1h 34m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a210fff754b84eefa2981cf43f4327cf : FAILURE in 2h 03m 47s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/d878c50842e74d7b8736d56bcd620e56 : SUCCESS in 1h 34m 43s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/a70d975be52d4e62947f09f710549d28 : SUCCESS in 1h 18m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/50172fbc0f33458cba06a9a7449bfe01 : SUCCESS in 44m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/95714b5d591d438a96ff0baca13fb377 : SUCCESS in 46m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bc002493d5d43c9a0daa6b75f205a12 : SUCCESS in 56m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15ca6d251201450ba8e67bb033fafd88 : SUCCESS in 57m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f32a331de69e43c898c178f8f6fd09d8 : SUCCESS in 53m 03s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9eea91044a494d428cb07dc21df00abb : FAILURE in 51m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4eeafd7bfdad4976a0b1cbc0052f4bf5 : SUCCESS in 1h 57m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/afff78e5a62549df989a8aedf459587b : SUCCESS in 1h 01m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/642efa0b1b2f44058693b8272181f15f : FAILURE in 44m 46s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/cfc5c5f8b42a4a6691969d51fc7d7149 : SUCCESS in 1h 15m 54s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/68d2e72701e74ae2b41c85f5f7c9b519 : SUCCESS in 2m 45s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ef89ba6dec37457c9f0dec3c4f9c8571 : FAILURE in 49m 49s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/11c25074fdca45f7b2dcedf4e3c05a66 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a7ad114889d3604b15c87ea51d3802e4408e953f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-21 09:14:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2946dd0457e3fc3d11c1c37a051dce3e401cba7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 11:53:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbe6c8c3a198413795be6468bbebf708\n\n- grenade https://zuul.opendev.org/t/openstack/build/b376acfb11ea4a6fa4cd0f1b15ed4102 : SUCCESS in 1h 11m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f06513a997bb4189acaf3e3a7fb68b9b : SUCCESS in 1h 45m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb46b48d01b7411197576f10a134ced1 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ff586fe4df4d4b9fee50d86f7d50bd : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0180ec136da4234bc373b03f90a8e17 : FAILURE in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5a18b6bab3d414fa00893a882921d52 : SUCCESS in 7m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6044bc92cc7f48e3a0f326ac3dff4381 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36be2a0c219b4a73b29250c4cb874ec7 : SUCCESS in 5m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d10ae16cc36541489364f109483f4f17 : SUCCESS in 5m 57s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/c9ee2f168d0c411e9cd00cde9e1b5e9e : SUCCESS in 1h 13m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b24cceb9977543a8ae9fb5919f63ce5a : SUCCESS in 2h 28m 51s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/9ffff1dd35be4b7696b2670816327820 : SUCCESS in 1h 05m 44s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/5266ec48cbee4d7396fd5c99be420b44 : SUCCESS in 1h 49m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0bd5e12aa6c5401f930cf664767936d9 : SUCCESS in 1h 22m 01s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c1b2c78240ca4edaa9ec462d71605734 : SUCCESS in 16m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bafe446c87b4f63b598647541ada97a : SUCCESS in 35m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4ef38984cb33413d9777761e34481b4e : SUCCESS in 1h 44m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f9afb9aa388644a9a22c806d1bd713f5 : SUCCESS in 49m 34s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/878229c937534efbb64c2238f3b46b91 : SUCCESS in 1h 20m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4bd48d2df27447a4aad955a4bbba13c5 : TIMED_OUT in 2h 34m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e094eeb18f1a4e699d21bd7f7ae7ea26 : SUCCESS in 56m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d4c81669216649d79163a09c13d56fbc : SUCCESS in 33m 59s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/4568970696d14df2a0e9db306dd408c0 : SUCCESS in 38m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7865bee5fb814c19bb06e6cbd37b59da : SUCCESS in 5m 03s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/901c1872e7944cd9889e782add0827a7 : FAILURE in 1h 01m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/626f76e612ac4774a1e5f0e60762b84e : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2},{"id":"68f5ef5f5f96db40d9149f3c94badf0460a79c45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-21 19:33:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2f451ef3476968534c09011949eea5aab3df7c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 21:29:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2696a0a522394c8da4d2dac529ed43c7\n\n- grenade https://zuul.opendev.org/t/openstack/build/cfa580c6552c4731860fd45cea3e08c6 : FAILURE in 32m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a83b85d8b0674d6a92c52982dda0c7c6 : SUCCESS in 1h 51m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62a39acd57904d5e8742a9f8956b2328 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7266f86abbe40f29d9c5360bf6e6a43 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3db8056291049d690af9e3aead28d7e : FAILURE in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/724b238c8c104b3a9ef655df93e12b28 : FAILURE in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93b1f0eface74648a0ab9ecacb001818 : FAILURE in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7e0431e130c498788a7ebe789372ca2 : FAILURE in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70ed2d91b44d48cbbbd7917e805df292 : SUCCESS in 5m 44s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/a8b610cd5bd74555a08b35dc9679b842 : SUCCESS in 1h 25m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/15e6804faa1848b696af49cbf3e82b4d : SUCCESS in 1h 18m 15s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/63a4f7aa3a9f415b9f44e804c80b7264 : SUCCESS in 1h 47m 37s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/09cae1d1845c4903be84ffec886b1ff0 : FAILURE in 11m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0af223caeb5c4f35b5e13f278b346a62 : SUCCESS in 1h 22m 11s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5a5eb8e98e0841dd9acc2318d939fcb9 : SUCCESS in 38m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/175cd06718644325aa5e82369fb97c7f : SUCCESS in 33m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9c75f4b2346b464aa7ce98bd028dbea3 : SUCCESS in 57m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/259892ccf0044c8c9abfa05dfe896443 : SUCCESS in 57m 28s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/414c3be5db5142c1892ab5e8e72862b7 : SUCCESS in 1h 53m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ac2997ed1c08427384232df7914cab6f : FAILURE in 28m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/de7881334b184ce2a63e49fbdda25f73 : FAILURE in 32m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f2ddcddf35c453e8d89c6f1a5c8cd87 : FAILURE in 25m 30s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/217d8b85fe6d47328b4a53f42458af3b : SUCCESS in 43m 53s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/03fa9c9c490a4f4fb8e9ccd83410baf2 : SUCCESS in 6m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f9c89462180d45e7a7344b877cdb1f9c : FAILURE in 1h 26m 37s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a6556fab9cf44f1f8fc4bf1e01074a74 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":3},{"id":"1fe65cdf06d048604223f0b25999c3da163cff59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-22 07:12:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f02fbfa1a3b752decc74540db7239c6a011bac09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 09:38:48.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7e269c5e8db49ba8475551bfcfde3d9\n\n- grenade https://zuul.opendev.org/t/openstack/build/b67faeb2b5fa43cbb18d269772331877 : SUCCESS in 1h 01m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b7d5087cc0f411ab6af23eda043c506 : SUCCESS in 1h 38m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc7ef81979de4f23b1cc9bf0cf66d115 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c02eeaab6df41eb8e9814bc0b7ad5be : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e5c67c6b614a2db8e33ec392e6bd9d : FAILURE in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e6fb35c64ec45cc9f8a028855588ef5 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/337c2c1f5a1c4f64973948b2e2d878f4 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b210e9274acf4f83a31f1364083056d5 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/210ce5e1499b46efb60be74583f426cf : SUCCESS in 4m 17s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/e3c009c50d794f53a4aefa4cdb7d76bb : SUCCESS in 44m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/17916b06b644402e85d5bc84303ebaff : SUCCESS in 2h 21m 51s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/c2cf1f4334024266a48062a2ce762a36 : SUCCESS in 1h 53m 19s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/36e97e9754a540e68fa5c8e04126c192 : SUCCESS in 1h 36m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ee64f1f4073046dbb32efb73727c0f06 : SUCCESS in 1h 27m 49s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/00a7c859899f45fbb7439d7d10dbc8b8 : SUCCESS in 16m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28d50a65bb1e4d0d94f41c5f54fb3517 : SUCCESS in 29m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d33fa1d318a4eea860964826c5280dd : SUCCESS in 1h 51m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bdb21e45adeb4f53afa66351d61a3130 : SUCCESS in 43m 58s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/77e7be30765f47c283ab07ceb225e994 : SUCCESS in 53m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e08aaf66195e4ff69e539bbf8863f199 : SUCCESS in 1h 50m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/00e03a85a2474e61886b3080c79e3e5b : SUCCESS in 59m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e82cc3b9d19b405d829668d34aa68e47 : SUCCESS in 1h 00m 35s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f62f43dc066944eb95d47136d93f4987 : SUCCESS in 1h 05m 21s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/943697089a5b40e08a138ca9262f30de : SUCCESS in 4m 39s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/54580dbc49a143a0b18550dbbdf844a8 : FAILURE in 1h 26m 01s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3001e056b73b4bf091b01809bb1cd970 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":4},{"id":"f01af57febd812aedf673d5ba025c81bae6c7b18","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-22 11:09:48.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"7dee01526b2f8ce68aba9f6178afb5bcbc82f700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-25 11:25:48.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"50e325dc424e20769ee763f282dc1ae3b9dd7eb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 13:29:21.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e13e5957146b4baf9308492ace2bd592\n\n- grenade https://zuul.opendev.org/t/openstack/build/8ca75f8b6bc84a7aba112fb833288383 : SUCCESS in 33m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/901dcdaa6aa4479f9a1d24947c273830 : SUCCESS in 54m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eca419686d6b405bbc0d5ec137c60e4b : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36feb11a4fb4ff8b8a77d2ee89ca8fc : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419ff0e667914c1bab122c337395b2c7 : FAILURE in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c62f4b07f104be78f9bb903d6f64ae5 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31a4592d68e84c4c9752d6b5aa204a3c : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8177b250dda243c497437b7ec1909a7a : SUCCESS in 2m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9641db457b244c6bb27fcc6400b435d7 : SUCCESS in 4m 34s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/8e7986f15219477bbed6cced2c1314df : SUCCESS in 42m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a9e3bda9886949faa89e49f0dfc6d98b : SUCCESS in 1h 30m 58s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/1917e8132770497a977068b4b0f204d5 : SUCCESS in 1h 29m 24s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/5c7eaee8a82049d2afcd6daa92c073b2 : SUCCESS in 1h 25m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b208b7ad9a2d4172bbef39639214c352 : SUCCESS in 1h 11m 10s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/dd8607b6f6534727bfbe0d80f653dbaf : SUCCESS in 42m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2164a5906934768aac32d2fe4be2390 : SUCCESS in 53m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/40aff41e9b5847ec97c21b47cb75fa5a : SUCCESS in 1h 43m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/44e802dbc36c4059b4d806b9bba9ff52 : SUCCESS in 51m 40s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a55b4aebbb514b53a4e0fbdfb78786bb : SUCCESS in 53m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f4a90ed52561420197573d2ef629780d : SUCCESS in 1h 57m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b358106d20bc4f3d9cc63009c388e3ca : SUCCESS in 57m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f9f8d357c88492899a8ac8759286028 : SUCCESS in 55m 47s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/cf6a54aec460453d85ef824c8ee5aade : SUCCESS in 1h 10m 23s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1f1969979d6c4bf69b28cfba6bb014bd : SUCCESS in 3m 46s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b226dbab31e941ca83a97717d244c891 : FAILURE in 1h 20m 02s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d02f8c49e501493eacb21bb60ee400eb : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":5},{"id":"69ef4bd74f33a9ab0526dd7d4c288bd2aaa3a036","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-28 12:41:20.000000000","message":"Patch Set 5: Review-Priority+1 Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"73c450ed7907a97364b33fcab3645a647b81bd03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-08-31 10:44:43.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ae27a8c14ba12e6ad1bac5870c05c9f079f9ed5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 13:02:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/194ae0321a674360842c6c39f03d00c9\n\n- grenade https://zuul.opendev.org/t/openstack/build/b3de9d898b6f42149b0fd6cf072a34eb : SUCCESS in 59m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cc779a6a10c4415bed10ce4b244b3ff : SUCCESS in 1h 35m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a87bdd322e884cfe9d0b864f991a013c : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15de29518de74907ad08139d120a6c4e : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a5774d8364742fd9c4d87808782b3ae : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c41276847208444d92337ae0073524f6 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1271eb0865d34c3f8edb0bad2b8e5ecf : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae470f089b9545beb05fe1e6c17ca73f : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c54a250c96e940ef88f7cb31bd6913bf : SUCCESS in 2m 47s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/866a86d231024c3d99dd07e58560762b : SUCCESS in 1h 24m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/af7074bcca334f4d917ca07737329b6f : SUCCESS in 2h 12m 48s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/c7ab09b0ffbd4f8bae06f1b98e1706c0 : FAILURE in 1h 21m 58s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/85b6e9434fda4d98910e984e6c22c2b1 : SUCCESS in 43m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/69e708468cd040daa7ebfbc91e06910a : SUCCESS in 1h 14m 35s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/19fbba951fce47fa8213dee9b6c5da80 : SUCCESS in 16m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2539f821a824fcab3c066efd8fd60ed : SUCCESS in 55m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd950a44bd2c4e5aafe846dd9cc0c066 : SUCCESS in 1h 29m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/09b6e881637e437d98330602c17a56a5 : SUCCESS in 32m 47s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d392636663044efeadf5fb209f330131 : SUCCESS in 1h 42m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ef74f2b2d5dc4724b2700ce9ed39beeb : SUCCESS in 1h 45m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c530237eb73477c8cc4e6f88cc6c9ff : SUCCESS in 29m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/51f02c609287460ca9354de447fe1f85 : SUCCESS in 50m 20s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/be34ff4d3be44005b9c9ddcb225c8ec6 : SUCCESS in 53m 49s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9062f010fe404c4aa93197e72ee5c5b6 : SUCCESS in 4m 25s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2e73cf2f53be44439401a1c6a78bade4 : FAILURE in 1h 29m 40s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1a82c373c97d4903ad0412dd97abf7f6 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":6},{"id":"92735d11958bbf75a6223235222d133f362c1cc8","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-09-01 06:14:50.000000000","message":"Patch Set 6: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"53c222313b31f64d0d40dbd2533e6b0d2b061719","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-09-01 06:15:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4fc920355e8f841916c2b736833142e7e374a615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-09-01 06:17:39.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"491720327755adeca81286e6374d9034d9cb1e9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 08:27:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3adf01ed4bd4d77a416d7ad57a376da\n\n- grenade https://zuul.opendev.org/t/openstack/build/8f5215198db24d06b19a039d95d84bb7 : SUCCESS in 45m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0640b15e4e1046ac818285d82e0304ac : SUCCESS in 1h 17m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df613ede945c483593a2e340d44c6b29 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e2d677c58e40c19845d8e05c3551c2 : SUCCESS in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c26139352b34ecc8710e6ca46cb9f82 : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b8fae48f8f14ecd8e9d27cb5b78fa7b : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c427713ce85a4ab18960783221e0851c : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d295fc9392a94cf09d62ee23dea56f49 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55bbc4874dca46aeba9db2db3cb285f8 : SUCCESS in 3m 34s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/94c27d4829cf4f8297de8c1d50ac8803 : SUCCESS in 1h 17m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d84eb3da1fb47b8a042d85728d186f1 : SUCCESS in 1h 22m 12s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/d22ff5bfde3a47f8823ce5ddc20e8fe3 : SUCCESS in 1h 26m 06s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/4b4ab040a8ce40c78e72d2fa93abd7b3 : SUCCESS in 1h 22m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c9f74ea08a62401bbb8cae6ce3cb4c9c : FAILURE in 1h 20m 45s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5b7e1203089b4e9aba8cfc697cf20479 : SUCCESS in 36m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59ece847e0b84785850696638faa28da : SUCCESS in 1h 00m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/225faf426a394a57a219ed20d9171fd5 : FAILURE in 1h 57m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f4249a43ecda4a9ba1b48d48537fd855 : SUCCESS in 46m 34s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bc7fd9a3d73a42bd9bf992d57f94cf39 : SUCCESS in 59m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/36ad85b23ea04bb884b28d92a44f067c : SUCCESS in 1h 21m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/46e606ff3fb64887bed8e3fca29ed666 : SUCCESS in 29m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/39e814833d64417e9a13c769c2b712f5 : SUCCESS in 1h 01m 13s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/3fb0e41b4c8e46518b3b0d2859d8d595 : FAILURE in 32m 02s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a7a00771b8ca4481b55383efdbd23d0c : SUCCESS in 3m 26s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c82e5bf12f064c5ca6814b602282d9ca : FAILURE in 45m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b3753bcc717e41a785c5443922a5bc74 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":7},{"id":"c4121f4387bb301bef8eac2d943adf0e9c66823f","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-09-04 11:06:35.000000000","message":"Patch Set 7: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cbd8d554a10ae65bbb8d693f4e1c21c787142c47","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-09-11 12:18:13.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"138e424e49025cea6119544fa74b98778a3f9581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 15:40:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba8a8f38feb44e258b8b9f281d33258b\n\n- grenade https://zuul.opendev.org/t/openstack/build/6213660f153044b198514145fe58ab4a : SUCCESS in 53m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/896f9039b71f49a7bbf43f86f1949b93 : SUCCESS in 2h 06m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac3d3449ce434d9ca195add74894edb2 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f32cf9936ff45dd89dbf5c913ae69b3 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b904f9ed13ce4d4fb6678b65d40d0cba : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/451c7b925ede4bb4b4b2cc35ba6647f4 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a6653624fde471980d9d1e13d232e63 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e0755d2bf184a869b07ed9a097b7537 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0590cfc5bbd54f0797544a6abc3d1809 : SUCCESS in 3m 58s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/eb04384b86cf4862bcf17c2f5a0d8fc0 : SUCCESS in 1h 02m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/051df3aa2bec43fc83aab9bcaf36ff70 : SUCCESS in 1h 11m 49s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/4fe9caf35e1949cdad7312fde0fd864e : SUCCESS in 54m 35s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/0233c501ae064beba82909fda03c217c : SUCCESS in 44m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7b52f8521662429785fd725933bc4cb6 : SUCCESS in 40m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/064fbe20e55c4455aae11550b9867e0d : SUCCESS in 44m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/484a473846b34aa0a5f6202fa71f9952 : SUCCESS in 31m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75902c0a6fe74c9289ee1e0ba3f8973e : SUCCESS in 50m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a499b9b5c002467cba9e30afde529957 : SUCCESS in 36m 01s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8ac4b85411834b78a134710549a053d4 : SUCCESS in 56m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38d0aeaa00cc4ac08bd8901573dbb808 : SUCCESS in 2h 01m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9842f4edb27c433f917bd9ad5a103369 : SUCCESS in 57m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0309e41c6c084f4d818eb48cf458bf54 : SUCCESS in 59m 40s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/8db96dc705cc4463bcd3c735b3a705c3 : SUCCESS in 1h 12m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/28d77347dedd4ad3963b421233a89d12 : SUCCESS in 3m 58s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/fed1b47237014a05bf77ffe1c913e397 : FAILURE in 1h 17m 27s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f3a9e625b8924d069f1556caa5454c9f : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":7},{"id":"131aefe3dedafc8064a00186fb611aa205babed9","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-01-21 04:32:47.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"615d36cb62a62231fadddb8df9dbadee51c21bec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 06:35:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3778770ce44d5dafea26ca3e313a86\n\n- grenade https://zuul.opendev.org/t/openstack/build/830b30d9ebe949a8ae3bd5af90aa0ce2 : SUCCESS in 59m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8303ce08b37e4548b51bf2be6547b16d : SUCCESS in 1h 44m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7032881b2ae9482196e24c087a88cd35 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec61753f56b49db99dc2e0fbf395b0f : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1110a26222834a919146d160bbd905b8 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9078ccc36e464240915576be545e0477 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1a199e4ae7849f68c5a363b19e44ea1 : SUCCESS in 2m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e27232bebf3141c1a4e210828450e92a : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fdec5645a4b41978799c55c05582650 : SUCCESS in 4m 27s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/f516ef1303374398a9bd0d06ba106468 : SUCCESS in 1h 05m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e7904e7bd5e04e1486b4840301bc00c4 : SUCCESS in 1h 20m 34s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/7b73da6d790449ecb9bf12eeba3db0b2 : SUCCESS in 1h 42m 09s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/d5c148c049e440f6af9fbed0963f440f : SUCCESS in 1h 17m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/62b1162434464dea916c79c8bfb8d12d : SUCCESS in 1h 15m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9be587f31df240189a67fe3c9d00a4be : SUCCESS in 17m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/393e4a6d5f994200abfa3bf74e315578 : SUCCESS in 58m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/431c8ce79e544a18a338cf4700aaf78c : SUCCESS in 1h 10m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b24ca893c48f4aa3b859819fb4252e5b : SUCCESS in 56m 16s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1b6971639b8f4ee3b804b3e9fa8e3285 : SUCCESS in 1h 08m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/136b08cbb33145eebe18c92af668d150 : SUCCESS in 1h 58m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5fa3dd7508a4458e8ac1b88fb8b55ee8 : SUCCESS in 1h 13m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bfadd30c4fb04f04b9a1e014adcbfbb8 : SUCCESS in 45m 20s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/3bb3e7cd5a094b92b956c0fd24364337 : SUCCESS in 1h 01m 13s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/831f73659f554d7f8437cb20075caa88 : FAILURE in 1h 01m 52s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/e462ddf877af4589912aded56c066a19 : SUCCESS in 36m 02s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7396757015a94e418018643bc463d262 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":7},{"id":"0fd61abc640f732a768d28b777d88832d0bf6fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-04-14 09:47:25.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"85816119e551c971d2c8ae8efc7b109a91c3c949","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-04-14 09:48:28.000000000","message":"Patch Set 7: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d14b9a95ffe4dd2f0b93c430903db4ad99e93143","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-04-14 09:51:12.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n","accounts_in_message":[],"_revision_number":9},{"id":"c419a498d1a6386f0e12a2634fdd1f5a2b730ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 12:15:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c4cf15ef0894e04aaf7930cbb42be13\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b42126fe0b84fb58f00fe10462615c3 : SUCCESS in 37m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f68e59fdb074432ad86108ea88ee5e2 : SUCCESS in 1h 57m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a17fcd07c7c4bb589a8affe502ed733 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85aa1f9b17814783accdc35e7332733d : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bdfa996eb194c1b8b0256380dc459dd : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a88e78f75ca24ca6907ad9390f7244db : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/601d3e80a80047c5b5859d217d6161f1 : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6add885c66c4248a8353cd2973b22c3 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d137c25c93464248b83c5010e49c0426 : SUCCESS in 5m 17s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/e6b21419bab24d3ba82fad981853e98f : SUCCESS in 1h 32m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/95e7840f3f024bb3ba984019259838fa : FAILURE in 2h 08m 17s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/56e4c420b5244e7f87562774eb24be8d : SUCCESS in 58m 36s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/03bca99a0d2840898d1cc0379f75f5b5 : SUCCESS in 1h 36m 23s\n- cinder-tempest-lvm-multibackend-2024-2 https://zuul.opendev.org/t/openstack/build/c5feba075c874d8998608c5449b1f6ad : SUCCESS in 1h 41m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aa9d663cd86f4467ae1a95aad1cf7adc : SUCCESS in 40m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a8faae55e5ea4a0189b30e180a09f5a2 : SUCCESS in 13m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1ab31378e864124ab8caf8f92f0dfca : SUCCESS in 33m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/211752205134427ca242e03cfcc741a0 : SUCCESS in 59m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b047ba6853da42798d4a0ff578917dcf : SUCCESS in 54m 55s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/deeb0957932e4d829fb69bd110542cb7 : SUCCESS in 1h 00m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d48f0389d0f14e7a996489c20816bded : SUCCESS in 1h 51m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/894558df7c1c4e83a46202e8a0cd1ff6 : SUCCESS in 1h 09m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57ef2517236742e69b8aa944bcd97918 : SUCCESS in 35m 08s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/37bb605ef7f945caa0aa971caea4e530 : SUCCESS in 1h 15m 43s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/7b39743fa18e467abbd9ba97e9309821 : SUCCESS in 1h 45m 57s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/85a54be8afee4d7f9016dc1072c01eca : SUCCESS in 53m 19s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ea4ea96af435422faf8ba64eab08e372 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":9},{"id":"3333596fedb311a3ce82355b5b141d10e50fcb50","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-14 18:34:16.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d35b0849042d7196d5dbbc6cbc1f40bd9a3bc5f5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-05-14 18:41:59.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e1365a431b543065b4739f6adb1f5d2253142f38","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-05-18 13:19:24.000000000","message":"Patch Set 9: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e90e090b96e8669abdef6edb0bb4b016b94324da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-05-19 17:47:56.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"b1cbbb36593590e98d6063e09c2507bf5478c2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 20:28:29.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f91eebd8129d45bf8f6e58dec0d4de1a\n\n- grenade https://zuul.opendev.org/t/openstack/build/3afb6513820240f8a3a2771a96380628 : SUCCESS in 49m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1af7746c9b4462fbcce78e0aae56dba : SUCCESS in 1h 31m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8dba527e79e4f619c159e953ca3074b : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52b608df18c84651b4e47fb4dd5abdef : FAILURE in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d39103b8c5a44bc4bcc4209032a40636 : SUCCESS in 24m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23498343458c4750b0e95c9e272c56ba : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68ed6fde4edc4dfc8eea1dcd463b98b6 : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1aa9da9659ae40d38e4a56e839b30b23 : SUCCESS in 8m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3b65679721a4cc89a810d8d2d7342d7 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cec4f8bb6dfa4b70befead7d3adba525 : SUCCESS in 4m 18s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/fcc74cf9b4c24bc9acdcaeda0f0e595c : SUCCESS in 1h 20m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/18cadc8812cf4ab8b1841905ff90c195 : FAILURE in 2h 32m 14s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/28e432637e334d139bf2b69840056db0 : SUCCESS in 1h 48m 26s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/f7e4487aa0d0471a8d3fdd7ff4e28ee7 : SUCCESS in 1h 43m 49s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/39419acbabb8405cab51b7179e95f1d1 : SUCCESS in 1h 39m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e50af39281cb4b679c6b4fb7a4276103 : SUCCESS in 42m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/869f2afef06b4edda7795f60fa3d5b27 : SUCCESS in 18m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e680fb9bf4a44929e47bf05402d3fd3 : SUCCESS in 57m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c6e1db9bf484b37ae735881d525cd99 : FAILURE in 50m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/98831c6a40c648e898e70321652d4308 : SUCCESS in 58m 56s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/736ed87439c74417b4fa5efaab0025a2 : SUCCESS in 1h 43m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f31cfad9b47b479c9a9de42230ecace5 : SUCCESS in 1h 04m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c039c61ee85449088beb296cff468897 : SUCCESS in 1h 11m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6150490c85eb44ef9c4e485047fcb8d3 : SUCCESS in 1h 18m 44s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a49b8432bc11450ab541594b76227733 : SUCCESS in 1h 18m 53s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/430e3bab82ed4420a291f4475b146dc3 : SUCCESS in 1h 27m 32s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f93a1a3b957f4da1947e9209b0a0ec4f : SUCCESS in 47m 21s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/70200f9c287447aa9cf0b65a2654ec7f : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":10},{"id":"f7e39c5b46075801e3009b30199037a980453391","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-05-26 08:38:52.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"c01076382154f9b895f0c5295cbe36dde037e650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 10:50:22.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d009e247f934b7585c01afea6aab321\n\n- grenade https://zuul.opendev.org/t/openstack/build/7e0bcd1e5b2f405dab3f1081d9bdf4b8 : SUCCESS in 59m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04ad9b10c75446adb9e5255ad66e1303 : SUCCESS in 1h 14m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c088bf82b4f44ae9b2fa3e38c7edea48 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fa3d208142471aa9a0b48b997a55a4 : FAILURE in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee833402554946f2847d17dd069f7944 : SUCCESS in 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a54582fac2034226951d7a6b0b524c42 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfa0ea08505641e7818406edf2872eb5 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9065465c2eb74b68af5a5067fab5e3e9 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2ec46bb04fc3428ab0b6e03c42e624c9 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ca4687af4e04bdb8fc44763177fd526 : SUCCESS in 3m 24s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/859fda54b83c460c911b145016d3903b : SUCCESS in 45m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/271ea4e612e34a01952d4682114dff08 : FAILURE in 2h 05m 33s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/7ece448aa673400ebd3cb3ca98671a73 : SUCCESS in 1h 43m 29s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/6733a7fe87ba4eae8cfcd65e34c89a3e : SUCCESS in 1h 40m 16s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/9eee2ba9f8e64c9dbefecfdbb2b5aa03 : SUCCESS in 1h 35m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e330a60d7e644ba4a022a4fdca1d21c9 : SUCCESS in 1h 26m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/09d36b77d8b041fb84a5889d47ab3f69 : SUCCESS in 9m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c455b2a054794d179a652c45ecdd31f9 : SUCCESS in 1h 13m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fae805ea3224b539547e05cbc2859e9 : SUCCESS in 1h 36m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f2d9f496a6eb4604bf43246d6104c6b0 : SUCCESS in 51m 28s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/45b397ae50254db59b212b4806ed61ab : SUCCESS in 1h 16m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0291de67e51d492aba75d383a15a32b8 : SUCCESS in 1h 47m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5818505ca9424ccf90e1cf925d43af58 : SUCCESS in 1h 14m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c345b96ef8bb4666ade60623993411fc : SUCCESS in 1h 05m 07s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/e2a6d15fd3ba4006b2f75b9081936fd1 : SUCCESS in 1h 11m 53s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/5d038090072443ffa633cd0d57796749 : SUCCESS in 1h 42m 14s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9dd036122a6241a8980f6b1b409617cd : SUCCESS in 33m 06s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f115443c1cae4e1380a49adb95e6e817 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":11},{"id":"097055e6c5382cef345289e789820e7b5b640cee","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-05-26 11:13:50.000000000","message":"Patch Set 11: Review-Priority+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c5974d8b229c434bdb583047a506663c1f4b4f9f","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-05-29 12:07:17.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"250a8e1aa9a6b4681c97500ceff00306bf06e63d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:31:42.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/faf05307d34f453ba45bb68ab2e89404\n\n- grenade https://zuul.opendev.org/t/openstack/build/dc95e7d305804c6ea60ae5b3e3d7394a : SUCCESS in 1h 05m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11dcfd3ad9e1458c8ab105bf19b90e9f : SUCCESS in 2h 18m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b00504703a3b4feabc01ebe68713e163 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f569ca5ab50a4ef68e17f4f5e9f4720b : FAILURE in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/817c7e9fe2eb4f9795a9a2f0bdf7fa0a : SUCCESS in 30m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f62b6452034ac6b2ff0f2a210ce7f2 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96215a69f2ee4f6e8c6f77842c11f8b7 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0047125adb04f77ba953ab2230d797f : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d7231f6030e4cc2a3cb584b079b6389 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f203efe5a50044fd870aa58dad97462d : SUCCESS in 3m 11s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/76f3dce5e86145bcacbca3807e686580 : SUCCESS in 1h 25m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1e1feab42c4b4f6fb1934615f5f18ee7 : FAILURE in 1h 56m 27s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/ae6c292357bf4814829dee7914f27825 : SUCCESS in 1h 43m 35s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/a5c57a779ab7488ca951c32d8002f337 : SUCCESS in 1h 40m 28s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/2504e5d781814cf580eaf01a1646c248 : SUCCESS in 1h 19m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4996c717fb8641499c89db4d605c10b7 : SUCCESS in 54m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4f9c19e16af346afb3ae063c39a36ad1 : SUCCESS in 9m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea2cc096798d42d3a05db821a8cd950e : SUCCESS in 1h 02m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b9f18f31a3d44b6b3488139d4627a83 : SUCCESS in 1h 47m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bede496bf1bd44d592e4f09c25c7d264 : SUCCESS in 59m 08s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c04a543c44484729b807d874c7055a3c : SUCCESS in 55m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d859a3cce8a0458386fed694a6aa2bdf : SUCCESS in 1h 00m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b3406fa4d2a411794d6a09bdbf87afc : SUCCESS in 1h 01m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad6d7af33a464a68a852db4aa1d85953 : SUCCESS in 1h 07m 56s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/fb9ace3d69ef472d84cafde974670658 : FAILURE in 1h 25m 05s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/5da86498ba5d4b6fa0d2cfe7199cbf33 : SUCCESS in 1h 18m 06s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/c7ca98ce53584138810990dd6c9126bf : SUCCESS in 1h 17m 49s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/ee4b4fd15f4948dbbef93348c5b75d9d : SUCCESS in 1h 04m 53s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/388177602cdc414db1de71546ba816a6 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":11},{"id":"30cf37955d0e18940c0889ba3fbd6ef23d8bf3a5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-05-31 19:20:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":11},{"id":"e02a82cae65ee9ea87f67b4d01ea4738d00c2312","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-14 12:12:37.000000000","message":"Uploaded patch set 12.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"e7d6cf511c5c62c4d04f2771d4025a6bfd823b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 14:49:42.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cec0d270fc704acfabb78d1213933106\n\n- grenade https://zuul.opendev.org/t/openstack/build/17dcc6e6fa4f4ed2924d0318057d27db : SUCCESS in 1h 11m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2815f101252479090c30d5379a332d0 : SUCCESS in 1h 37m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ad277ae2c01494580c0ad2167cfc3ee : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394b209617aa434db8e714ccb294e5f0 : FAILURE in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f7e470b80684687a561c02b06663cf7 : SUCCESS in 28m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/262959334dbb4c6492626b57a7ffb1ff : FAILURE in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/641ba9d6c87541a58eff34ebca0cffa7 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44c9a48d1d184121a496f2c27d5c6984 : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0e68426f79e43c49e5571c86d82c46e : SUCCESS in 4m 07s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5a19567aae96497890f0a896bb5a2ab9 : SUCCESS in 1h 17m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bd2f7030764f4b0f972ae7c4c8e6e587 : FAILURE in 2h 29m 13s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/9e25e62f92d640af8a6131b2f327b398 : SUCCESS in 1h 40m 49s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/10f3382459d74f1bb2abfddf556da5e1 : SUCCESS in 1h 42m 26s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/701849cd85824b98a08f44ed5b956d05 : SUCCESS in 1h 39m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c6e18ffb221d48c0b5db7c9678f610f2 : SUCCESS in 58m 38s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c0b63fd802b24ff2bb09550ef3be1c0d : SUCCESS in 15m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27167369dc26405391722113abd00679 : SUCCESS in 29m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f81f886f0c2542fcbaaf99e7ec76b2b4 : SUCCESS in 1h 25m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/04b8c4e1419948c49bf85f2f99457ec3 : SUCCESS in 28m 27s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0b408b5828ba464eaa4193563828ef97 : SUCCESS in 1h 47m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/65e2427e1c7c4b79ace9c393abf842bc : SUCCESS in 1h 42m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/efe6283682c94b01a627a4a984348aac : SUCCESS in 52m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b7bc57bb81004033aebe782efff7bc71 : SUCCESS in 1h 05m 18s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f195312d2cf94bff939e32331d3962d0 : SUCCESS in 56m 18s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/26c0ca21c39245aa9c8d1d4f7250f846 : SUCCESS in 1h 41m 25s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/1bb6b259dd8a426d9f49d542b55cc418 : SUCCESS in 1h 26m 35s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4884922b07db4c07b1de9e6fd0156381 : SUCCESS in 1h 03m 00s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/63dfe3e82b474d6ebf4a05c2ca11d3b6 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":12},{"id":"681f2801cb02db83f257571d3ec7e0459d78ce12","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-14 18:34:56.000000000","message":"Uploaded patch set 13.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"4841bc489114aa8a1d64b9adc9dbc4e19135fb68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-14 19:24:13.000000000","message":"Patch Set 14: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n","accounts_in_message":[],"_revision_number":14},{"id":"507195df53d5bf2b378441a71bb0ad959f09c303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-14 19:24:17.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4605feaafcbe4774b5be97c2b39c85ab\n\n\nWarning:\n  Dependency cycle detected and project openstack/tempest doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":14},{"id":"26e8e8a51ec9f3318953a8ef0bb1b1ce0dbee2ff","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-15 06:44:06.000000000","message":"Patch Set 14: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"dc164d2fedd57f1846be0b1fe03d4768b8a2e5d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 09:28:15.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0aa706eaa88b4ea9a2a8c39452c85b11\n\n- grenade https://zuul.opendev.org/t/openstack/build/c1dc029a2be04126b1af5c9703072a69 : SUCCESS in 30m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ecaa7f811e804ad699da624d878b6edf : SUCCESS in 1h 15m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7323277787184ddc9067e8459aac7233 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a700fc325dcc42ee8a0082d164acb228 : FAILURE in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5de81fa63354edf8e21843c0121193c : SUCCESS in 29m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50de2d01d824468ba8604c8b87efa03a : FAILURE in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1190a3cf0d949a0a1e4931a4a4a9dfe : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b233591791934f69b36a3da94caf90b9 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47b56a6a1f1643e595a7964b1a3371e1 : SUCCESS in 3m 24s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/333d761117f54f59a5f75c856d567eaa : SUCCESS in 1h 05m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2db0db56e2704dfc9c437bb27607deab : FAILURE in 2h 37m 59s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/bc8590058fa64e16a71f5acb2080feb9 : SUCCESS in 1h 35m 11s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/86d2c257884249918341faad056ebdd8 : SUCCESS in 1h 43m 00s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/19a86f277c444fe7bbedefc96e31111b : SUCCESS in 1h 38m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5f7675ccf9104b5eb50d2d99913cfbbd : SUCCESS in 40m 18s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f7af2e573b084450a0310e4e7fffcae2 : SUCCESS in 15m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c930289a55ca4777a590261ebc988462 : SUCCESS in 42m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5c0a3470a2454287bf49f9ac8ac5d276 : SUCCESS in 1h 39m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a651fecbd6754c2dab68ac6e73c90362 : SUCCESS in 26m 42s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aff371444a1b4246a93be5021eaaa6f4 : SUCCESS in 1h 54m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e31d6567657844febee5e58b4b4289eb : SUCCESS in 57m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/05bc5a53a1fc438a86beb80dedbfb62d : SUCCESS in 1h 08m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c9ab2f055ade41f0a18db32e02063b82 : SUCCESS in 35m 56s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/bf18e6649b884acc9784bafef50db7f2 : SUCCESS in 48m 12s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/0854d42527dc451a88e51d25107450a1 : SUCCESS in 1h 40m 13s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/665d7e3675a746bcb7ccfcbaa7356b9c : SUCCESS in 54m 47s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/570e504ea2d144a2b31526a1e6d9cf1a : SUCCESS in 1h 06m 03s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4111b7376ccb4d3fb995af8a9a932073 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":14},{"id":"f0320c4786bdbb3d5b902e285cd134f908e6bda7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-15 10:00:29.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"48f7a7fc71acc6b31236120bc6ca45cce90c8422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 12:46:54.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a86b139ac68434cab518e737c32e380\n\n- grenade https://zuul.opendev.org/t/openstack/build/4dce8f2e43864980af08d1d531024fbf : POST_FAILURE in 1h 02m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/491266ee6c954eeeba7a10608f0f9309 : SUCCESS in 2h 00m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b67fee2908d648849bd2eb61d3f463dc : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c11a59412d48c78f533b18f4251405 : FAILURE in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f716b70d8a29483889092a81adc445e3 : SUCCESS in 28m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afdac90d8e7b4c4aae705eb4d2306bf6 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17452091c4ff43019a8393e544d2ab75 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/857e6f066efe498ab91049fd973a3941 : POST_FAILURE in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7cffc3898fa14f7cadeca65722a95c3b : SUCCESS in 3m 31s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/021d38342d7649128fdce1ce1968197d : SUCCESS in 1h 25m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f7116ae7876e42fe9ecf80f39b0d8af0 : FAILURE in 2h 29m 43s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/02608ef37c05402d81cdafff2afc16fd : POST_FAILURE in 1h 34m 03s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/12e7b27feb464f9e93a6b6113b3c8a12 : SUCCESS in 58m 50s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/1bfc507112864501988231a008a473e4 : SUCCESS in 1h 14m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6ec86d98a6564d59b663a9d9c97f4ddf : POST_FAILURE in 1h 11m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/572d7f69784d465a87c3fec69b3b62ff : SUCCESS in 18m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca3e10655a5b409581d790c1f4928861 : SUCCESS in 1h 05m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7708804d175e47328fc321b2950d7e1a : SUCCESS in 1h 40m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5ab080b2144747e28a922d38df261e55 : POST_FAILURE in 57m 31s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fc71bacbd08143d5978c9bb86c4567f5 : POST_FAILURE in 1h 36m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/675e6edaeca44c3d80c85c01263f88f6 : SUCCESS in 1h 44m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0ccdd52a9f8400cae9d9bc0ed4aae2b : SUCCESS in 1h 04m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a201820bcb234bde9cdb42c3ca3e5112 : SUCCESS in 1h 01m 51s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/19fa7a35a1b6407a9fbc0fd9297c99af : SUCCESS in 1h 16m 36s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/dc69128f84a54c59b53c5e7d217091b3 : SUCCESS in 1h 45m 21s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/cfb72eb4bbde4baeace5ee451d62bc13 : SUCCESS in 1h 27m 58s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9ee15eea48584335af412a024daf3e30 : SUCCESS in 1h 01m 20s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/86fda3a34cea4d5e946225ddf5af0b14 : POST_FAILURE in 3m 31s","accounts_in_message":[],"_revision_number":15},{"id":"4cd746728fac33fd4a4aea955f6b800d1c19d57f","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-16 04:03:13.000000000","message":"Patch Set 15: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"67e9f607816edf1e78026bc56a94901b9fab4859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 06:08:08.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b10d1b8bb71346f9967eaeb46ba0711e\n\n- grenade https://zuul.opendev.org/t/openstack/build/96e9b9cc14cb4bc9903f50151bdaf0fd : SUCCESS in 38m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9475317f13c847cd992128d0f50b5bb5 : SUCCESS in 57m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03d1f17f53544cd6b97536b8745e0898 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5436967e934aecaf25f43dfa83ad51 : FAILURE in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ca9dccb97b04a0682c2ebc0f0449a92 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b007ad58cc72457cb66ea106cc400bc5 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/508ac252b12242bb9922642f86bb2fb4 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56adfd6fe88f4ea0853183a0d449caf3 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b78de8650d374f8da9459e9c9ad25c55 : SUCCESS in 4m 13s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/494ca9744cca4e04bb8eb447da2a4cb8 : SUCCESS in 1h 06m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a1c0dafb374e4885b4c4e3775ef056a2 : FAILURE in 1h 19m 14s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/93899fff25214d4a860c4a5b4917b11f : SUCCESS in 1h 01m 51s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/463a73fbaadc4f438b9eb6760b15a6c3 : SUCCESS in 1h 16m 47s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/e67953c2f8714a8190fcf94c37141137 : SUCCESS in 1h 08m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5d2ff88631bf44e19404612345ebb1f1 : SUCCESS in 45m 52s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1e960e12613e43329bded33cb46142f0 : SUCCESS in 15m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce203392fa3141a68cbe2eb2dbd17ae7 : SUCCESS in 1h 12m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5c27210021334b7e958ee8b9b6e1833d : SUCCESS in 1h 39m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4173a66c51304ceb80d6ed8ffc5ab06c : SUCCESS in 29m 23s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/41aabc750b5c40ac93b756a7db707c0a : SUCCESS in 1h 19m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/85d28e00d7064b35a22a028d0f60a2fe : SUCCESS in 1h 08m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf0f1dab4dec4adebc9aa8155aa42a11 : SUCCESS in 41m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a99585426c15446dacd9b87619c48e30 : SUCCESS in 57m 12s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/448c3d3f0a2143e99b0b7cd7604a329f : SUCCESS in 1h 14m 38s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/d00ee4e3b2184e5e843119f3a08e571b : SUCCESS in 1h 57m 51s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/e98308707bdc42e09ca09f7956c6329f : SUCCESS in 1h 13m 25s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f544e184800f403a8756fd70bc46cb6b : SUCCESS in 1h 06m 42s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4c463f050b644717868764a9b869c003 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":15},{"id":"caa6744c11c6ab109b83c40c9b7a0864a6c79e2d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-16 09:05:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":15},{"id":"3e89d0883283d28dab64f13b20e5a96b73ac1db9","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-06-18 13:18:16.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"30d7890679f3541cbaeaf4700ac1550a7c72f1ae","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2026-06-18 13:21:43.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a4ee104f8b3b9f9670ad671fc1f55a02226eb14f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-22 05:05:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":15},{"id":"1459bbcca8d0d8b295998961c7510ff837f2a923","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-22 05:56:26.000000000","message":"Patch Set 16: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"6a636e40abadf1d6b0e998ec22d6677af1dacfb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 08:03:38.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f30e732dff314150b86ab5af8a5c17ea\n\n- grenade https://zuul.opendev.org/t/openstack/build/d3060e57a4514bbdb4aee6548708c56e : SUCCESS in 1h 07m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/faced016461c46b5b16dfcc8fa14f9ab : SUCCESS in 1h 46m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70321b2f14e4478e856212075a1eac34 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48d7d06781784a67b5d852781b467f44 : FAILURE in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2988a525582d4d11ab38548dae5bca7f : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d2aa3ca973e4cd7bb8083b713fe236b : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b378ff22ba4450abaaea243fcd5c414 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49d20c9024ac4d11974dbf2607bb4e4c : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67bc503115244ff2a82f6ac0aaf6307f : SUCCESS in 4m 05s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/dffb120ed88a46fbb05cbe89fa186637 : SUCCESS in 53m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/742ab1bdb89f404a99eb87ddd4d73d5c : FAILURE in 1h 51m 55s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/8f4ad8d91bea46d2a36c15483c35b3cf : SUCCESS in 1h 23m 45s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/1cee882135be47bf8b225e609b711508 : SUCCESS in 59m 38s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/cb54cdeb0fd443b3af11e1a589593c3b : SUCCESS in 1h 35m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/49c40e4010864e428d2cad82c66433ec : SUCCESS in 1h 12m 33s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2e7624d6aaa24d10b164d67af4fc469d : SUCCESS in 7m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bde5705a8914e3bb3ffede563b32477 : SUCCESS in 58m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aac0041ba5a84152bfd86cc52df693a6 : SUCCESS in 1h 29m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4334aeff0e414831b94ebf2c59818c19 : SUCCESS in 56m 43s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a2be1f26d7af4cd882dc460eb278f2b4 : SUCCESS in 1h 43m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ead8daf67eef48c49db37256284b699a : SUCCESS in 1h 46m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7801bee6f17a43eb977ed03e94e71182 : SUCCESS in 1h 15m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6695ee2d03a149549378f97b2675c4c5 : SUCCESS in 1h 05m 24s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/cd217b7b1bc44024bfc1b5740da2608c : SUCCESS in 44m 51s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/e47b061b16b547048617b85d9295eb26 : SUCCESS in 1h 52m 40s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/53b2c46b983b42008d08e97638dfd538 : SUCCESS in 1h 25m 14s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b50d310709ff4624a20c80f501932e0a : SUCCESS in 58m 43s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ca33502d4bc04439a1ee27da7d660b18 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":16},{"id":"7bb02e69d65ffe1800a79a93d3742261dcdee1d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-22 09:47:07.000000000","message":"Uploaded patch set 17.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"368f4db083897e92a3f102f68fcd1f4ca481cdd1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-06-22 09:49:24.000000000","message":"Patch Set 18: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n","accounts_in_message":[],"_revision_number":18},{"id":"abbbd5251b52b3a297b5d62d2926c6818b72c75e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 12:11:37.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a8de004a1f442c39a428ba2c7e6ac5a\n\n- grenade https://zuul.opendev.org/t/openstack/build/682ea3fc4e74415a83e0929f75830030 : SUCCESS in 1h 04m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a21bafc8e7974b2cab1a0282f6b96f57 : SUCCESS in 1h 46m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27aafef6beb0487393aaa7ca2d3c9a6c : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e73da44b170408697a86e1cb2296b35 : SUCCESS in 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80cd19ec60c94edf94a4d0a6511c29e1 : SUCCESS in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319d5c446776471b852a581ab29b97f4 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/732a991fc4704908b42fbaee94695f8a : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d0da09487b8452ebddcc1187b9ab3d8 : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/191adcf87725429a8d5fa031d55578a8 : SUCCESS in 4m 30s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5c3f3a61126445d994ccab2dbe7c5406 : SUCCESS in 1h 21m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/684c09079d7d4ccf84da318b04929c3c : FAILURE in 2h 19m 15s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/471c5e3fd260461091f322ee1300ec3d : SUCCESS in 1h 36m 40s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/0bfa9110b7cc40fbb6a684e45ca89696 : SUCCESS in 1h 41m 05s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/1483b5d678e6495a8cbb4c98dd24e9c4 : SUCCESS in 1h 40m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cc0d356ac4f04cd599b8f3803a29484f : SUCCESS in 40m 46s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/36cc189ef9604ffdaa9b3c3ed03c2ca1 : SUCCESS in 20m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ee72db20bd04f1fbf805b8b9e15e42b : SUCCESS in 1h 00m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0ea046696f9a4452b2fd8222d16b2107 : SUCCESS in 1h 36m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ae5342dac44d4581aeb1cf70d3193b39 : SUCCESS in 59m 45s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1373c0bde3ee4e3388e7ddea67fa0e9b : SUCCESS in 1h 42m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2a7462cb03c2400da649e3c5d5e50243 : SUCCESS in 2h 06m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/28425a44e76745b0922f1de4428579b8 : SUCCESS in 55m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6538c6f64e2c406e9a82715929f57bca : SUCCESS in 45m 35s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/e6d012cb6ad94c88b7ee0b11af1ca8bb : SUCCESS in 1h 08m 10s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/46f8f2e20d434f7ba7c6a49c15fcacf5 : FAILURE in 50m 07s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/29bd06058f97404bae30f38cc28645c9 : SUCCESS in 1h 20m 35s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7f24de6273294c9f8f5e352081c1ba4a : SUCCESS in 27m 44s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/27ef7d649d6a4ca4bd848afb88adc651 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":18},{"id":"5b399d455d18db59fb56056521729de76bd9d7af","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-05 09:47:52.000000000","message":"Patch Set 18: Review-Priority+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":18},{"id":"691962aa12e203e147bc5c0945232a80e93f2093","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-14 10:48:14.000000000","message":"Uploaded patch set 19.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"ed1ac65fff2a4bfcd8a2aa84b3a55893c231dba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 13:06:47.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8009288c63124f00b126e901a63fe462\n\n- grenade https://zuul.opendev.org/t/openstack/build/00a6a7eac9d3418c8b8254f279e7d445 : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/435aee0df13e4a759cf3d7637d8c8897 : SUCCESS in 1h 27m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8afdba494339432b9a9bbc1be3329c28 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e796576dbb384f2f91fdbb5058372e8a : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faa5cc751fc24c1cb3381399db8f4e61 : SUCCESS in 28m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caf0950501314f0e965c977af885a884 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ccc6bf6b3df64dfca8cbd5d8114ca1cc : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8c6364f134d4ed1b841785c4738b45c : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3984c9aae59f4249a398fc431334028e : SUCCESS in 4m 07s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/09b52bfac85444b4955e04b4d08cec1c : SUCCESS in 55m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e92a1e30512f4294b8b01f0953361a87 : FAILURE in 2h 12m 10s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/60a85108401d4f6ea630c7975ba88570 : SUCCESS in 1h 50m 32s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/6b06efac141a4aa486fc3442008d700d : SUCCESS in 1h 44m 46s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/13990a45dc234d3d95125b7bc41f9c15 : SUCCESS in 1h 35m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1dce88c516674d9fbdcc920e09c51ecf : SUCCESS in 1h 06m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/93e2f23cb3da4387a20e6d4201cb8e93 : SUCCESS in 15m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b50a1e06d9cb4f668a2012a4ff7f5134 : SUCCESS in 42m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbf39b459a90488780c21846d7241d95 : SUCCESS in 1h 09m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/88637ecdd08240d1a2c5eaa720bc8a0b : SUCCESS in 50m 28s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/c6833fae7bba447284a72cb167c13ecc : SUCCESS in 1h 36m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/db018ad73acc46ef8edf91908efe1f0a : SUCCESS in 41m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/81867ed483f64bcf8a6769496c450dfe : SUCCESS in 1h 11m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/50c8d3205d454405a173236cfa9b9a86 : SUCCESS in 1h 01m 50s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1b03064c44cc4d42819a2eb780b86f43 : SUCCESS in 31m 43s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/38f9def30131499382d5a6289de62430 : SUCCESS in 1h 18m 12s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/970f274050bd4f66aa1dac6700f766bd : SUCCESS in 1h 04m 08s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/c6adab224fd9432d954ec3894d6e83fa : SUCCESS in 27m 01s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8f267bfd43784b3f8b0a93f12474a044 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":19},{"id":"49f987efc7af8d98a1560a7d23177035347b5a3f","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-07-14 13:07:48.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7dfa09405c110d0e1f5d1ca954518d90c29bc3e6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-15 07:11:36.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":19},{"id":"06f1b0cad9efb0cfac631e82317e7a3b0cdb4584","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-15 07:12:16.000000000","message":"Patch Set 19: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"f5dafaf6d0747c95fa7420864e653f3a70ce938a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-15 07:39:41.000000000","message":"Uploaded patch set 20.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"5bd68501b14d94aebb2d097bd4f31fb523e0341a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-15 07:42:51.000000000","message":"Uploaded patch set 21.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n","accounts_in_message":[],"_revision_number":21},{"id":"7b45a8b948fc25cca778feeb0e91754dbb927c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 09:40:16.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e2a8e06bec84348b5cb78d0cf288c60\n\n- grenade https://zuul.opendev.org/t/openstack/build/dea9c979afff4bc797d1fd8e877374c1 : SUCCESS in 1h 05m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d455abe0ba664b7c91331b9f7b623a8d : SUCCESS in 1h 23m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46dfee72558f40579e6c748ca089efac : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b40731e3d6fc4dcc93168e41ff68c898 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edfbda157e5a4021b93113a6a465e2c0 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d610017cf4c94e5ba4f8f05074c888c8 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ebdaf5c5cfc4536928995edfcee31c1 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af641176eb1641ea8d6ec899780b5c23 : SUCCESS in 5m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aabfda90777f4cb9bd4ebcb2c76dd648 : SUCCESS in 5m 58s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/736bac51f81f47c8bcff385df10d35c2 : SUCCESS in 35m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/751abb07c672471d9d84e96214a2a438 : FAILURE in 1h 06m 40s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/a76906bb3ed64cb988f4b7f1ea85bf41 : SUCCESS in 1h 03m 26s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/81c502fce36f4e79902c27bc028fdc22 : SUCCESS in 1h 33m 08s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/369eb116223a42da8230ee0ac48a78d5 : SUCCESS in 1h 44m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2bda01d46ea14b73b7be38dfd61ca5ea : FAILURE in 1h 06m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/07e7b0bccd65474caad6ca646717e761 : SUCCESS in 15m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d982adfef4848e89e7d87770f20d7f7 : SUCCESS in 30m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f3291fd9850342408233d3de74463ed8 : FAILURE in 1h 27m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a4332a907fea40b1ab8f41f10b6315fe : SUCCESS in 52m 45s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/e857c9055cd940338479a5fe5ce800f3 : SUCCESS in 1h 40m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5273ac4900b74d88be3977cb1bf48514 : SUCCESS in 1h 02m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dec6a17cbfff478388dec7f8a523136d : SUCCESS in 1h 19m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61f904e4ebf7452498e25b2b147fc18d : SUCCESS in 35m 26s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f6988043d81944fe91d55a689631870f : SUCCESS in 54m 43s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/b4507aca44e6418a95b716362e430f85 : SUCCESS in 1h 35m 45s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/d29293ca80204be2b47382977f305e4f : SUCCESS in 36m 53s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/c240ac7638c24ea19f5c8768ac7a258e : SUCCESS in 59m 30s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/66bd98a3595c42c58edbf2765da09f07 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":21},{"id":"3fcefcc3c0ec6263c2bdb64e5180192456e13ad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-15 12:18:25.000000000","message":"Patch Set 22: Patch Set 21 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"4d691c8c88224a28b1b6106787bceccd036f2c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 14:11:38.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/264c1836525944a0b11e25e156de6969\n\n- grenade https://zuul.opendev.org/t/openstack/build/abba4580e95d4763955e4b28fbcc1fe9 : SUCCESS in 1h 03m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd6cabf6595e498d9e5474b70a2796bb : SUCCESS in 47m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5be64c6018c4fdb8f9b535e518631e1 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ccc657bb6fd4b64bf09eac78662e47e : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c404d0f8ab4d4a0db11cf4e851bf28fa : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6531ccae254a4ca47783d7706972b2 : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/849d7255c8e8466f9f04e156bb79f0fc : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3a6c2a3fc3e4e4ca1788a90764c1ad0 : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1257500d59b9459ba8d220ba5c00ccd7 : SUCCESS in 2m 49s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/b08106cacddc415fbef91497705e26dd : FAILURE in 1h 15m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f14725ed07224726b71e06724a77af10 : FAILURE in 1h 42m 41s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/6b3da040db5843e8b3bc4105b703e066 : SUCCESS in 53m 23s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/e6484614cf924ee38024a4d78e95225f : SUCCESS in 1h 46m 31s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/b922f589bc14498ba38702b51f626abc : SUCCESS in 1h 34m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7a55eced2edb42d89d9766b831ee3488 : SUCCESS in 39m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e738a2885970420496227d11b3b31bce : SUCCESS in 13m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e83801abc04449248ab2ca74fb4fab45 : SUCCESS in 56m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9104c879a5b14bfbb8283478f0f468fa : SUCCESS in 1h 41m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/777a3c2ddd1f4837b6a6ad8988c1b2ef : SUCCESS in 52m 42s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/c73ecab7f8f4454ca0963c09339f30c0 : SUCCESS in 48m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d04b0901f4844d55ba1ca7bcb27af19a : SUCCESS in 47m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a8c077c7cac479c8cc9346d5b130217 : SUCCESS in 1h 08m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb46040ab7c84d3d9e73f0b5ca132477 : SUCCESS in 1h 05m 33s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/9a3fd9d596d34347b671c06f0114b569 : SUCCESS in 1h 06m 15s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/a7e7f53bfa784c01be461b7007abf617 : SUCCESS in 39m 46s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/180946c5e2a9400ca06fea3e5383743d : SUCCESS in 1h 11m 20s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/6b3aba354327499dadf07c8bc14733fa : SUCCESS in 1h 01m 50s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9832c5693b84464ab2101aefdae69269 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":22},{"id":"a8ec1a62843fb84c9873cb599ae7d607ff731d8a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-07-15 14:55:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":22},{"id":"36d25961a1f4280146685e0b95cdbec7bf38a953","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-08-05 12:22:50.000000000","message":"Uploaded patch set 23.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"4f35d0e8169f330aa1dc5c3085df1c220361ef36","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-08-05 12:23:06.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n","accounts_in_message":[],"_revision_number":24},{"id":"323dc96858f6521cb45ede865de1db3950551d5a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-08-05 12:23:41.000000000","message":"Patch Set 25: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**is:ANY**\")\n","accounts_in_message":[],"_revision_number":25},{"id":"4ba6baf2d399cbf5ef47360a721d8ed251fe3cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 14:14:27.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e6d4dad9ea64bd59fa42f84575016e2\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd437ee851e94d2b94f59bfaaa241240 : SUCCESS in 33m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f20447a93a54d4694560fa2e35818b9 : SUCCESS in 41m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81aa6b4e8b124b489769e0aadde9e0cf : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ee39b66c304c308b0e379eadc8b46f : SUCCESS in 13m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07b245977df647fd90b94aa0009daa4f : SUCCESS in 32m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f52479b180a642e19ee3075bd04158f8 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e52ea8ce3764c4bb2d82dc6b5c12b31 : SUCCESS in 6m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2f72e96719b43dd9204532bb6e09c6b : SUCCESS in 6m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/636c6f86f25c450db6d66ce54494230d : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/51a21c9b3d9e448bb8f0d47856941ae2 : SUCCESS in 6m 58s\n- tempest-full-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f6e02c0d2a154c4dbe6561097d41b004 : SUCCESS in 1h 19m 07s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/a952be6bc69e4c5aa0171015a572ad05 : SUCCESS in 48m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2b471df5461741d2849638121947ed4a : SUCCESS in 59m 03s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/6ba37196351844a69f9ad8b7b7164f8b : SUCCESS in 1h 45m 09s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/1ba455eac06d4de2bed6fbaee6fcb387 : SUCCESS in 1h 03m 15s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/bce94a72824a4791b563aa568d704c68 : SUCCESS in 1h 22m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4d9e2295b2bf44a09d85847ab0fd27ef : SUCCESS in 1h 19m 02s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ab938031df2046ab8a0373edc9a1f434 : SUCCESS in 7m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaa36c18575b41258d096df93a6a7908 : SUCCESS in 53m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/427730e16f5d4b9489881b8812bf3d70 : SUCCESS in 49m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ea5d87e25dea47ffb50f198d98f8e6e5 : SUCCESS in 29m 50s\n- tempest-full-rbac-old-defaults https://zuul.opendev.org/t/openstack/build/7b2b4d10894f46d788b275c066bc5599 : SUCCESS in 1h 32m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/80c208dc240c4948b7bb32cbbf97c5aa : SUCCESS in 44m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2a76cbdf7c347389669ade43cedcb1f : SUCCESS in 1h 14m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e19dd7dea5ee4572b747fce0e2de2d85 : SUCCESS in 1h 16m 38s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/b1e0f5669ab148d3ac939b273ce3aad7 : SUCCESS in 42m 25s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/aca8510af3944f32973fdea4964fea49 : FAILURE in 33m 23s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/464298f0c8bc463fa45739503100a06a : SUCCESS in 1h 16m 11s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/6776a9cb012b491cbcca6bb7f497d56c : SUCCESS in 33m 47s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/41026b03b969416eae5a1b72b4f10f1f : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":25},{"id":"957d60310dd7b5567845391b9e54d2479b0e261d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-08-06 09:55:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"869b81ef37fc8ae2339e844d2ec631f64a0fbc2c","revisions":{"c9e9542a86924d70a8a726808979f43e899eb808":{"kind":"REWORK","_number":1,"created":"2025-08-17 19:12:20.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/1"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 19:10:51.000000000","tz":180},"subject":"Add test cases for Glance image cache operations","message":"Add test cases for Glance image cache operations\n\nThis patch introduces new image cache tests covering image cache\noperations in Glance listing cached images and cache cleanup.\nThese tests aim to validate correct behavior and improve coverage of caching\nworkflows with volumes and instances.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c9e9542a86924d70a8a726808979f43e899eb808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c9e9542a86924d70a8a726808979f43e899eb808"}]},"branch":"refs/heads/master"},"6c366930ea42646951d9c077c0f4ebba522c1f2d":{"kind":"REWORK","_number":2,"created":"2025-08-21 09:14:15.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/2"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-21 09:07:39.000000000","tz":180},"subject":"Add test cases for Glance image cache operations","message":"Add test cases for Glance image cache operations\n\nThis patch introduces new image cache tests covering image cache\noperations in Glance listing cached images and cache delete\nThe tests aim to validate correct behavior and improve coverage of caching\nworkflows with volumes and download images from glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6c366930ea42646951d9c077c0f4ebba522c1f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6c366930ea42646951d9c077c0f4ebba522c1f2d"}]},"branch":"refs/heads/master"},"2bf46c11ab3186cafbd45cd39439d8226b0e9220":{"kind":"REWORK","_number":3,"created":"2025-08-21 19:33:48.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/3"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-21 19:27:31.000000000","tz":180},"subject":"Add test cases for Glance image cache operations","message":"Add test cases for Glance image cache operations\n\nThis patch introduces new image cache tests covering image cache\noperations in Glance listing cached images and cache delete\nThe tests aim to validate correct behavior and improve coverage of caching\nworkflows with volumes and download images from glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2bf46c11ab3186cafbd45cd39439d8226b0e9220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2bf46c11ab3186cafbd45cd39439d8226b0e9220"}]},"branch":"refs/heads/master"},"04fa691cb9b15594f098e7304593d98d1b20c657":{"kind":"REWORK","_number":4,"created":"2025-08-22 07:12:28.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/4"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-22 07:09:52.000000000","tz":180},"subject":"Add test cases for Glance image cache operations","message":"Add test cases for Glance image cache operations\n\nThis patch introduces new image cache tests covering image cache\noperations in Glance listing cached images and cache delete specific\nimage\nThe tests aim to validate correct behavior and improve coverage of caching\nworkflows with volumes and download images from glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/04fa691cb9b15594f098e7304593d98d1b20c657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/04fa691cb9b15594f098e7304593d98d1b20c657"}]},"branch":"refs/heads/master"},"bd5bb03bae703bc1271a6c3209bc3327a8a06b6d":{"kind":"REWORK","_number":5,"created":"2025-08-25 11:25:48.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/5"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-25 11:24:51.000000000","tz":180},"subject":"Add test cases for Glance image cache operations","message":"Add test cases for Glance image cache operations\n\nThis patch introduces new image cache tests covering image cache\noperations in Glance listing cached images and cache delete specific\nimage\nThe tests aim to validate correct behavior and improve coverage of caching\nworkflows with volumes and download images from glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bd5bb03bae703bc1271a6c3209bc3327a8a06b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bd5bb03bae703bc1271a6c3209bc3327a8a06b6d"}]},"branch":"refs/heads/master"},"f79eda88a9f691005e92cf26d57e32188288dd90":{"kind":"REWORK","_number":6,"created":"2025-08-31 10:44:43.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/6"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-31 10:43:09.000000000","tz":180},"subject":"Add test cases with image cache operations","message":"Add test cases with image cache operations\n\nThis patch introduces new image cache tests covering image cache\noperations in Glance listing cached images and cache delete specific\nimage\nThe tests aim to validate correct behavior and improve coverage of caching\nworkflows with volumes and download images from glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f79eda88a9f691005e92cf26d57e32188288dd90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f79eda88a9f691005e92cf26d57e32188288dd90"}]},"branch":"refs/heads/master"},"830fc70bbdecc8c4be06bc4b4f795c34b067d2ab":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-01 06:17:39.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/7"}}},"commit":{"parents":[{"commit":"290069b157d948b76ab04993d8b6b5ae9e74e212","subject":"update tests for bug 2112187","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/290069b157d948b76ab04993d8b6b5ae9e74e212"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-09-01 06:17:39.000000000","tz":0},"subject":"Add test cases with image cache operations","message":"Add test cases with image cache operations\n\nThis patch introduces new image cache tests covering image cache\noperations in Glance listing cached images and cache delete specific\nimage\nThe tests aim to validate correct behavior and improve coverage of caching\nworkflows with volumes and download images from glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/830fc70bbdecc8c4be06bc4b4f795c34b067d2ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/830fc70bbdecc8c4be06bc4b4f795c34b067d2ab"}]},"branch":"refs/heads/master","description":"Rebase"},"8c3bd0ae3c526fd6238935eb12e2c7c8e7add2fd":{"kind":"REWORK","_number":8,"created":"2026-04-14 09:47:25.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/8"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-04-14 09:46:56.000000000","tz":180},"subject":"Add test cases with image cache operations","message":"Add test cases with image cache operations\n\nThis patch introduces new image cache tests covering image cache\noperations in Glance listing cached images and cache delete specific\nimage\nThe tests aim to validate correct behavior and improve coverage of caching\nworkflows with volumes and download images from glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8c3bd0ae3c526fd6238935eb12e2c7c8e7add2fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8c3bd0ae3c526fd6238935eb12e2c7c8e7add2fd"}]},"branch":"refs/heads/master"},"e44416b75ecebd61111a2592310e6c2a1e81eae6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-04-14 09:51:12.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/9"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-04-14 09:51:09.000000000","tz":180},"subject":"Add test cases with image cache operations","message":"Add test cases with image cache operations\n\nThis patch introduces new image cache tests covering image cache\noperations in Glance listing cached images and cache delete specific\nimage\nThe tests aim to validate correct behavior and improve coverage of caching\nworkflows with volumes and download images from glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e44416b75ecebd61111a2592310e6c2a1e81eae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e44416b75ecebd61111a2592310e6c2a1e81eae6"}]},"branch":"refs/heads/master"},"ccdc8b9290c318a6aeb143f1f06afa41286aa057":{"kind":"REWORK","_number":10,"created":"2026-05-19 17:47:56.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/10"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-05-19 17:29:04.000000000","tz":180},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and downloading\nimages from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ccdc8b9290c318a6aeb143f1f06afa41286aa057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ccdc8b9290c318a6aeb143f1f06afa41286aa057"}]},"branch":"refs/heads/master"},"65141dd066121354c0122ef5dcee2a9cdb29dccd":{"kind":"REWORK","_number":11,"created":"2026-05-26 08:38:52.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/11"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-05-26 08:25:16.000000000","tz":180},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and downloading\nimages from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/65141dd066121354c0122ef5dcee2a9cdb29dccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/65141dd066121354c0122ef5dcee2a9cdb29dccd"}]},"branch":"refs/heads/master"},"9e01d6ed48cfd2859edf6689df90c78c8e3e950c":{"kind":"REWORK","_number":12,"created":"2026-06-14 12:12:37.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/12"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-06-14 12:11:47.000000000","tz":180},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9e01d6ed48cfd2859edf6689df90c78c8e3e950c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9e01d6ed48cfd2859edf6689df90c78c8e3e950c"}]},"branch":"refs/heads/master"},"89ac976c2955268e10ce658a5645d239278d9bf2":{"kind":"REWORK","_number":13,"created":"2026-06-14 18:34:56.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/13"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-06-14 18:31:25.000000000","tz":180},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/89ac976c2955268e10ce658a5645d239278d9bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/89ac976c2955268e10ce658a5645d239278d9bf2"}]},"branch":"refs/heads/master"},"00bb6a0dc2fbc54cd26ebbd5a42fefeda7282fde":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-06-14 19:24:13.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/14"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-06-14 19:24:13.000000000","tz":0},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/993270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/00bb6a0dc2fbc54cd26ebbd5a42fefeda7282fde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/00bb6a0dc2fbc54cd26ebbd5a42fefeda7282fde"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4b10f72e0d6af3da6e27b4773d343d43ed99259b":{"kind":"REWORK","_number":15,"created":"2026-06-15 10:00:29.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/15"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-06-15 09:59:57.000000000","tz":180},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4b10f72e0d6af3da6e27b4773d343d43ed99259b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4b10f72e0d6af3da6e27b4773d343d43ed99259b"}]},"branch":"refs/heads/master"},"d85b73d1aed09e7cec7e312ae32e7fd7a9f59203":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2026-06-22 05:56:26.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/16"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-06-22 05:56:26.000000000","tz":0},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/993270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d85b73d1aed09e7cec7e312ae32e7fd7a9f59203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d85b73d1aed09e7cec7e312ae32e7fd7a9f59203"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1e346e51a8103bff9268d4b257346d553896f4d9":{"kind":"REWORK","_number":17,"created":"2026-06-22 09:47:07.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/17"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-06-22 09:39:56.000000000","tz":180},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1e346e51a8103bff9268d4b257346d553896f4d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1e346e51a8103bff9268d4b257346d553896f4d9"}]},"branch":"refs/heads/master"},"980b8499e241765b374fd0f56e7367dd3e9460b7":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2026-06-22 09:49:24.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/18"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-06-22 09:49:24.000000000","tz":0},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/993270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/980b8499e241765b374fd0f56e7367dd3e9460b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/980b8499e241765b374fd0f56e7367dd3e9460b7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f2eb47c4d6489a287db84ea2462c599ad40c90ec":{"kind":"REWORK","_number":19,"created":"2026-07-14 10:48:14.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/19"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-07-14 10:47:27.000000000","tz":180},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/993270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2eb47c4d6489a287db84ea2462c599ad40c90ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f2eb47c4d6489a287db84ea2462c599ad40c90ec"}]},"branch":"refs/heads/master"},"ab0aa14b94854b00c3fe9bccd9c36c3a64e2a589":{"kind":"REWORK","_number":20,"created":"2026-07-15 07:39:41.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/20"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-07-15 07:39:21.000000000","tz":180},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/993270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ab0aa14b94854b00c3fe9bccd9c36c3a64e2a589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ab0aa14b94854b00c3fe9bccd9c36c3a64e2a589"}]},"branch":"refs/heads/master"},"a55a2d349da00c919af2f54780e2ead1166a1706":{"kind":"REWORK","_number":21,"created":"2026-07-15 07:42:51.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/21"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-07-15 07:42:38.000000000","tz":180},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/993270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a55a2d349da00c919af2f54780e2ead1166a1706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a55a2d349da00c919af2f54780e2ead1166a1706"}]},"branch":"refs/heads/master"},"87ebeee95eec28876cb1cb96cc3c601a7d56b677":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-07-15 12:18:25.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/22"}}},"commit":{"parents":[{"commit":"48141b5d7a1377cc4bd5e779c6e6355663fe816d","subject":"Revert \"Disable test_extend_attached_encrypted_volume_luksv1 due to 2116852\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/48141b5d7a1377cc4bd5e779c6e6355663fe816d"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-07-15 12:18:25.000000000","tz":0},"subject":"Add test cases for image caching and fix image base class","message":"Add test cases for image caching and fix image base class\n\nThese tests cover caching workflows with volumes and delete specific\nimage from the Glance cache.\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/993270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/87ebeee95eec28876cb1cb96cc3c601a7d56b677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/87ebeee95eec28876cb1cb96cc3c601a7d56b677"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"70c4143230559059ed0839956a2a153b36573c29","ours":"a55a2d349da00c919af2f54780e2ead1166a1706","theirs":"48141b5d7a1377cc4bd5e779c6e6355663fe816d","merge_strategy":"recursive","contains_conflicts":false}},"ef8ee2e9d08d3ed572a2c97312b091a81824b581":{"kind":"REWORK","_number":23,"created":"2026-08-05 12:22:50.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/23"}}},"commit":{"parents":[{"commit":"70c4143230559059ed0839956a2a153b36573c29","subject":"Merge \"Add response schemas for show_instance_action 2.84\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/70c4143230559059ed0839956a2a153b36573c29"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-05 12:21:41.000000000","tz":180},"subject":"Add test with delete specific image from cache","message":"Add test with delete specific image from cache\n\nNew test with delete specific image from cache\n\nFix image base class\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/993270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ef8ee2e9d08d3ed572a2c97312b091a81824b581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ef8ee2e9d08d3ed572a2c97312b091a81824b581"}]},"branch":"refs/heads/master"},"7977ee688c053cd0b21b7aeaf4f95a96df382d84":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2026-08-05 12:23:06.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/24"}}},"commit":{"parents":[{"commit":"fa340b1f5fff8f57a018be206644c81b6d36b3d4","subject":"Fix v2_98 schema chain for share APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fa340b1f5fff8f57a018be206644c81b6d36b3d4"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-05 12:23:06.000000000","tz":0},"subject":"Add test with delete specific image from cache","message":"Add test with delete specific image from cache\n\nNew test with delete specific image from cache\n\nFix image base class\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/993270\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7977ee688c053cd0b21b7aeaf4f95a96df382d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7977ee688c053cd0b21b7aeaf4f95a96df382d84"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"70c4143230559059ed0839956a2a153b36573c29","ours":"ef8ee2e9d08d3ed572a2c97312b091a81824b581","theirs":"fa340b1f5fff8f57a018be206644c81b6d36b3d4","merge_strategy":"recursive","contains_conflicts":false}},"869b81ef37fc8ae2339e844d2ec631f64a0fbc2c":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2026-08-05 12:23:41.000000000","uploader":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"ref":"refs/changes/12/957612/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/12/957612/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/12/957612/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/12/957612/25"}}},"commit":{"parents":[{"commit":"fa340b1f5fff8f57a018be206644c81b6d36b3d4","subject":"Fix v2_98 schema chain for share APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fa340b1f5fff8f57a018be206644c81b6d36b3d4"}]}],"author":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2025-08-17 12:41:22.000000000","tz":180},"committer":{"name":"Maxim Sava","email":"msava@redhat.com","date":"2026-08-05 12:23:41.000000000","tz":0},"subject":"Add test with delete specific image from cache","message":"Add test with delete specific image from cache\n\nNew test with delete specific image from cache\n\nFix image base class\n\nChange-Id: I3a593b4acf1322ef954661dc026548d79134cffd\nSigned-off-by: Maxim Sava \u003cmsava@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/869b81ef37fc8ae2339e844d2ec631f64a0fbc2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/869b81ef37fc8ae2339e844d2ec631f64a0fbc2c"}]},"branch":"refs/heads/master","description":"Edit commit message","conflicts":{"base":"70c4143230559059ed0839956a2a153b36573c29","ours":"ef8ee2e9d08d3ed572a2c97312b091a81824b581","theirs":"fa340b1f5fff8f57a018be206644c81b6d36b3d4","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
