)]}'
{"id":"openstack%2Ftempest~957753","triplet_id":"openstack%2Ftempest~master~I72fdfe98c5c8b11d76bd2b212174d5617b081e25","project":"openstack/tempest","branch":"master","topic":"bug/2112187","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-08-19 00:25:31.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2025-08-19 00:25:31.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-08-18 23:00:44.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I72fdfe98c5c8b11d76bd2b212174d5617b081e25","subject":"update tests for bug 2112187","status":"MERGED","created":"2025-08-18 16:04:33.000000000","updated":"2025-08-19 00:27:12.000000000","submitted":"2025-08-19 00:25:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957753-bug/2112187","meta_rev_id":"ab16509105864b90df70189cdf8870629c623aee","_number":957753,"virtual_id_number":957753,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-19 00:25:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2025-08-18 23:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2025-08-18 23:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2025-08-18 20:56:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2025-08-18 23:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-18 16:14:24.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-08-18 16:26:24.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-08-18 20:14:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a16357113b4ead6cb740cc8d9df5174b28410b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-18 16:04:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"757064f46f83a3016def3d9e2571c9fe7db82c65","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-18 16:14:24.000000000","message":"Patch Set 1: Code-Review+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb8182a7e0ffc8559dc889f91e7ea54f63ff67c3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-18 16:26:24.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc5215a358ba1c4858d0f92a31911bea19da0dcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-08-18 17:18:32.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"12e79754aebd542b02ec4f0f1586ebadb679002d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-18 17:28:14.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e599033038bc403decf16f0a66a652eda3e21a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 20:14:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6102278795e490aafe25d4c11b1abcd\n\n- grenade https://zuul.opendev.org/t/openstack/build/4d30c14aedb149a194097e7ba4ae7481 : SUCCESS in 54m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ec6cd59f8e342f990bb87029ade735f : SUCCESS in 1h 29m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/635a8933c0c34a71b34b2d5ac0982c54 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/060126c350e04dca86a86d9585ad3d04 : SUCCESS in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637ed2adf77d45cdafb8b2ee0fa29a48 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0746c784f081450ab60a1197bf66d58a : SUCCESS in 7m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94d5abbd5f78401f8057b43f66a2813b : SUCCESS in 7m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/24c80d68a4514a719264d47f2b6b4b02 : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37034ae8fae44f529c3cd3d6d1c4f8be : SUCCESS in 5m 11s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/b585e389e4254e1eacdef418f58930c7 : SUCCESS in 1h 00m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/01ce9287cfb84a7fbf3a431293df1c3a : SUCCESS in 1h 36m 21s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/b9f1e74a1ae84fa09a22162b7be3252b : SUCCESS in 1h 07m 11s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/7ceb7e1d27114bbeb856ab6a51b451b0 : SUCCESS in 1h 26m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a4219a00279046f1b0ddd7370b17728d : SUCCESS in 59m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6293e798f1554de894f37cc63e3a1e42 : SUCCESS in 44m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37ae5fec5fb74218ad0a5a7f39cb972c : SUCCESS in 29m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b971e77d1109478ab8825cbd82d12ba3 : FAILURE in 2h 46m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/96d47c7519584a3ab4b8f1a3bd43a91c : SUCCESS in 51m 59s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0f3978004a754736830b553f678f5309 : SUCCESS in 1h 36m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/92eb1d58fd9943b3a45cf0206643e605 : SUCCESS in 1h 05m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d49b0d0b5bca4158aa1520e273b782a8 : SUCCESS in 30m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ad974aa854a4b6b9d4baba95ade3046 : SUCCESS in 45m 40s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/8bda30d9f5014e11bf2103be4fe6faec : SUCCESS in 59m 24s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/6694cf070f774b049865d0059311a8b7 : SUCCESS in 2m 37s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/548e1839e1cf4ed99d811e117bf11f0a : FAILURE in 1h 23m 58s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/19d9fd9a5e5946ca90d353bd6e431a37 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b693d970ae4198cfbcfb024e78a6f57c15d06516","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-18 20:48:54.000000000","message":"Patch Set 2: Code-Review-1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5999eed2c5e44e3426b2757ce4085bd1bd7fdc36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-18 20:55:47.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f383929346f11218b63931ac86a2f6ee812ab5e6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-08-18 20:56:04.000000000","message":"Patch Set 3: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1bf9045cba81db18ab7020a035f90846dd2520ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 22:57:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2039ea3eb5b946779dabd6f3f262a5f9\n\n- grenade https://zuul.opendev.org/t/openstack/build/f880ff33f33a4237aa5aaf4994b251b4 : SUCCESS in 1h 00m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36b63c65a13640198713ebf238847c0f : SUCCESS in 1h 58m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/919387294f6a4bd8ac82763987529db2 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202f5187f7964e7684359ecc74bfcb71 : SUCCESS in 18m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e24e92ba5d374579ba0c8439818720ec : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9647a1d28e8440e294b83faf875c4cca : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/722dd0e16c774a95bbef2516addb211a : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6790cb3c09af4c35a2fcfa3de02ca898 : SUCCESS in 6m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9c8ae3385774b7fb1fdd81ff344ce12 : SUCCESS in 2m 39s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/ed7f85277a7d4af89fdce93366c003e0 : SUCCESS in 1h 29m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d7939e8645a24f529066cffac27e16d0 : FAILURE in 1h 52m 15s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/cec4ec837ddf4eb6947de4483a88963d : SUCCESS in 1h 55m 01s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/1891351d7d984a899e029733465596cc : SUCCESS in 1h 25m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/02d857d746434795bd230d41c191c196 : SUCCESS in 46m 35s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/aa3325795c2947a1a03025c0eb00baf8 : SUCCESS in 29m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1f3b5ee19a547fcb046449e1252f8d2 : SUCCESS in 45m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8255de83770f4087bf1873abde548bd8 : SUCCESS in 1h 19m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f927e193907c47b0812a4259c4be5f46 : SUCCESS in 43m 10s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/921a346ac1af4ba68395d5a20c48d559 : SUCCESS in 1h 47m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c66643f3f20473986b42dbedc02839d : SUCCESS in 1h 36m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4349f249f4464c268073958dfa0889a5 : SUCCESS in 54m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/095ac15fcdca4558a5e1a1c12cc70c63 : SUCCESS in 1h 02m 58s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/d7fad5a8a4474b5589358af1dd6f05ef : SUCCESS in 59m 30s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5ed3cde428f74f90884aad917c746c5c : SUCCESS in 3m 21s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3918d9317cbb45cfb1c8b396d9e84b1a : FAILURE in 1h 02m 17s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1bfbb628e6974f689234f7302153976a : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ede37cf473d1b7ac413f6cd828736f97219bd882","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-08-18 23:00:44.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"75c9482b3aa16d0bea5c675ac8dda3a2a2dc4463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 23:01:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2b79193e9183d4fd3ef1ba71f8f1ffa5adefc4fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 00:25:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e137a5e0ffae416289dc987f2efaf4c3\n\n- grenade https://zuul.opendev.org/t/openstack/build/9788ca3b22a54102a719a8f34e5a11ef : SUCCESS in 55m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26b8e6c11d7346dd8ee7ee95071c019f : SUCCESS in 1h 01m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d515d41a4d2436cb278a6d3b9075a64 : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/727a5d052153415c8575191f963c458d : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99608e6ed6524f28ad5d944c6e22da5f : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5df3140b180c46dc81e62108788cf74f : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19f4312edaf2445e94a2caba8f1440b8 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e678a57ea75490ea30c15cf05a7c3ab : SUCCESS in 3m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b8a2c7ce14aa40fdbe592ed00c15f985 : SUCCESS in 1h 23m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee14f8dad53a46149f5d975f651da3c7 : SUCCESS in 45m 47s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/b15d94544a1b4c148bd8f62893274e3c : SUCCESS in 39m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f6fb56dcdf944bf794952574f8e4c786 : SUCCESS in 45m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80eaac55878a453e8962ac247c312299 : SUCCESS in 34m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cf921304a1ae459e8081be9afb3a1aa4 : SUCCESS in 47m 13s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fe0e32b8665b4a6cbbed383a6a603b86 : SUCCESS in 56m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e6791b88437843a59d3b58d47edadf02 : SUCCESS in 1h 21m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6963814450794faea261c572df9ad800 : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":3},{"id":"6d91f7000760efd66697180574935fa461f824b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 00:25:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ab16509105864b90df70189cdf8870629c623aee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 00:27:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5adf85846727426cbe5617638771e5d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1cf1f4e7564f589f00d5e59acd823e : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"290069b157d948b76ab04993d8b6b5ae9e74e212","revisions":{"6bb3a69ab1f6650c5cf55ee2493a44a7c52e93c9":{"kind":"REWORK","_number":1,"created":"2025-08-18 16:04:33.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/53/957753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/53/957753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/957753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/957753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/957753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/53/957753/1"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-07-08 23:18:05.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-08-18 16:04:10.000000000","tz":60},"subject":"update tests for bug 2112187","message":"update tests for bug 2112187\n\nThis change updates:\n\ntempest.api.compute.admin.test_volumes_negative.\nVolumesAdminNegativeTest.\ntest_update_attached_volume_with_nonexistent_volume_in_body\n\nand\n\ntempest.api.compute.admin.test_volume_swap.TestVolumeSwap.\ntest_volume_swap\n\nto account for the fact that only cinder may call attachemnt\nupdate aka swap_volume.\n\nThe tests now assert the expected 409\nand are annotated with the skip_because decorator\nto allow the nova change to be backported\nwhile allowign tempest to pass on both master and stable branches\nby skiping the affected test until all branches are in sync.\n\nChange-Id: I72fdfe98c5c8b11d76bd2b212174d5617b081e25\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nRelated-Bug: #2112187\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6bb3a69ab1f6650c5cf55ee2493a44a7c52e93c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6bb3a69ab1f6650c5cf55ee2493a44a7c52e93c9"}]},"branch":"refs/heads/master"},"aa963ad49436830f994345a0e9ee1426bee121f1":{"kind":"REWORK","_number":2,"created":"2025-08-18 17:18:32.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/53/957753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/53/957753/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/957753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/957753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/957753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/53/957753/2"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-07-08 23:18:05.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-08-18 17:15:26.000000000","tz":60},"subject":"update tests for bug 2112187","message":"update tests for bug 2112187\n\nThis change updates:\n\ntempest.api.compute.admin.test_volumes_negative.\nVolumesAdminNegativeTest.\ntest_update_attached_volume_with_nonexistent_volume_in_body\n\nand\n\ntempest.api.compute.admin.test_volume_swap.TestVolumeSwap.\ntest_volume_swap\n\nto account for the fact that only cinder may call attachemnt\nupdate aka swap_volume.\n\nThe tests now assert the expected 409\nand are annotated with the skip_because decorator\nto allow the nova change to be backported\nwhile allowign tempest to pass on both master and stable branches\nby skiping the affected test until all branches are in sync.\n\nChange-Id: I72fdfe98c5c8b11d76bd2b212174d5617b081e25\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nRelated-Bug: #2112187\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aa963ad49436830f994345a0e9ee1426bee121f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aa963ad49436830f994345a0e9ee1426bee121f1"}]},"branch":"refs/heads/master"},"290069b157d948b76ab04993d8b6b5ae9e74e212":{"kind":"REWORK","_number":3,"created":"2025-08-18 20:55:47.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/957753/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/53/957753/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/957753/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/957753/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/957753/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/53/957753/3"}}},"commit":{"parents":[{"commit":"e7e3947694b4412a09c7d151c78a3fd47033adfc","subject":"Merge \"Add more server migration tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7e3947694b4412a09c7d151c78a3fd47033adfc"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-07-08 23:18:05.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2025-08-18 20:51:52.000000000","tz":120},"subject":"update tests for bug 2112187","message":"update tests for bug 2112187\n\nThis change updates:\n\ntempest.api.compute.admin.test_volumes_negative.\nVolumesAdminNegativeTest.\ntest_update_attached_volume_with_nonexistent_volume_in_body\n\nand\n\ntest_multiattach_rw_volume_update_failure\n\nand\n\ntempest.api.compute.admin.test_volume_swap.TestVolumeSwap.\ntest_volume_swap\n\nto account for the fact that only cinder may call attachemnt\nupdate aka swap_volume.\n\nThe tests now assert the expected 409\nand are annotated with the skip_because decorator\nto allow the nova change to be backported\nwhile allowign tempest to pass on both master and stable branches\nby skiping the affected test until all branches are in sync.\n\nChange-Id: I72fdfe98c5c8b11d76bd2b212174d5617b081e25\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nRelated-Bug: #2112187\nSigned-off-by: Balazs Gibizer \u003cgibi@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/290069b157d948b76ab04993d8b6b5ae9e74e212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/290069b157d948b76ab04993d8b6b5ae9e74e212"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
