)]}'
{"id":"openstack%2Ftempest~964163","triplet_id":"openstack%2Ftempest~master~I310819900fc48d9cb3d31d4b6cafe05bfd1fc3c7","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"35834":{"account":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"last_update":"2026-01-05 11:27:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I310819900fc48d9cb3d31d4b6cafe05bfd1fc3c7","subject":"Test coverage for Cinder upload volume back to glance","status":"ABANDONED","created":"2025-10-15 12:34:18.000000000","updated":"2026-01-05 11:27:23.000000000","total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"b44a5a63e13fc0a1617d9013054f90c1ff67211b","_number":964163,"virtual_id_number":964163,"owner":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-16 15:27:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-10-20 09:07:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"CC":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-15 14:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-20 09:07:16.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"},{"updated":"2025-12-25 12:54:34.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"CC"}],"messages":[{"id":"ded6c2a3e72c89a6db3875b209bdb4119734c89d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"date":"2025-10-15 12:34:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cfa1404771ec89ac94c326f18d8b5ae60bd1e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 14:36:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd8b76c0171349d591878a48d96aabb6\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b0c7d401a4a4a588c9458232f824f9c : SUCCESS in 32m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/540d32ae84154a458647ede9dae43bb7 : SUCCESS in 1h 11m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9816f76e6b74c729e16056a166667c6 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aff18f215e4418eac2a417b3971d140 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a987f25fa16c400ab4af9d593f55b1e9 : FAILURE in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13a958ebfedc48afae20b0139456fd9b : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b962571b0724582b73140a42337a83a : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d625d90c7314ed08827e0a4351c5cce : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e85b7de6e684766ac9b67df7f1d709f : SUCCESS in 4m 05s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/4e63b0cd5b46405e939abe158e1dca8a : SUCCESS in 1h 03m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5a53c996abb049118d8fcaaab7b511e9 : FAILURE in 1h 49m 07s (non-voting)\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/4dd16a809dc2447eb3836877141bc396 : SUCCESS in 1h 15m 10s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/11bac07b032349eeac67a9c38dc55834 : SUCCESS in 1h 09m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/67f720c9a3f14bd784cc0d6c8ef73b04 : SUCCESS in 1h 08m 36s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3a86e457a52c4cadab5fbb5408f5b1b4 : SUCCESS in 24m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4378aa0c589a4fc58e5165eccbf08fda : SUCCESS in 1h 03m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57ade511d644437b9032a3b4b191a265 : SUCCESS in 1h 47m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ed988c1baa3d40748691c428164f9b4b : SUCCESS in 36m 25s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f1cc74b34c214018bc0ededfec1eff52 : SUCCESS in 1h 39m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/01100d81328f444ab01976946cf75766 : FAILURE in 1h 20m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/74955caf725c4eb09a746b84e443ce44 : SUCCESS in 56m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e5094315afb14d629735dd29b0dcd077 : SUCCESS in 1h 04m 59s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/cca6f8d633f94d72b1b84fdac8c04754 : SUCCESS in 1h 20m 06s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c042aa7145f7425ca99eb2b39c28ffd6 : SUCCESS in 3m 23s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c926e61af15a4a99950cc88e20f5e77e : FAILURE in 1h 05m 33s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/78fdf5199f4d4c28af52cb25a36d444b : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6bdaacbc1b2a310536a248f5674838e5ea73a8df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"date":"2025-10-16 06:35:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"34606d7715346513396d890dd3dee2eb65d53b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 09:05:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/69314fa3449f412584b464e5642d0857\n\n- grenade https://zuul.opendev.org/t/openstack/build/47d50454178e4ba8a712ed60d4f07c3c : SUCCESS in 1h 02m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/513379f6e8d54c58a48e6b139a1b8df4 : SUCCESS in 1h 22m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a816a00241b4ca4ac43758d50f22fc2 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38c421797a64322bf1cd83f309bc95f : FAILURE in 30m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e7fe627ab27407292476ea9b5793742 : FAILURE in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9e7b9eea83a46ee8e15205cb191ac02 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f32565e475154602ac02fb2ca59509dc : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a0a1309431948ceb988da026531a623 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00f51d1a517743e28081ac0b9ca9efde : SUCCESS in 4m 38s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/a7e67b1ebcdf4a0286256b4cb84aa3be : SUCCESS in 1h 13m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d4aad6874dcf42f6ac4551a51aadbf2c : SUCCESS in 2h 23m 55s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/1d06b2ac0c234e6f8515dad2dfd9974d : SUCCESS in 1h 19m 09s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/8d02d3b47c5240d5a6441905ddfac143 : SUCCESS in 1h 31m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bc86fa96a7024a23880469275abc0b59 : SUCCESS in 1h 17m 05s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4148ca7bea854af69173079bd1552d71 : SUCCESS in 47m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61f1ac0fbb6245cc8492eb3b930e83ed : SUCCESS in 56m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c063f7426694c47a6868e75872d5d57 : SUCCESS in 1h 38m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a1e7fbb72fad4dabbd7a3dba9677f6ed : SUCCESS in 54m 09s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f12b6b7aeac4190bff615bc168ae57a : SUCCESS in 1h 39m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d9d1c1907e8149bbb312f2528ac6bc82 : SUCCESS in 1h 53m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/86932341fab54597b405f9ec3e1ffe2e : SUCCESS in 55m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85f8785a8c3c42ceb101aeaf61353397 : SUCCESS in 51m 33s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1d84e895a2664c54abc645dccb15b80e : SUCCESS in 39m 06s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a240f2547d22428ab4942f46abb962e6 : FAILURE in 1h 07m 33s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bb390558412e468cb3ea1139687247a7 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"7b406d32defd1945a67f59796a62d03c2d03fe04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"date":"2025-10-16 13:31:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7b3857e013dc4920eec3e9c17db074b09da90fee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 15:27:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/264f63220f0544458561a221179c2a19\n\n- grenade https://zuul.opendev.org/t/openstack/build/f38e5e49532a47d4a765202948f1f865 : SUCCESS in 1h 02m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d52b51b2e5ec4a81ae4708d718467efc : SUCCESS in 1h 43m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6dd85c47c044b72828b4892602022c8 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115ab8cdb47747a7a633ea59eb589227 : FAILURE in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a535aaf1eec4ff78f9fc85d2aa51969 : FAILURE in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/937b9e90533c45939010391c6b7a2b32 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/289f2d0b22ab4658a36db5c06f45102d : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76b66c428105460d920d4d33a27e572b : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8d2193c8ab043eca679b4a17673b718 : SUCCESS in 4m 55s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/0360b76a9a8945d384117a23aa8bfa47 : SUCCESS in 59m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/49270bc28b9c430abb9591d856908658 : SUCCESS in 1h 54m 32s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/9aa9ef8f6e334754a52d9544540e7dc0 : SUCCESS in 1h 21m 20s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/2d539030597048718aa57368c2dabe22 : SUCCESS in 1h 33m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2e9c36cd56cf4e00b58660e0628697a6 : SUCCESS in 1h 06m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/79c507408c0f44efbed54bd9954ce3d1 : SUCCESS in 29m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6aab73fa0cb442f4989e0f33108f1104 : SUCCESS in 29m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eed3d9bc096e4f14b7efa04da3398881 : SUCCESS in 1h 02m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a91b15d338c4401f95b7e27526cddbbe : SUCCESS in 37m 49s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/698b4be28ffb4942842c105fd829e109 : SUCCESS in 1h 39m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89232517ee77421899968b86101f16ab : SUCCESS in 1h 51m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/38e50fc954444c4690deb2019bcb511a : SUCCESS in 59m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a81162d9702b4d76bf36b09b9dc6f7c9 : SUCCESS in 1h 02m 23s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/d7de9c4eb3c449d1950b7e5324043ac2 : SUCCESS in 41m 30s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c7b570916d7649dfb5e36d632a7c135c : FAILURE in 1h 21m 01s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b49062b5197641c088f753506985d327 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"a7e94f4c2f67b250c7d4f1a1fa17677ec0b314b8","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2025-10-20 09:07:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a17db5aa2d426a7d8f69694eb1d173892e9735d","author":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"date":"2025-12-24 11:13:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10737d7e25b4f9e1f7dc704e816f5d619f2bca3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 13:33:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31999f34d0e74ab3b4e5a477d3b71d47\n\n- grenade https://zuul.opendev.org/t/openstack/build/0df3b004d1034546ab538b22cbc7dcc8 : SUCCESS in 1h 01m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bba0d71f53094ed399590af916193369 : SUCCESS in 1h 41m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ccde2584657489db6cd5ae8332b2b34 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3959dfedb04faab675e53be6b73449 : FAILURE in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c640bbc6e1874f659e34be99c2614f39 : FAILURE in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39e0e53f92d94fedb6c4d0483cc41452 : SUCCESS in 2m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/284e1f391427447eaca2415301043cea : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84d0e65dd0904bad9746d63b24bd04a4 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/594b04bef36f4043b2b96dc3028f8ffb : SUCCESS in 4m 09s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/858cb04756dc4657acfd57e5ded374c6 : SUCCESS in 40m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9bea5e9c6e4a4f29946fb35aacdd2835 : SUCCESS in 2h 14m 39s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/3f74f906e3d142be9a0283581db266c9 : SUCCESS in 1h 46m 07s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/3b70f841e612401bb8fd064c995e4871 : SUCCESS in 1h 35m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/82d32d8cd57044b5b918c7c018973896 : SUCCESS in 54m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/8fbc18c5bf4e418d95a53277094759e2 : SUCCESS in 34m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abb52b05b2cb42629b5380b5cba08a35 : SUCCESS in 1h 05m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ce44102bfaf4a8ba1b39a75b76d00a4 : SUCCESS in 50m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6c68f2079da942358cd5eb559b4771e3 : SUCCESS in 29m 41s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/32cc53265fd44be7944c41e0d45f6068 : SUCCESS in 1h 43m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c2adce7f6c5d4a3083c71e92996ad3c5 : SUCCESS in 1h 24m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc79b08b5b5243e297bfb0f372a17b9a : SUCCESS in 1h 11m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b43e446ecd9e49d783732cf8520b4408 : SUCCESS in 54m 54s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/430dfae2a1854fbe9201432e730542d8 : SUCCESS in 37m 12s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e002d5e29b684530bfaa4a684c7dfbc1 : FAILURE in 1h 04m 12s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/2b286065e0d6406b9e37a646f629e365 : SUCCESS in 45m 57s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c5d59f1c8f1b438fbb5f75dd69fd5de1 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":3},{"id":"3b09d1046bf95d4ea121f21f7bce38c5efe319c9","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2025-12-25 12:54:34.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b44a5a63e13fc0a1617d9013054f90c1ff67211b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"date":"2026-01-05 11:27:23.000000000","message":"Abandoned\n\ntest moved to openstack/cinder-tempest-plugin https://review.opendev.org/c/openstack/cinder-tempest-plugin/+/972089","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"180d9eff4ff9971cdd1bf0ee74e2ecc0332d2701","revisions":{"7f8377b179b9fdd2f702e857a274360972da9142":{"kind":"REWORK","_number":1,"created":"2025-10-15 12:34:18.000000000","uploader":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"ref":"refs/changes/63/964163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/63/964163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/63/964163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/63/964163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/63/964163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/63/964163/1"}}},"commit":{"parents":[{"commit":"d14ac45f67b831bcd0765b276404053db5fda495","subject":"Create releasenotes page for tempest 44.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d14ac45f67b831bcd0765b276404053db5fda495"}]}],"author":{"name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","date":"2025-10-15 12:32:36.000000000","tz":180},"committer":{"name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","date":"2025-10-15 12:34:16.000000000","tz":180},"subject":"Test coverage for Cinder upload volume back to glance","message":"Test coverage for Cinder upload volume back to glance\n\nThis test copies image to another store, create a volume from that image\nand then upload the created volume back to glance.\nThis test requires multistore enabled\n\nChange-Id: I310819900fc48d9cb3d31d4b6cafe05bfd1fc3c7\nSigned-off-by: Yosi Ben Shimon \u003cybenshim@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7f8377b179b9fdd2f702e857a274360972da9142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7f8377b179b9fdd2f702e857a274360972da9142"}]},"branch":"refs/heads/master"},"f66e4e54478b3f55eb5e56e3b624238e7b77cda6":{"kind":"REWORK","_number":2,"created":"2025-10-16 06:35:59.000000000","uploader":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"ref":"refs/changes/63/964163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/63/964163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/63/964163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/63/964163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/63/964163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/63/964163/2"}}},"commit":{"parents":[{"commit":"d14ac45f67b831bcd0765b276404053db5fda495","subject":"Create releasenotes page for tempest 44.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d14ac45f67b831bcd0765b276404053db5fda495"}]}],"author":{"name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","date":"2025-10-15 12:32:36.000000000","tz":180},"committer":{"name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","date":"2025-10-16 06:35:46.000000000","tz":180},"subject":"Test coverage for Cinder upload volume back to glance","message":"Test coverage for Cinder upload volume back to glance\n\nThis test copies image to another store, create a volume from that image\nand then upload the created volume back to glance.\nThis test requires multistore enabled\n\nChange-Id: I310819900fc48d9cb3d31d4b6cafe05bfd1fc3c7\nSigned-off-by: Yosi Ben Shimon \u003cybenshim@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f66e4e54478b3f55eb5e56e3b624238e7b77cda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f66e4e54478b3f55eb5e56e3b624238e7b77cda6"}]},"branch":"refs/heads/master"},"180d9eff4ff9971cdd1bf0ee74e2ecc0332d2701":{"kind":"REWORK","_number":3,"created":"2025-10-16 13:31:38.000000000","uploader":{"_account_id":35834,"name":"Yosi Ben Shimon","display_name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","username":"ybenshim"},"ref":"refs/changes/63/964163/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/63/964163/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/63/964163/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/63/964163/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/63/964163/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/63/964163/3"}}},"commit":{"parents":[{"commit":"d14ac45f67b831bcd0765b276404053db5fda495","subject":"Create releasenotes page for tempest 44.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d14ac45f67b831bcd0765b276404053db5fda495"}]}],"author":{"name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","date":"2025-10-15 12:32:36.000000000","tz":180},"committer":{"name":"Yosi Ben Shimon","email":"ybenshim@redhat.com","date":"2025-10-16 13:31:28.000000000","tz":180},"subject":"Test coverage for Cinder upload volume back to glance","message":"Test coverage for Cinder upload volume back to glance\n\nThis test copies image to another store, create a volume from that image\nand then upload the created volume back to glance.\nThis test requires multistore enabled\n\nChange-Id: I310819900fc48d9cb3d31d4b6cafe05bfd1fc3c7\nSigned-off-by: Yosi Ben Shimon \u003cybenshim@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/180d9eff4ff9971cdd1bf0ee74e2ecc0332d2701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/180d9eff4ff9971cdd1bf0ee74e2ecc0332d2701"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
