)]}'
{"id":"openstack%2Ftempest~966203","triplet_id":"openstack%2Ftempest~master~Id1bf48b356fe9d0edbe316c91c94e9c4d3b37fa8","project":"openstack/tempest","branch":"master","topic":"uefi","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-05 18:41:00.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Id1bf48b356fe9d0edbe316c91c94e9c4d3b37fa8","subject":"wip: Add scenario test to verify vTPM support","status":"NEW","created":"2025-11-05 16:42:59.000000000","updated":"2025-11-06 15:46:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97602a850c35429ebc02a6d56ae6301749435901","_number":966203,"virtual_id_number":966203,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-06 15:46:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-05 18:41:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e76a5cff25dddc66aee297608441c2f54feaf381","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-05 16:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62bad1dbc9e7284d3c7b37093eb12bc9d2e9b906","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 18:41:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d98aa931c98f401f96b1370e751bcb8e\n\n- grenade https://zuul.opendev.org/t/openstack/build/03ea3182930e4726b21974c3d6b3404b : SUCCESS in 50m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dafe7f03ec945909ab51638ecf6ff7c : FAILURE in 1h 52m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d954edba9434b05830c1c1b49396eb9 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc935d2029c94c338d8d681b1f0e022a : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ed3ba0b2c74f2cb5e16303fb32c0c1 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/849b87a9aff24eb4a9b7c61374b87398 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fec3d94d35b948d5b723e63b5f894b89 : SUCCESS in 6m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bfde776f19df41a688e400c0f3aae5eb : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f239fa2ae6a2483f80a90953d8867083 : SUCCESS in 7m 45s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/828ebdb8456c48e1bfeee0b5b61c7137 : SUCCESS in 1h 22m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e36e58b831c94094ad0f948fe632d97e : FAILURE in 1h 50m 28s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/bc0502f2fc9e41c0ab2bc4bcb1f6afdd : FAILURE in 1h 36m 01s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/64df82129851419ab1f37cdb17bb41bb : FAILURE in 14m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2aa27acef97f4ac0995d50943b22c26a : SUCCESS in 1h 10m 51s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c0c3ad300279481985d5555f8fb19410 : SUCCESS in 32m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a05169dfacd441a9a1c58dfa608a26b9 : SUCCESS in 58m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28464f05049a4d9ba1e2d7b41fa8b310 : SUCCESS in 1h 47m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5bd884937c44457ea371e5937d45c35b : SUCCESS in 58m 44s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b66103746c964afe9b37559721378fdc : FAILURE in 1h 48m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ec6e96d62f5426d97fe3ea29f60d8f4 : SUCCESS in 1h 48m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3874a454a0004e46a2824dfa47a15d98 : SUCCESS in 1h 08m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/77ab12c6316c40ae96dd4adf8a05cfd2 : SUCCESS in 55m 05s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2e7b0543f742472fadcd1a101a039e42 : FAILURE in 1h 13m 23s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ee672411f64b4bccbe24f7f9eaa773ca : FAILURE in 1h 12m 42s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e9265202b2c840a49067b5e04ca25d4b : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":1},{"id":"58423ca13aa948e9f7a578b1cd1d7682275f54c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-06 04:56:11.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fdd3ec91d060ed0d3a509eae669fda50695c1d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 07:39:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8799e6f8ca994dd3b12d017ba14208bd\n\n- grenade https://zuul.opendev.org/t/openstack/build/b61f29822fd04143aaa5ea0abeccbff2 : SUCCESS in 57m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eabfae3e172e4674a59a088ed26143a9 : FAILURE in 1h 43m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66909e761c424c7e8183d20897ac74bd : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea261f70c3304ba4b11b8542072540c2 : SUCCESS in 13m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84e18e0b6c134a0ba264b5159d8dbe59 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a6cd1f5f789477798a1e3dc7e410353 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a71086ca39c24209a7670982160d09a0 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0588203761144f0967e7c50aee20fa7 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/763b57250f924d158d1a7d28fb896e99 : SUCCESS in 7m 00s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/c9d40a3b8e7d49e58c60af83ffc93ae8 : SUCCESS in 1h 18m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2e5fa6a90ccb4e438bd4def084971fa1 : FAILURE in 2h 37m 50s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/d296d2b534a54d3d855d42ae1ca3a3e0 : FAILURE in 1h 37m 36s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/524c62893d9c463f8c3c3583b5c544dc : FAILURE in 9m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1391fb577fc74a9ca7e2ef4740c58907 : SUCCESS in 1h 09m 23s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c5ae4dafcbb94760aa3ccdaf9b0ccf34 : SUCCESS in 16m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fb8907c8d8948b5aae03e2996cd324f : SUCCESS in 1h 01m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b4c165ba698144ef8371ab26486cc17a : SUCCESS in 1h 46m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/34fa47533eea49c4bbfd9c16b28c2ac7 : SUCCESS in 53m 49s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4411a8b0e0fc4bce9c6246c5b13f0c66 : FAILURE in 1h 54m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b7dbf935b16144249d6e544dd63efc28 : SUCCESS in 1h 53m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d68797307b34c218a0c5617a1af149e : SUCCESS in 57m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/10bb0019c54c40868d1b411aee62ce72 : SUCCESS in 53m 58s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/58958deeca544b429acbf40a76e2a81b : FAILURE in 1h 09m 09s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4d3ea7d3f8e5480993e276bdd4929a08 : FAILURE in 53m 18s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/60cd6bf002194e039b9ff33d47ebae19 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"16c8b8116076a8695d55b836f69a48eff7de08cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-06 12:37:51.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"97602a850c35429ebc02a6d56ae6301749435901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 15:46:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6425e69dbdb8420d9f92470171cae048\n\n- grenade https://zuul.opendev.org/t/openstack/build/18375d5e789d435aa9a96493a984ca8d : SUCCESS in 58m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5ca1f2e6bb94bca993d05edca453da0 : FAILURE in 1h 57m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18de8cf80dcb4ddaadbb6d27f3687ea3 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e03e708cb9c44e4aadd62997133f2a7 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09f52d7698242d58c2bb20daaaa3bd3 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fbda08e9bb6464ea65d33e851309a1d : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6155bc6354724d27af044b6765c9754a : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7144e73a29b04f14bb8ff77993a17315 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e66022f20224756b3921f50773c7082 : SUCCESS in 7m 18s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/799439b7153a465d9c73392544029801 : SUCCESS in 1h 25m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/75f84951db50405b950c6fa4df78b91c : FAILURE in 2h 57m 13s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/883f9fe618584fa5b6af1f5cd745c729 : FAILURE in 1h 42m 32s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/8f08a201d43a49b2a2b97796084e0bb8 : FAILURE in 11m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d63c48737d314e23a8bef4aff3a472e3 : SUCCESS in 1h 21m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d00e927e0e9c47ab89e71daf9df8119f : SUCCESS in 36m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bede127ee49424ba71f3bb9d1d6aae7 : SUCCESS in 1h 02m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4a847ff29c6e40b3805a2aa0c8fd7a14 : SUCCESS in 1h 43m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/142fe6378d6442fd8f4c467b541c066c : SUCCESS in 1h 00m 47s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e00a363eacd14ddc919930e4d266ed09 : FAILURE in 1h 36m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b148ea46e3cd445e98e4febe8d81ff23 : SUCCESS in 1h 47m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a04e96680294239a094fde7ad5e874f : SUCCESS in 1h 15m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7babf04a69442c3b6d89e0f07360562 : SUCCESS in 1h 00m 45s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/54e5ba95023d41b2a19de09a5fe4adf9 : FAILURE in 1h 22m 05s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e29f61499e0445139e8c7c621cd6034d : FAILURE in 1h 35m 06s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/21953d3c2dd34864904351dea15f3008 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"550c97ca0d0cacc114e54d26dd7f027d55083dad","revisions":{"1a401e36c0e447d09a907f6ff428c0ba56b90ea5":{"kind":"REWORK","_number":1,"created":"2025-11-05 16:42:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/966203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/03/966203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/03/966203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/03/966203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/03/966203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/03/966203/1"}}},"commit":{"parents":[{"commit":"1ca57c3d447d6fba5e3ff1e1b3355aeef50d70da","subject":"Add tests with UEFI boot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1ca57c3d447d6fba5e3ff1e1b3355aeef50d70da"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-05 16:42:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-05 16:42:48.000000000","tz":540},"subject":"wip: Add scenario test to verify vTPM support","message":"wip: Add scenario test to verify vTPM support\n\nChange-Id: Id1bf48b356fe9d0edbe316c91c94e9c4d3b37fa8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a401e36c0e447d09a907f6ff428c0ba56b90ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a401e36c0e447d09a907f6ff428c0ba56b90ea5"}]},"branch":"refs/heads/master"},"8fc2c7bb44366d53b89fc2b41798a392ee62bdfb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-06 04:56:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/966203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/03/966203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/03/966203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/03/966203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/03/966203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/03/966203/2"}}},"commit":{"parents":[{"commit":"1ca57c3d447d6fba5e3ff1e1b3355aeef50d70da","subject":"Add tests with UEFI boot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1ca57c3d447d6fba5e3ff1e1b3355aeef50d70da"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-05 16:42:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-06 04:56:11.000000000","tz":0},"subject":"wip: Add scenario test to verify vTPM support","message":"wip: Add scenario test to verify vTPM support\n\nDepends-on: https://review.opendev.org/c/openstack/devstack/+/966236\nChange-Id: Id1bf48b356fe9d0edbe316c91c94e9c4d3b37fa8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8fc2c7bb44366d53b89fc2b41798a392ee62bdfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8fc2c7bb44366d53b89fc2b41798a392ee62bdfb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"550c97ca0d0cacc114e54d26dd7f027d55083dad":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-06 12:37:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/966203/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/03/966203/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/03/966203/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/03/966203/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/03/966203/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/03/966203/3"}}},"commit":{"parents":[{"commit":"1ca57c3d447d6fba5e3ff1e1b3355aeef50d70da","subject":"Add tests with UEFI boot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1ca57c3d447d6fba5e3ff1e1b3355aeef50d70da"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-05 16:42:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-06 12:37:51.000000000","tz":0},"subject":"wip: Add scenario test to verify vTPM support","message":"wip: Add scenario test to verify vTPM support\n\nChange-Id: Id1bf48b356fe9d0edbe316c91c94e9c4d3b37fa8\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/550c97ca0d0cacc114e54d26dd7f027d55083dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/550c97ca0d0cacc114e54d26dd7f027d55083dad"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
