)]}'
{"id":"openstack%2Ftempest~966315","triplet_id":"openstack%2Ftempest~master~I506b44b97413bb3318cac865036abd0ad045767d","project":"openstack/tempest","branch":"master","topic":"rework-tempest-install","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-01-31 19:36:44.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"removed_from_attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-02-17 14:16:30.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-01-31 19:36:44.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"34510":{"account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"last_update":"2026-05-17 07:00:42.000000000","reason":"\u003cGERRIT_ACCOUNT_34510\u003e replied on the change","reason_account":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}}},"hashtags":[],"change_id":"I506b44b97413bb3318cac865036abd0ad045767d","subject":"Remove deprecated regex opts","status":"NEW","created":"2025-11-06 17:41:49.000000000","updated":"2026-05-17 09:04:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c79aae3a86e2d2d0ad6432be0212d3f09284923d","_number":966315,"virtual_id_number":966315,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-30 13:57:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"all":[{"value":1,"date":"2026-02-17 14:16:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":1,"date":"2026-01-31 19:36:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 17:54:16.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-11-06 19:45:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-11 18:53:21.000000000","updated_by":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"reviewer":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"state":"REVIEWER"},{"updated":"2026-02-17 14:15:12.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"40a8b43faf1670f626783a480ab4202c3518b9d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-06 17:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d0f4a2a6ba89fc81d49941eeed2ec51189feffa","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-06 17:54:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f598bf6ddc10e259eb7216c604e23fda75644f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 19:45:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8a5a395a304460aa226f1f5acc91508\n\n- grenade https://zuul.opendev.org/t/openstack/build/64c7345901be4193adbe7b47dd10e6d1 : SUCCESS in 49m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c086ae7d82774cf09da718e01eaa05fe : SUCCESS in 1h 43m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e9ed70be1e648f4b117d4a572068ee2 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633ddf979fe44f6fa3b7fa50c10eb7cb : SUCCESS in 13m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59f09e6cf2294a838aeaaba03e9b18db : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cd9fb098ecd4df2a7e9351a003a16d5 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78cc089176e54f81b9336c07682423fa : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/218aaf8e90104729991c300535dd93fe : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e1ccbdf37ae4fbd93568b4686b3b5cf : SUCCESS in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1661cf56e767436a8d9e1bcd0227f7dc : SUCCESS in 5m 40s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/8220a8e833344fab9763c0a9c397a094 : SUCCESS in 1h 33m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0ea0c0046ed0436da7b901901496f729 : SUCCESS in 1h 17m 04s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/bb853be2c0724c2dbd34211752e090b3 : SUCCESS in 1h 37m 57s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/103dd913c98741e88d147b18e314eb2c : FAILURE in 11m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c08a71dfcefa4a4391f6aa754a805fcc : SUCCESS in 56m 44s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/fe0080afb8bf44e0aa089206deb79b4e : SUCCESS in 41m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a2905afb3cb4340b56f90cff141d7f2 : SUCCESS in 52m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/29fc5b18483541c3b57b412d5b7e56ed : SUCCESS in 1h 13m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7cd1ad969617411aa8df163fb5063f56 : SUCCESS in 45m 53s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4d1866a8fbca4122b8f67d0cacecbbc7 : SUCCESS in 1h 39m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/03c3984e19d348ea83fc0efcb2a70b76 : SUCCESS in 1h 52m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/16ebe75ef75248cfa56dcaa8f48c096d : SUCCESS in 1h 05m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2da3b75416dd4bb79a54793c95cd3d49 : SUCCESS in 46m 02s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/91591b768efe47dcb31efdeffa11591c : SUCCESS in 1h 16m 05s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3a22ba65cd304cb1b9665d951cde1499 : FAILURE in 1h 00m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3b4be712298648cb82338b85f6bc2265 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a53cf00bd4ffe84e5c4f8ae44bc7578ad273b86a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-07 10:49:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"eb4c67f0784be4556e32b2228e2cef8d4fce0f49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 13:24:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa7e63cacbe84d688382d7ec0cddeda2\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e592001beb44b9eb476a26340b81823 : SUCCESS in 1h 15m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96273232906642b5a14a56f23d21360b : SUCCESS in 1h 33m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9041d7fd254d45bb9b6e2bb0a7b303ce : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70dc5020bba4519a09079c8a214a5b3 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9beaf1bf304f45d5b6c293d884a19a83 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eec3617dadd4534b94cd4e2b8f29f82 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f06555be31474a5b957cdd8d24db1186 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d2d683940a2451f8e45e57ebeced2b3 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb63ec83dcfa4fe6b73515cc702f49a8 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0e4c248c3674267a30ce208d103f93a : SUCCESS in 6m 32s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/607bd723537c479dab87bca2cf822a4e : SUCCESS in 1h 21m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7facb04a92b74573bbdce115ca3c32b5 : SUCCESS in 2h 28m 05s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/7e1820dc7927417db3e5650760358fe6 : SUCCESS in 1h 30m 42s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/acc934c6ae5c484fbab6986ceafbbe09 : SUCCESS in 1h 22m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0fe4ed3961c9471aa88b8520854fbccc : SUCCESS in 1h 15m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/078d9a2f5059485aa481d1fdf75446ed : SUCCESS in 36m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28d75d5347874880831c4023052dff9c : SUCCESS in 58m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1e469960e7d845f7bbc45beadf3004a5 : SUCCESS in 1h 53m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8403c7f3a3024082b377f0c7478bb972 : SUCCESS in 52m 30s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/35949faa2fbd43529a95a43d9846231b : SUCCESS in 1h 47m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ad230e5f63b45d1b0a45fd714394ebf : SUCCESS in 1h 36m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c76a32d5e214e33b6c2c27d878fca6c : SUCCESS in 1h 10m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6df9a9ce0e744649212399979e06ee9 : SUCCESS in 58m 45s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/522e0d4ec492403ea9c563fcd9b80018 : SUCCESS in 1h 20m 33s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a5d749c6f1da47c08a5003e3ae168f6b : FAILURE in 1h 20m 14s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/ada9da37ced6459c83a9ab6da45dafcf : SUCCESS in 1h 12m 54s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e5290b80a1e4481cb67c92f6a45d107c : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8f157cbbaec8d166e20c18fffa8896916beea860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-30 11:38:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"35f1ff2604b0d8fa8919e6787a786b14f1e5f0ae","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-30 11:42:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ac60f391b5d623f6ec680b5184cc272a8c43782f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 13:57:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba69959d4d04ee280f6bb69690add28\n\n- grenade https://zuul.opendev.org/t/openstack/build/202a5d484e8c4311a32f60ba6e2cda81 : SUCCESS in 30m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d86db99ac724405a905d05cc2f4511d1 : SUCCESS in 1h 51m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/387d32e3b7704c63b649a7e6532a4457 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c68ee14198b4a78abb4b8a13726647a : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2726d2223af14442bcb3c6b568d07fdc : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a5df63a32cb49089d56767fbed80ea2 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7904c917091843fc922d17ac3e343113 : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7beb4230fc804f5fa496ea6efdea0feb : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99cedaf733c34c5483a9b09d02f45972 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87e636c57bab4959b1f6ca7e9203c817 : SUCCESS in 3m 39s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/787fb762ab194d678fbc68b5529cb3d5 : SUCCESS in 48m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/20daadfdae1342b19020d902417fbca4 : SUCCESS in 2h 15m 27s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/d3cb20ed826842c49619945f2c6c6b36 : SUCCESS in 55m 47s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/ba091c1093964a9bb26196f668af4628 : SUCCESS in 1h 51m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/11b3c545ba594e1186dd4047aaff8016 : SUCCESS in 41m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0e62a6ddd98c493cbc3569addca12139 : SUCCESS in 26m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/016019858aae4219a088dac4867393e4 : SUCCESS in 58m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aeeb4d79664c4a5d969494bfd7e59170 : SUCCESS in 49m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ba0f4c365cde499898e3e23b89016909 : SUCCESS in 55m 33s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b5b66168a13545708f6ed394240af38d : SUCCESS in 1h 36m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6696f6a8f2e944178e101c9527426c05 : SUCCESS in 1h 36m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/17441d42530b4175a976cff40d7d0f85 : SUCCESS in 45m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fea8a4712ed84f78821ca026ff3f8956 : SUCCESS in 35m 14s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ad3fae38c8a34c28a29f7f0e02ba44b0 : SUCCESS in 1h 03m 03s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0cc40724004f428c87525ac2f1746fd1 : FAILURE in 58m 03s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/5ce447e79718447b84445e78956b4cfa : SUCCESS in 1h 11m 16s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/17e50e8023d24b5a8a718e7e6c0b9202 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":3},{"id":"e7e1db885a4f38b418463cfbcda9d0b377c486a8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-01-31 19:36:44.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3a7fa82716be63398db5576a03252f91d0d37d44","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-02-17 14:16:30.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60c1fe244c9ce37ace3678b19916d28da0f0fb6f","author":{"_account_id":34510,"name":"Maxim Sava","email":"msava@redhat.com","username":"msava@redhat.com"},"date":"2026-05-17 07:00:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c79aae3a86e2d2d0ad6432be0212d3f09284923d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-17 09:04:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3623ad198f1f42948a464dab265dc384\n\n- grenade https://zuul.opendev.org/t/openstack/build/d8656e45e0d5453c8edc7c4094091fb5 : SUCCESS in 34m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2915c0137aaa47b3a8ef0da8d1eeb425 : SUCCESS in 1h 45m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2175c8fa16bc4966a3f2b338ece75952 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1876fdd9a142db92540a4d87e40293 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3769ef745eb4d65b0d3fa3b11cb827a : SUCCESS in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4646dbfb9154db69485b17735326eaf : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53191126c091446d8b6ab15ef7319001 : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3dd91f7caf4a4e828261a10c182b6e36 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ff00ba3129643e28bd460897ad01732 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7048dbb99d4a4cd5ab58f66133c311b2 : SUCCESS in 3m 44s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/1f73e2efd00842928711a47513991a0f : SUCCESS in 1h 16m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f8eab95b481d48c98ad2c43891b282b3 : SUCCESS in 1h 23m 18s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/0e83df0f99884643bc956a4967924978 : SUCCESS in 1h 17m 24s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/1ade4e9109c8465dbe2000324bae5029 : SUCCESS in 1h 43m 06s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/32dab7a621324498a19ce0ae0696366c : SUCCESS in 1h 20m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f0bce731dcc74475b20241bfc8c59918 : SUCCESS in 1h 26m 38s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c8161a8ddb5c445ab15185f2b5ecc68c : SUCCESS in 18m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c014cfd7b94498bb2322d1cc41a6c44 : SUCCESS in 1h 08m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1fe95c2071794e14986a6a5e9e179f29 : SUCCESS in 53m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/851a7e36adb44f9bb86340fa1c657f8c : SUCCESS in 46m 15s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d446dd12c26141c1aa45549dd6c132e2 : SUCCESS in 1h 57m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c7a4ca6138fc4e3e9b0b3269a3e5928f : SUCCESS in 1h 38m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c9bfee800e54b779cdb0c9c137f7075 : SUCCESS in 1h 13m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/09b231252bea436c84a76ad146effbb9 : SUCCESS in 1h 04m 21s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/c8378bde1cfe4dfaaa1dba87d1a6d58d : SUCCESS in 1h 14m 55s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/318ba47cd974458f9ce671f42cc57d09 : SUCCESS in 1h 15m 42s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/6ea9d401c6e04cb79a8a4af3245cea02 : SUCCESS in 1h 06m 11s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9b4e5b27c4074f4c9434e3c71ba756c4 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3da869573ff437f3877b08bc8c7488cb49f40df5","revisions":{"87f429baa4d3bcc623e64b872c0038720ede75b7":{"kind":"REWORK","_number":1,"created":"2025-11-06 17:41:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/966315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/15/966315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/966315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/966315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/966315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/15/966315/1"}}},"commit":{"parents":[{"commit":"a2d9b429cac1f36d2fcec42aa2c98f9a18cad160","subject":"Remove run-tempest-26 role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a2d9b429cac1f36d2fcec42aa2c98f9a18cad160"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 17:00:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-06 17:41:41.000000000","tz":0},"subject":"Remove deprecated regex opts","message":"Remove deprecated regex opts\n\nWe also remove conditionals for long-since deleted stable branches.\n\nChange-Id: I506b44b97413bb3318cac865036abd0ad045767d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/87f429baa4d3bcc623e64b872c0038720ede75b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/87f429baa4d3bcc623e64b872c0038720ede75b7"}]},"branch":"refs/heads/master"},"b5683f4ad48b8df03663a3ef790d2e12441bd01a":{"kind":"REWORK","_number":2,"created":"2025-11-07 10:49:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/966315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/15/966315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/966315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/966315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/966315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/15/966315/2"}}},"commit":{"parents":[{"commit":"da54cbd92323c7ff05c16eea8dc4be235a98a502","subject":"Remove run-tempest-26 role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/da54cbd92323c7ff05c16eea8dc4be235a98a502"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 17:00:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-07 10:10:07.000000000","tz":0},"subject":"Remove deprecated regex opts","message":"Remove deprecated regex opts\n\nWe also remove conditionals for long-since deleted stable branches.\n\nChange-Id: I506b44b97413bb3318cac865036abd0ad045767d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b5683f4ad48b8df03663a3ef790d2e12441bd01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b5683f4ad48b8df03663a3ef790d2e12441bd01a"}]},"branch":"refs/heads/master"},"3da869573ff437f3877b08bc8c7488cb49f40df5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-30 11:38:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/966315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/15/966315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/966315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/966315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/966315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/15/966315/3"}}},"commit":{"parents":[{"commit":"c5a189f9e37011b455a74262ace09415623494e4","subject":"Remove run-tempest-26 role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c5a189f9e37011b455a74262ace09415623494e4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 17:00:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-30 11:34:05.000000000","tz":0},"subject":"Remove deprecated regex opts","message":"Remove deprecated regex opts\n\nWe also remove conditionals for long-since deleted stable branches.\n\nChange-Id: I506b44b97413bb3318cac865036abd0ad045767d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3da869573ff437f3877b08bc8c7488cb49f40df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3da869573ff437f3877b08bc8c7488cb49f40df5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
