)]}'
{"id":"openstack%2Ftempest~966317","triplet_id":"openstack%2Ftempest~master~I208b67c447b664b203dfe625eee5f490d4f1cbb5","project":"openstack/tempest","branch":"master","topic":"rework-tempest-install","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-06 19:45:03.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-02-17 14:49:40.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}}},"hashtags":[],"change_id":"I208b67c447b664b203dfe625eee5f490d4f1cbb5","subject":"Move exclude lists into package","status":"NEW","created":"2025-11-06 17:41:49.000000000","updated":"2026-02-17 14:49:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c8d5d7c3c2cd6db050fc3302689309130977e923","_number":966317,"virtual_id_number":966317,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-30 14:55:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"all":[{"value":1,"date":"2026-02-17 14:49:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 17:55:31.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2025-11-06 19:45:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-17 14:17:22.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"3f8371ebce1035fca7528a7535e0da6f340a45fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-06 17:41:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48c3e63ba3f547ade010024260f1f318ad4ff878","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-06 17:55:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7541a9d82407647423b6787727792cfddf65e3ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 19:45:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b73d7b3c8f614bd7b689050079e57409\n\n- grenade https://zuul.opendev.org/t/openstack/build/cad92feff0204af4ac7aa5cfae5d1895 : SUCCESS in 58m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e20fa74aea9b4060890084ec41bc340e : FAILURE in 27m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34fb5cc2e1d4412896d89c9d61c17550 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/030b05bcc7144b3888ac13ac8bac5071 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c02abbdf1b3d4af39396c711b85ac774 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3443376bd97b4db4aa6dbffe9eabee55 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/454da3309f4e4d1192efe8118e89bde9 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6fea1f7966942ff8a13fb307356994f : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36a0d03baf5945a4b94b1c9676e07690 : SUCCESS in 6m 25s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/22de55ea7a5b4f589fa67ae8a9fb42de : FAILURE in 30m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4885aabc0fc64f759b497d35f9958bf6 : FAILURE in 28m 19s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/9f2d14a9603549e9a573e05989f47181 : FAILURE in 27m 45s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/dd3108dee0044abc91aa5bcd862874e5 : FAILURE in 10m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/18d831d8e9734804abd10f5caa2736d3 : SUCCESS in 1h 07m 44s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ca1454f4bda74715be656d0a67df911c : SUCCESS in 38m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2653faeb3493491598f8249ef9453d3c : SUCCESS in 1h 04m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d8f457ad5fd94b7ba5104f263ec60710 : SUCCESS in 1h 29m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aceb8a05e8bb4d87a5b21704a80c45cf : SUCCESS in 56m 58s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d3301b80f75a4bc0a448522810edd299 : FAILURE in 14m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e2fec84ad441477aa7717e6e010a13f2 : SUCCESS in 1h 54m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/10d3c92d9ddb41de89c25231484b48b3 : SUCCESS in 1h 13m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91f53fb3ddc74d269a12b52834e771e3 : SUCCESS in 1h 08m 27s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/64f6ffcfcd9c47ed8995e16ca541ffcc : FAILURE in 25m 23s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a73dc75d99ec494eb64d2691f0319bf0 : FAILURE in 33m 28s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5caa944ed0f247c9844518cdd7ac9dfd : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1},{"id":"578e11ff16330196329b4f90b131f43afbb76be3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-07 10:49:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e78de66b5dd8c2367b89d168e30cfa926f56508e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 13:03:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40530928360747b691b23a0a366ff3df\n\n- grenade https://zuul.opendev.org/t/openstack/build/0028e2028d69427c9842a1b2d8d8f420 : SUCCESS in 34m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb5aae57e8e94a6f9ca780a273b715a2 : SUCCESS in 1h 36m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/943a0296ded04d1fa46533825dac0868 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28accec833e54dc7b4f022cb6cbb85c3 : SUCCESS in 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f4762db78354bb48749b1d9119b8b04 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9269c4cdc2964bbea9003a43ca8178a6 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b82ab7897ac04ad59b9f5fd8fb717b42 : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2f003d8be9f468c8b180b12f13d86f7 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c1d618ea1bd4f8ca4043acc934343e0 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/daedaad5adb84205b336fa4c31261f22 : SUCCESS in 6m 56s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/9939d53367cc4a90be757aa7703c47ad : SUCCESS in 1h 29m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/327ac3fc6e094d5e9f57aeef1e693617 : SUCCESS in 2h 07m 51s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/099f7848e8dd4eba8edc1f195e6f64df : SUCCESS in 1h 28m 18s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/e1c6b88d6dc241ce90d4b5b598524712 : SUCCESS in 1h 35m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fdf78f37b2e2457fa6fb538f88bd0398 : SUCCESS in 1h 31m 09s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4b77bdb503bd4fe7beb72d9b3b8c4ac4 : SUCCESS in 37m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cd2474a81f04932ba7e21e46d77c1af : SUCCESS in 1h 03m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bbeb7fae1eed48a68c51d009bab87c1d : SUCCESS in 1h 47m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8b9a5bb051c74028a1ac7eb3331180ac : SUCCESS in 49m 42s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/28e64b6c5fa34303b84a3ed0661a4dc7 : SUCCESS in 1h 57m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/39019d27ff414b1881235a327f890250 : SUCCESS in 1h 44m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a461c63b7f924ea8a1f8b4e3497650bc : SUCCESS in 46m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49abe609978c42e28fe9f5acf4a987df : SUCCESS in 1h 04m 26s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/756d9160a5b24624bbb4fde081ec0327 : SUCCESS in 1h 36m 28s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a2abaf54f15a43149279375e01731217 : FAILURE in 1h 33m 06s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/8a1e7025bbd640a784783b052754535d : SUCCESS in 1h 03m 25s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/03b4f84f9fbd45d794dbcc02d7ac3d23 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b43ccaabde9fdf759895f0a5f0a47e698ddd228a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-30 11:38:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5a97a51ce69b95026c1ae465ab2730b73a65beec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-30 12:36:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f5fa7f684b54a7da9d6b75fa956593746f3b6f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 14:55:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ab78f192e27468fba3ef8340cfd8f72\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd7f005c2b784b6b9b5958d9532d7378 : SUCCESS in 29m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afe5fb00cf8443eb9169e527fe0ef8d9 : SUCCESS in 1h 49m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bf3381f5ee640f4a9356f49be7e5d55 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/853a27f17ffa4b1bafd4cdbe906185ce : SUCCESS in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bc212f4b0814b26bf72caa303ad9171 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa2fb578dfa4e02a0865427f3dd7303 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7ff55a818ed4a77a23d095162f3ca17 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/027c95098bb648e7abdca24f51fa50ef : SUCCESS in 6m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f94b3c782704fa0b576b70e08cf6213 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7ceac6c218d485b828573d460a2379b : SUCCESS in 2m 40s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5f7dba3af8e146548eb0fffca97047da : SUCCESS in 1h 15m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/62429b4676c449959aae45e4710a9adb : SUCCESS in 1h 14m 57s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/c32514d958d64ac6a417296dbac9a08c : SUCCESS in 2h 10m 09s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/ad957fd81d7d4c2eae49b789dde91fe9 : SUCCESS in 1h 34m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f4dec8d98a7445e5b87ededac6a27592 : SUCCESS in 1h 16m 58s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/59e0597bd5924f02a9032f9d62f2b9e6 : SUCCESS in 25m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/385d8c1265ac4e109f4d7b59c45f9e19 : SUCCESS in 40m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/22d5e150feba49d8b053478eaf94a4f4 : SUCCESS in 1h 58m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/24260b8dae4344fcba8699ebeb5b750a : SUCCESS in 1h 06m 28s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/09f62d3866de4a22aac22ca3e68824cc : SUCCESS in 1h 43m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/da2ffb9b85204f8fb7db70bb978c5b74 : SUCCESS in 1h 45m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f3b2db1234764bc4a502d35c7f27aea8 : SUCCESS in 43m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92d9e690b3f846f09fd4f2b32c097376 : SUCCESS in 1h 07m 15s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/856019351b8140e9ba1b4d7c6e56ab65 : SUCCESS in 41m 17s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d678b4f027f9446eb05720c298ec115c : FAILURE in 38m 15s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/089b7841c4ac476ebf94f98a3b8c1672 : SUCCESS in 1h 06m 25s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/51d6692c86a94b8fbdecc1fd5da9a334 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":4},{"id":"c8d5d7c3c2cd6db050fc3302689309130977e923","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-02-17 14:49:40.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"54b7b79ab6824d6ca506e9daa9cdbcebb92c0128","revisions":{"7e633b4ab19466dff012ca12cac5ee4f221613a4":{"kind":"REWORK","_number":1,"created":"2025-11-06 17:41:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/966317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/17/966317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/17/966317/1"}}},"commit":{"parents":[{"commit":"ffc4f4914d1d7f3e45990a76a44b2c10c87d7b67","subject":"tools: Remove alias for old exclude-list file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ffc4f4914d1d7f3e45990a76a44b2c10c87d7b67"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 17:04:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-06 17:41:41.000000000","tz":0},"subject":"Move exclude lists into package","message":"Move exclude lists into package\n\nThese aren\u0027t available when tempest is installed as a package otherwise.\nSymlinks are provided to the old location to avoid breaking anyone.\n\nChange-Id: I208b67c447b664b203dfe625eee5f490d4f1cbb5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7e633b4ab19466dff012ca12cac5ee4f221613a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7e633b4ab19466dff012ca12cac5ee4f221613a4"}]},"branch":"refs/heads/master"},"3bba1d9e6124c497fad0c7a2a6eb089be86c2243":{"kind":"REWORK","_number":2,"created":"2025-11-07 10:49:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/966317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/17/966317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/17/966317/2"}}},"commit":{"parents":[{"commit":"3f5816497f038d1f97ae3bce254e8f0bf92d2d27","subject":"tools: Remove alias for old exclude-list file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3f5816497f038d1f97ae3bce254e8f0bf92d2d27"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 17:04:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-07 10:15:08.000000000","tz":0},"subject":"Move exclude lists into package","message":"Move exclude lists into package\n\nThese aren\u0027t available when tempest is installed as a package otherwise.\nSymlinks are provided to the old location to avoid breaking anyone.\n\nChange-Id: I208b67c447b664b203dfe625eee5f490d4f1cbb5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3bba1d9e6124c497fad0c7a2a6eb089be86c2243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3bba1d9e6124c497fad0c7a2a6eb089be86c2243"}]},"branch":"refs/heads/master"},"de316fee125aece009ac4cf19ed62b75cd306cbe":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-30 11:38:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/966317/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/17/966317/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/17/966317/3"}}},"commit":{"parents":[{"commit":"7c16a3ccc085a3909ee1c269ef786c880866b676","subject":"tools: Remove alias for old exclude-list file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7c16a3ccc085a3909ee1c269ef786c880866b676"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 17:04:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-30 11:34:05.000000000","tz":0},"subject":"Move exclude lists into package","message":"Move exclude lists into package\n\nThese aren\u0027t available when tempest is installed as a package otherwise.\nSymlinks are provided to the old location to avoid breaking anyone.\n\nChange-Id: I208b67c447b664b203dfe625eee5f490d4f1cbb5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/de316fee125aece009ac4cf19ed62b75cd306cbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/de316fee125aece009ac4cf19ed62b75cd306cbe"}]},"branch":"refs/heads/master"},"54b7b79ab6824d6ca506e9daa9cdbcebb92c0128":{"kind":"REWORK","_number":4,"created":"2026-01-30 12:36:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/966317/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/17/966317/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/17/966317/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/17/966317/4"}}},"commit":{"parents":[{"commit":"3da869573ff437f3877b08bc8c7488cb49f40df5","subject":"Remove deprecated regex opts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3da869573ff437f3877b08bc8c7488cb49f40df5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-04 17:01:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-30 12:34:47.000000000","tz":0},"subject":"Move exclude lists into package","message":"Move exclude lists into package\n\nThese aren\u0027t available when tempest is installed as a package. Move\neverything bar a legacy alias for the exclude-list file into \u0027tempest\u0027\nto address this. Symlinks are provided to the old location to avoid\nbreaking anyone.\n\nChange-Id: I208b67c447b664b203dfe625eee5f490d4f1cbb5\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/54b7b79ab6824d6ca506e9daa9cdbcebb92c0128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/54b7b79ab6824d6ca506e9daa9cdbcebb92c0128"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
