)]}'
{"id":"openstack%2Ftempest~966339","triplet_id":"openstack%2Ftempest~master~I5d252c580a3853877fe480e73676a5e907c756dd","project":"openstack/tempest","branch":"master","topic":"readd-ironic-job","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-11-07 05:54:38.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-11-07 03:53:21.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I5d252c580a3853877fe480e73676a5e907c756dd","subject":"Re-add the ironic job in gate","status":"MERGED","created":"2025-11-06 18:38:15.000000000","updated":"2025-11-07 05:56:21.000000000","submitted":"2025-11-07 05:54:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966339-readd-ironic-job","meta_rev_id":"ae88c1d7a766ac0775d4b23206fbdf103893f4d8","_number":966339,"virtual_id_number":966339,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-07 05:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2025-11-06 21:08:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2025-11-07 03:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2025-11-07 03:53:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 20:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-06 21:08:53.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"965fba28437f142a2c905697fcc741a6d10e8832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 18:38:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf25ae66e842b61e29007beb524acce4bf34222f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 20:38:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f041d0ae7cb47f6bb764655a4e5c2c3\n\n- grenade https://zuul.opendev.org/t/openstack/build/58a42f98b35d461785c2f93ff7cc1119 : SUCCESS in 35m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cb4f51bca704c3eb23810d728426c48 : SUCCESS in 59m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f960a918ca3f4d3ab585c30dba918178 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/646084bec849454d90022c8c9a734ac6 : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e7a7bc4be74f7dad7c7ad0b492fe40 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/688a10a124694e6f87ef35730bdfdbc2 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aaef04f63eb8433da6e0bb5356a3b8bb : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01300ad58bcd4c858565594edacc966c : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/add87b23a4764fa687b393e518577512 : SUCCESS in 8m 31s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/daff68b3ae4741279bf0bfcd7cb5e131 : SUCCESS in 1h 22m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/69c9f98f305d49bb95c3bbe69497fad6 : SUCCESS in 1h 57m 23s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/1f89bb7464d34c55838c2ad196c505ef : SUCCESS in 1h 30m 42s\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/30c81d5bea5b41719c0cdae67336ba49 : FAILURE in 10m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8a89c85d80e849a286abb932be7981bc : SUCCESS in 1h 09m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0b47c896ac2e4b259d8d18ad9e3d713d : SUCCESS in 17m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a8380935bc64a4cbda1bf37d2a232f3 : SUCCESS in 1h 00m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9fa2a79b05254a9cab57a7fde835c01e : SUCCESS in 1h 46m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/300e201672b447eab5788775c007d3b4 : SUCCESS in 50m 49s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a4ab1ae639bf4bca9ac5fafe8951c5da : SUCCESS in 1h 48m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3ee9cd7eb2a14cd9b7b2d4177d9b7a93 : SUCCESS in 1h 27m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/899e1e7f99b340e38f7739433569d8db : SUCCESS in 58m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/efdca94fd9804e12aa8b3185c2b7bfe4 : SUCCESS in 33m 12s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/5f405938057249be9db94ad8ccab71a0 : SUCCESS in 1h 11m 48s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ea6d16689d784d719ad4d367f5e251b3 : FAILURE in 1h 19m 29s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/824d22b072df4195b15f49aa7594a2a9 : SUCCESS in 50m 21s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/37bea3f42ad341ceb081837d4aa7543d : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b401fdd2d87b2be318b894e69bad0dad7f2fea1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 20:42:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9f8534497819d7f222d935d9a23ace5948b49f01","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-06 21:08:53.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"945d70131e9d71b3f0d269c331565388bd87d2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 23:01:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd9d1381a8d94f76abda4872d22eeffe\n\n- grenade https://zuul.opendev.org/t/openstack/build/8823064b92bf43018e9dded7bb6c41da : SUCCESS in 53m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c59bfb78c1d4149b5dcda7d9fca6203 : SUCCESS in 1h 05m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/594c57d2d7874cb29b9cbd870884dc91 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6ee86880274ec5a7ec923981646da5 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cec0c790f6f431c89564e0c1faedc3c : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbfe8de5641f468c908e7afdf7ab7e34 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3c86038762e46f0bdd1f931edba5ff3 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb2276f506204291971f2f5c18702095 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29e0aa70e199450cb321192e10ec389a : SUCCESS in 4m 47s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/dc227ebe62074d7d9db36d6a89a6e09b : SUCCESS in 1h 39m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7d2f544f5f3a4a6895f579d80d08e107 : SUCCESS in 2h 13m 05s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/da46124a71cf45f2aec92a20501fc801 : SUCCESS in 1h 31m 05s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/afbbea02febd4ad4992d7c2f7ca6e480 : SUCCESS in 1h 37m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/87e11e74b9764c8cb9ea83a5cc31cf05 : SUCCESS in 1h 22m 04s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9c856bef1bcf405aa1641a47a71f12ef : SUCCESS in 44m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4658f1a4f4ba4e3598b32a0f870e2a5e : SUCCESS in 59m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f5d49b9a2edb46e78b530e19a2615935 : SUCCESS in 1h 33m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6749c1d6316e41c39a04f8f16e34ecb5 : SUCCESS in 52m 00s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e158f504d6ed4d068c541f2c98bdc345 : SUCCESS in 1h 52m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c1b1cc37d2ca489d9af071ecf5bcee35 : SUCCESS in 1h 44m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f3e31e398ec4f0a9019a7b8f0c39a52 : SUCCESS in 1h 09m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c64a772928e440c3a9f92a45552a3e53 : SUCCESS in 1h 00m 54s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0194eff1984349d690fd8d241718fa16 : SUCCESS in 1h 11m 28s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/65a60320b256470c8f8b7e5ab7223ddd : FAILURE in 1h 28m 10s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/e3768150876449cda0638ef7d8b04481 : SUCCESS in 1h 13m 10s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a650a8faef6b475386ec46199578fb88 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cb4be50a091006e21420e28f2f18c3b844e85615","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-07 03:53:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cd29fdfad873fdcabe77a0819c1a326a876c7c03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 03:54:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e6a06abff661e6b667351deccfc029cbccd7dbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 05:54:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a4056c633c4799ab02a2b1e4c77239\n\n- grenade https://zuul.opendev.org/t/openstack/build/221b8568a0354c3fbcb56e942a644b49 : SUCCESS in 1h 01m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a5c2b16f1134153a014d2515a02c327 : SUCCESS in 1h 40m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578fef5af4a14146a1c964b2c3b45ac4 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/445b48dd5eff4ae58a58455dc18c6b85 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7ef119146f045a59c3be242d5a88b95 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/085ffcff17124cf199cb7e3bb1d72c87 : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11861cf450d74bc28afc7d671114497b : SUCCESS in 6m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4057e64cb424cc384295ae9bf9dea05 : SUCCESS in 7m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a06a17b387ed4a168d6e96b66660e7e0 : SUCCESS in 1h 24m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d26110bcb0274dbbb97145a825500aaf : SUCCESS in 1h 05m 47s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/b13ca48b8b3c4aa1bdb391bb0553c881 : SUCCESS in 1h 05m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/539e8358baf047c8a2ce509ab5184658 : SUCCESS in 1h 05m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0aedf10c0b7543719b249a2981c8258c : SUCCESS in 42m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e59d6fd321114fd98c6bca9aeea9eac8 : SUCCESS in 1h 16m 02s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f25cfad59a3467a9c738ebb1edcabad : SUCCESS in 1h 51m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38154ac6e05042d8bd4b7b2fdb6972ed : SUCCESS in 1h 54m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/63f65b5aacbb47b489fcc1b212c3b9a2 : SUCCESS in 32m 17s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/0a723a2668b34693a672615a8d878ba7 : SUCCESS in 31m 16s","accounts_in_message":[],"_revision_number":2},{"id":"30da4c882f9ba81e5343b3c116969f1762175df4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 05:54:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ae88c1d7a766ac0775d4b23206fbdf103893f4d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 05:56:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fe0cf7cfab448fcb393076fb920eec9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de60ea7e3305459f907862f09e9b1e9a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f5253aa23581e7b14a83a8f70079322cc08da41f","revisions":{"265f8ddc3883e692ef5c6ab3953a656d7150f5e5":{"kind":"REWORK","_number":1,"created":"2025-11-06 18:38:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/39/966339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/39/966339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/39/966339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/39/966339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/39/966339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/39/966339/1"}}},"commit":{"parents":[{"commit":"eb3f2dabde91a835ab48ca4be2de9ed2a162c747","subject":"Merge \"Fix formating and releasenote filename for 43.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eb3f2dabde91a835ab48ca4be2de9ed2a162c747"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 18:36:40.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 18:36:40.000000000","tz":0},"subject":"Re-add the ironic job in gate","message":"Re-add the ironic job in gate\n\nIronic job ironic-tempest-bios-ipmi-direct-tinyipa\nhas been renamed to ironic-tempest-bios-ipmi-direct\n- https://review.opendev.org/c/openstack/ironic/+/950192\n\nZuul did not give any error or warning when this jobs was\nignored to run and it went unnotice until I manually checked\nthe gate jobs.\n\nChange-Id: I5d252c580a3853877fe480e73676a5e907c756dd\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/265f8ddc3883e692ef5c6ab3953a656d7150f5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/265f8ddc3883e692ef5c6ab3953a656d7150f5e5"}]},"branch":"refs/heads/master"},"f5253aa23581e7b14a83a8f70079322cc08da41f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-06 20:42:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/39/966339/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/39/966339/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/39/966339/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/39/966339/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/39/966339/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/39/966339/2"}}},"commit":{"parents":[{"commit":"2b2ac77728f8a9b42bab9f9de947c80dbb6353c3","subject":"Pin stable constraints for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2b2ac77728f8a9b42bab9f9de947c80dbb6353c3"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 18:36:40.000000000","tz":0},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-11-06 20:42:44.000000000","tz":0},"subject":"Re-add the ironic job in gate","message":"Re-add the ironic job in gate\n\nIronic job ironic-tempest-bios-ipmi-direct-tinyipa\nhas been renamed to ironic-tempest-bios-ipmi-direct\n- https://review.opendev.org/c/openstack/ironic/+/950192\n\nZuul did not give any error or warning when this jobs was\nignored to run and it went unnotice until I manually checked\nthe gate jobs.\n\nChange-Id: I5d252c580a3853877fe480e73676a5e907c756dd\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f5253aa23581e7b14a83a8f70079322cc08da41f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f5253aa23581e7b14a83a8f70079322cc08da41f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
