)]}'
{"id":"openstack%2Ftempest~966350","triplet_id":"openstack%2Ftempest~master~Ibfd3a005bdf80ed173529be189beb47ba5b29714","project":"openstack/tempest","branch":"master","topic":"2024-1-last","hashtags":[],"change_id":"Ibfd3a005bdf80ed173529be189beb47ba5b29714","subject":"Pin stable/2025.2 constraint in tox to release 46.0.0","status":"MERGED","created":"2025-11-06 20:46:51.000000000","updated":"2025-11-07 03:50:40.000000000","submitted":"2025-11-07 03:47:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966350-2024-1-last","meta_rev_id":"4882717cadae960d2433e0ac4763f698ca86d214","_number":966350,"virtual_id_number":966350,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-07 03:47:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2025-11-06 23:32:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2025-11-06 23:32:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-06 23:00:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"612401c443051f9015c3311e62b9a8ea90948f55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 20:46:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c2c7dc9970dee6ba7626d6dcc0793b3adb073bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 23:00:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fb3eb9dab2e4d618b6d25fc1c4422c9\n\n- grenade https://zuul.opendev.org/t/openstack/build/7d3a7851141f4e7d83df2389890df5f7 : SUCCESS in 45m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a2b718c7718477aa447a4a7809c6714 : SUCCESS in 1h 59m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b672aa8ffc94b2c8db47cb3604d7736 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e76c41537948468c096940a0a01d36 : SUCCESS in 14m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5edbc6de67e48f893799ec7ef4125f4 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/289b0621d377432085a5fbc0c5db64b3 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95b9017d800f4e988fa903eaef5a3587 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e4ebe03ef204fdab24764b85125b024 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e978b3d0d97940b89f99e4fa047ea03d : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ef41052328014eb5a4bea1915b0c06b2 : SUCCESS in 6m 23s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/889e7ab26aaf4972b088b851a8c38102 : SUCCESS in 1h 04m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f42d2eb46371492abe01a9a59bfb08ae : FAILURE in 2h 08m 38s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/70d81a169c8b4f6fa1e86554455b1373 : SUCCESS in 1h 16m 00s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/53ffeda69cc0404baa810b4d1b2b477e : SUCCESS in 1h 32m 35s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d447b8af9e7446e0b11ed5996bf91663 : SUCCESS in 56m 49s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ee08b512b94a4dac96b736cde332df86 : SUCCESS in 37m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f98439696ac04386bf314e3a13ff691c : SUCCESS in 57m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75fcb337e2ec474dbb69f1955a099927 : SUCCESS in 1h 39m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8585f832d9b34195b908e1442fd05358 : SUCCESS in 51m 42s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3c3ca94a032e480db44f175df3ee10b3 : SUCCESS in 1h 56m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c750351e109c4592b299cfbef04f989e : SUCCESS in 1h 12m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a20e3aef499d437099f207ec80d81c96 : SUCCESS in 1h 08m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/579a743dfc644c43b56a283e69b2479f : SUCCESS in 53m 24s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/141e49d4bd4246c88968220063d08ca9 : SUCCESS in 51m 22s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2d967b981968417cac2879f4284bff54 : FAILURE in 1h 08m 29s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d9058093b01b405db526c31258edbb9e : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a0aaa3d5eab01ef6463df91d19115f45d3b544bc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-06 23:32:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebafc922efdcf0431fd21e12e9e8c4109633e7f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 01:31:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66917713c46d2cfe4b16032269ce9dbaca0c6627","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 03:47:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d1663ddbad14ce7bb1f582912303d9c\n\n- grenade https://zuul.opendev.org/t/openstack/build/d7b6b28b76b843708e24e48955ceccc7 : SUCCESS in 1h 03m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbb7389a206c43e99b82ffdd03fc18e2 : SUCCESS in 1h 44m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e9500003114265ac6a8f3fcae4b598 : SUCCESS in 13m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99a0e1039c7c44d692a210f3f82f9eee : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b278387502d492eaa59e9b9a9168048 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbf22af94fb840fdadd117440da4e5a2 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af078cd6e1e14ad5a689921be39a1a06 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/844bdbdaa35f4d7c8f40fad980e9382f : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c81bd9cce3c549eba523958ab7535c4c : SUCCESS in 9m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3f4f4ac9dc90467dbfd1336ed9d4aeae : SUCCESS in 2h 07m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5894c5bda9948cbbae2e1f13d354c36 : SUCCESS in 1h 09m 22s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/4d616e6d1cb049858a6eb23069da7f85 : SUCCESS in 56m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec73e972dd8347cab883272ea1ac5403 : SUCCESS in 1h 01m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab9611d6ed6546fb9f417b9392df6f39 : SUCCESS in 46m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/83de383227d94d7d91232223ad44cdac : SUCCESS in 1h 24m 56s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f4b7f304e7e84d96a632d6efb9e8482d : SUCCESS in 1h 50m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/452a461c15eb414f8a55d134ab3339fa : SUCCESS in 1h 54m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/46b05d540b994ebdb4371995e036c5c1 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2ac359b31f012e5b52f68759986ee4137566866c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 03:47:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4882717cadae960d2433e0ac4763f698ca86d214","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 03:50:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6e92efcdc934fd9ab572212116a8641\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e5fcbaf9d041fdb52b69126b374a72 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96d020d00cf74ff58c4cad5292a140f3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f26bd1be19731f114271b10ae543d0d4cffcb382","revisions":{"f26bd1be19731f114271b10ae543d0d4cffcb382":{"kind":"REWORK","_number":1,"created":"2025-11-06 20:46:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/966350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/50/966350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/966350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/966350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/966350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/50/966350/1"}}},"commit":{"parents":[{"commit":"2b2ac77728f8a9b42bab9f9de947c80dbb6353c3","subject":"Pin stable constraints for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2b2ac77728f8a9b42bab9f9de947c80dbb6353c3"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 20:45:40.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-06 20:45:40.000000000","tz":0},"subject":"Pin stable/2025.2 constraint in tox to release 46.0.0","message":"Pin stable/2025.2 constraint in tox to release 46.0.0\n\nWe are going to release Tempest new tag 46.0.0 to declare\nthe end of support for stable/2024.1. stable/2025.2 constraint\nis latest stable constraints tested with current tempest hash\nwhich is going to be used for 46.0.0 release.\n\nFurther information:\n https://docs.openstack.org/tempest/latest/requirement_upper_constraint_for_tempest.html\n\nChange-Id: Ibfd3a005bdf80ed173529be189beb47ba5b29714\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f26bd1be19731f114271b10ae543d0d4cffcb382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f26bd1be19731f114271b10ae543d0d4cffcb382"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
