)]}'
{"id":"openstack%2Ftempest~969504","triplet_id":"openstack%2Ftempest~master~I48ae55577f5a3f80a9f60899094dffbc2e862cea","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"I48ae55577f5a3f80a9f60899094dffbc2e862cea","subject":"DNM try to run test_extend_attached_encrypted_volume_luksv1","status":"NEW","created":"2025-12-03 17:17:33.000000000","updated":"2025-12-03 19:16:18.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8223b2159cbac86cd739e15579060f49e9ae8bf1","_number":969504,"virtual_id_number":969504,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-03 19:16:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 19:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0af36f5e85dd1321b05245a6a777d26117acf234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-12-03 17:17:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8223b2159cbac86cd739e15579060f49e9ae8bf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 19:16:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7f3d8148af9483dbc32656a85fbc090\n\n- grenade https://zuul.opendev.org/t/openstack/build/1aece748fc73405e883b6e2641779db0 : SUCCESS in 1h 00m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3e39ea7738d4f3388a048b935052592 : SUCCESS in 56m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fc0a1e420cf40daac2175aefff1c9ad : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4921ef8c15754fb2ae0eb2a38226b4a3 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3bab9cd89ad4463a0528fcba85aee9d : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/133506e1a090469ca4d2384b52cbfca8 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f0c9661040a4bc29fe4f63f5b8dce11 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d2ca8f43b237450083d726f2cbc47c87 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/377d5f0810704dd9891f283e9dbd1430 : SUCCESS in 6m 59s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/8567c57afbf04476b2dd8c687f1356dc : SUCCESS in 1h 00m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/00a0f807f6d4429ab5d7d2f407b7d1ac : SUCCESS in 1h 37m 40s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/abff080c36fd4c8393453ced79b79666 : SUCCESS in 1h 46m 18s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/a43e65909d2445969b1446d2312fd530 : SUCCESS in 1h 29m 41s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/339e1b7be2ed4a58a81f1f8f6fb0784b : SUCCESS in 1h 29m 32s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/17e8684aa0a84ccab1927c46cf561eba : SUCCESS in 34m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9066f0da418f4d3c8137730132e74d84 : SUCCESS in 59m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7e082d85d767411490a625ffc8661215 : SUCCESS in 1h 44m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0af362987bff4ba2be03d706faac128c : SUCCESS in 59m 53s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/64a995ae33584ad08b158018dfb1145a : SUCCESS in 1h 46m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/21bfafd9a0c84fdeb73c5588e8e33f4c : SUCCESS in 1h 56m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c0b0b70517c4809acc65ac50cebd20e : SUCCESS in 1h 09m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cfe18c36077b4ac49467fb461f2fd1a2 : SUCCESS in 54m 31s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/7f3feb5e9a6540178fbcdc98683dea48 : SUCCESS in 1h 10m 16s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ed116282e1f24ff0b560eab42b4baaf8 : FAILURE in 1h 07m 16s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/442d6d630b934114b715175914c04037 : SUCCESS in 1h 13m 43s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/90df15ccb2bd4eada1a52e7487e60137 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5030f159078332c87a283697c2fcd98d035ff77c","revisions":{"5030f159078332c87a283697c2fcd98d035ff77c":{"kind":"REWORK","_number":1,"created":"2025-12-03 17:17:33.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/04/969504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/04/969504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/04/969504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/04/969504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/04/969504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/04/969504/1"}}},"commit":{"parents":[{"commit":"8ccefd4bea48d6b38f7e16c50f97931b3fc2a5e2","subject":"Merge \"tox: Speed up bindep target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8ccefd4bea48d6b38f7e16c50f97931b3fc2a5e2"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-12-03 17:03:43.000000000","tz":-480},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2025-12-03 17:17:21.000000000","tz":-480},"subject":"DNM try to run test_extend_attached_encrypted_volume_luksv1","message":"DNM try to run test_extend_attached_encrypted_volume_luksv1\n\nChange-Id: I48ae55577f5a3f80a9f60899094dffbc2e862cea\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5030f159078332c87a283697c2fcd98d035ff77c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5030f159078332c87a283697c2fcd98d035ff77c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
