)]}'
{"id":"openstack%2Ftempest~969511","triplet_id":"openstack%2Ftempest~master~I3bbdab71ec47f9c6ac99698aa1225239e7226706","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"37924":{"account":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"last_update":"2025-12-04 00:22:21.000000000","reason":"Change was submitted"},"34373":{"account":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},"last_update":"2025-12-03 19:12:35.000000000","reason":"\u003cGERRIT_ACCOUNT_34373\u003e replied on the change","reason_account":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-12-04 00:04:10.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I3bbdab71ec47f9c6ac99698aa1225239e7226706","subject":"Fix tests for testtools 2.8.0 compatibility","status":"MERGED","created":"2025-12-03 17:50:38.000000000","updated":"2025-12-04 00:23:11.000000000","submitted":"2025-12-04 00:22:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969511","meta_rev_id":"ad84ce0e7dea7d96cc123814ac3c14207fe060af","_number":969511,"virtual_id_number":969511,"owner":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-04 00:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},"all":[{"value":1,"date":"2025-12-03 19:12:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},{"value":2,"date":"2025-12-04 00:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},{"value":1,"date":"2025-12-04 00:04:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},{"value":1,"date":"2025-12-04 00:04:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 17:51:09.000000000","updated_by":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"reviewer":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},"state":"REVIEWER"},{"updated":"2025-12-03 17:51:09.000000000","updated_by":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-12-03 18:01:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1e618efdc962144d59147fbaeaf4636ded1fb57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"date":"2025-12-03 17:50:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"576ede305432615e007b244a425af98bbf9bdfaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 18:01:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d20e5b73bbf49de852ac321c1b61a4b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c18ef9680e749e5b470700703e0f886 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b59c9eb4c3a4acf95e5579c2fdfa811 : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c66d4db58e540b5a3fe8b1c97cba930 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f417a56581b450d94b1fab84b9b3fc7 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ea0ce736a9b4e9ba57da9a4057db8d7 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3753c2ae948d4385ad8985c1bd913541 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4494fc7aaff9404d9061b6af838bc4db : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"576f8f6f62bf099abdc71f9bcdfa3a034eb56129","author":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"},"date":"2025-12-03 19:12:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"27f2dff2939038fd1e07235e3d488d9450cd4f9b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-12-04 00:04:10.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4996b6884d33c536ef95c6214b8033036bce7d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 00:04:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fcda3273a71939aed7a39d5d1437aca4ca338f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 00:22:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7986ba0ed65451fbb3256cb2c0a2094\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ecc0cd507a479389e4e4e180bbeb09 : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/350dd63a329246c6afbdaaa5f41a3fce : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0384be56f9bc40fd888dc378c4885e81 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a0bce3043db4a698095704394d356d6 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b17de5308f55458bb18dcaed10d69ba9 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/633476b9faaa482aa8322020c34b7c9e : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b1de469e39612b923a2bad2af2e0376c2530776a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 00:22:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ad84ce0e7dea7d96cc123814ac3c14207fe060af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 00:23:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6957cc6c1cfa4dbf8f09fc0ea15fe502\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628bb8d0033847aab60036a75ebee547 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e2e854051b247ce914982ab0c3c0ed5e2f01b09","revisions":{"8e2e854051b247ce914982ab0c3c0ed5e2f01b09":{"kind":"REWORK","_number":1,"created":"2025-12-03 17:50:38.000000000","uploader":{"_account_id":37924,"name":"Abhishek Bongale","display_name":"abongale","email":"abhishekbongale@outlook.com","username":"abhibongale"},"ref":"refs/changes/11/969511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/969511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/969511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/969511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/969511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/969511/1"}}},"commit":{"parents":[{"commit":"7cc9ef2b9670aac0be95a6dd6902ac2a1e91964e","subject":"Merge \"Add SRBAC alt manager persona to dynamic credentials\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7cc9ef2b9670aac0be95a6dd6902ac2a1e91964e"}]}],"author":{"name":"Abhishek Bongale","email":"abhishekbongale@outlook.com","date":"2025-12-03 17:11:06.000000000","tz":0},"committer":{"name":"Abhishek Bongale","email":"abhishekbongale@outlook.com","date":"2025-12-03 17:33:28.000000000","tz":0},"subject":"Fix tests for testtools 2.8.0 compatibility","message":"Fix tests for testtools 2.8.0 compatibility\n\nFrom 2.8.0 testtools, testcase.TestSkipped has been removed we are\nadvice to replace with testcase.TestSkipped with unittest.skipTest\n\nThis PR trying to fix tests that are using TestSkipped by replacing it\nwith unittest.skipTest..\n\nCloses-Bug: #2131281\nChange-Id: I3bbdab71ec47f9c6ac99698aa1225239e7226706\nSigned-off-by: Abhishek Bongale \u003cabhishekbongale@outlook.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8e2e854051b247ce914982ab0c3c0ed5e2f01b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8e2e854051b247ce914982ab0c3c0ed5e2f01b09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34373,"name":"Jason Paroly","email":"jparoly@redhat.com","username":"jparoly"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
