)]}'
{"id":"openstack%2Ftempest~972257","triplet_id":"openstack%2Ftempest~master~I5d14e85b7a7cf9a5a8a5073dc27785db8587ecbb","project":"openstack/tempest","branch":"master","topic":"centos9_stable","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-01-10 02:07:23.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2026-01-10 06:45:07.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-01-09 17:54:23.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I5d14e85b7a7cf9a5a8a5073dc27785db8587ecbb","subject":"Use default python3_version in 9-stream stable jobs","status":"MERGED","created":"2026-01-06 05:35:57.000000000","updated":"2026-01-10 06:46:55.000000000","submitted":"2026-01-10 06:45:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972257-centos9_stable","meta_rev_id":"68900973ae2099a104b43a9e24daa6ebf2643bc6","_number":972257,"virtual_id_number":972257,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-10 06:45:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2026-01-09 17:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2026-01-09 17:54:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-06 07:21:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-08 08:15:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2026-01-08 08:15:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"9ad1f75d9d91281091fe904277b805ba5bbbf337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-06 05:35:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10fa669949bd763b48108ef4c4d5f747d15558bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 07:21:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b06b50f8334a68a3bcce60046e9ff7\n\n- grenade https://zuul.opendev.org/t/openstack/build/3233a274ab8c4daa8c217e7ef0bb1fe5 : SUCCESS in 58m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/169048bf42ff42d6b1ee0ab747a5249d : SUCCESS in 57m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39688329d296468483d321220d8d20c6 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5516519f91542e0a81836a298062657 : SUCCESS in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02493c16abf04a23a4693d08961d8eee : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ec3979a313044af9222843b4f82983f : SUCCESS in 6m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d855b880805d420f99274fe58aa8d880 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50d62ec4013f47de9b647e332b73d88e : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf2e4288b19642d5a03822348d6954a9 : SUCCESS in 7m 03s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/ec357c0189d84fe9b03c7dd3356d6545 : SUCCESS in 40m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/eb0bc386233644928b28327b2f9cfaea : SUCCESS in 1h 30m 34s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/2ff1e1e358e8440b9def2ad14fb2c5bf : SUCCESS in 1h 35m 27s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/fc75569fbea64889b056e9f66dea4810 : SUCCESS in 1h 19m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f59e47f384174c37ad177816ae189444 : SUCCESS in 1h 18m 35s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3a53670d8d374b538179dd79597e8a18 : SUCCESS in 32m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c09abb5515124440a456869c87da606e : SUCCESS in 53m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1a2cd60887074e4a90f3ef0da50aa74f : SUCCESS in 1h 39m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/677a2dfe0b844bcdb317027007a9d597 : SUCCESS in 54m 34s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/54cddef3af544891aa7e27c3f3ac40ee : SUCCESS in 1h 38m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/69daaf096278432fbb9b72019feb846d : SUCCESS in 1h 14m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/517b07e6e55841a99640c50eb8446c45 : SUCCESS in 49m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ba75b13be15a42169cfbdd401982fbfc : SUCCESS in 43m 17s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/bac2905a38214af8834ab265b107a27c : SUCCESS in 1h 15m 45s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7d48e9fa3a4344b5b30e9af8bcd2db99 : FAILURE in 1h 06m 22s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/b2e9f5b7aae24d2ba062a685eb854277 : SUCCESS in 46m 30s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/44426a1356294e55a2a6d5392f5f1baa : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6929695e0f05896fa74dc2e6baa69656afb0893d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2026-01-08 13:49:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f0e3fba6830404d80e5d835900a0c37ac840d5d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-08 13:54:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a24819fafc91bc3f64ed4364de389d69c492888","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-01-08 19:55:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0d9758ccc595ae08cfac5d886aa281b02f04308f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-09 07:14:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f0613ee10aded1c1dfbca8494f5d0792386ef2a6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-09 07:15:17.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"29c33d54d83ac4f97a865466fc99d2968145c4a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 08:31:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e601f5931cb49c1984505a0c4bd189f\n\n- grenade https://zuul.opendev.org/t/openstack/build/497ed602e852445eb96960aa0494e5a6 : SUCCESS in 56m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/daae0c40c0504b3a9520acca1a34e3df : RETRY_LIMIT in 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14adcab05623446587d177f58cac29a5 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff7d969e6b340619210f2a08b2e2d82 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df327a00e86140bebef2dc5be2b13119 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2042b4d9a96d45e5ac9b2cfd551efc7f : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2426abeef454c0c88806ad547a9ca9c : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85759eb4850e4fa1928496a6ade5698d : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f34022d1ed5649a3b850ff0d378567cb : SUCCESS in 5m 29s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/95edbfe827ae4cec8955503681eb23a2 : RETRY_LIMIT in 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/95f00e3ae4e44261ada5cff5c6518d68 : RETRY_LIMIT in 33s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/02f200686f944616b5353b3df4950663 : RETRY_LIMIT in 1m 09s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/4682549484b747a5a22cd1bcd94780b2 : RETRY_LIMIT in 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2b00bf4255614c69baedb3c5513f2572 : RETRY_LIMIT in 52s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/539ef88775624f838d4f65da130c2451 : SUCCESS in 39m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/768336a63354414cb4d671ed7aaeb8d5 : RETRY_LIMIT in 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5c7df108afed41f4b376c2e1de2069b4 : RETRY_LIMIT in 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/22cceea0102b49e3be4a9cc548117c27 : RETRY_LIMIT in 22s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9e67a75e436340e29ea92a9c51ea8fc7 : RETRY_LIMIT in 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/09d04b5993164467b20f150009da9bab : RETRY_LIMIT in 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0cca23ee6f54730bb8c29b05c01328e : SUCCESS in 1h 06m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a56ca272039f4a65a15b8cc94649c4ef : SUCCESS in 56m 27s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ddca9d59d62f4b059344b9d2107ba0c6 : RETRY_LIMIT in 41s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7cbbbf0f7a2a42d3b3c7c223a163a820 : RETRY_LIMIT in 29s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/3d9c3e915b9441d885f7b261de48a9f9 : RETRY_LIMIT in 35s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/14b91848712c4715809d7dddd437b4b1 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"74edd387d3fc71f3cb779b8fda5c173bf6d4b4af","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-09 09:50:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"760cbe269c03d0ce941cebabfd5c9fa26efd7149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 12:13:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f87e456392144d99adcf26100774a93\n\n- grenade https://zuul.opendev.org/t/openstack/build/cb12e1b0021047c283715f218af22bed : SUCCESS in 58m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e56619d17d28440ea59b41a97f91efdc : SUCCESS in 1h 54m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9b2150d03aa4106852b7646686c30f9 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f6e849218842b792c594e548bf2f2f : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa078bb093547aeaca787a33122bdf8 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7b1ccf83fe94b0d96c8fc9e47a4148b : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/abf657ec726840648db541e5849354a5 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7189a14a58854e26834629b3416bbd51 : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1036ddf715040fe8afff69ab9397627 : SUCCESS in 6m 41s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/57d4b68f04cc4a47a30d9f50f1695a27 : SUCCESS in 52m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c72f739df4745009b00c15d3b8c98ae : SUCCESS in 2h 17m 45s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/f4df784cbe4c44aea2b3e9c6237b93b9 : SUCCESS in 1h 34m 55s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/a0cc0f9e0e2548fd90f1bacff808ffa9 : SUCCESS in 1h 38m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e7324d100d714e4ebb0edf90c87cd7e4 : SUCCESS in 1h 20m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/267c99b3a0b1489b87e6696a373f060d : SUCCESS in 31m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a387878cdb3442c83abfdc6f307ce78 : SUCCESS in 56m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c6c8ffc892f4d5ea3bd709fa2b2aa74 : SUCCESS in 58m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cd271d178ce64d6eb7a00e84c306c90a : SUCCESS in 26m 39s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/da4e4f773ae44787b9430879df398bb5 : SUCCESS in 1h 39m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4da9197dc9434a90a756b412a5ebbaa4 : SUCCESS in 1h 49m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4d7a817b0d146daac653dadb4c91283 : SUCCESS in 1h 13m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/42dbaec26b2344a8a8332306339bf548 : SUCCESS in 1h 03m 57s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a348a46567b54a4ebd5bd16d1c22d669 : SUCCESS in 1h 00m 44s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8637145d4b9e4d3ebd88768e438c80ad : FAILURE in 1h 16m 31s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/614680f12a2642879ddb983faa85f3d5 : SUCCESS in 59m 50s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/aedd1b6518684365aa46c782d5c59533 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6f0fe3c1c1a4d1b5a5eed85a9f0576b6b6491daa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-01-09 17:54:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"472dcc1bce06b93e39b71c0719ae50eff21eab85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 17:55:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30ed7d04f56efa8cb2d976da626d69f85413bb5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 20:47:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3934e9fd5d74a64a668447ce4468061\n\n- grenade https://zuul.opendev.org/t/openstack/build/0e5fea52ee9d4686a87d1d7661ca5f31 : SUCCESS in 59m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13a189c4f62947daa6bcbba6d89d126a : SUCCESS in 1h 18m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfdc4e10b94a4b17a6dce62ec5d45027 : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e86b03b2d2524e8bae444ffe721bde2c : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0711c2c5da7b477ab014cc07544e920c : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd9f3919e5774670ac7e822c5b5beb65 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1db4e5a9c17b4813a3bf319b273c5944 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f69a23d61bbd4e7fb06ae81baac58ac3 : SUCCESS in 4m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/baf7720eb7344e6aa42b84a1b9371e66 : SUCCESS in 1h 40m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/274ac8fcde3b463f94c0175b144e1ed3 : SUCCESS in 1h 02m 39s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/57c001eb0594458eb11427e5d8f96a42 : TIMED_OUT in 2h 36m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc8c73932edb4df3ab98d05251b03977 : SUCCESS in 58m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15da50e4aea8434e82d5da71f7b3bbec : SUCCESS in 1h 42m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/556af1a7609742d8a968507c29bcbf81 : SUCCESS in 1h 13m 21s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/52d88e211e844ebfac9285e7abac36e5 : SUCCESS in 1h 42m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f572b34179f14944bbbd746e76c2178b : SUCCESS in 2h 15m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fba451ae040b483887a709419920a43f : SUCCESS in 52m 25s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/e84ebd4b01474ef28355d8ed1787927e : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":2},{"id":"176eea8e7995bb7118800c4b104b1c414f1b8b0c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-01-10 02:07:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4d1d26f395fbdc43e2dae87b8fa19419af7cb0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 04:37:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5efde49a50c47b189bd74b25191fc94\n\n- grenade https://zuul.opendev.org/t/openstack/build/64480da2713546d5bdd2084f6483a837 : SUCCESS in 31m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6aaab0ef570144ca96eebbba16649c25 : SUCCESS in 1h 00m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0b7db66b2b54c56819af0ba4f8c61f6 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ad078213ba42d6b06588b7bb304a07 : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ccf9e32cf54009a782c014992e2e22 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/799848cd7f18468ea2535608b69d9cd5 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3291e7412a9c4456865f3ea9b21d3f56 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e1e755b44a84b57bd2f203486236a1d : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/74bfc395684741398f1192e653b69e31 : SUCCESS in 6m 45s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/1d17f46bf63b4e2b8f19b633d991c7d9 : SUCCESS in 1h 24m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/80783f7af42e455aa2359d992cf0395d : SUCCESS in 1h 55m 37s (non-voting)\n- tempest-full-2025-2 https://zuul.opendev.org/t/openstack/build/b953054a90384848a08ed1372e432314 : SUCCESS in 1h 20m 57s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/b6db20a1069e4a7ebc3ee4237127ed2c : SUCCESS in 1h 30m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9492997c223a4b0f8ff48fa45b45ffe1 : SUCCESS in 59m 22s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/66479e4dfc7941e0990d4d5ae10852e2 : SUCCESS in 33m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eab364289b143328fd7d8fab3815cd0 : SUCCESS in 50m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/32a7f43fe4a841408307f2c55f707f92 : SUCCESS in 1h 30m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b6d272f72e2b40d78cceb9ee1b093c01 : SUCCESS in 53m 44s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ad13f53e5eb64351b52f7f9556c8df7c : SUCCESS in 1h 33m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40b8d7d0aa4245b28a5a530bdbe66a69 : SUCCESS in 1h 27m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f3cdf74d4ad4af2a81aa1f9f7c5b58a : SUCCESS in 56m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fdd5d98179334ffcb1ba8d127e4931d1 : SUCCESS in 33m 41s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/129df74cfcc747f8bb69aca9dae63fc9 : SUCCESS in 1h 11m 13s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0b95dfea3fcc4778aef45d5e17e52e17 : FAILURE in 1h 36m 06s (non-voting)\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/d536a12a978a4ab095f1248fcc9f5a58 : SUCCESS in 51m 14s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/891b4c02d3af408cb0fd67fd39b4f183 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8a97b58003075abd573ecdf4eeeafa2461d6b331","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 04:38:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a75ea7ae301426bcde4fe9f0692dcd4da15a65d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 06:45:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72a8123be35941a08962ba6acfbd58d0\n\n- grenade https://zuul.opendev.org/t/openstack/build/7cc15363263546e3aaccac191fea2147 : SUCCESS in 39m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05a1269063e548219e86de318d09591a : SUCCESS in 1h 45m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf939a1827548bfadc3bc18045c8f69 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/863c1f998ca24e749fba0b6396be0ac4 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b4b8c1289a841b68776862a08cb00a9 : SUCCESS in 7m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19bce5d502d847618ec867f4c43c0962 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a679c9d74c334e6d93cb25211d0fa87e : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7da0fdc5cd744d0d8cbb396457b52a19 : SUCCESS in 8m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/625ec9614c4f4547a1dae579b5c2f3c3 : SUCCESS in 1h 13m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e54598b3aef34ef3bffd88f784c8ebb2 : SUCCESS in 40m 44s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/37e033099f9e4e94a53c24b1919ee7cf : SUCCESS in 1h 25m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27a0664d241242229782c8b4764c2801 : SUCCESS in 1h 02m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91d280af9a4f4e76990c2049f718448b : SUCCESS in 38m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a24fbfae02954d618835406d5559b58f : SUCCESS in 1h 29m 21s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2ed1c493f6714aa5a09a6b16dd6cdc7f : SUCCESS in 50m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/111fe0eae6204d4da12391d2c6dfa0d7 : SUCCESS in 1h 52m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8969376d80fe44a7bae729e1b87bd20f : SUCCESS in 59m 35s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/5d4a4cc9b3bf49058619c9fdefc963d0 : SUCCESS in 1h 05m 59s","accounts_in_message":[],"_revision_number":2},{"id":"9f5e91a103768c0f4376998a5d3ceb057df113af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 06:45:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"68900973ae2099a104b43a9e24daa6ebf2643bc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-10 06:46:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18100259c3124222964ddff97292b7c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb2ab709e0c4dd9b22fc3773c5930ba : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5bf7cdb69b5659ce8bec281bd4b1fc0ed8c29b22","revisions":{"1461d90b423dbdc475d258e549b577db6d74799d":{"kind":"REWORK","_number":1,"created":"2026-01-06 05:35:57.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/57/972257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/57/972257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/972257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/972257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/972257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/57/972257/1"}}},"commit":{"parents":[{"commit":"1b54854f98e94476301a5985a3c4f8ef29ea2901","subject":"Merge \"Break out novnc connection validation to a mixin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1b54854f98e94476301a5985a3c4f8ef29ea2901"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-01-06 05:34:02.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-01-06 05:35:30.000000000","tz":330},"subject":"Use default python3_version in 9-stream stable jobs","message":"Use default python3_version in 9-stream stable jobs\n\nStable branches missing fixes to support non default\nPYTHON3_VERSION[1], stable branches works fine with\ndefault PYTHON3_VERSION i.e 3.9, so created job\noverrides for this.\n\n[1] https://review.opendev.org/q/topic:%22bug/2109591%22\n\nChange-Id: I5d14e85b7a7cf9a5a8a5073dc27785db8587ecbb\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1461d90b423dbdc475d258e549b577db6d74799d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1461d90b423dbdc475d258e549b577db6d74799d"}]},"branch":"refs/heads/master"},"5bf7cdb69b5659ce8bec281bd4b1fc0ed8c29b22":{"kind":"REWORK","_number":2,"created":"2026-01-09 07:14:39.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/57/972257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/57/972257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/972257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/972257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/972257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/57/972257/2"}}},"commit":{"parents":[{"commit":"1b54854f98e94476301a5985a3c4f8ef29ea2901","subject":"Merge \"Break out novnc connection validation to a mixin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1b54854f98e94476301a5985a3c4f8ef29ea2901"}]}],"author":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-01-06 05:34:02.000000000","tz":330},"committer":{"name":"Yatin Karel","email":"ykarel@redhat.com","date":"2026-01-09 07:14:04.000000000","tz":330},"subject":"Use default python3_version in 9-stream stable jobs","message":"Use default python3_version in 9-stream stable jobs\n\nStable branches missing fixes to support non default\nPYTHON3_VERSION[1], stable branches works fine with\ndefault PYTHON3_VERSION i.e 3.9, so created job\noverrides for this.\n\n[1] https://review.opendev.org/q/topic:%22bug/2109591%22\n\nChange-Id: I5d14e85b7a7cf9a5a8a5073dc27785db8587ecbb\nSigned-off-by: Yatin Karel \u003cykarel@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5bf7cdb69b5659ce8bec281bd4b1fc0ed8c29b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5bf7cdb69b5659ce8bec281bd4b1fc0ed8c29b22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
