)]}'
{"id":"openstack%2Ftempest~981991","triplet_id":"openstack%2Ftempest~master~I801632d437bbaad49168df15b4763887e35af408","project":"openstack/tempest","branch":"master","topic":"qa-2026-1-release","attention_set":{},"removed_from_attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-03-30 11:53:54.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-03-30 16:33:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I801632d437bbaad49168df15b4763887e35af408","subject":"Add Tempest gate job for stable/2026.1","status":"MERGED","created":"2026-03-24 19:20:11.000000000","updated":"2026-03-30 20:47:07.000000000","submitted":"2026-03-30 20:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981991-qa-2026-1-release","meta_rev_id":"2963bd702b70063fe429e63b8b6795b3941fcc29","_number":981991,"virtual_id_number":981991,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 20:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"all":[{"value":1,"date":"2026-03-30 11:53:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":2,"date":"2026-03-26 00:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":1,"date":"2026-03-26 00:56:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-24 21:20:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-30 11:50:53.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"}],"messages":[{"id":"4dadf86b988298c5121d3ee3b46c2d2eaf983b65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-03-24 19:20:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ac608f360d18669a3caa4ace3c7bb7c21da7b3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 21:20:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eae3c0facde4a08b19e75059a91a37f\n\n- grenade https://zuul.opendev.org/t/openstack/build/c21456b637104065b683ebf0686a6d16 : SUCCESS in 44m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/03fc013d862e4df38acec6a97139ea1c : SUCCESS in 1h 45m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77b4f72c5835475dbab99a852be9f6d7 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06edf3d2fd20485eb28442cca56a68b0 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab5eda8b218343898c7f2b6591cb7c41 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba9144477e704527806205dbecee932c : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0ff302420884df8b07b2eacbf60807f : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ebcb2891760a4aa495ecbc874e2c8592 : SUCCESS in 7m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6591ac61cc14d1cb3a9b9dd46b6ec69 : SUCCESS in 4m 22s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/b240951ae2914b8c983d0cea160cb23d : SUCCESS in 40m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/300a94c52495400d94fc9df230db6061 : SUCCESS in 1h 55m 35s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/38b1562ee3d24f31be2fe045d544395d : SUCCESS in 1h 57m 46s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/3cec582909454e628300f8e63de415a7 : SUCCESS in 1h 38m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7886436cae2744498341c2625d965ff3 : SUCCESS in 47m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/26d9293c40b2437981f287a5debcd24d : SUCCESS in 53m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f78b92a13c3455e9f70b70ebf2c70b1 : SUCCESS in 31m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/417ccb1c3f2f4ae0b686015035ab0cfb : SUCCESS in 1h 48m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b9f647480a8649d1b4a02f7cf6d5cbae : SUCCESS in 54m 23s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/304237a69b8b4e8eb02b0e66f021cf3b : SUCCESS in 1h 00m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/61619717d7a345fa86f38d4fcb8b45ab : SUCCESS in 1h 04m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f4b554ce80d490a946f7207a1fd7ab9 : SUCCESS in 1h 11m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/25a93019f393425b96ebac18b0f0b476 : SUCCESS in 1h 00m 27s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6d99f4838fd24f8e80771fe89fcec677 : SUCCESS in 44m 58s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1bb5f45091f441f19befb2700781dc49 : FAILURE in 38m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/93e7c59d9a5646ceaf3da496d8a6c337 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"60671f83331328c063c485ffd8593e3a8e2aad6a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-03-26 00:56:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b3bbf4c5308e68b6f2e3635d53a7a9f9edda3e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 00:57:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"719d7d15ebda0aafea2150d83a44469ee813424e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 02:57:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82d1f19b12c74fd1931111736d61a3b1\n\n- grenade https://zuul.opendev.org/t/openstack/build/014e32714a774178b8a7e981c20f7195 : SUCCESS in 45m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c5948020c68488496b0330bf24686b9 : SUCCESS in 1h 43m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bae364241db24a40b9aa2b7a249357e6 : TIMED_OUT in 31m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52985999586483b84aee35525b49d25 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49d19e3825474f76b961dc13ca9309f0 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bef2d3af289a48d69fa1b66d350dbcef : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/940a488d479146da92e10baefb65a770 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8a4ab40aabc4298b8d7a6fbefb8d34b : SUCCESS in 2m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2eef6980e0c44fecb03cb955297bc322 : SUCCESS in 1h 54m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d79fbb97649b424fa8df6f9511d6216a : SUCCESS in 1h 05m 41s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/b8463f44b0b448ed9bcb30b41bd93062 : SUCCESS in 1h 22m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d4571a82dbd4ff0bfca9b273cc2d161 : SUCCESS in 1h 06m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb83e50c7a674fdd9b3b95e5be686788 : SUCCESS in 1h 02m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/80d32c9dc4fa4ba08d2053215be25146 : SUCCESS in 1h 35m 16s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2350bd4582d0419ba23e1ffc15281097 : SUCCESS in 1h 48m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40349bc7d84346f3a9d401bdabd56c2f : SUCCESS in 1h 41m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7a795e5294704be3b7e4b464113fa0ed : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":1},{"id":"39bcd1d8718b8b884c49ae0c9eeed0dc2654e637","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-03-27 16:21:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cec616b6e7bbab58ce6bdee01f6a57647a62cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 19:03:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7634596d80424d2aaf40116b80cc4040\n\n- grenade https://zuul.opendev.org/t/openstack/build/584092c079c94f8284a28968560a87d2 : SUCCESS in 1h 04m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/465de1c69e574424be1060de5c0f1d6e : SUCCESS in 1h 51m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f67cb27879f40b3961fb25847299d09 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edb780d865a4710afa39240171e6061 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf6d81a160f49258323f05c365f384e : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b3248ddcac34c5d9c237348f93c023b : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fdd07dc648749618e1a902ae3e99225 : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd797f728b9e4a9186f44d1472bfdbc1 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d415202b038d41289a81f7c0fa118f94 : SUCCESS in 2m 54s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5fea82c7bf07498dae73035e7fe54291 : SUCCESS in 1h 24m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c3223c74846b4ab6af4b760ee13f950c : SUCCESS in 2h 36m 01s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/0d323b0118a34a6ba24f27007da3c157 : SUCCESS in 1h 45m 15s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/de72cf8a19584406bcb4a088d2e5ec0c : SUCCESS in 54m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/248c2aba5cda47f48c062394ba41d271 : SUCCESS in 1h 40m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b4888575484b433aa15e3c62aa1f6dd6 : SUCCESS in 41m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3d66b7aa30649e994217f157fa20845 : SUCCESS in 58m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08bf85ca28e5422db11ebd6b9df3881b : SUCCESS in 1h 49m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b3930ba59f1d4ce093c31be7b3c708da : SUCCESS in 40m 19s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/57ce44c5ebfd4c199d9c376376a2361f : SUCCESS in 1h 45m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b011222e53b349d0b46ab11ea2254842 : SUCCESS in 1h 45m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/32b7619704a64fbcae57e0731db180ae : SUCCESS in 1h 10m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8d12965d6a244a7ba9940a557dbbe04 : SUCCESS in 34m 20s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/5d5977c0423f4285981fb5eb3afa97c9 : SUCCESS in 1h 21m 21s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f07314b251d34ff4956d94c5b9508a06 : FAILURE in 40m 57s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1bc9196f0eb24373a0b7521be21556ea : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"568834c470f8b6769ea2c601fdec7de2eb8aa122","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 19:04:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76a3d57bdcbff83a12537ee2d25d1f946319b341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 20:47:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8491e92ed40849acbf42204ac9da016a\n\n- grenade https://zuul.opendev.org/t/openstack/build/89c8d539dbc54d7598514f68d39b611f : SUCCESS in 34m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9744340610a04502a0525a69c59c5d43 : SUCCESS in 1h 16m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/694f3f90aa5a4446b6080eaf243d4f2a : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/229cedbea9e14dd48f2d86fcca31903e : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b466909049cc496b89c6bf90515e3396 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d955d0e53841438c85bd8aafeb7bcd59 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/330a0ad1eebd4912b7d95642de6e0dc5 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cbfe3be90aa4516a296f3e5d516f450 : SUCCESS in 4m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/01da74994049429c81a2d38e1b4cf9d2 : SUCCESS in 1h 37m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/81e93b791e974b4fb66b2b2595c528f1 : SUCCESS in 53m 39s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/44305504091e48a4bffe19ee01b7e0e8 : SUCCESS in 59m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d11f4d06f7a47daaa43142636b62202 : SUCCESS in 35m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa4c4801380f4e0bb4227197bd8a7371 : SUCCESS in 1h 02m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0ab0154d56ae45e0bacfbc899dae7f84 : SUCCESS in 1h 02m 30s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f71aeedf46544bd88f0efbd9e1749da7 : SUCCESS in 1h 42m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/79cce413c20c48e6a456bbe4ef859826 : SUCCESS in 1h 06m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f9b8fcfce93845f184fccd713269bff3 : FAILURE in 38m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6339645eded2e66a7ed2bd395296f3657a310f9a","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-03-30 11:53:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7e02e2d1aaffeb4943a8bb29c6106797d230679","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-03-30 16:33:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6222c9da5e0bcbf2a8b1e3727686b42afcc643c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 18:39:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3e133c307f741bdac9662725455091c\n\n- grenade https://zuul.opendev.org/t/openstack/build/adcda8dd317844c1ace402dec53d0fa0 : SUCCESS in 49m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df422b46043746c9b04e68000f8ff5f4 : SUCCESS in 2h 00m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fecfcbf05e6b40a486d46e0f8733da2d : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e182182911724e568d395c9670c43abf : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21bc1e52daa346d09751b6014b7c2df6 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6d6b6d0741b4147be02bc807b1eaa22 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a3c9ca001a8464197943ffc19aee200 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0276598451a84109a4220c8ebc0eb41a : SUCCESS in 6m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf2dcb3e259b4904a812d2f774472aa5 : SUCCESS in 5m 08s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/2ba3e8fc7037439bb94a90e1809f9627 : SUCCESS in 1h 07m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5215c8754f584780b2ff51b63ea42057 : SUCCESS in 1h 32m 14s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/1d396e716b414bb8ba15a9a675839cf8 : SUCCESS in 1h 00m 44s\n- tempest-full-2024-2 https://zuul.opendev.org/t/openstack/build/1d9770b73dad483584eaa079695fef87 : SUCCESS in 1h 41m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/13c1d6e201d3471db38ca2f5e227b6b2 : SUCCESS in 1h 36m 54s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2261b7e247024bb2806f3274902d8fed : SUCCESS in 32m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/043a2dd05e6e4b319c336266e237d6b6 : SUCCESS in 1h 03m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4051f6b211914ab59d8af0fd01d27c58 : SUCCESS in 54m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e0898bd1ce894ac9b9e916b31f39d41e : SUCCESS in 58m 36s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/91bd8b19db4a469d8d69885ba05d4a6b : SUCCESS in 1h 58m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/872efdaa5dcc465280e1d870895a2d63 : SUCCESS in 1h 22m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae444c8fb16e41b6a314608e9ab90c8e : SUCCESS in 43m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/10207488b00a4bf89241c28fc7100227 : SUCCESS in 1h 05m 23s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6aebf7581636495ba36188302a79b814 : SUCCESS in 1h 14m 31s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a41de9404ae7408981e4db70c2cb7219 : FAILURE in 1h 16m 52s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f505cd4cf3944cc491512c390764beca : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b63d915f61da082385d8eabb38b0a3c38e48ac06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 18:40:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66961ddb98278384b1bbee886c6aaf7b097747c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 20:43:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb036b97a0844fc1aeb67aaa3155900e\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2daddb501214c1f991221ab55f86ceb : SUCCESS in 1h 01m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e83ff86826ab4e3aa4b2238dc0865316 : SUCCESS in 1h 01m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f80a8162eeb4b388dd9bce9d3f99c72 : SUCCESS in 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9443b4abb9804b468862d0eb36df6948 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5dd7d4530ed4d37bed705c8532cc880 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/017f202add6b4f768d808b4a283eb8a6 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/263cd513d0e84c5cb003ec5c6612b302 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6300ed7a0de64d8bbcac4132f2a341f8 : SUCCESS in 3m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d8e14284c37e482cbe0cc1b31ee03575 : SUCCESS in 1h 50m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/349e33a0e655462c933cc83808e09d49 : SUCCESS in 39m 37s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/9ec6a78165984a2b9103b57e9bd3734c : SUCCESS in 1h 22m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8e3a34d79db40cf8ace540af3cde9a4 : SUCCESS in 35m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/124e7ef7080a4a1d9f267314c599ef70 : SUCCESS in 1h 02m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/52856e20dc6f4b54acaa1a4eb96fda7a : SUCCESS in 1h 17m 45s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/be88cbc7a8494c689c2ccad80cab40c4 : SUCCESS in 1h 33m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/487cc46fa5a24f5eaa6161bd1222d795 : SUCCESS in 1h 53m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b924fdb0b94a4d88890b40ed9fc31860 : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7e19b27464301b89f8a06f34417f4115b4365554","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 20:43:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2963bd702b70063fe429e63b8b6795b3941fcc29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 20:47:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e38d91ffb78c4dedbd3bbb592080aee2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f22d6e39a844cff893f3db09c1b83ed : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f735305a98d70b1a1bfc4a3ffdccad58df2680c","revisions":{"8f735305a98d70b1a1bfc4a3ffdccad58df2680c":{"kind":"REWORK","_number":1,"created":"2026-03-24 19:20:11.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/91/981991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/91/981991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/91/981991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/91/981991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/91/981991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/91/981991/1"}}},"commit":{"parents":[{"commit":"ec35a2023dd23e7c2193608638267a61b71379a7","subject":"Switch to the master constraint in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ec35a2023dd23e7c2193608638267a61b71379a7"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-03-24 19:19:10.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-03-24 19:19:10.000000000","tz":0},"subject":"Add Tempest gate job for stable/2026.1","message":"Add Tempest gate job for stable/2026.1\n\nNow we have stable/2026.1 branch ready for devstack\nand so does for all service projects.\n\nThis commit adds the Tempest testing for stable/2026.1\nby adding new jobs running on stable/2026.1 version of\nopenstack.\n\nChange-Id: I801632d437bbaad49168df15b4763887e35af408\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8f735305a98d70b1a1bfc4a3ffdccad58df2680c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8f735305a98d70b1a1bfc4a3ffdccad58df2680c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
