)]}'
{"id":"openstack%2Ftempest~988611","triplet_id":"openstack%2Ftempest~master~I25c8d3c7efb467387806f6aa8f04bcf01dbde2c6","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-05-15 13:59:45.000000000","reason":"\u003cGERRIT_ACCOUNT_32704\u003e replied on the change","reason_account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-15 23:10:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I25c8d3c7efb467387806f6aa8f04bcf01dbde2c6","subject":"doc: Stop messing with sys.path","status":"MERGED","created":"2026-05-14 12:38:58.000000000","updated":"2026-05-15 23:12:31.000000000","submitted":"2026-05-15 23:10:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988611","meta_rev_id":"5aa706a30caca779648a2908abd6f494373aa011","_number":988611,"virtual_id_number":988611,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-15 23:10:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"all":[{"value":1,"date":"2026-05-15 13:59:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-15 20:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-15 20:54:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-15 20:54:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}],"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-14 13:26:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-14 15:54:15.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2026-05-15 13:43:09.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"},{"updated":"2026-05-15 20:54:59.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"4cc7f938de718487eaae5b80c9ba06f92a0bcf91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-14 12:38:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6894cc717d44f74cf46370d9f157e1874a2a4484","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-14 12:44:30.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71b340cd9a72963722270f90c2965485113d88e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 13:26:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0019acf31e774fbeb5fd6ed0046b8b12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae89931a9664ba99f6f2ac3ac7b7b49 : FAILURE in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7659f67f2134fdab5276c277142dd4c : SUCCESS in 26m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/191ab223f4834cdf849129582a9bd51c : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d0376a0397cd13c14671f98f05447cd5aeda9452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-14 14:29:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"58fe9ad8568c772fe0ad89c1b24be22b91dfdbab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:29:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfaeba39b27e484c873567e5a369e0f1\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff6864e9881b4f84810f0cfa6ce1bad9 : SUCCESS in 1h 06m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6202f7d10b1c43588b1fbd34c0934525 : SUCCESS in 1h 56m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7d5af7ea4e34acd91163f8ae087a17b : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb98f029ea44b61881cc103028dfe84 : POST_FAILURE in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7041b561f52141e2b77f6c9741087885 : SUCCESS in 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7bd187d37d34f8b95fbc4c59738b9b2 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f3ee951d06345409481f40aad5e9058 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2c3123d1abe4910aa35b19f9512cb26 : SUCCESS in 5m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/900a226969ab4ff5a1c7e2542e550e57 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3da4b022feda469ab466af185e9c2353 : SUCCESS in 4m 50s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/c7040ecaee2e408bbb0ec3db867cc0ac : SUCCESS in 1h 35m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e1cb1a3e0a304b16b1547523d69a5c57 : SUCCESS in 2h 30m 00s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/cd10507e428b4e48a0e855e24af7bf1d : SUCCESS in 1h 38m 03s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/6ca9d1ebbeca4b82b093c1fd985ce12b : SUCCESS in 1h 46m 38s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/6897169cd2df41c7b96055bdf556855b : SUCCESS in 1h 47m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/70c109a0eea54ccaa9832dcfbb54fae2 : SUCCESS in 1h 19m 36s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/cc3dcda0ee3f49e6a38ad52cb21132d0 : SUCCESS in 23m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70023c927ebe474ca5c7f152e38549b7 : SUCCESS in 56m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c89c2d5f83c34779b0e1a7bc385f2d91 : FAILURE in 1h 50m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ebbee344593242aa816fd9745d6eda5b : SUCCESS in 1h 03m 41s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2a56e25a2a8244028c658ea79bfa53b2 : SUCCESS in 2h 10m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1bf31c31120e4505b80d7549983eefb5 : SUCCESS in 1h 42m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9b4b385326f34c6ca0a2683cdcf7b780 : SUCCESS in 1h 05m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9fe05c5767684966835d18571cb7da53 : SUCCESS in 1h 08m 19s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/cc8720d0a7b04eba8472dfbde716e497 : SUCCESS in 1h 13m 51s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/c6468e86d0e24d2dba8d31bfbc5e29f0 : SUCCESS in 1h 57m 25s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/564822ade61f4e0fab82595c247b8518 : SUCCESS in 1h 06m 28s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/530e5260e26b47e6affcf9d9e1979490 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"26b019c917bfa8d0115a95f1f0dddd00d23ec9b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-14 17:49:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"095ae740699d0b7a68406cea8a4e40ccae4cbd4b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-14 17:49:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"825853afe22f8531360f15be602ecf3e078cf3f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-14 17:50:12.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0ca45ce1b09391ee40e7e35ca4114bfd3cdac12b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 19:49:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54847083df0346a1bbd89f97e4dd08ea\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd9ab49760144bd7b3561d1c5a908e5f : SUCCESS in 1h 12m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/456a9dadbb92446dbc184c7aeceec937 : SUCCESS in 1h 51m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/975aec055ea54377b1074124f07bc902 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde22ab249b34971bbc83b130ce1762f : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d8aef463a224a6c96d3fb860d57f921 : SUCCESS in 29m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91d3ddd01cb049c58b0d091b7e5ec3fb : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/372529a54af74ab68f2e48fbfd6377f5 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1367ae30c8242b5bc4e1dc0bdc2e328 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15bd3845c7aa48c2838a35c498d149da : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e3736684a9c4a4ba96eb59c87f872f9 : SUCCESS in 3m 26s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/a1a6b3c484bb40dfb8df4f4ec63ae830 : SUCCESS in 1h 34m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e5e2e2206d314be8a9df536789a8649a : SUCCESS in 1h 11m 43s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/19c0a5f2b47d44d2a37da5e636523777 : SUCCESS in 1h 24m 14s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/c947cc0fafe34fb09f33de70cfd932af : SUCCESS in 1h 05m 18s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/c38751cb5da2402c89d80b9d204f5d39 : SUCCESS in 1h 51m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f3155f2cd2da420497d11de9a47c25fa : SUCCESS in 41m 46s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/47da99e2f0f7483086f9cd1b7f5f84da : SUCCESS in 18m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f703a14588e74a00a8d53cd9aec18871 : SUCCESS in 58m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d022058be8104773b888c2309a11bf09 : FAILURE in 1h 43m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/25e11c1d122a417392481e1bb432c53f : SUCCESS in 1h 03m 36s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ed5c4123ceb34971b99e521369383579 : SUCCESS in 1h 49m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bdb5f61c795347ad8beb5dfd6b7e8516 : SUCCESS in 1h 03m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/32bb03028ec842cca52e4b28fa84e1ae : SUCCESS in 41m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3affa4cf2c11447fa957286ac907dd6f : SUCCESS in 44m 56s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a00870f4241141f0a0713b71f2e6906e : SUCCESS in 54m 29s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/86b0e54caa824ae4b45d09a0f1f8b5a9 : SUCCESS in 53m 59s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/784c4e2dd3fd45018100ff8643525078 : SUCCESS in 1h 01m 33s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/130ce52a36a649388bd2d881e6b426c2 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"67964ac9a86ec278c596131fd5078ed290738be7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-14 22:02:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"63ef0c71691d3e55985f7b416138e92559aaec2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 00:28:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d862243935042239146a07190a9ba18\n\n- grenade https://zuul.opendev.org/t/openstack/build/b355ac4396c347369f3c88d5daeb4e63 : SUCCESS in 29m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14267c5c1a104afc95127c5b389c8003 : SUCCESS in 1h 40m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7fd476dd54344c699907127c1f4a1f1 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aae03b4fff5434ea2c59d8bb3ae9962 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbaca6f368af45e4968e79e0a4c8f90e : SUCCESS in 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70e0c87c4804144aa5612a22ff73f3d : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cbb03f629954db594d045669bf581f6 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea5e87d030c1466e8d16ee82de8f8b82 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12df4a1df62d461394ddd810495f80e5 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1762ea48599845f290d5b82a9d71b9ff : SUCCESS in 3m 50s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/d3a107bf6f8a413f82e351e68ca5f3a5 : SUCCESS in 1h 12m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ec1478c839e948aea418f1d6ce8d3746 : SUCCESS in 2h 23m 07s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/0351abc2444b4f008907bd8d9bf23767 : SUCCESS in 2h 02m 00s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/0ee413f028f8434db8357cb4ddb5c007 : SUCCESS in 1h 40m 42s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/779cbe7b47944eadbe680b5d7c0dae29 : SUCCESS in 1h 32m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0638d9aeb2f8491f9013610ef9a8b65e : SUCCESS in 1h 27m 46s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e675c07fb9ec4b839ca978d366cddc47 : SUCCESS in 7m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44af07f5fedf48118842e32610ab84b4 : SUCCESS in 39m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3cdf54170314decaf931c9ee3818597 : SUCCESS in 1h 19m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/094b9311702f4b45a26b35a01c9810e0 : SUCCESS in 48m 12s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8fe9873fdf714bb3b56fcc6ba087150a : SUCCESS in 1h 25m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f1d43314335a41b9a0664d295b545fd1 : SUCCESS in 1h 05m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/65bf7c3ebf684e74b258fa5b5653e512 : SUCCESS in 1h 02m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b899bde61074a8e879c9b094f7d75f8 : SUCCESS in 1h 15m 42s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ebd8103eb0d9446985068cdc44b49af6 : SUCCESS in 48m 12s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/227058c7aefb4706aecc273b91217c2c : SUCCESS in 1h 06m 10s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/5177cf88ed1147c6a19bc913a7baaf29 : SUCCESS in 42m 07s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/48f761afa9204cafbad66a0324c21bf8 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":3},{"id":"3e03018dbbb8b53355d6b8e9ff6b060a926b1d1c","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-05-15 13:59:45.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"12bc014f8b71a553205c54899777ee5ba9f88eee","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-15 20:54:59.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15c6c24c74ee0f42991c104f9f45b9cce7ed4323","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-15 20:55:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6bf19517880cf97c9339538cabccb8d247585368","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 20:55:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"983ed0731071063a934f4cef13f663876a83dbee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 23:10:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3188a8a050f94e76814587357d431911\n\n- grenade https://zuul.opendev.org/t/openstack/build/06dcaa30dd754c63ac2bf3d80021f81c : SUCCESS in 1h 00m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b3e1f4dc0a743b6b2fa56dc37bbef0a : SUCCESS in 1h 45m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6a9fe0f47f4186836fa0c1ebadd0b5 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1aba732ef874790861ca66a0e0a9436 : SUCCESS in 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28698067b88d4580b63cea05f8153a89 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad2af5bae0bb4752be0b441d9212eea4 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29a36f4f0c97433da47ef6ab9787add0 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd8c2de56896459b96d745000b0d9a29 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bd59f54dfc34c1ca64e15e8297b529b : SUCCESS in 4m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3451fa7eee354e93a18de409589c5e93 : SUCCESS in 1h 43m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/81a3197c33534ba18230898ea1875602 : SUCCESS in 1h 14m 11s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/2e4093575a104c81a8bfc589bc5746b0 : SUCCESS in 1h 00m 31s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/6ea7fa1f272945f586852be60d064aba : SUCCESS in 1h 22m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b4de53356dc04a1fbc3964485d5bb48f : SUCCESS in 1h 01m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44a063cdbf2c4c768e5e6ea53ecb7262 : SUCCESS in 42m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5cb9b2146f75420ebe66d844850a3b5e : SUCCESS in 1h 26m 59s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9219a43c4d8f4a1592bc1c25a3d75a49 : SUCCESS in 1h 44m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37d7b462a68c475ba2cb546bc403aef1 : SUCCESS in 58m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/81b035b9db304379b30c157c5b05cd03 : SUCCESS in 1h 05m 29s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9d08fdfd78724b288de602b69c3fdf66 : SUCCESS in 27m 50s","accounts_in_message":[],"_revision_number":3},{"id":"90f011dee0cf39e4d6a81def0caf4f7ba2c62ad9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 23:10:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5aa706a30caca779648a2908abd6f494373aa011","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 23:12:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da2e717fbb2f44e5aa3d353fc9f1fd13\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5cfd6760a634cb59834f960796af149 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7dc98751776c458c8bab33f919bf9d4c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"628f2f5ee769aa05c923665accaea68ad1f51d5a","revisions":{"b596728c2b3baca1af07216b7132ca7e308f2142":{"kind":"REWORK","_number":1,"created":"2026-05-14 12:38:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/988611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/988611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/988611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/988611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/988611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/988611/1"}}},"commit":{"parents":[{"commit":"32f3e019f369cdfdb51f8e0133324a769162f213","subject":"Merge \"Switch back the tox constraint to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/32f3e019f369cdfdb51f8e0133324a769162f213"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-14 12:28:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-14 12:34:33.000000000","tz":60},"subject":"doc: Stop messing with sys.path","message":"doc: Stop messing with sys.path\n\nWe were unnecessarily and *incorrectly* messing with sys.path. By\ninjecting \u0027../../tempest\u0027 into sys.path, we cause shadowing of anything\nthat is already present on PYTHONPATH. One such thing is cmd, which is\nalready a stdlib library [1]. Previously we were getting away with this\nbecause Sphinx was importing pdb, which imports cmd and causes it to be\ncached in sys.modules, as part of the sphinx-build command. This is no\nlonger the case as of sphinx v8.2.0 [2], which has caused this latent\nbug to rear its head.\n\nFix this by simply dropping the sys.path modifications. They are\nentirely unnecessary given we already install the package under test\n(tempest, in this case) in the virtualenv.\n\n[1] https://docs.python.org/3/library/cmd.html\n[2] https://github.com/sphinx-doc/sphinx/commit/404e4ffbedb316fe2ba61e3f610437ecc0194211\n\nChange-Id: I25c8d3c7efb467387806f6aa8f04bcf01dbde2c6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b596728c2b3baca1af07216b7132ca7e308f2142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b596728c2b3baca1af07216b7132ca7e308f2142"}]},"branch":"refs/heads/master"},"833eea3b52eaccd62cd85efd9f3169d03b8ac33a":{"kind":"REWORK","_number":2,"created":"2026-05-14 14:29:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/988611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/988611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/988611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/988611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/988611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/988611/2"}}},"commit":{"parents":[{"commit":"32f3e019f369cdfdb51f8e0133324a769162f213","subject":"Merge \"Switch back the tox constraint to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/32f3e019f369cdfdb51f8e0133324a769162f213"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-14 12:28:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-14 13:48:55.000000000","tz":60},"subject":"doc: Stop messing with sys.path","message":"doc: Stop messing with sys.path\n\nWe are generating api documentation using sphinx-apidoc. There\u0027s a\ndesign issue in this tool, where if you call it with e.g.\n`sphinx-apidoc tempest/api/compute` then the generated automodule\ndirectives will use `compute` as the top-level module, giving us entries\nlike:\n\n  .. automodule:: compute.admin.test_availability_zone\n\nrather than:\n\n  .. automodule:: tempest.api.compute.admin.test_availability_zone\n\nTo work around this, we were messing with sys.path. However, this comes\nwith its own issues. By injecting \u0027../../tempest\u0027 into sys.path, we\ncause shadowing of anything that is already present on PYTHONPATH. One\nsuch thing is cmd, which is a stdlib library [1]. This was not an issue\npreviously because Sphinx was importing pdb, which imports cmd and\ncauses it to be cached in sys.modules, as part of the sphinx-build\ncommand. This is no longer the case as of sphinx v8.2.0 [2], which has\ncaused this latent bug to rear its head.\n\nThe fix for this has two parts. Firstly, we need to drop the sys.path\nmodifications. Then we need to address the sphinx-apidoc issue. Rather\nthan hacking on the generated files, we opt to switch from\nauto-generating api docs on each run and instead generate them once and\nmodify the output.\n\n  sphinx-apidoc -f -o doc/source/tests/compute tempest/api/compute\n  sphinx-apidoc -f -o doc/source/tests/identity tempest/api/identity\n  sphinx-apidoc -f -o doc/source/tests/image tempest/api/image\n  sphinx-apidoc -f -o doc/source/tests/network tempest/api/network\n  sphinx-apidoc -f -o doc/source/tests/object_storage tempest/api/object_storage\n  sphinx-apidoc -f -o doc/source/tests/volume tempest/api/volume\n  sphinx-apidoc -f -o doc/source/tests/scenario tempest/scenario\n  sphinx-apidoc -f -o doc/source/tests/serial_tests tempest/serial_tests\n  sed -i \u0027s/automodule:: compute/automodule:: tempest.api.compute/\u0027 doc/source/tests/compute/*\n  sed -i \u0027s/automodule:: identity/automodule:: tempest.api.identity/\u0027 doc/source/tests/identity/*\n  sed -i \u0027s/automodule:: image/automodule:: tempest.api.image/\u0027 doc/source/tests/image/*\n  sed -i \u0027s/automodule:: network/automodule:: tempest.api.network/\u0027 doc/source/tests/network/*\n  sed -i \u0027s/automodule:: object_storage/automodule:: tempest.api.object_storage/\u0027 doc/source/tests/object_storage/*\n  sed -i \u0027s/automodule:: volume/automodule:: tempest.api.volume/\u0027 doc/source/tests/volume/*\n  sed -i \u0027s/automodule:: scenario/automodule:: tempest.scenario/\u0027 doc/source/tests/scenario/*\n  sed -i \u0027s/automodule:: serial_tests/automodule:: tempest.serial_tests/\u0027 doc/source/tests/serial_tests/*\n\nWe then integrate sphinx.ext.coverage to ensure documentation is added\nfor any new tests going forward.\n\n[1] https://docs.python.org/3/library/cmd.html\n[2] https://github.com/sphinx-doc/sphinx/commit/404e4ffbedb316fe2ba61e3f610437ecc0194211\n\nChange-Id: I25c8d3c7efb467387806f6aa8f04bcf01dbde2c6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/833eea3b52eaccd62cd85efd9f3169d03b8ac33a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/833eea3b52eaccd62cd85efd9f3169d03b8ac33a"}]},"branch":"refs/heads/master"},"628f2f5ee769aa05c923665accaea68ad1f51d5a":{"kind":"REWORK","_number":3,"created":"2026-05-14 17:50:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/988611/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/988611/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/988611/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/988611/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/988611/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/988611/3"}}},"commit":{"parents":[{"commit":"32f3e019f369cdfdb51f8e0133324a769162f213","subject":"Merge \"Switch back the tox constraint to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/32f3e019f369cdfdb51f8e0133324a769162f213"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-14 12:28:13.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-14 17:50:11.000000000","tz":0},"subject":"doc: Stop messing with sys.path","message":"doc: Stop messing with sys.path\n\nWe are generating api documentation using sphinx-apidoc. There\u0027s a\ndesign issue in this tool, where if you call it with e.g.\n`sphinx-apidoc tempest/api/compute` then the generated automodule\ndirectives will use `compute` as the top-level module, giving us entries\nlike:\n\n  .. automodule:: compute.admin.test_availability_zone\n\nrather than:\n\n  .. automodule:: tempest.api.compute.admin.test_availability_zone\n\nTo work around this, we were messing with sys.path. However, this comes\nwith its own issues. By injecting \u0027../../tempest\u0027 into sys.path, we\ncause shadowing of anything that is already present on PYTHONPATH. One\nsuch thing is cmd, which is a stdlib library [1]. This was not an issue\npreviously because Sphinx was importing pdb, which imports cmd and\ncauses it to be cached in sys.modules, as part of the sphinx-build\ncommand. This is no longer the case as of sphinx v8.2.0 [2], which has\ncaused this latent bug to rear its head.\n\nThe fix for this has two parts. Firstly, we need to drop the sys.path\nmodifications. Then we need to address the sphinx-apidoc issue. Rather\nthan hacking on the generated files, we opt to switch from\nauto-generating api docs on each run and instead generate them once and\nmodify the output.\n\n  sphinx-apidoc -f -o doc/source/tests/compute tempest/api/compute\n  sphinx-apidoc -f -o doc/source/tests/identity tempest/api/identity\n  sphinx-apidoc -f -o doc/source/tests/image tempest/api/image\n  sphinx-apidoc -f -o doc/source/tests/network tempest/api/network\n  sphinx-apidoc -f -o doc/source/tests/object_storage tempest/api/object_storage\n  sphinx-apidoc -f -o doc/source/tests/volume tempest/api/volume\n  sphinx-apidoc -f -o doc/source/tests/scenario tempest/scenario\n  sphinx-apidoc -f -o doc/source/tests/serial_tests tempest/serial_tests\n  sed -i \u0027s/automodule:: compute/automodule:: tempest.api.compute/\u0027 doc/source/tests/compute/*\n  sed -i \u0027s/automodule:: identity/automodule:: tempest.api.identity/\u0027 doc/source/tests/identity/*\n  sed -i \u0027s/automodule:: image/automodule:: tempest.api.image/\u0027 doc/source/tests/image/*\n  sed -i \u0027s/automodule:: network/automodule:: tempest.api.network/\u0027 doc/source/tests/network/*\n  sed -i \u0027s/automodule:: object_storage/automodule:: tempest.api.object_storage/\u0027 doc/source/tests/object_storage/*\n  sed -i \u0027s/automodule:: volume/automodule:: tempest.api.volume/\u0027 doc/source/tests/volume/*\n  sed -i \u0027s/automodule:: scenario/automodule:: tempest.scenario/\u0027 doc/source/tests/scenario/*\n  sed -i \u0027s/automodule:: serial_tests/automodule:: tempest.serial_tests/\u0027 doc/source/tests/serial_tests/*\n\nWe then integrate sphinx.ext.coverage to ensure documentation is added\nfor any new tests going forward.\n\n[1] https://docs.python.org/3/library/cmd.html\n[2] https://github.com/sphinx-doc/sphinx/commit/404e4ffbedb316fe2ba61e3f610437ecc0194211\n\nChange-Id: I25c8d3c7efb467387806f6aa8f04bcf01dbde2c6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/628f2f5ee769aa05c923665accaea68ad1f51d5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/628f2f5ee769aa05c923665accaea68ad1f51d5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
