)]}'
{"id":"openstack%2Ftempest~990999","triplet_id":"openstack%2Ftempest~master~I7c352ea9b5de99d2e482f647c246e499c277f5e8","project":"openstack/tempest","branch":"master","topic":"bug/2154775","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2026-06-11 03:53:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c352ea9b5de99d2e482f647c246e499c277f5e8","subject":"Wait for volume \"in-use\" in test_encrypted_cinder_volumes_resize_revert","status":"MERGED","created":"2026-06-02 00:25:49.000000000","updated":"2026-06-11 03:55:21.000000000","submitted":"2026-06-11 03:53:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990999-bug/2154775","meta_rev_id":"ae2fec77377cb3183e9363a434d114573a81fd26","_number":990999,"virtual_id_number":990999,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-11 03:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":1,"date":"2026-06-04 11:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":1,"date":"2026-06-09 13:25:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2026-06-11 01:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":1,"date":"2026-06-11 01:40:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-02 02:49:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 12:53:00.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"},{"updated":"2026-06-09 13:25:33.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2026-06-11 01:40:52.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"fe34f0119592bf96772215beae2c3fdb766cbfe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-06-02 00:25:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fc5bf7946e520ef55cd7a8093f1a5a438b86aa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 02:49:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8df71b59c9d416a9d64d1ce8eb70635\n\n- grenade https://zuul.opendev.org/t/openstack/build/89eb940caa4a40cdb6b455421bb6898e : SUCCESS in 54m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef1b9ad8d4ae4ba0a83391c56db464e1 : SUCCESS in 1h 30m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acad545a948045129a3455fd61194f49 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619e62fb5e334265b545814d71e59cfb : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0601cba0ac5c49cd8dfe48aa6e124a6c : SUCCESS in 22m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67ec11384fd5488b84e3287104aec037 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/523debe8fcb84760b987f497cce03829 : FAILURE in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b16122c551f94719abef42b360523583 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a51ab16d22641a4b4a4a37462a3cd62 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2163fd5b484041a8832ef4026774c691 : SUCCESS in 4m 35s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/b995a0e337184f89a466f857473b718c : SUCCESS in 57m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c427b09f00c341f8ae27a2edc7691fee : SUCCESS in 2h 06m 29s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/2e5622bda0244e178c84d3463e7278ed : SUCCESS in 1h 52m 55s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/23c01f5e88ad4881ae5b9819b712ffde : SUCCESS in 1h 42m 12s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/7a930af4799545608a6d95b7dc4660ed : SUCCESS in 1h 15m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4488496afcf14f1b91a9a3d50169b50e : SUCCESS in 1h 26m 35s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e88047f2ad354d4f9a3488cb2cee8697 : SUCCESS in 15m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/599cd4b69c64434889febff0a381d056 : SUCCESS in 58m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b9d3b25843024df3a107b01052e8fe97 : SUCCESS in 1h 49m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/62ee6ef2b0cb424095a90ee567dea67d : SUCCESS in 56m 31s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8c7351e930c44e43a5e0918ebccc83fa : SUCCESS in 1h 25m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b68631e9b80140a3ace39f19a7a0bdcd : SUCCESS in 2h 17m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/257d3a27e9ce440ab2ab88ee3421d34b : SUCCESS in 1h 06m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/87e7ce6898e440b5a81474a4a9f0feaf : SUCCESS in 45m 40s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/69aa5723857c452082894c157d02efaa : SUCCESS in 1h 30m 02s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/18949ef39ab346c49b5431d538c26aca : SUCCESS in 1h 07m 02s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/286e07f6f5884e23bd576e13a2247d4d : SUCCESS in 58m 45s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/35686466648d4d0c8477149055bb8695 : SUCCESS in 1h 05m 22s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ef0070dac61a49a68b19ceff00066e44 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c0be700919abb3b4bcc505b6c67f24a30e78678c","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-06-03 12:53:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0471061a4d1a1e91063209c1a812787f4dfcd9fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 14:59:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42cfbac131e840e6b6405d0761a2df31\n\n- grenade https://zuul.opendev.org/t/openstack/build/641a9c3e3f6b486582f03cfea85e6d0e : SUCCESS in 1h 06m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c4ba0d58d6e48c09be31fb87129d585 : SUCCESS in 1h 29m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5098f9fe34194ab9abd64a7ec7195f03 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96daec433cf42c7bced4688579d0bd6 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e089f361cd004a398a7655fe527a6eb1 : SUCCESS in 29m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5739f1a45038442393d476a6e271ced0 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e65de38cb37e4c56bc1f6be211d48d6c : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43dae2aab4b2495491b613c58786a8d3 : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02f3556c9d3d4d1eaff7eef2d9df51bb : SUCCESS in 4m 47s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/4d4ee74dc2544171b8a36723a595335b : SUCCESS in 1h 32m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e5cec94c524145ae946f894b233d6103 : SUCCESS in 1h 48m 20s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/49368d8a0dd942b1bd928aac89e577ab : SUCCESS in 2h 00m 42s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/e8a7332bb6f34532b3d9e5f979cf5e16 : SUCCESS in 56m 00s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/7a102c0f15224f84b2a76b685ecb23b3 : SUCCESS in 1h 46m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/81263ea132194ac29812419a0156d965 : SUCCESS in 1h 20m 22s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ba9f913048eb47c1a804d180cd44a794 : SUCCESS in 8m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00dcca6af24947a0aee49f43b6f8a1a0 : SUCCESS in 59m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c3b3f23f35e3494a825f428da995df55 : SUCCESS in 1h 24m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/37b29089752a43adb33cbe72e255d1ff : SUCCESS in 1h 01m 37s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/27e3e33b101f4aa48557a2e1986e8b50 : SUCCESS in 1h 48m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2d5f9fa1f8974e229a10577ee95385b4 : SUCCESS in 1h 19m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/488a3204f5ff41a4a948d2012b131864 : SUCCESS in 1h 05m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fbaa0758bcfc41f3a19f6f8aae42f7e4 : SUCCESS in 1h 10m 22s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/116ff83fdff44eea942b9299cc361325 : SUCCESS in 1h 14m 31s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/f8361a23d10744c48a96197da524824d : SUCCESS in 54m 43s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/937965075f994152a5372ae9b6858df8 : SUCCESS in 49m 30s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/09914fabbbc44c25b86803e7d1bff7e6 : SUCCESS in 1h 03m 34s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/79529fa003c4402fa7990ebac17432e9 : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7a384e859b6cbec8ac21e92c5298992257066d54","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-06-04 11:08:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d52737beb5a1759529a8176afc851cff0db2eb20","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2026-06-09 13:25:33.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f4608993cc14febfa70cfd6f8b62fe242b41e6b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-11 01:40:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ce5c8bf9565df6dd92ec912d74fd72a677cd06e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 01:41:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"075789743397e4a8798929956601e7b26282fed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 03:53:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c066740f7c5d45e78d36e4a76c9cdefd\n\n- grenade https://zuul.opendev.org/t/openstack/build/2a64558046894000b9d16d4d605d342e : SUCCESS in 32m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fddb00ddea6d437a80c5fe46ba844325 : SUCCESS in 2h 07m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cdcb2c1273c4b94b01d88053a619d3f : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77b51f145e564988bf39b4cc733c39e5 : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/267d1bc7f05e41f891f4d547b0b0e0ad : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6cb6ffd2635d479a9b16ec6644b3cdb1 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57074ebbf66a4626b5523fb85dc6b87a : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f6e33bba9b447a2abb3d2f17c8e923d : SUCCESS in 3m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/69b23e17e5a24d9f833eb4d31e1b4703 : SUCCESS in 1h 47m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b8fcccd8497493e90600d5be000f5e8 : SUCCESS in 1h 13m 38s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/525014fbae0b49ecb454207ed2412710 : SUCCESS in 2h 03m 14s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5917e99f9598467faeb69f4673047e83 : SUCCESS in 46m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a3ee66736c634e4dafe6904985fae60c : SUCCESS in 1h 02m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3eaae6ef3b84aa8b8dfa5d9cbcbc276 : SUCCESS in 58m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8d2140ae814b4840819a1bc1feb1a0bf : SUCCESS in 52m 57s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/43dd065587e84899acb2670172a37ea1 : SUCCESS in 2h 01m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e08913d78c8743ffb9e850a1e5fdb99a : SUCCESS in 1h 53m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e1709e8109b2497c9185f02ff839fe96 : SUCCESS in 50m 56s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/03e1ce97cf304f39afecdbd584b36e7e : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":1},{"id":"77f804707c178d9ac533c454670486d574f2fad5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 03:53:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ae2fec77377cb3183e9363a434d114573a81fd26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 03:55:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b00c3c2cbf54de0b737795e28bee257\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0540f940bf41549cdce32642cd6b12 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a4588d549cfad67cddfb3b5e41644e46dca31b6","revisions":{"5a4588d549cfad67cddfb3b5e41644e46dca31b6":{"kind":"REWORK","_number":1,"created":"2026-06-02 00:25:49.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/99/990999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/99/990999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/99/990999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/99/990999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/99/990999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/99/990999/1"}}},"commit":{"parents":[{"commit":"529e273fc25518327e06547897cbd63c27161b44","subject":"Stop running upgrade testing for EOL branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/529e273fc25518327e06547897cbd63c27161b44"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-06-02 00:22:59.000000000","tz":-420},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2026-06-02 00:22:59.000000000","tz":-420},"subject":"Wait for volume \"in-use\" in test_encrypted_cinder_volumes_resize_revert","message":"Wait for volume \"in-use\" in test_encrypted_cinder_volumes_resize_revert\n\nI saw an intermittent failure with this test where it failed to detach\nthe volume at the end of the test due to the volume not being \"in-use\",\nhowever shortly after the test failed I saw the same volume detach\nsucceed during test tear down.\n\nThis leads me to think there is a race where the instance can become\nACTIVE after the resize revert is done but the volume isn\u0027t quite back\nto \"in-use\" yet.\n\nThis adds a volume resource status wait on \"in-use\" before the test\nattempts to detach the volume at the end.\n\nCloses-Bug: #2154775\n\nChange-Id: I7c352ea9b5de99d2e482f647c246e499c277f5e8\nSigned-off-by: melanie witt \u003cmelwittt@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5a4588d549cfad67cddfb3b5e41644e46dca31b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5a4588d549cfad67cddfb3b5e41644e46dca31b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
