)]}'
{"id":"openstack%2Ftempest~994406","triplet_id":"openstack%2Ftempest~master~I66ff13ee08e4036ec27f71256161bdcdf8333f04","project":"openstack/tempest","branch":"master","topic":"test_resize_volume_backed_server_confirm","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-06-24 17:32:45.000000000","reason":"Change was submitted"},"32704":{"account":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"last_update":"2026-06-24 17:32:45.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-06-24 09:31:25.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-06-24 15:32:31.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I66ff13ee08e4036ec27f71256161bdcdf8333f04","subject":"Mark ``test_resize_volume_backed_server_confirm`` as unstable","status":"MERGED","created":"2026-06-23 07:14:36.000000000","updated":"2026-06-24 17:33:51.000000000","submitted":"2026-06-24 17:32:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994406-test_resize_volume_backed_server_confirm","meta_rev_id":"7f9007075035295c72550a535fc30da856a2d80e","_number":994406,"virtual_id_number":994406,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-24 17:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-24 09:31:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-24 15:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2026-06-24 09:06:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-24 15:32:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-06-24 09:31:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-24 15:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},{"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-23 07:15:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2026-06-23 07:15:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2026-06-23 09:29:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-23 12:20:35.000000000","updated_by":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"reviewer":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"state":"REVIEWER"},{"updated":"2026-06-23 16:22:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-06-24 09:06:30.000000000","updated_by":{"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"},"reviewer":{"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"},"state":"CC"},{"updated":"2026-06-24 09:06:57.000000000","updated_by":{"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"},"reviewer":{"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"},"state":"REVIEWER"},{"updated":"2026-06-24 09:31:25.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"2b467840e9ea387e6689e5e076685b32359caaa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-23 07:14:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79c58a5466fe8ffa023a20cb1f651d4e3c3c6f22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-23 07:15:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37d68e3534852b12747df10da5696cfb46bd0564","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-23 07:22:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e0916ff391817397255cf9e26d0b53b33952838c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 09:29:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/217754e51448488396d7a469eb4bca68\n\n- grenade https://zuul.opendev.org/t/openstack/build/037dab62ea0a4c83ae635f00f81f74a9 : SUCCESS in 1h 06m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f24783ae50e48099f3179a8def88cf0 : SUCCESS in 1h 49m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4537f2f4a4244bdaa6a63a26b2861f0d : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6111661f3f9e4347b3388b8f9c5bbc6c : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec26720665a048e89021479aba15e42a : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf10b69732ff49b19d230f0faba70724 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97beae5164c44eba800e5974763d5864 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e179bacb7f942fbac3cbab0e3b010ef : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e19a8478010643da8983eb85af806d34 : SUCCESS in 3m 53s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/919c8e7fd7b74c5ca7865eb9cff91312 : SUCCESS in 1h 23m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cf84240079b74ff6aa13365a14652c4e : SUCCESS in 1h 14m 46s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/9f1c21910159473292f6e5f218202d93 : SUCCESS in 1h 50m 21s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/ec73f83a35ee4b73a30673732b644724 : SUCCESS in 1h 43m 02s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/29c87a582ac7473c97c889ca9b52e3db : SUCCESS in 1h 33m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1d595d2238a14d6d89128dd3976cdaaa : SUCCESS in 40m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3dced876465542a89766a93d0a6cae41 : SUCCESS in 19m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82377185855a4394981ccf023ab20d9b : SUCCESS in 33m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/687b28ee1e53486888e23460573aed91 : SUCCESS in 1h 55m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/54dec4637b974b458777208018c96bfd : SUCCESS in 50m 32s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/11f9c278c9ed455194d395f03cd3de2a : SUCCESS in 55m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1211c7fe9d4f4129983a5617967dd221 : SUCCESS in 1h 29m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f71e1f9bbb844075a555c955a5556241 : SUCCESS in 1h 15m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/765793e4e14c421ea0844a7bbb989e8f : SUCCESS in 35m 25s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6095c3d466af49fa96cff1f7c717bbb1 : FAILURE in 1h 07m 18s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/b6cf6e3c508449b5bac9d81baf5a0d6b : SUCCESS in 2h 07m 30s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/4aeccac43c864a02bd55c855ae56d1ad : SUCCESS in 1h 16m 16s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/00efe7dd65ef451b992f084662e2c611 : SUCCESS in 1h 09m 59s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d3ebc0e2e1ee4744bdcd79df0d8cdac2 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e21f279023e4cfaabaa74122fb58032691061a4e","author":{"_account_id":32704,"name":"Alfredo Garcia","display_name":"Alfredo Garcia","email":"alfrgarc@redhat.com","username":"alfrgarc","status":"Senior Software Quality Engineer @ Red Hat"},"date":"2026-06-23 12:21:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c3bb735183170f0568946463a96f6db376203da","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-23 15:17:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8badfe50347e48fd56700f37ec345cfe4ce91a34","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-23 16:22:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"058de9c5da4dbaa3535c12c37ccebd6df2529eca","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-23 17:38:18.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b8248cee81b563611c29f1d88f34af39e1ef612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-06-24 08:12:00.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ecfbcd1bebfce6b644563bee60b1f9c1a770878d","author":{"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"},"date":"2026-06-24 09:06:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea13b9f59228b02f1ef10c05aaf1ed4761e2ac4c","author":{"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"},"date":"2026-06-24 09:06:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6d6537b17a57455b1e1068d5ab99e194cfa3910f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2026-06-24 09:31:25.000000000","message":"Patch Set 3: Code-Review+1 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"771522de0a7516285c9971a9f258bbe9e60f1149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 10:26:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7d30d4facc84141a1bda7097664673c\n\n- grenade https://zuul.opendev.org/t/openstack/build/faf7c5c6803e485aa157c1a6eaeca0a3 : SUCCESS in 1h 00m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf3897602d9a4208850a57d2f2374d78 : SUCCESS in 1h 35m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16585faee48c4a4a94799fd0856a7442 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1892c80d9a5149049f02b1654658bbab : SUCCESS in 15m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6508673ee9074d6bbd25dcddb4183edb : SUCCESS in 25m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/663dd8940bce4cc8ab47d8d4c61a7fd6 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bcc950870eb841c49f2bc1723ebfa200 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2fe0a247294b467997d8476032e5441e : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e958aac070243c5a88e2604649f9425 : SUCCESS in 5m 28s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/a030b9940e7e442481463aa756ba5267 : SUCCESS in 1h 15m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b537561870a14e89b3df56a7b1938bcd : SUCCESS in 2h 04m 16s (non-voting)\n- tempest-full-2026-1 https://zuul.opendev.org/t/openstack/build/5fd7eae427b94ae7ac8b14530b024fd8 : SUCCESS in 1h 44m 54s\n- tempest-full-2025-1 https://zuul.opendev.org/t/openstack/build/18ee98a8643b42f2a1ae1c4779afdf49 : SUCCESS in 1h 42m 34s\n- cinder-tempest-lvm-multibackend-2025-1 https://zuul.opendev.org/t/openstack/build/6caecd4237714047a9df3c627aaa9aea : SUCCESS in 1h 29m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/377e4f917cb941f9a5d0fa84af435d7a : SUCCESS in 1h 15m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f500b56a3ae6484684ded3b6adb24e1e : SUCCESS in 19m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1868665d69e4dd586ed264e13f0bea0 : SUCCESS in 1h 00m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4abfcb40cf75439090ab706041d26b26 : SUCCESS in 1h 50m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a089356371db4801bc8febadd541c380 : SUCCESS in 56m 46s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8d255937316e46e9a7923bb7d676f9ef : SUCCESS in 1h 38m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/23e443c4bd1a4b8a94a40f8838cb4fd3 : SUCCESS in 1h 06m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/31ccb1eb51f5485bbb32b24ddb7beb89 : SUCCESS in 1h 10m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d4be0e584bf4b6bb2b49d991620a2a5 : SUCCESS in 1h 08m 06s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2b4d4d028b8a4ab0adf6fdadb2dc6205 : SUCCESS in 52m 11s (non-voting)\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/40ac2fbfd228468cb3ab94766396f9a2 : SUCCESS in 1h 53m 33s\n- tempest-full-preprov-creds-no-admin https://zuul.opendev.org/t/openstack/build/a954f46d7cee45aba3765735c9d6663f : SUCCESS in 48m 44s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7b8b08c451564f09961c24321a538107 : SUCCESS in 1h 10m 04s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5357efe9754541999797339f45cf9e12 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":3},{"id":"9dd2f2fe3ff4ca220dcc0f26f6a152dbee7b0b8a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-06-24 15:32:31.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1cf504c157587c9d93eecb02d13fe439f777a77c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 15:33:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7b35508722e3e2145f4106f7837b322c8bc2fd78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 17:32:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a8a8ef31574e2f8813b904bd732e73\n\n- grenade https://zuul.opendev.org/t/openstack/build/61414f1186d641bc9ac03fc1a8778fce : SUCCESS in 58m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0e1ed0349d5405f8cc1e8b98c45a32d : SUCCESS in 1h 35m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08902355478f47edb26da42d89693c79 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83dedea0ca964c748f43242890c35a55 : SUCCESS in 31m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f11186749f2941fbb3ab1fa792ab6e51 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d079352cd9ec4de9a27150c35f00e4cc : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33423d9fa7454da2b54cb4a5fb9affe5 : SUCCESS in 4m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cc89ab0bf834c4a96b07f981cd0a975 : SUCCESS in 4m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/576c250b44084c88a86b1bc274132aaf : SUCCESS in 52m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7911ebc101354e979dbe7b40ec0c0f4e : SUCCESS in 1h 17m 07s\n- tempest-full-preprov-creds https://zuul.opendev.org/t/openstack/build/3d3345e70a44487386a3b7a934591792 : SUCCESS in 1h 46m 17s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/ac1bf0042faa4f45b8591fff8383ac61 : SUCCESS in 1h 20m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dad607d77e9a4988b6748706e55f13ce : SUCCESS in 1h 12m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/163d9a6a9b7d45a7be8f95d358c1835b : SUCCESS in 58m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/14692ca678a444f4910096d14550e8ff : SUCCESS in 1h 16m 36s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fbaf6dcb159449a3aca63b84bc973228 : SUCCESS in 1h 42m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f686010c5434ef7a7ec032711f932a3 : SUCCESS in 1h 54m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ed04183bd7f1410692f818f93dd5a224 : SUCCESS in 54m 39s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/63a784b20081420bac2b147321e01d5f : SUCCESS in 1h 05m 51s","accounts_in_message":[],"_revision_number":3},{"id":"fcdf10357db51591a041377b9b116141aa0bf564","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 17:32:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7f9007075035295c72550a535fc30da856a2d80e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 17:33:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7891967ac65445ab27ea4f1e895d3aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ce104d652cd4848a50f8744db56e7cf : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1a502cabe95e31b87b74fcabe8c343a4458da3a8","revisions":{"bcf81363f51ccaa3302443ad93e86a5aad19044b":{"kind":"REWORK","_number":1,"created":"2026-06-23 07:14:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/994406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/06/994406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/994406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/994406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/994406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/06/994406/1"}}},"commit":{"parents":[{"commit":"a3aac772e637e114c0a60a840949642725a613b5","subject":"Merge \"Add object symlink tests for Swift\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a3aac772e637e114c0a60a840949642725a613b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-23 07:13:53.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-23 07:13:53.000000000","tz":120},"subject":"Mark ``test_resize_volume_backed_server_confirm`` as unstable","message":"Mark ``test_resize_volume_backed_server_confirm`` as unstable\n\nThe test ``test_resize_volume_backed_server_confirm`` is marked as\nunstable because of [1].\n\n[1]https://github.com/cirros-dev/cirros/issues/140\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I66ff13ee08e4036ec27f71256161bdcdf8333f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bcf81363f51ccaa3302443ad93e86a5aad19044b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bcf81363f51ccaa3302443ad93e86a5aad19044b"}]},"branch":"refs/heads/master"},"d8ee056278acea1cec42a59d66c5e6defc83f2d5":{"kind":"REWORK","_number":2,"created":"2026-06-23 07:15:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/994406/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/06/994406/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/994406/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/994406/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/994406/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/06/994406/2"}}},"commit":{"parents":[{"commit":"a3aac772e637e114c0a60a840949642725a613b5","subject":"Merge \"Add object symlink tests for Swift\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a3aac772e637e114c0a60a840949642725a613b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-23 07:13:53.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-23 07:15:48.000000000","tz":120},"subject":"Mark ``test_resize_volume_backed_server_confirm`` as unstable","message":"Mark ``test_resize_volume_backed_server_confirm`` as unstable\n\nThe test ``test_resize_volume_backed_server_confirm`` is marked as\nunstable because of [1].\n\n[1]https://github.com/cirros-dev/cirros/issues/140\n\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I66ff13ee08e4036ec27f71256161bdcdf8333f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d8ee056278acea1cec42a59d66c5e6defc83f2d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d8ee056278acea1cec42a59d66c5e6defc83f2d5"}]},"branch":"refs/heads/master"},"1a502cabe95e31b87b74fcabe8c343a4458da3a8":{"kind":"REWORK","_number":3,"created":"2026-06-24 08:12:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/06/994406/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/06/994406/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/994406/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/994406/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/06/994406/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/06/994406/3"}}},"commit":{"parents":[{"commit":"a3aac772e637e114c0a60a840949642725a613b5","subject":"Merge \"Add object symlink tests for Swift\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a3aac772e637e114c0a60a840949642725a613b5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-06-23 07:13:53.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2026-06-24 08:12:00.000000000","tz":0},"subject":"Mark ``test_resize_volume_backed_server_confirm`` as unstable","message":"Mark ``test_resize_volume_backed_server_confirm`` as unstable\n\nThe test ``test_resize_volume_backed_server_confirm`` is marked as\nunstable because of [1].\n\n[1]https://github.com/cirros-dev/cirros/issues/140\n\nRelated-Bug: #2158063\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: I66ff13ee08e4036ec27f71256161bdcdf8333f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a502cabe95e31b87b74fcabe8c343a4458da3a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1a502cabe95e31b87b74fcabe8c343a4458da3a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38173,"name":"Tanvi Nautiyal","email":"tnautiya@redhat.com","username":"tnautiyal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
