)]}'
{"id":"openstack%2Ftenks~655083","triplet_id":"openstack%2Ftenks~master~I8241e8207c5caee7cb9c479bcca6196fedbc7fa3","project":"openstack/tenks","branch":"master","topic":"655083","hashtags":[],"change_id":"I8241e8207c5caee7cb9c479bcca6196fedbc7fa3","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:40:37.000000000","updated":"2019-04-26 17:18:16.000000000","submitted":"2019-04-26 17:18:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655083-1556299096727-0411a076","meta_rev_id":"3f7def26a775f47d7193130c37eac6de0e685a52","_number":655083,"virtual_id_number":655083,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-04-24 13:59:52.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-04-26 17:18:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-04-26 16:08:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-04-26 16:08:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 11:12:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-04-24 13:59:52.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-04-26 16:08:44.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-04-26 17:18:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"377973d68528fba288fc5b296c7ac7138e4edd45","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:40:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46b608fe50f555a5191ce1b60c4293181e367e80","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-23 12:52:15.000000000","message":"Patch Set 1: Code-Review-2\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"060435d0f6b374a2790d4b032ea2dae45034865a","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 13:16:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0ff8a5ed4ee908abe4fcd0f508567c2b7cef60fe","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 13:17:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"123c0a1562201f40d0c68ffe757bf4a7fd9c3372","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 13:17:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6dee7a4e2802a9de887641f8742cb09e52688050","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-23 13:38:47.000000000","message":"Patch Set 3: Code-Review+2\n\nbetter, thanks :)","accounts_in_message":[],"_revision_number":3},{"id":"b50df47000c302c6f17d5da9b77ce871a1626617","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 14:01:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/83/655083/3/check/openstack-tox-cover/f58838f/cover/ : SUCCESS in 7m 45s\n- openstack-tox-pep8 http://logs.openstack.org/83/655083/3/check/openstack-tox-pep8/85bc01e/ : SUCCESS in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/83/655083/3/check/openstack-tox-py27/ae5ea92/ : SUCCESS in 6m 49s\n- openstack-tox-py35 http://logs.openstack.org/83/655083/3/check/openstack-tox-py35/2f49b81/ : SUCCESS in 6m 34s\n- openstack-tox-py36 http://logs.openstack.org/83/655083/3/check/openstack-tox-py36/c116ad2/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/83/655083/3/check/openstack-tox-docs/b05405e/html/ : SUCCESS in 6m 06s\n- build-openstack-releasenotes http://logs.openstack.org/83/655083/3/check/build-openstack-releasenotes/26a5062/html/ : SUCCESS in 5m 09s\n- tenks-tox-ansible-lint http://logs.openstack.org/83/655083/3/check/tenks-tox-ansible-lint/d9f3bbe/ : SUCCESS in 5m 50s\n- tenks-deploy-teardown-centos http://logs.openstack.org/83/655083/3/check/tenks-deploy-teardown-centos/c8d4c33/ : RETRY_LIMIT in 2m 39s\n- tenks-deploy-teardown-ubuntu http://logs.openstack.org/83/655083/3/check/tenks-deploy-teardown-ubuntu/c486fb0/ : RETRY_LIMIT in 1m 54s\n- kayobe-overcloud-centos http://logs.openstack.org/83/655083/3/check/kayobe-overcloud-centos/c60d2ed/ : RETRY_LIMIT in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"750aa868f8f72a0829db8b5f3e9c27cd23756f36","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 16:46:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6562c6ea8075338a0e518617ee2ed8f57df69692","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 17:38:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/83/655083/3/check/openstack-tox-cover/ddff72f/cover/ : SUCCESS in 6m 29s\n- openstack-tox-pep8 http://logs.openstack.org/83/655083/3/check/openstack-tox-pep8/e1cfece/ : SUCCESS in 6m 22s\n- openstack-tox-py27 http://logs.openstack.org/83/655083/3/check/openstack-tox-py27/b54c9aa/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/83/655083/3/check/openstack-tox-py35/15471d2/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/83/655083/3/check/openstack-tox-py36/01aac88/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/83/655083/3/check/openstack-tox-docs/e756dfe/html/ : SUCCESS in 4m 53s\n- build-openstack-releasenotes http://logs.openstack.org/83/655083/3/check/build-openstack-releasenotes/c6c96cb/html/ : SUCCESS in 7m 37s\n- tenks-tox-ansible-lint http://logs.openstack.org/83/655083/3/check/tenks-tox-ansible-lint/e2ec43f/ : SUCCESS in 4m 24s\n- tenks-deploy-teardown-centos http://logs.openstack.org/83/655083/3/check/tenks-deploy-teardown-centos/6c2ef7b/ : RETRY_LIMIT in 1m 55s\n- tenks-deploy-teardown-ubuntu http://logs.openstack.org/83/655083/3/check/tenks-deploy-teardown-ubuntu/f169753/ : RETRY_LIMIT in 1m 57s\n- kayobe-overcloud-centos http://logs.openstack.org/83/655083/3/check/kayobe-overcloud-centos/3fa5e43/ : RETRY_LIMIT in 2m 23s","accounts_in_message":[],"_revision_number":3},{"id":"bd37dcc2c727a838b20283b039eba5682fa9c6d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-24 11:11:47.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"abbf443cfcad5a94434dab29a265a6eb9fec312b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-04-24 11:12:14.000000000","message":"Patch Set 4:\n\nHopefully fixed the Tenks job. I think we\u0027ll need to fix the kayobe job in kayobe.","accounts_in_message":[],"_revision_number":4},{"id":"0e610043e51f809cd7acd0787f6ffad6a684049d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 11:38:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/83/655083/4/check/openstack-tox-cover/65abc88/cover/ : SUCCESS in 4m 32s\n- openstack-tox-pep8 http://logs.openstack.org/83/655083/4/check/openstack-tox-pep8/607675c/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/83/655083/4/check/openstack-tox-py27/c0a336e/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/83/655083/4/check/openstack-tox-py35/e417218/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/83/655083/4/check/openstack-tox-py36/a7ab328/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/83/655083/4/check/openstack-tox-docs/e8262d4/html/ : SUCCESS in 4m 11s\n- build-openstack-releasenotes http://logs.openstack.org/83/655083/4/check/build-openstack-releasenotes/1d6a72f/html/ : SUCCESS in 3m 53s\n- tenks-tox-ansible-lint http://logs.openstack.org/83/655083/4/check/tenks-tox-ansible-lint/5e8c90c/ : SUCCESS in 4m 52s\n- tenks-deploy-teardown-centos http://logs.openstack.org/83/655083/4/check/tenks-deploy-teardown-centos/9463699/ : SUCCESS in 7m 38s\n- tenks-deploy-teardown-ubuntu http://logs.openstack.org/83/655083/4/check/tenks-deploy-teardown-ubuntu/81383c3/ : SUCCESS in 4m 28s\n- kayobe-overcloud-centos http://logs.openstack.org/83/655083/4/check/kayobe-overcloud-centos/aebe6db/ : RETRY_LIMIT in 2m 20s","accounts_in_message":[],"_revision_number":4},{"id":"ec622d84191584eefce9e5af085d401d6996edb8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-24 12:21:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2a318f6df269ca59baa59f92d48ff89a055249f3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-24 12:22:32.000000000","message":"Patch Set 5:\n\nadded one more change and should be all","accounts_in_message":[],"_revision_number":5},{"id":"9fb4169ecedf415190db281403804a41fa6c26ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 12:44:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/83/655083/5/check/openstack-tox-cover/718ca79/cover/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/83/655083/5/check/openstack-tox-pep8/67c930e/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/83/655083/5/check/openstack-tox-py27/d65dc53/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/83/655083/5/check/openstack-tox-py35/fad5276/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/83/655083/5/check/openstack-tox-py36/61c7641/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/83/655083/5/check/openstack-tox-docs/4e72374/html/ : SUCCESS in 4m 40s\n- build-openstack-releasenotes http://logs.openstack.org/83/655083/5/check/build-openstack-releasenotes/3bd0e51/html/ : SUCCESS in 3m 54s\n- tenks-tox-ansible-lint http://logs.openstack.org/83/655083/5/check/tenks-tox-ansible-lint/9c87e72/ : SUCCESS in 4m 47s\n- tenks-deploy-teardown-centos http://logs.openstack.org/83/655083/5/check/tenks-deploy-teardown-centos/3383f26/ : SUCCESS in 5m 49s\n- tenks-deploy-teardown-ubuntu http://logs.openstack.org/83/655083/5/check/tenks-deploy-teardown-ubuntu/73300bd/ : SUCCESS in 4m 40s\n- kayobe-overcloud-centos http://logs.openstack.org/83/655083/5/check/kayobe-overcloud-centos/5a18efe/ : RETRY_LIMIT in 3m 23s","accounts_in_message":[],"_revision_number":5},{"id":"dfb7ffbf134d1bc42b03b63ced6fdce7d9f9144d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-04-24 13:59:52.000000000","message":"Patch Set 5:\n\nlet\u0027s wait for Kayobe side then before moving forward then","accounts_in_message":[],"_revision_number":5},{"id":"ab5636ebcf36cf8419bb87e19dfb87a685534583","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-26 14:20:51.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d91266f620a264c54ea60dfc9d027f1e7dd8c5f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 15:56:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/83/655083/5/check/openstack-tox-cover/4dd71bf/cover/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/83/655083/5/check/openstack-tox-pep8/bb33a58/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/83/655083/5/check/openstack-tox-py27/e9ed62a/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/83/655083/5/check/openstack-tox-py35/9971de3/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/83/655083/5/check/openstack-tox-py36/9e89896/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/83/655083/5/check/openstack-tox-docs/c4b69f7/html/ : SUCCESS in 4m 28s\n- build-openstack-releasenotes http://logs.openstack.org/83/655083/5/check/build-openstack-releasenotes/ea45ab5/html/ : SUCCESS in 4m 13s\n- tenks-tox-ansible-lint http://logs.openstack.org/83/655083/5/check/tenks-tox-ansible-lint/3034d58/ : SUCCESS in 4m 35s\n- tenks-deploy-teardown-centos http://logs.openstack.org/83/655083/5/check/tenks-deploy-teardown-centos/ee7dd94/ : SUCCESS in 5m 35s\n- tenks-deploy-teardown-ubuntu http://logs.openstack.org/83/655083/5/check/tenks-deploy-teardown-ubuntu/7dd325d/ : SUCCESS in 5m 09s\n- kayobe-overcloud-centos http://logs.openstack.org/83/655083/5/check/kayobe-overcloud-centos/8b04baf/ : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":5},{"id":"e2b5e50c8ab23f237188fba8eec2db919b446f7a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-04-26 16:08:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"fee3435265ad90eeb8ccd18f15569d845e91029b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 16:08:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a684c04d3099cd75e0884805a7fa5cce65640588","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 17:18:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/83/655083/5/gate/openstack-tox-pep8/cde430b/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/83/655083/5/gate/openstack-tox-py27/0ed1733/ : SUCCESS in 4m 45s\n- openstack-tox-py35 http://logs.openstack.org/83/655083/5/gate/openstack-tox-py35/973ff4a/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/83/655083/5/gate/openstack-tox-py36/c05769a/ : SUCCESS in 5m 48s\n- openstack-tox-docs http://logs.openstack.org/83/655083/5/gate/openstack-tox-docs/9379ad8/html/ : SUCCESS in 6m 06s\n- build-openstack-releasenotes http://logs.openstack.org/83/655083/5/gate/build-openstack-releasenotes/b19e68a/html/ : SUCCESS in 5m 02s\n- tenks-tox-ansible-lint http://logs.openstack.org/83/655083/5/gate/tenks-tox-ansible-lint/32a59b8/ : SUCCESS in 7m 51s\n- tenks-deploy-teardown-centos http://logs.openstack.org/83/655083/5/gate/tenks-deploy-teardown-centos/34e2909/ : SUCCESS in 10m 06s\n- tenks-deploy-teardown-ubuntu http://logs.openstack.org/83/655083/5/gate/tenks-deploy-teardown-ubuntu/ea6a46e/ : SUCCESS in 6m 40s\n- kayobe-overcloud-centos http://logs.openstack.org/83/655083/5/gate/kayobe-overcloud-centos/0ac866c/ : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":5},{"id":"e413ae8f88b3fafc6a0b274c5aa0ba86affc60f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 17:18:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d7142c04d23f348da55202a2d8696d3eddb18699","revisions":{"30e59af36cfd19a0499b22477885af07df67ac69":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:40:37.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/83/655083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/83/655083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/83/655083/1"}}},"commit":{"parents":[{"commit":"aa0bda25ac0a34193ba2b4f65a854cdd2745379b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/aa0bda25ac0a34193ba2b4f65a854cdd2745379b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:40:09.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:40:33.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I8241e8207c5caee7cb9c479bcca6196fedbc7fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/30e59af36cfd19a0499b22477885af07df67ac69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/30e59af36cfd19a0499b22477885af07df67ac69"}]},"branch":"refs/heads/master"},"ee564548b9f1116f3e24c14f91c605d620eb6097":{"kind":"REWORK","_number":2,"created":"2019-04-23 13:16:30.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/83/655083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/83/655083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/83/655083/2"}}},"commit":{"parents":[{"commit":"aa0bda25ac0a34193ba2b4f65a854cdd2745379b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/aa0bda25ac0a34193ba2b4f65a854cdd2745379b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:40:09.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 13:16:22.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I8241e8207c5caee7cb9c479bcca6196fedbc7fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/ee564548b9f1116f3e24c14f91c605d620eb6097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/ee564548b9f1116f3e24c14f91c605d620eb6097"}]},"branch":"refs/heads/master"},"48dc6fa5bd81b4859c4394ae60fc3337bdd32f23":{"kind":"REWORK","_number":3,"created":"2019-04-23 13:17:35.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/83/655083/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/83/655083/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/83/655083/3"}}},"commit":{"parents":[{"commit":"aa0bda25ac0a34193ba2b4f65a854cdd2745379b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/aa0bda25ac0a34193ba2b4f65a854cdd2745379b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:40:09.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 13:17:29.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I8241e8207c5caee7cb9c479bcca6196fedbc7fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/48dc6fa5bd81b4859c4394ae60fc3337bdd32f23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/48dc6fa5bd81b4859c4394ae60fc3337bdd32f23"}]},"branch":"refs/heads/master"},"d9d4cdda22e27f49680d1f49596ac81c90b6b28f":{"kind":"REWORK","_number":4,"created":"2019-04-24 11:11:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/655083/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/83/655083/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/83/655083/4"}}},"commit":{"parents":[{"commit":"aa0bda25ac0a34193ba2b4f65a854cdd2745379b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/aa0bda25ac0a34193ba2b4f65a854cdd2745379b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:40:09.000000000","tz":480},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-24 11:11:42.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I8241e8207c5caee7cb9c479bcca6196fedbc7fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/d9d4cdda22e27f49680d1f49596ac81c90b6b28f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/d9d4cdda22e27f49680d1f49596ac81c90b6b28f"}]},"branch":"refs/heads/master"},"d7142c04d23f348da55202a2d8696d3eddb18699":{"kind":"REWORK","_number":5,"created":"2019-04-24 12:21:50.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/83/655083/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/83/655083/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/83/655083/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/83/655083/5"}}},"commit":{"parents":[{"commit":"aa0bda25ac0a34193ba2b4f65a854cdd2745379b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/aa0bda25ac0a34193ba2b4f65a854cdd2745379b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:40:09.000000000","tz":480},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-04-24 12:21:48.000000000","tz":120},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I8241e8207c5caee7cb9c479bcca6196fedbc7fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/d7142c04d23f348da55202a2d8696d3eddb18699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/d7142c04d23f348da55202a2d8696d3eddb18699"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
