)]}'
{"id":"openstack%2Ftenks~710100","triplet_id":"openstack%2Ftenks~master~I54388f713e17a47601be2db63a26fc43ca7ac6c8","project":"openstack/tenks","branch":"master","hashtags":[],"change_id":"I54388f713e17a47601be2db63a26fc43ca7ac6c8","subject":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10","status":"MERGED","created":"2020-02-26 17:51:49.000000000","updated":"2020-03-11 17:40:06.000000000","submitted":"2020-03-11 17:36:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710100-1583948198818-eb552905","meta_rev_id":"4e231a4f54d351d11f8669a9640329592166abc3","_number":710100,"virtual_id_number":710100,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 17:36:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-11 16:55:59.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-11 16:55:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-03-11 16:56:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 15:20:19.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-03-06 14:11:54.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-03-06 15:34:59.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-03-11 17:36:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e25822098a7f594e0a014ee42edfb93f43a8bd8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-26 17:51:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0115df37615f22a1a85910ab69e9c20bca488f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 19:07:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/604e457a27514bacb09b984294add321 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812c781ef50843a0b10540c7598c3746 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31a4361b08104010b3303637deb472ec : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/373b5c52577149b8a47ddffaa484b598 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74b784456919465380b85ea2e37c2f7c : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34649df832124f80bb4932cd5db34bb3 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8125b8b391134cb6bed5e9b02ae46d81 : SUCCESS in 4m 38s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/dc5bd4f7bff04e0994ae7c3be75bdb52 : SUCCESS in 6m 07s\n- tenks-deploy-teardown-centos https://zuul.opendev.org/t/openstack/build/e37f546a9588494a85ed27e1589fdae9 : FAILURE in 11m 55s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/ca2969ac41e243a4a576cca07f092fe3 : FAILURE in 6m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b207853509024924b92e86bd3f011856 : FAILURE in 1h 07m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fc3d45dd21fdc64dd7f2f2128d77726319f774a6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 09:42:15.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cb3a512441fd3d695b3cef002e4b5ef2a35a3d03","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 10:17:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"69e1738d0a469f345291e9bcdb64ea3cf0c210fc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 10:22:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8b7dcdbe8800353817142dc474f16557a5cd154a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 15:50:07.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d202a8e93d6246fc4189d11faa67a0464a581dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 21:18:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28a44344c7f2468da1d10b6589b7c35d : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef5e68141d08438686b3c53c7bece6c8 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a35b7b93b2a7409293d8612b1c24f905 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed697d15e1fa44699cfe162be9eaa296 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a1d82ebdf234415a5a87e302e88c7e6 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea0c6797a0c466ba359fbc59afd393f : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aefac30d22d84ff19a49824499e88005 : SUCCESS in 7m 31s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/f24637f60b70415faee46b826782b9f8 : SUCCESS in 7m 36s\n- tenks-deploy-teardown-centos https://zuul.opendev.org/t/openstack/build/2d42785187ba4bcbab34b1212a9efdf4 : SUCCESS in 13m 34s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/6cd1d8d873114e5d972a624d87dde14c : SUCCESS in 8m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/015c6218a13740e4b1deb73710007feb : SUCCESS in 1h 43m 44s","accounts_in_message":[],"_revision_number":5},{"id":"3e5669bac5b438007cd5df8c50a14b2521ca6b3b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-06 14:11:54.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e1481d1e7b230825db4b04b087e77fdb2c810a76","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-03-06 15:34:59.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"cdda223c59887a6a2cbc3f84daf3f775f66f335e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-10 17:41:11.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1ac897346da0c8c0075ada6385d4094352e1a326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 17:41:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"694bd74047fd923e75ccd804aed39fa3f080bd3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 18:56:11.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c32bb75c8f64407aaf67a4209699b882 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19807ed254594adab565c0f4e8fa5f69 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09574c17ec064f0db34549ce1bb4641f : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b673ebe6b94ce6a2d3779b8d301d7d : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8747b3cc40124bd799d0722e02c66da5 : SUCCESS in 4m 04s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/b18bb32b570e486e93543fe0f256dc3b : SUCCESS in 5m 21s\n- tenks-deploy-teardown-centos https://zuul.opendev.org/t/openstack/build/413318fe55e345a9942c721630a43ab0 : FAILURE in 10m 34s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/9edc39407282445c8de6214bcb7906fa : SUCCESS in 6m 44s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/811d150d70f54305858c7d40ad06aa7a : FAILURE in 1h 09m 32s","accounts_in_message":[],"_revision_number":5},{"id":"5476a5129fd4d451be84e8ee006d02ee38a1944a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-03-11 10:24:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"03a3ed7146edc7ee2c2b92050c04b2e19fdcc13f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 11:40:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6577cf61078464e8d90e47f2a85dd8d : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44417f8004ca4ac5baebbbf7be81a872 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9448a132eb649f0885bddfebc536a86 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a1c63e2be4946a1a66e50cc5aca3f89 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7915e1596c284f2e9482c40b88cba172 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9c7c587c9a4dd9a47c83c717ced37a : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e25cabb2445747848b069fde2f17b0e6 : SUCCESS in 4m 06s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/48f46b007f3d43be8f93e92bc87c5df6 : SUCCESS in 5m 00s\n- tenks-deploy-teardown-centos https://zuul.opendev.org/t/openstack/build/74efc7f475a9440eb24091ec4c0074ca : FAILURE in 11m 59s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/d5ee69978b364beca1646b573b5bbe4f : SUCCESS in 6m 18s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bdacb32859bc4355a1e66c96709d571c : FAILURE in 59m 19s","accounts_in_message":[],"_revision_number":5},{"id":"eec4469e3213089788a5dee415001392e434f3ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 16:55:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"1f98aa9c2d70b02f6a2f2c27ab4993154d3ac271","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 16:56:32.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0ad9758e1ccc397769f36f65e2f6015088d5411a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:17:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07ee5911aae64fa5b6e3283034e65b50 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b046969cea934fc398c4ca211bd44de4 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b832c25ae6954302a130ffe4cb876292 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/317c2796ef974a0cb75534ef345e3c4e : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed0731a0472748928b36157f8158537c : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81cae87727dd479b9559a721cbb81168 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d07d8737efc47bcb9b472833d62970e : SUCCESS in 4m 44s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/c40cf5eb7e2d44d1b2a056cd623bbc45 : SUCCESS in 6m 48s\n- tenks-deploy-teardown-centos https://zuul.opendev.org/t/openstack/build/a369b3bf5ccb43c8ba821203f77d278f : SUCCESS in 12m 22s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/835025f6879d43a798b8705104e64907 : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":6},{"id":"333236b3356f6543fe6b45026b5baeebc981a474","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:17:29.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7a404478c8e628a6f419a04b23ccdf3457d73378","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:36:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"366f42e6cd86514dacf2ff4531fa1e5180592e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:36:38.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7bd452826234cbdab7b542551b1cc8f : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/688b2b06da094881921635972359cee9 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92c3c55c99a14080b474414cc3b856f7 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698e447ed5954640a3238609761b643d : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a04a62f8219425a8cd8531650e05d4a : SUCCESS in 4m 57s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/473f4eb635404d5f8e0dddebc14a1525 : SUCCESS in 6m 21s\n- tenks-deploy-teardown-centos https://zuul.opendev.org/t/openstack/build/70b99d9332984832b9925d926073cbc7 : SUCCESS in 12m 39s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/0147b17403094e97905a7018faf4c62b : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":6},{"id":"4e231a4f54d351d11f8669a9640329592166abc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 17:40:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507d0f08eefa45beb56b3e2eb9f0baac : SUCCESS in 1m 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e99f6cb867d425aa77fe34017be77b4 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"512174a828039d10dcd15f46a8de170595ec0f16","revisions":{"ad8a2c970409591bfdfbb636db165018b5397a30":{"kind":"REWORK","_number":1,"created":"2020-02-26 17:51:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/710100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/710100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/710100/1"}}},"commit":{"parents":[{"commit":"e4dae3dd4a4ceedb04f4f994e76bb2de278757ac","subject":"Install virtualbmc to system when SELinux is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/e4dae3dd4a4ceedb04f4f994e76bb2de278757ac"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-26 17:42:17.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-26 17:49:19.000000000","tz":0},"subject":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10","message":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10\n\nThis means that ansible 2.9 is typically used, and causes ansible-lint\nto fail. Two workarounds have been added to address this:\n\n* Use .get() to avoid passing an undefined object through to_nice_yaml\n  filter\n* Create a dummy tenks_update_state.py module to avoid ansible-lint\n  being confused about our action plugin.\n\nChange-Id: I54388f713e17a47601be2db63a26fc43ca7ac6c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/ad8a2c970409591bfdfbb636db165018b5397a30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/ad8a2c970409591bfdfbb636db165018b5397a30"}]},"branch":"refs/heads/master"},"7967a8b3e4bdd2e17b97e564a1fe1e262859f6a3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-27 09:42:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/710100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/710100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/710100/2"}}},"commit":{"parents":[{"commit":"ff1691d57ecf84fe5c28ab9d65f0a184de87df93","subject":"Install virtualbmc to system when SELinux is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/ff1691d57ecf84fe5c28ab9d65f0a184de87df93"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-26 17:42:17.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 09:42:01.000000000","tz":0},"subject":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10","message":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10\n\nThis means that ansible 2.9 is typically used, and causes ansible-lint\nto fail. Two workarounds have been added to address this:\n\n* Use .get() to avoid passing an undefined object through to_nice_yaml\n  filter\n* Create a dummy tenks_update_state.py module to avoid ansible-lint\n  being confused about our action plugin.\n\nChange-Id: I54388f713e17a47601be2db63a26fc43ca7ac6c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/7967a8b3e4bdd2e17b97e564a1fe1e262859f6a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/7967a8b3e4bdd2e17b97e564a1fe1e262859f6a3"}]},"branch":"refs/heads/master"},"876e3a3b9ad032f41a7e48e009122a5e8e0ea9f1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-27 10:17:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/710100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/710100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/710100/3"}}},"commit":{"parents":[{"commit":"e76b6e31bfed2fc36c351b8f955459453f58e335","subject":"Install virtualbmc to system when SELinux is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/e76b6e31bfed2fc36c351b8f955459453f58e335"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-26 17:42:17.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 10:16:54.000000000","tz":0},"subject":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10","message":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10\n\nThis means that ansible 2.9 is typically used, and causes ansible-lint\nto fail. Two workarounds have been added to address this:\n\n* Use .get() to avoid passing an undefined object through to_nice_yaml\n  filter\n* Create a dummy tenks_update_state.py module to avoid ansible-lint\n  being confused about our action plugin.\n\nChange-Id: I54388f713e17a47601be2db63a26fc43ca7ac6c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/876e3a3b9ad032f41a7e48e009122a5e8e0ea9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/876e3a3b9ad032f41a7e48e009122a5e8e0ea9f1"}]},"branch":"refs/heads/master"},"68ad1b77d553167448669fd247302442136fe2fb":{"kind":"REWORK","_number":4,"created":"2020-02-27 10:22:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/710100/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/710100/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/710100/4"}}},"commit":{"parents":[{"commit":"e76b6e31bfed2fc36c351b8f955459453f58e335","subject":"Install virtualbmc to system when SELinux is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/e76b6e31bfed2fc36c351b8f955459453f58e335"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-26 17:42:17.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 10:18:16.000000000","tz":0},"subject":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10","message":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10\n\nThis means that ansible 2.9 is typically used, and causes ansible-lint\nto fail. Two workarounds have been added to address this:\n\n* Use .get() to avoid passing an undefined object through to_nice_yaml\n  filter\n* Create a dummy tenks_update_state.py module to avoid ansible-lint\n  being confused about our action plugin.\n\nChange-Id: I54388f713e17a47601be2db63a26fc43ca7ac6c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/68ad1b77d553167448669fd247302442136fe2fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/68ad1b77d553167448669fd247302442136fe2fb"}]},"branch":"refs/heads/master"},"7d5ed7a8d587fe1b2978e08d261a5d5d95a0aaaf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-02-27 15:50:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/710100/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/710100/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/710100/5"}}},"commit":{"parents":[{"commit":"13de0edd6c4699cf9a3096f4ed2803efdc30a38a","subject":"Install virtualbmc to system when SELinux is enabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/13de0edd6c4699cf9a3096f4ed2803efdc30a38a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-26 17:42:17.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 15:47:44.000000000","tz":0},"subject":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10","message":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10\n\nThis means that ansible 2.9 is typically used, and causes ansible-lint\nto fail. Two workarounds have been added to address this:\n\n* Use .get() to avoid passing an undefined object through to_nice_yaml\n  filter\n* Create a dummy tenks_update_state.py module to avoid ansible-lint\n  being confused about our action plugin.\n\nChange-Id: I54388f713e17a47601be2db63a26fc43ca7ac6c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/7d5ed7a8d587fe1b2978e08d261a5d5d95a0aaaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/7d5ed7a8d587fe1b2978e08d261a5d5d95a0aaaf"}]},"branch":"refs/heads/master"},"512174a828039d10dcd15f46a8de170595ec0f16":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-03-11 16:55:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/710100/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/710100/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/710100/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/710100/6"}}},"commit":{"parents":[{"commit":"3db82fd5517f90b3a39d28bfb62910d3a4ea34c3","subject":"Pin virtualbmc to \u003c2 when using Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/3db82fd5517f90b3a39d28bfb62910d3a4ea34c3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-26 17:42:17.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-11 16:55:09.000000000","tz":0},"subject":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10","message":"Bump ansible version constraints to \u003e\u003d2.7,\u003c2.10\n\nThis means that ansible 2.9 is typically used, and causes ansible-lint\nto fail. Two workarounds have been added to address this:\n\n* Use .get() to avoid passing an undefined object through to_nice_yaml\n  filter\n* Create a dummy tenks_update_state.py module to avoid ansible-lint\n  being confused about our action plugin.\n\nChange-Id: I54388f713e17a47601be2db63a26fc43ca7ac6c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/512174a828039d10dcd15f46a8de170595ec0f16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/512174a828039d10dcd15f46a8de170595ec0f16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
