)]}'
{"id":"openstack%2Ftenks~714082","triplet_id":"openstack%2Ftenks~master~Ibef91d9ef499c8741b61a170672a23f530a600bb","project":"openstack/tenks","branch":"master","hashtags":[],"change_id":"Ibef91d9ef499c8741b61a170672a23f530a600bb","subject":"Fix node scheduling with pipe in hostvars","status":"MERGED","created":"2020-03-20 11:03:12.000000000","updated":"2020-04-02 17:17:26.000000000","submitted":"2020-04-02 16:50:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714082-1585846227248-9ed293ab","meta_rev_id":"a56bad40d546217a58c7bc30848ce5b21809e413","_number":714082,"virtual_id_number":714082,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 16:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-03-20 12:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-04-02 14:52:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 12:54:55.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-04-02 16:50:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d2c2b5985164dbf04fff5e9ccada893bd11e10e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-20 11:03:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3868bbfb9f398f87277659050acfda0f14c32b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 11:23:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f297a471993461bb7a4610907d796b0 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e91ccb0281cb46d9b46f876c237c28d6 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d54981a4904391a16ea837a4958c1a : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7d56e923b144d08bec303a1e7992bd0 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782567f7ae644251a6a1c96bf9ba19fe : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05541942454647c5b5881f196ed6e447 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa57d8daeae1436e8d9869fd3899c0eb : SUCCESS in 4m 21s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/2f704707ee45440faeb837586f1251b1 : SUCCESS in 4m 41s\n- tenks-deploy-teardown-centos7 https://zuul.opendev.org/t/openstack/build/ed9f7efd632342f392ee872850f752de : SUCCESS in 11m 36s\n- tenks-deploy-teardown-centos8 https://zuul.opendev.org/t/openstack/build/feec0db36f0c44cab8a306523c9bba70 : SUCCESS in 8m 54s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/1b088ac9648a406e9a18c7f99859366d : SUCCESS in 5m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/645a207e4b2e410db3d78e9290253238 : FAILURE in 18m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d6d98d0c0099916f7ab9f8eeeffab8492fe1ba45","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-20 11:33:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c0fc098ddce18f3c297587044e903688a2b3d5f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 12:47:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da2855b0159d45698e127116ab34af59 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede1cf1299554e9d958ac2825ca11c0e : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f7fcfb7f68f4d17ace3ab2d3588ada2 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56becf8987ef4518a80db8333b4a3723 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/576a5e67840e4f5693d6938120f1ea1a : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0d3876df1a34f9a9d99c8abdd72cb3c : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cf191c33ab94ba2aa3d602adc2b98ac : SUCCESS in 3m 41s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/a631ddac1d1d43fda65672924a24117e : SUCCESS in 4m 31s\n- tenks-deploy-teardown-centos7 https://zuul.opendev.org/t/openstack/build/7c8a9450d9fa4951899043d11efe9cff : SUCCESS in 11m 28s\n- tenks-deploy-teardown-centos8 https://zuul.opendev.org/t/openstack/build/03e7b61b7464410c80c50a1ad7d84c05 : SUCCESS in 7m 31s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/9dd2f4dde0cf4a9dab97171348031064 : SUCCESS in 6m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fefccc7476cf41c292fa7d5aa112f502 : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d401d1b7f80f22ed0a053fb6196e7e16b6a30908","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-03-20 12:54:55.000000000","message":"Patch Set 1: Code-Review+2\n\nSeems reasonable to me.","accounts_in_message":[],"_revision_number":1},{"id":"a96f12fc3673315312f0218c998bbfbbe9fecb2e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 14:52:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e727bd798d55169c1859f6a6fbd835a166f52e30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 14:52:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e294feb5312307a4955af105b2612fb8d2fd8d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:50:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb80e216673d4bdda86d42178f603e10 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e5f2cc8d60b46118d7fc9b228c06077 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f272ce3e439448684921cf079592993 : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3fc9f61310487b9663ebacd6d4b6f0 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1c2324828d64ee1994046bab07c970e : SUCCESS in 5m 18s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/e39879b2624a4ef7a4d74d82549edf22 : SUCCESS in 6m 23s\n- tenks-deploy-teardown-centos7 https://zuul.opendev.org/t/openstack/build/9b3114e4e685434da841e6b33c46a4cf : SUCCESS in 13m 18s\n- tenks-deploy-teardown-centos8 https://zuul.opendev.org/t/openstack/build/3a7c27d744814f6f92950b3e305820c2 : SUCCESS in 9m 44s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/cd7206a336a24a8fa59f0b0f59502fce : SUCCESS in 8m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/81a3e7d19bd44fc8b0c0cfdbedc5e449 : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"46de0b2d768ed85e9be2985c1d86a6e48a1d49f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:50:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a56bad40d546217a58c7bc30848ce5b21809e413","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 17:17:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a11717321047688a1bc96ed5eba671 : SUCCESS in 2m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f70ad476f3e4463ca3167ddeeb56c42f : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8df4a68fb1e7d939a7489fc059386b68651f34fa","revisions":{"8df4a68fb1e7d939a7489fc059386b68651f34fa":{"kind":"REWORK","_number":1,"created":"2020-03-20 11:03:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/714082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/82/714082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/714082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/714082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/714082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/82/714082/1"}}},"commit":{"parents":[{"commit":"bcea11dc60fce4976d3d33451565ea65681a9350","subject":"Merge \"Stop gzipping logs in get-logs.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/bcea11dc60fce4976d3d33451565ea65681a9350"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-20 10:59:30.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-20 10:59:30.000000000","tz":0},"subject":"Fix node scheduling with pipe in hostvars","message":"Fix node scheduling with pipe in hostvars\n\nIf Ansible hostvars contain a pipe (|) character, this can cause\nproblems during scheduling as Ansible fails during Jinja templating.\nThis is probably a bug in Ansible/Jinja.\n\nThe particular case where this was hit was when using screen, the\nTERMCAP environment variable gets set to something beginning with\n\u0027SC|screen|VT\u0027.\n\nThis change addresses the issue by moving the capture of hostvars inside\nthe tenks_update_state action plugin rather than evaluating them in a\nplaybook.\n\nChange-Id: Ibef91d9ef499c8741b61a170672a23f530a600bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/8df4a68fb1e7d939a7489fc059386b68651f34fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/8df4a68fb1e7d939a7489fc059386b68651f34fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
