)]}'
{"id":"openstack%2Ftenks~721177","triplet_id":"openstack%2Ftenks~stable%2F1.0~I8b550e5b5208afdf4c1aaa97ceaa5805238cb334","project":"openstack/tenks","branch":"stable/1.0","hashtags":[],"change_id":"I8b550e5b5208afdf4c1aaa97ceaa5805238cb334","subject":"Cap openstacksdk for compatibility with Python 2","status":"MERGED","created":"2020-04-20 07:11:41.000000000","updated":"2020-04-20 10:15:17.000000000","submitted":"2020-04-20 10:13:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721177-1587377611278-7130f63b","meta_rev_id":"5c6456625adc8358ad4a8a99407c5bc134a85fa4","_number":721177,"virtual_id_number":721177,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 10:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-20 07:39:09.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-20 09:52:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-20 09:52:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 09:52:49.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-20 10:13:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b3aa2390aad772c959fe0c5467641995e270af7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-20 07:11:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5f3b368df996f4f32db2e56b75c16b7b56866ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 07:36:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e648b158648a47b0aa81da899422edda : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57643d4fb8964053bb5c9c093ddcfccd : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d1503778e21468aa047830d99bccd49 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d39d6a30a9464db3b0668bc7579f7c89 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10c6e62a0d234c928a775231381a250c : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7050dbb5fd964d488f30e255b62c60d4 : SUCCESS in 6m 25s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/2337779a3209442da184e6ec460de091 : SUCCESS in 4m 36s\n- tenks-deploy-teardown-centos https://zuul.opendev.org/t/openstack/build/15b4f3ed64564434b4074d925ec12809 : FAILURE in 23m 48s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/b9bbdd572c854fd7a29d32c8b7743235 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"fc5ec4516ea496cf65c81c7e4fb86266fe992142","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-20 07:39:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b4f97fb1211d3e8bf26d1beb6488c742e00e8a19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 07:54:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b05bfbe6cb8248698d207e7c041d63db : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759d9855253b45cfb229adb56ed246c0 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3607e4ffb78c46c9b562d23395829568 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/548697e264c54310b7b12fc3b59f5b78 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cce6ba9f1914b4a85fe5acef45fc7f6 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00c982b1a254732bef20e5f154b3bde : SUCCESS in 7m 58s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/bf044c8c92f84588852d8133e9e96a07 : SUCCESS in 5m 19s\n- tenks-deploy-teardown-centos https://zuul.opendev.org/t/openstack/build/a3df9e0261a245389c82b72ef94e2711 : SUCCESS in 10m 53s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/ba673afb658142609e2a8e3d0d3019f4 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5d885113cdb50000ac6c8e81e2f1edcb5a6a9f57","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-20 09:52:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06071d4ce80fd4c03aa3d6730693de563eea9a72","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-20 09:52:49.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ca554d4a3e1242e2bb120acfd65f0686ce3127c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 09:53:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2004d29e725776eae6f99b106e846a56b6a10001","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 10:13:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68cc64b6f9164fe2916155bbd05ad115 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21fb7faceb7c41d49c846e636b6dabf0 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3893f64ae1646b39a1c518dab7acb76 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fd6a4ec5fba4c9984d546f9e14d8620 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a7e4f7fc94c439bbe857c7a9c0c1093 : SUCCESS in 7m 12s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/5503c80c2b664ff1b96f9b56b37a6203 : SUCCESS in 5m 47s\n- tenks-deploy-teardown-centos https://zuul.opendev.org/t/openstack/build/bb5afc44a1b04eb09d26b5ff1374f9cf : SUCCESS in 11m 03s\n- tenks-deploy-teardown-ubuntu https://zuul.opendev.org/t/openstack/build/e3ffe8964eda44cfa961aec0534f401b : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"9fc204787fc306333bf8e47c4b26f581dc51a4b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 10:13:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5c6456625adc8358ad4a8a99407c5bc134a85fa4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 10:15:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5501f5368c4e35b55f521a678ffc94 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9bf85a5e9ecfcf26eeb5b56c4d8a629943d9b116","revisions":{"9bf85a5e9ecfcf26eeb5b56c4d8a629943d9b116":{"kind":"REWORK","_number":1,"created":"2020-04-20 07:11:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/721177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/77/721177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/77/721177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/77/721177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/77/721177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/77/721177/1"}}},"commit":{"parents":[{"commit":"90224cf33101ce5f35ecee1483215b8d6c150b09","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/90224cf33101ce5f35ecee1483215b8d6c150b09"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-20 07:11:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-20 07:11:20.000000000","tz":120},"subject":"Cap openstacksdk for compatibility with Python 2","message":"Cap openstacksdk for compatibility with Python 2\n\nChange-Id: I8b550e5b5208afdf4c1aaa97ceaa5805238cb334\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/9bf85a5e9ecfcf26eeb5b56c4d8a629943d9b116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/9bf85a5e9ecfcf26eeb5b56c4d8a629943d9b116"}]},"branch":"refs/heads/stable/1.0"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
