)]}'
{"id":"openstack%2Ftenks~736249","triplet_id":"openstack%2Ftenks~master~I52350a54e58042f390eb2ad84b220194ca2ef4b9","project":"openstack/tenks","branch":"master","hashtags":[],"change_id":"I52350a54e58042f390eb2ad84b220194ca2ef4b9","subject":"Install python setuptools \u0026 virtualenv","status":"MERGED","created":"2020-06-17 14:54:11.000000000","updated":"2020-06-19 05:54:15.000000000","submitted":"2020-06-19 05:52:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736249-1592545963950-0900ee8c","meta_rev_id":"d4e9ee0aa9f17a783ca2a878ba46ff9952820684","_number":736249,"virtual_id_number":736249,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 05:52:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-06-19 04:48:00.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2020-06-18 14:56:41.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2020-06-19 04:48:00.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 14:56:41.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-06-19 04:48:00.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-06-19 05:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61e75fbc95e03592c112ec22ea404c6ccef766ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 14:54:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80346f6ed402788bce6a4429cd008f91a6cbe592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:14:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca03f7225d214efb8f7bc87376e632e9 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e0a72a4890a4c8fb22f80c0ba731144 : SUCCESS in 10m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c50be202542441a6883b7c1d71d11376 : SUCCESS in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df6547b4a8fa4853ab6760e78c4aaf71 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c92e3e370e945ff88aaf2f2f29444b6 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e024b436e54d4ea6a8eb1b0d8ea318a7 : SUCCESS in 11m 09s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/b3f307b1b55041b8afaf5df0cefa9e29 : SUCCESS in 10m 53s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/ba8d718c31424695a5561d4ae651d3e2 : RETRY_LIMIT in 5m 55s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/1960e132dcc94d35b578306f343643bf : RETRY_LIMIT in 5m 56s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/5b10a22f874e4fff8fc544812d42866f : RETRY_LIMIT in 5m 46s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/a7027c51786b4689bb001379c648f45c : RETRY_LIMIT in 4m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/41b834f1893f4d6d9a36bbf97d5154dd : SUCCESS in 1h 15m 04s","accounts_in_message":[],"_revision_number":1},{"id":"dd1cef4a3720a6a466b35778da7f988008886644","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 10:01:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"058ea87fcc331f46c8d8bf593f45b42a95bd80aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 11:12:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96dfbf3bdae44ba5b9eb1e8e24b65d04 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4306806690f40b8b3b8f28ce24e7810 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/784d672aa339480d96b2f7a7dcf6912d : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c40a2ca219904fa1a451b971f4f3a6c3 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74f37720fc5b4007bc061f34e5e460b8 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/870204c0ce2040c985b521f20f048d1f : SUCCESS in 6m 52s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/7542934eed394447bc3fb169671141b2 : SUCCESS in 4m 31s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/ef688ee4b96e40ce9ee4fc678e7f5087 : SUCCESS in 7m 43s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/1e33d71d946f44d5ae2efd1bdccaaece : SUCCESS in 7m 59s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/d0914beb67f34658b2de0c76061370a8 : SUCCESS in 9m 04s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/18525028e16d45418a909257f682ac1b : SUCCESS in 8m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/198774aa4be64d1cb0d6c8b621ed1e47 : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b4451eeee1b3aadec7cdcc4f7cb4b70e095ee6d8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-18 14:56:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ddc144786d383a22fc4e85c87fe3f19a7afb0a2a","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-06-19 04:48:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThank you Mark.","accounts_in_message":[],"_revision_number":2},{"id":"6880522448de0413cf5795c2978186cdfab82cdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:48:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1a3cf50f16c1abedddb190d2164849f984bf277d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 05:52:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"988dccb27264e6370176b7d3cc07a7c3f8be20a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 05:52:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfce53068974691b26ebad5839eea36 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2caf0610b2a64271a3b7b7ae5e4980e6 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8d3718c7ffc4726853eebb4a41bf224 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d38629f5061448b6a9fa15d891a5d29e : SUCCESS in 5m 51s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/50e4cdbb8a83487a8d370ce20b2a4e8e : SUCCESS in 5m 17s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/cebd2ee83167423db752b76b4665a9c5 : SUCCESS in 6m 52s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/ba4c38976abf4324afcd76d85660788f : SUCCESS in 7m 44s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/c6318650f1cd424fa1d001bbdb38ebf2 : SUCCESS in 7m 57s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/9f51de29c166451992becd05a6b0b777 : SUCCESS in 5m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f8c4ddcef9d246c2bcf4e4b0634a29af : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d4e9ee0aa9f17a783ca2a878ba46ff9952820684","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 05:54:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb3c476fd56945c89d3458d461f69bdc : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d2cb3e2e9348da30e93a80d6ae952e697bee2068","revisions":{"1a56bf0f59c00af03f3fc1c22bfa4fe8d1e71035":{"kind":"REWORK","_number":1,"created":"2020-06-17 14:54:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/736249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/49/736249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/49/736249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/49/736249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/49/736249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/49/736249/1"}}},"commit":{"parents":[{"commit":"288d663912009ffefbb89f207b81e0ac8649016a","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/288d663912009ffefbb89f207b81e0ac8649016a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-17 14:54:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-17 14:54:05.000000000","tz":60},"subject":"WIP: test CI","message":"WIP: test CI\n\nChange-Id: I52350a54e58042f390eb2ad84b220194ca2ef4b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/1a56bf0f59c00af03f3fc1c22bfa4fe8d1e71035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/1a56bf0f59c00af03f3fc1c22bfa4fe8d1e71035"}]},"branch":"refs/heads/master"},"d2cb3e2e9348da30e93a80d6ae952e697bee2068":{"kind":"REWORK","_number":2,"created":"2020-06-18 10:01:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/736249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/49/736249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/49/736249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/49/736249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/49/736249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/49/736249/2"}}},"commit":{"parents":[{"commit":"288d663912009ffefbb89f207b81e0ac8649016a","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/288d663912009ffefbb89f207b81e0ac8649016a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-17 14:54:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 10:01:47.000000000","tz":0},"subject":"Install python setuptools \u0026 virtualenv","message":"Install python setuptools \u0026 virtualenv\n\nThese are no longer installed in the nodepool image provided by infra.\n\nChange-Id: I52350a54e58042f390eb2ad84b220194ca2ef4b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/d2cb3e2e9348da30e93a80d6ae952e697bee2068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/d2cb3e2e9348da30e93a80d6ae952e697bee2068"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
