)]}'
{"id":"openstack%2Ftenks~748818","triplet_id":"openstack%2Ftenks~master~Id301c0c6ffacea81e4771219beb6cbf8462c1dc4","project":"openstack/tenks","branch":"master","hashtags":[],"change_id":"Id301c0c6ffacea81e4771219beb6cbf8462c1dc4","subject":"Use importlib to take place of imp module","status":"MERGED","created":"2020-08-29 17:05:11.000000000","updated":"2020-09-14 14:08:57.000000000","submitted":"2020-09-14 14:02:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748818-1600092133509-7e3b2230","meta_rev_id":"2b6e088890f9b198daeaeb67cb569309ba1a7a80","_number":748818,"virtual_id_number":748818,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 14:02:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-09-14 12:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-09-14 12:56:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 08:00:36.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-09-14 12:56:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-14 14:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"017cd44976b90b87ea5ea5b2bf3601098dd5bfab","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-08-29 17:05:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"140ccbc97f4289399e27fc28e4b99d748ee8fe50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 18:10:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d6dda1a55ca48999da8da42cb2b6b6c : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30af4e04d0a4f8ab64c833f0eb0148b : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a50bffc53db4b499dd10627c63dcb48 : FAILURE in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6040a21e2a814e25b51832b1eafdb9e9 : FAILURE in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/396ac87f6dcf47a2934d5248a65ff10a : FAILURE in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0187ef8bc24d2a8bea90325a5d06f1 : SUCCESS in 4m 59s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/0e04db0f227442bfa761c0b7422f220e : FAILURE in 2m 59s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/9e3a222bd79f45a3aac35c25b715a744 : SUCCESS in 6m 11s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/dc2d9bba8b604f0190d728dc63edf41e : SUCCESS in 4m 31s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/22f939dafae34cbfbb867f669f63cbd7 : SUCCESS in 5m 38s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/abd5cc237bbf477490d003a27139bf2f : SUCCESS in 4m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c94411d7380a416db9b0119515b86d71 : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b2967c2be92c4e1021eb83a27018023d732606c4","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-08-30 11:27:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03f8477d67bae18bf6c0cbbed2dfd9eacf1449bf","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-08-30 11:38:36.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9fe55e29eaceedda1247c66d6e0b4d604433fbe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 12:44:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35b4b5e677c44c89a2b9326206841967 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524c9d5e6064459086102057695906cc : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b461de038949a594527a99ccac907d : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b2e379788a945d6b622a1757039c522 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70dc28e22be849ad808008b96f02b743 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f4e4181e3c44d5941141fe53d59024 : SUCCESS in 8m 02s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/b2ff22d61b1e4056a1457c01fa7bc5cd : FAILURE in 4m 09s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/292f613d4d464d48ba094285dccfdc6d : SUCCESS in 6m 23s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/bca03ace58564a8caedb5665686ace55 : SUCCESS in 5m 06s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/ba71bec638b94f0995a5a23d0f94cc7d : SUCCESS in 6m 08s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/752c4b0502df405da766cdbc60dcc05a : SUCCESS in 4m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5d94d63f38ac4d1085baeae2b010a95a : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":3},{"id":"654e0334eddf758eabfb1ec4688643517ca0ea8d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-09-01 08:00:36.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"609a65d756cdd53384dc1fa7b10f80e7e59ae7fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-01 08:06:27.000000000","message":"Patch Set 3:\n\nimportlib looks like a step backwards to me if we need an extra function.","accounts_in_message":[],"_revision_number":3},{"id":"d96b0f06f9db716cfd1c8032ba058b1472640259","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-09-09 01:10:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2049ec477f25c86369949d15f955a7c1c846f762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 01:46:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2013f50f42e544789d6753fc443acc55 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d12bd950c804446bbac1bc385166e8df : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1cd38559539406ab21312606299e0cf : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/688d364b88c94cf587c1230859822d21 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387af3be8f3446eaab6f45df8a287fc4 : SUCCESS in 6m 56s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/5409311706f74c83853beb2bc03bbc01 : SUCCESS in 3m 16s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/ff1ec0fe161b4838acf21d1afe1981e9 : SUCCESS in 6m 54s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/1083e10cb24c41bab15c5bd6d080f479 : SUCCESS in 5m 15s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/32f30d8593464aac9fd08cd2ca03307e : SUCCESS in 6m 33s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/01cde610beb34258b7133405df3de4f0 : SUCCESS in 5m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f31e201994ab49fc9c949a7d15ce5432 : FAILURE in 33m 39s","accounts_in_message":[],"_revision_number":4},{"id":"fc5f67df14630285e1c973b10d6bb48095449997","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 11:17:23.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ef2d9926a53c54768d0135189f43ac15f8a6700e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:24:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6b0c482be5c4b44bc6f78be17f4560b : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9853c829face4253b5312c39ab63fe02 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13ae18dd97474ebb94652fa04280f334 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e7ebd4fd7df4fc3b50c20d6e3d9a838 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd98ce76c4a64307b0da49c9f7123d2e : SUCCESS in 16m 46s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/a41b7e2b116a460db732bef086ae0023 : SUCCESS in 3m 16s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/fce616b5229e49d9a0df9c6f9ecf2bec : SUCCESS in 6m 30s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/4346fc95ad9c475f9b10cb99ee906e8e : SUCCESS in 5m 26s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/8eabbe0e5f3442c7bd5daf78277aa864 : SUCCESS in 7m 38s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/b54bccbc8b0440fc90f400eea3079e30 : SUCCESS in 5m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bde2ea72c4ea490f9bdda53ad5e45631 : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":4},{"id":"1f55fc25516e3ca46a944246876940153f2fe257","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-09-14 12:56:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":4},{"id":"7ed59da621493f72861e4e0cfc9c22257f385e10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:57:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"61c3759ea78936963a647509562a6f43495a29a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 14:02:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e26a084fa7e45e29d50ea1129155182 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14f6ba67bc064b469b7deac4a44f627a : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea959f9aa21b46b18e99d410558751e4 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8cd7298507416a8924aba3fac3e3c5 : SUCCESS in 15m 34s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/c1c7da5df5b34c3aa99e42c800f1d097 : SUCCESS in 3m 35s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/0f1889b80390418299863726065dabfa : SUCCESS in 6m 32s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/265ea3b7397642c28466621c1cacfb58 : SUCCESS in 6m 41s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/237362c7413c4a0ea1b01bf8f72e1267 : SUCCESS in 6m 45s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/636080c5310f421b963a22618f8008ef : SUCCESS in 5m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9850ec109e6349a3a7c6345dbf79b0be : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":4},{"id":"8b1359b3c9ebdf0d0662328a141af7c1999ff715","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 14:02:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"2b6e088890f9b198daeaeb67cb569309ba1a7a80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 14:08:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31f491593cb41b685418bce6b1ea131 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9ed29ce4e05d33c73d6f15e1bb7a8fcca803a263","revisions":{"308812b11487800bc27ce375c2cba815f238ffc7":{"kind":"REWORK","_number":1,"created":"2020-08-29 17:05:11.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/18/748818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/18/748818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/18/748818/1"}}},"commit":{"parents":[{"commit":"6f1e4dc8fcb5cfa0f31458c77a476f21790d8682","subject":"Merge \"Trust receive filters of baremetal nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/6f1e4dc8fcb5cfa0f31458c77a476f21790d8682"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-29 17:04:44.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-29 17:04:44.000000000","tz":480},"subject":"Use importlib to take place of im module","message":"Use importlib to take place of im module\n\nThe imp module is deprecated[1] since version 3.4, use importlib to\ninstead\n\n[1]: https://docs.python.org/3/library/imp.html#imp.reload\n\nChange-Id: Id301c0c6ffacea81e4771219beb6cbf8462c1dc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/308812b11487800bc27ce375c2cba815f238ffc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/308812b11487800bc27ce375c2cba815f238ffc7"}]},"branch":"refs/heads/master"},"18f25b2281ff47392e1b197a1e6efd201eb06e0c":{"kind":"REWORK","_number":2,"created":"2020-08-30 11:27:24.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/18/748818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/18/748818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/18/748818/2"}}},"commit":{"parents":[{"commit":"d17101a437722ed994462737ed699d067ae34f15","subject":"Set min version of tox to 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/d17101a437722ed994462737ed699d067ae34f15"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-29 17:04:44.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-30 11:27:23.000000000","tz":480},"subject":"Use importlib to take place of imp module","message":"Use importlib to take place of imp module\n\nThe imp module is deprecated[1] since version 3.4, use importlib to\ninstead\n\n[1]: https://docs.python.org/3/library/imp.html#imp.reload\n\nChange-Id: Id301c0c6ffacea81e4771219beb6cbf8462c1dc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/18f25b2281ff47392e1b197a1e6efd201eb06e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/18f25b2281ff47392e1b197a1e6efd201eb06e0c"}]},"branch":"refs/heads/master"},"f057ea80332f3b8ff9a256dfaf4ef011b321eb77":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-08-30 11:38:36.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/18/748818/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/18/748818/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/18/748818/3"}}},"commit":{"parents":[{"commit":"d17101a437722ed994462737ed699d067ae34f15","subject":"Set min version of tox to 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/d17101a437722ed994462737ed699d067ae34f15"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-29 17:04:44.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-30 11:38:25.000000000","tz":480},"subject":"Use importlib to take place of imp module","message":"Use importlib to take place of imp module\n\nThe imp module is deprecated[1] since version 3.4, use importlib to\ninstead\n\n[1]: https://docs.python.org/3/library/imp.html\n\nChange-Id: Id301c0c6ffacea81e4771219beb6cbf8462c1dc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/f057ea80332f3b8ff9a256dfaf4ef011b321eb77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/f057ea80332f3b8ff9a256dfaf4ef011b321eb77"}]},"branch":"refs/heads/master"},"9ed29ce4e05d33c73d6f15e1bb7a8fcca803a263":{"kind":"REWORK","_number":4,"created":"2020-09-09 01:10:58.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/18/748818/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/18/748818/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/18/748818/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/18/748818/4"}}},"commit":{"parents":[{"commit":"4557c983caa481fbae743545ba0268fb6cb9f08c","subject":"Use victoria tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/4557c983caa481fbae743545ba0268fb6cb9f08c"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-29 17:04:44.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-09-09 01:10:52.000000000","tz":480},"subject":"Use importlib to take place of imp module","message":"Use importlib to take place of imp module\n\nThe imp module is deprecated[1] since version 3.4, use importlib to\ninstead\n\n[1]: https://docs.python.org/3/library/imp.html\n\nChange-Id: Id301c0c6ffacea81e4771219beb6cbf8462c1dc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/9ed29ce4e05d33c73d6f15e1bb7a8fcca803a263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/9ed29ce4e05d33c73d6f15e1bb7a8fcca803a263"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
