)]}'
{"id":"openstack%2Ftenks~774474","triplet_id":"openstack%2Ftenks~master~Ibd61e090611b3b7a7e0670c854362b512454bf3c","project":"openstack/tenks","branch":"master","hashtags":[],"change_id":"Ibd61e090611b3b7a7e0670c854362b512454bf3c","subject":"Fix virtualbmc installation after release of cryptography 3.4","status":"MERGED","created":"2021-02-08 14:55:10.000000000","updated":"2021-02-08 17:06:22.000000000","submitted":"2021-02-08 17:04:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774474","meta_rev_id":"3fc8ac15fb7883b59c5c60a70459a9a3cf3ca46d","_number":774474,"virtual_id_number":774474,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 17:04:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2021-02-08 14:58:37.000000000","_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2021-02-08 15:18:48.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2021-02-08 15:18:48.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-08 14:58:37.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-02-08 15:18:48.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-02-08 15:56:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15a8c2a475cc4038a7460da93a581f90c2e74e54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-08 14:55:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a46bdfe374629fcd06b5eafae16b30a5fb0b1909","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-02-08 14:58:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7564f2179adb9ead30bc9fc6b7453f00bf6a0678","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-02-08 15:18:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b29b9bf77fcc2cfa036740d520a0891acf6562dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:56:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62db006141f64eec8aace27be914a56d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218fde2d8d70441fa51f0b5934818039 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/278d992246bd401e9d2038812cc782d6 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4a8b15c77804b66bc549da57b9dcf4d : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b81348eb0844c99b148a0ce5ab6f11 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4cdd34e3f244f5393fdb896ed0fcc33 : SUCCESS in 3m 42s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/7320f6f8ec074a8f80283dee22444930 : SUCCESS in 4m 30s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/0c27003987e046bbb2718f2c8053da11 : SUCCESS in 6m 36s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/5f7dfbb070204a25b19c88dca7be5877 : SUCCESS in 5m 16s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/44192e5191af472285d3ee90cd1a226e : SUCCESS in 7m 21s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/08e56de5d7be4e1ca7b48aa204c1d20c : SUCCESS in 5m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/66dc8b6e52ae40149699fb6749011963 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":1},{"id":"eaa8fd66a10888300241a3200cac99d1ce796c4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 15:57:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38f8379010e2695d3d020b6ef88d785a5830e311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:04:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c25997cd782249248046613f8bb5775c : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/120c842e6c1d470289494fc05fa6d0bb : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f861cf489a1487c9c0ceac693b3625e : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfdb531f228b4273bbd15b4c4993bf7e : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd62df1b130141b992c909f1e74bcf91 : SUCCESS in 4m 18s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/511e6bbe1bc143a5bacc7a3ac99f9469 : SUCCESS in 4m 32s\n- tenks-deploy-teardown-ovs-centos8 https://zuul.opendev.org/t/openstack/build/de31ac2538cf4cb8aeeda2beee5935f4 : SUCCESS in 8m 25s\n- tenks-deploy-teardown-ovs-ubuntu https://zuul.opendev.org/t/openstack/build/5c499da2300949589191ea4f7a088194 : SUCCESS in 7m 09s\n- tenks-deploy-teardown-linuxbridge-centos8 https://zuul.opendev.org/t/openstack/build/d7733a76d0274e92858f27ac065a97c1 : SUCCESS in 8m 05s\n- tenks-deploy-teardown-linuxbridge-ubuntu https://zuul.opendev.org/t/openstack/build/97f32d50f5e54b5b8f56242ff15fef62 : SUCCESS in 6m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b01441f790444c89b5c7fe6cb72a3706 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":1},{"id":"754168544d097a3ff31afc870ebfdb4296a22fb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:04:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3fc8ac15fb7883b59c5c60a70459a9a3cf3ca46d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:06:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9e7f069d8a040cf87223a86b8904aab : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/657047f995f940009fe2112a8963eb77 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f34bb772a018d3ee3a2872358542ad51c22fc06e","revisions":{"f34bb772a018d3ee3a2872358542ad51c22fc06e":{"kind":"REWORK","_number":1,"created":"2021-02-08 14:55:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/74/774474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/74/774474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/74/774474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/74/774474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/74/774474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/74/774474/1"}}},"commit":{"parents":[{"commit":"0719f06a6bcf920a575161518b9603150ac81ee2","subject":"Merge \"Fix invalid argument formatting in the exception message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/0719f06a6bcf920a575161518b9603150ac81ee2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-08 14:54:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-08 14:54:17.000000000","tz":60},"subject":"Fix virtualbmc installation after release of cryptography 3.4","message":"Fix virtualbmc installation after release of cryptography 3.4\n\nInstalling virtualbmc system-wide on CentOS 8 fails with:\n\n    ModuleNotFoundError: No module named \u0027setuptools_rust\u0027\n\nThis error appeared following the release of cryptography 3.4, which now\nincludes Rust code. It can be installed without Rust using a Python\nwheel, but only with more recent pip than version 9.0.3 available as RPM\non CentOS 8. The cryptography bug report [1] recommends pip\u003e\u003d19.1.1.\n\nAlso ignore PyYAML when installing system-wide to avoid conflicts with\nan existing RPM package installation.\n\n[1] https://github.com/pyca/cryptography/issues/5753\n\nChange-Id: Ibd61e090611b3b7a7e0670c854362b512454bf3c\nStory: 2008607\nTask: 41788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/f34bb772a018d3ee3a2872358542ad51c22fc06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/f34bb772a018d3ee3a2872358542ad51c22fc06e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
