)]}'
{"id":"openstack%2Ftenks~830182","triplet_id":"openstack%2Ftenks~master~I4831769909c341c72bb178daf8df3309773a56ad","project":"openstack/tenks","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-28 12:59:18.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-02-28 12:59:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4831769909c341c72bb178daf8df3309773a56ad","subject":"Use ansible_facts to reference facts","status":"MERGED","created":"2022-02-21 14:44:23.000000000","updated":"2022-02-28 13:00:48.000000000","submitted":"2022-02-28 12:59:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"830182","meta_rev_id":"2aaee69f4969d9bd104d4b8ac6869594c90e0400","_number":830182,"virtual_id_number":830182,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-28 12:59:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2022-02-28 11:47:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2022-02-28 13:00:48.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-24 09:46:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2022-02-28 11:47:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-21 16:12:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-21 16:42:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-02-21 18:17:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-02-24 09:43:47.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-02-24 09:46:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-28 11:47:16.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"20ccf83d9cac8d1dc3280dd1b67b290cb7718eed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 14:44:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0ca0312eadf9c3284adce37e0f9481cd2a61061","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 16:12:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18190e50fd3b4e6395cf11daf30950e3 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd10e2b50744f059be5e0d9915a19ba : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c10409c9084e4abef2e5f2ee829bce : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ce94914fdd34ab2bec4c6b3eb2b1260 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e5de47dc0e864cca8155d02f4ef03d71 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7753663bae24493e8b320c86e859a805 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ca3e422074e42e0aa62cd79d7f66b37 : SUCCESS in 7m 47s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/5aec4b01b4bc4b9b92d49c3dcb4ac3ef : FAILURE in 5m 19s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/bd4f083eee344fa5898ef1d2236821d9 : RETRY_LIMIT in 6m 33s\n- tenks-deploy-teardown-ovs-bios-ubuntu https://zuul.opendev.org/t/openstack/build/3fa65dd097034ff58ea4155c7dcedcd7 : RETRY_LIMIT in 4m 00s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/b91a680024d24eb1b206ae5a089531a0 : RETRY_LIMIT in 6m 55s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu https://zuul.opendev.org/t/openstack/build/eac78a38173c47e99491706b4996030f : RETRY_LIMIT in 4m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4f7a49c9ec80435ebf788117674a995f : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e94ee803d719d855abae2ace632287d73c060813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 16:27:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f5a95f47f0c1597df847c5a750d86cac4d644345","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-21 16:42:28.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0f274b2b22fcff0be1aeee8592566956c5bff820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 17:52:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aed868b3d2146ef89039a44175d357a : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e31dc3035740039fc39bba6c87aab4 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394d7635e1a349d39391260e15398e56 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc0221ece1b344ee98b8f962dd42c682 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7140110a89454d8287371df006608f41 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1aff2699019b48cb981052c8c4522182 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fae41d236d64fabaaa86c624daef55f : SUCCESS in 4m 24s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/29964d3a4b514c91928a827ba14e65e4 : SUCCESS in 3m 30s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/ee2d40e80cec4506a9fb106eff637fa6 : FAILURE in 4m 16s\n- tenks-deploy-teardown-ovs-bios-ubuntu https://zuul.opendev.org/t/openstack/build/c4719edae2b746aca566b7dc4c4ccd8e : FAILURE in 3m 19s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/91d6937cae264d2cbe21ce8af8335b95 : SUCCESS in 9m 47s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu https://zuul.opendev.org/t/openstack/build/3464196f6e9c4e6a8af1391d2eb7d51c : SUCCESS in 6m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8da938900dba46d084ea0028b2385eba : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":2},{"id":"38cd2bf479dc96e18296250b541bc7046625c5d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-21 18:05:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cd38de997bdca4e3097149ff02a63b47914c8cdc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-21 18:17:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4d634aa21ef3dce2838fc47618499fc460418d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-21 18:41:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3292acc646ec0a25d81c6dbf8099d5d4b9161c87","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-21 18:44:13.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"77f4f68afc17bfb8b79997f05c58016f581013ac","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-21 18:45:02.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"625a092bbb1d66bffd1a7d4729123ba6e210b7f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-21 19:57:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2069ab924194a19a101c64ae29af2f5 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137a8ac1d8c348a184ecfde229b6485b : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e80401ee9ec4cfb91053d834e1bfbdc : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb726cfa156a41a99fb8db7041c7539d : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/535566b548a142ca88a01c793ae2e1b8 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c89d25157dba4652bfde7b484b39e9af : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/717d3770edb045eb926141fbafa4d574 : SUCCESS in 5m 54s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/5effcf992f2749ecbd24e882ee8440bf : SUCCESS in 5m 26s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/3785f28199a8464594aca3dfe096f429 : SUCCESS in 12m 50s\n- tenks-deploy-teardown-ovs-bios-ubuntu https://zuul.opendev.org/t/openstack/build/c691742d5f8741a0b3d7183eebacc715 : SUCCESS in 9m 03s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/274068e329a2430cb4b87c18d15cce94 : SUCCESS in 10m 52s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu https://zuul.opendev.org/t/openstack/build/63a83adc788d49fa806bc141eab70de3 : SUCCESS in 9m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/446aff3a3e1d4b7ea4000f98f687ce4b : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":4},{"id":"c1db37236e40f15f039c9b12817093a28a757a19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-02-23 18:17:43.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"0ecb50e50f50e7af3951655c55d299c9bf308406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 19:28:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9208d448f2f34bb3a6636c2f973cde92 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9e4ca4398da4688b31099626d86d74e : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a29badbf8354f7e8c51a3c960fd349c : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f43218f3f69848e19e3746fbac5b8ffc : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/758fe1027442417bbc3b69893e4bd3f1 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feef14bc850a45da8aa768ae1d4e4534 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f03b46d4e0f046788740bc9b26bd9d61 : SUCCESS in 3m 59s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/03fff0b01d4b4ecf9e7d3b0ecf01970d : SUCCESS in 4m 04s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/37c43b1ad1b34e748401849a4c5760d0 : SUCCESS in 10m 19s\n- tenks-deploy-teardown-ovs-bios-ubuntu https://zuul.opendev.org/t/openstack/build/6efc9b0612e942f084d90500050d17a6 : SUCCESS in 7m 15s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/8deb380b4d034918aaeb3782e8680c25 : SUCCESS in 11m 07s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu https://zuul.opendev.org/t/openstack/build/f00134aadd624a738132a7b5c5420813 : SUCCESS in 7m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5bb449307c724011af2e1cf161e43a87 : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":5},{"id":"6767cf3a251ca14be782ddc532d1f15dd1c58d9d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-24 09:43:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0e757f53c427753cd33eab832f0746dfcae4b620","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-24 09:44:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"043e1a49b41a7780c258760daac2c5351b04fd98","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-24 09:46:02.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"7840c9a591601b86094a3eadb764b47a1bcee195","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-02-28 11:47:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"731fb3c37a17fbe0b4b30871a486f33f46236e96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 11:48:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"51fb2bc24456e103b84a37a08693c0b94937394b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 12:59:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20418c605f98419d940a1795c84eebe5 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbfac746390c4c4b9a793e785c38e08d : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb8cd6b36a754cbfb3c5f7accd7aaf7d : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c5bfb0a324b49548eee646a9fb8c459 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c118fa12feef48ee9e3edc0147f2f8ae : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62abafb44675406e994e5e624cccd785 : SUCCESS in 5m 11s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/45bfd5fd2e784ae191b70dac76cdab2d : SUCCESS in 5m 04s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/9f5ed32a0a6442a79b859e33b51b847c : SUCCESS in 10m 28s\n- tenks-deploy-teardown-ovs-bios-ubuntu https://zuul.opendev.org/t/openstack/build/e7955603e2e440748a96a6dcfa89cbe0 : SUCCESS in 8m 57s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/1c7fcec161dd464da55f4cbcad52eb5b : SUCCESS in 9m 50s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu https://zuul.opendev.org/t/openstack/build/470f6e973cb04e389bfdc5244b034eee : SUCCESS in 8m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4dad414916514c0ab509ceaab7254023 : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":5},{"id":"4c246a8dacc460ccaf2b091b04ed1fc17a40e839","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 12:59:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ccd39cb45eb4d941a939b00613f61014d541579c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 13:00:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cfc9c4fed3546d3acf308b04bb37f6e : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9111328e8a715d2dd5ff2a1e95d996d538300e82","revisions":{"a8603743c567deb4aa21dae051ca88c07f886ba0":{"kind":"REWORK","_number":1,"created":"2022-02-21 14:44:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/830182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/82/830182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/82/830182/1"}}},"commit":{"parents":[{"commit":"fc1b42c60d19e148a746bc171a71ec13dd39ecfd","subject":"Support UEFI boot mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/fc1b42c60d19e148a746bc171a71ec13dd39ecfd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:43:28.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:43:29.000000000","tz":0},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Tenks\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nChange-Id: I4831769909c341c72bb178daf8df3309773a56ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/a8603743c567deb4aa21dae051ca88c07f886ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/a8603743c567deb4aa21dae051ca88c07f886ba0"}]},"branch":"refs/heads/master"},"c2eabf55c490d9c6a0121de5aedbf44aa258d651":{"kind":"REWORK","_number":2,"created":"2022-02-21 16:27:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/830182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/82/830182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/82/830182/2"}}},"commit":{"parents":[{"commit":"fc1b42c60d19e148a746bc171a71ec13dd39ecfd","subject":"Support UEFI boot mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/fc1b42c60d19e148a746bc171a71ec13dd39ecfd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:43:28.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 16:26:58.000000000","tz":0},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Tenks\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nChange-Id: I4831769909c341c72bb178daf8df3309773a56ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c2eabf55c490d9c6a0121de5aedbf44aa258d651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c2eabf55c490d9c6a0121de5aedbf44aa258d651"}]},"branch":"refs/heads/master"},"1e718cdf044b77a59e1905274a64331b66c2a35f":{"kind":"REWORK","_number":3,"created":"2022-02-21 18:05:29.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/82/830182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/82/830182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/82/830182/3"}}},"commit":{"parents":[{"commit":"fc1b42c60d19e148a746bc171a71ec13dd39ecfd","subject":"Support UEFI boot mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/fc1b42c60d19e148a746bc171a71ec13dd39ecfd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:43:28.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-21 18:05:12.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Tenks\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nChange-Id: I4831769909c341c72bb178daf8df3309773a56ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/1e718cdf044b77a59e1905274a64331b66c2a35f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/1e718cdf044b77a59e1905274a64331b66c2a35f"}]},"branch":"refs/heads/master"},"c00de2755ec99276536aad174046f3cd3305a6b2":{"kind":"REWORK","_number":4,"created":"2022-02-21 18:41:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/82/830182/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/82/830182/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/82/830182/4"}}},"commit":{"parents":[{"commit":"fc1b42c60d19e148a746bc171a71ec13dd39ecfd","subject":"Support UEFI boot mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/fc1b42c60d19e148a746bc171a71ec13dd39ecfd"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:43:28.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-21 18:41:00.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Tenks\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nChange-Id: I4831769909c341c72bb178daf8df3309773a56ad\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c00de2755ec99276536aad174046f3cd3305a6b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c00de2755ec99276536aad174046f3cd3305a6b2"}]},"branch":"refs/heads/master"},"9111328e8a715d2dd5ff2a1e95d996d538300e82":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-02-23 18:17:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/82/830182/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/82/830182/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/82/830182/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/82/830182/5"}}},"commit":{"parents":[{"commit":"d25067dd1af09e617f733d8e77ec1b0489d1c1ce","subject":"[CI] Optimise irrelevant files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/d25067dd1af09e617f733d8e77ec1b0489d1c1ce"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-21 14:43:28.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-23 18:17:43.000000000","tz":0},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Tenks\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nChange-Id: I4831769909c341c72bb178daf8df3309773a56ad\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/9111328e8a715d2dd5ff2a1e95d996d538300e82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/9111328e8a715d2dd5ff2a1e95d996d538300e82"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
