)]}'
{"id":"openstack%2Ftenks~856296","triplet_id":"openstack%2Ftenks~master~Icd60aca084323cede944ee261d2cf5941b320e4b","project":"openstack/tenks","branch":"master","topic":"venv","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-09-09 19:21:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-09-09 20:30:13.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-09-09 20:30:13.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-09-09 20:30:13.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-09-12 08:49:23.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-09-09 20:30:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd60aca084323cede944ee261d2cf5941b320e4b","subject":"Use venv instead of virtualenv","status":"MERGED","created":"2022-09-07 15:48:26.000000000","updated":"2022-09-12 08:49:23.000000000","submitted":"2022-09-09 20:30:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856296-venv","meta_rev_id":"08742b2f67bbe00c9b6e5c18bdda449564ff68c4","_number":856296,"virtual_id_number":856296,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 20:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2022-09-09 19:21:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2022-09-09 19:21:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 15:55:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-09-07 15:55:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-09-07 18:10:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-09 08:25:04.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-09-09 08:29:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-09-09 08:36:19.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2022-09-09 19:21:39.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"cffb4b27b7d0f5b17dc624a2c5445244ad04c979","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 15:48:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83ad23cb0fa0a7ee95bd8692d1412cab532d09f4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 15:55:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93f547f0cc3d7554ffde4098c17c160a21d89060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 15:57:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"455182a9f745b41b396a19bbd7746b244a5e2eb0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 15:59:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dff65bde786cfd402037aafdb9e17ec700f37162","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 16:11:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ff2ebf5fbacc049574fae45eaacabad534223812","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-07 16:30:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"11d1cca6058fcb66548bb127ff58739eea6c2a83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 16:34:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"602c4001383046a5d0ce0781c17a2bec26308210","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 16:35:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e4ef776190bdfca5aee2d031286bebbe05cc36b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 16:55:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bee084851fb35c7643e9a793575cf87a3c42cea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 16:58:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"315262bc65ae7a5e6a377c541f9483fa834ad2c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 17:17:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"77a2610b0a34c4e3ce20a5b2d091f26fbcab44bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 18:10:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3b76dd4dead4133b6a269fd71c9a77e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c91f9ce049db4c81868398251f8f9873 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b3fc3a658e48eaa38e5bd1ad65036f : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0cfdd67944f4cbd8e3741568f31e868 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a783214a782542e6b5631ed380d94ccf : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daf9b33868b648c1bd8ce37fcfa32971 : SUCCESS in 8m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5311dc3a59f34f5faf995c176809805d : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8194350b4c384c21a95aac63855c82ea : SUCCESS in 5m 28s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/95a0327aba314d0dae73115f9414bded : SUCCESS in 5m 10s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/ae726609a23d45edbe54b7635bffb2fa : FAILURE in 8m 49s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/87cf50589cbb4a2d8aaddce91ea1c9bc : SUCCESS in 11m 28s\n- tenks-deploy-teardown-ovs-bios-debian https://zuul.opendev.org/t/openstack/build/6238c0c30ad44bcdbc4cb32d7b159522 : SUCCESS in 9m 32s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/6900fd023c5342f39fd4b2247154b41c : FAILURE in 8m 59s\n- tenks-deploy-teardown-ovs-bios-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0811a2a197e441e94ce7036632068b6 : SUCCESS in 9m 48s\n- tenks-deploy-teardown-ovs-bios-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6199f22aad8e49e7a8dcd2e5286a8ae2 : SUCCESS in 8m 46s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/edcff5db159a403db149fd69b4eab92c : FAILURE in 8m 27s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/9dc141ad9405459ab3659285cb82f68c : SUCCESS in 9m 52s\n- tenks-deploy-teardown-linuxbridge-uefi-debian https://zuul.opendev.org/t/openstack/build/88ba4850437944fcad608e755efbb151 : SUCCESS in 8m 11s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/30862f8a259a4eb6a26afe4e2fc2e3ed : FAILURE in 8m 32s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd033dd3e56a486ab1c9ecde94b190ab : SUCCESS in 8m 25s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/307957e5fc9043128eb494aac6c9f1e4 : SUCCESS in 9m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c91580c7fc3245ee9b40ea946f16c09f : FAILURE in 51m 54s","accounts_in_message":[],"_revision_number":7},{"id":"cc4cf89a4bd873e0f47a5dec8f79a376439b069e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 18:25:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4155e15e8c6cca60d00b82b00d9a392892f302b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 19:46:17.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cb2d544866541ef9bfbffdc526793e6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6524ae0916e4453c85b57b2dbe5c9b1e : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f51460cf1247dcb5a3165669ea4889 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78add4ae27144b57ab02caaf365ba595 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/752dc2b0f7064ff4ad3f798582816e24 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98f7957e4db14e488bdd889a10c78172 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ce8425ea45e4ae8bbec5ebacd3b1c0b : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd8cd6dc367d4eda8013d535a139b3c3 : SUCCESS in 6m 55s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/0f374283e8d8432fa0ae7c31b111d6f1 : FAILURE in 6m 30s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/579c0ee0eed240ea9bb4b88e6a8b22f9 : SUCCESS in 14m 16s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/a8539c5562ca4f3bbadc934838e396a4 : SUCCESS in 13m 03s\n- tenks-deploy-teardown-ovs-bios-debian https://zuul.opendev.org/t/openstack/build/449105cb655745b0b67eeff6b9b7c12a : SUCCESS in 9m 55s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/1710285fd55042e1810da43fe10d0a9d : SUCCESS in 12m 01s\n- tenks-deploy-teardown-ovs-bios-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c53e71918e6456bb8df5fbacc35e646 : SUCCESS in 10m 11s\n- tenks-deploy-teardown-ovs-bios-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16b48795d6fb488e9c4e7e4dc2823f99 : SUCCESS in 9m 44s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/f2570a9075f14544bdfbffcd533db197 : SUCCESS in 13m 28s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/ae1988bbb0b04d4789de7743cacbc9f3 : SUCCESS in 12m 28s\n- tenks-deploy-teardown-linuxbridge-uefi-debian https://zuul.opendev.org/t/openstack/build/e1b934d1d98f4f59ae8ce3f0e5272fb2 : SUCCESS in 9m 49s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/58584d0021a641fab7d85c64e044ded7 : SUCCESS in 11m 03s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3ed6a4665e647cebc42fedae45f9aa3 : SUCCESS in 10m 09s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c091d096a884edc983616b6016c0f82 : SUCCESS in 9m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/56d7a5401220404c8a7d4b9841972617 : SUCCESS in 1h 15m 24s","accounts_in_message":[],"_revision_number":8},{"id":"5d7fb5307c9a0e68b9d4a7b39e445cda1d3b2330","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 19:48:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"aef6e4cc61a460c3bfd877ed537e21c2dd7fddf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 20:00:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f5befc4418184ae4c99b66e6e2fea9a10e2e54ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-07 20:01:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"35ebd264f7735e2bb27f2872839188eaa46038e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 21:15:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58a0cacb7f2a41469cd8d4d2a5ec2edb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79d9359c88164ddab3b912d525137e82 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22bfe5255b1441228751b7dc70b5113e : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fc04c26d9884fd69323345dfc71d255 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3260f987e3854e9fae3f4e48b8015b32 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c1bca03f80948ce9e0c7544f59a0850 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6245f4e465e42ed87480c2f13d7b6c6 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8323442cd6d84619b43fbfb96153415c : SUCCESS in 6m 21s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/d6578fccdf4c4ad58e494eba431a1e26 : SUCCESS in 5m 13s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/6ff70a0c79df4a478e74e2209493f5a5 : SUCCESS in 12m 41s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/3544afb608ed4a118392156d4ba28383 : SUCCESS in 12m 24s\n- tenks-deploy-teardown-ovs-bios-debian https://zuul.opendev.org/t/openstack/build/f07d667da37f4066aea08d1e6ea8e9ed : SUCCESS in 8m 00s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/e1a6de02974846b883aadcb0be6e8719 : SUCCESS in 13m 30s\n- tenks-deploy-teardown-ovs-bios-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2295f97fa59248b2be509ff3a4076aec : SUCCESS in 7m 30s\n- tenks-deploy-teardown-ovs-bios-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f920fae3a8be443d9140630ded90816c : SUCCESS in 11m 59s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/8d09e8afbf9d4c22a8276f5eb763712d : SUCCESS in 12m 15s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/96c5421b50b4483fabcf112fe27b1a93 : SUCCESS in 9m 13s\n- tenks-deploy-teardown-linuxbridge-uefi-debian https://zuul.opendev.org/t/openstack/build/95f8180915ae4ae586db13d0d90ca78b : SUCCESS in 9m 55s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/86f6b9b4cf094d2cb66a9c45fe5266bb : SUCCESS in 12m 19s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65dc44a3d4ea4196b8d654a4efc66ac7 : SUCCESS in 9m 06s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f405dbbfd1ab4520a912b88387cb6b62 : SUCCESS in 8m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a00b55123a37445c85f63b2f047b699b : SUCCESS in 1h 13m 06s","accounts_in_message":[],"_revision_number":9},{"id":"9187d3406ac3f43af6d1723643fb257eb6304156","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-09 08:25:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"615165bf5fac6bff4261d249c0f592aa00df6733","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-09-09 08:29:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"306d3f8e8a106dfd5459de2f3ce847352735c4b3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-09-09 08:36:19.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7d883645900896e14acaf30ce6461d57d0ebf1a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-09 15:56:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"13f94a1d5b8a57c96b83465f08889f49453194d3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-09 15:56:40.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"9860bf8346194909cb5fa00b1b4b17fc6c7006b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 17:17:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95eb0d85c275462395be1bb85bdbab3d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/986801ff2074496c96bab5b3aeb5cf3e : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0874fffa95914a94b1065e9a74c34acd : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c824fd7846548d7a0ba90d352932a0a : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26c16f0c082f427f99a58d3cce012a1e : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9edced8f26924bc4a29295ecf1e14910 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/403d5e0f595b45fd888222e9696aeca9 : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fac5ec009244de4b488f75402ef9ba0 : SUCCESS in 5m 44s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/9118e2ed02a04cd1b5c445e2314a70c7 : SUCCESS in 6m 38s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/2e822411c3a44730af67bb3dfebebe2b : SUCCESS in 14m 22s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/efc7003aad094e96b0917b568e17d121 : SUCCESS in 11m 33s\n- tenks-deploy-teardown-ovs-bios-debian https://zuul.opendev.org/t/openstack/build/2d7029585f60475cbfd2d292b571f0e9 : SUCCESS in 10m 19s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/04c55cdb1d5f4b4a9e9911c04945dccd : SUCCESS in 10m 50s\n- tenks-deploy-teardown-ovs-bios-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f02787dbccce4451ae937357fb4ee49f : SUCCESS in 10m 34s\n- tenks-deploy-teardown-ovs-bios-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f50d181ac3324b559c465b0bf2696389 : SUCCESS in 8m 48s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/c7f323b08a6f40e49fc649fed9edc644 : SUCCESS in 13m 52s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/ef902023d5894f66beedfef2562f3303 : SUCCESS in 10m 41s\n- tenks-deploy-teardown-linuxbridge-uefi-debian https://zuul.opendev.org/t/openstack/build/e63049677f684a2d85fda2e6a5067b73 : SUCCESS in 10m 33s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/a1f3b1dc3d0b4fd99f7d524aeb725080 : SUCCESS in 9m 55s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5647b08603d44f75b7d433c2a844b226 : SUCCESS in 9m 01s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ebafb2f127c4041a8c533a7f57b55cc : SUCCESS in 10m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fb728bd55912431ea4a2b08fab887c91 : SUCCESS in 1h 12m 24s","accounts_in_message":[],"_revision_number":10},{"id":"1a61cd58c8d2109257d36c462436b5f7d9ca2794","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-09-09 19:21:39.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"e28cfd472fc43d9c83c93df808c2f750158a377d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 19:22:09.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"dea4553c335d26988b411a2c1e03f811b9aa25f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 20:30:13.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d31f4ddd4394769a1372b4125681b3a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111cdff6557c41a6878211b44f9111e4 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea93eca6c1d448b950129a13ceca23d : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50046d58644f4c4297be658f9e43a927 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8845a417c7dd4fe79e0f880601949772 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f9a13965d1e438b82a151e36a0e6d03 : SUCCESS in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5078ef98e07433fb1ae6b1b974f0d31 : SUCCESS in 6m 24s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/a4ceeb4b241e4ebc9c56bf23337d5d74 : SUCCESS in 5m 49s\n- tenks-deploy-teardown-ovs-bios-centos8s https://zuul.opendev.org/t/openstack/build/5f0f589cec774a889edf8e6f970c00dd : SUCCESS in 12m 19s\n- tenks-deploy-teardown-ovs-bios-debian https://zuul.opendev.org/t/openstack/build/35fdd69f2b484b9eb4777e2694a85563 : SUCCESS in 10m 30s\n- tenks-deploy-teardown-ovs-bios-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69eafa9c0cf447dc86aad99228e27f12 : SUCCESS in 9m 16s\n- tenks-deploy-teardown-ovs-bios-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a64f8e9ba0d1463c9f5b2aa8d0c6aa6f : SUCCESS in 10m 21s\n- tenks-deploy-teardown-linuxbridge-uefi-centos8s https://zuul.opendev.org/t/openstack/build/401b3f66099e42deb94ebe63d694c488 : SUCCESS in 12m 03s\n- tenks-deploy-teardown-linuxbridge-uefi-debian https://zuul.opendev.org/t/openstack/build/51b0011497994d1a86a3068ac8fa06bc : SUCCESS in 9m 51s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95d08785ff404bc39051202ae74fdc13 : SUCCESS in 9m 08s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3d70626b79346ae982eb9c5f36e51b6 : SUCCESS in 10m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1dd6cdd707f8441cadf2c330d2f82d86 : SUCCESS in 1h 07m 25s","accounts_in_message":[],"_revision_number":10},{"id":"7bac11900d269bd16d73d559a4eb9b2855337985","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 20:30:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"1cab8478e81ea8f1de6d57b7a36368b4c2fbb0e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 20:31:13.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ef0e01574674659a05117cad82aeccd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa73dd3d4344717b13afd3ebe99ddb4 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":10},{"id":"94893371f34e7c210250c04a6755151f6363e491","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-09-12 08:44:11.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"08742b2f67bbe00c9b6e5c18bdda449564ff68c4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-12 08:49:23.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"6c59da3daf416fa4211858ef3a5154d0894fa85f","revisions":{"e34d256da96c7ab4a92b78bb390fa7300b4902b3":{"kind":"REWORK","_number":1,"created":"2022-09-07 15:48:26.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/1"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:48:24.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/e34d256da96c7ab4a92b78bb390fa7300b4902b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/e34d256da96c7ab4a92b78bb390fa7300b4902b3"}]},"branch":"refs/heads/master"},"f692519c33536c2730a08ae7ddb2b8b56e19722b":{"kind":"REWORK","_number":2,"created":"2022-09-07 15:57:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/2"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:57:25.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/f692519c33536c2730a08ae7ddb2b8b56e19722b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/f692519c33536c2730a08ae7ddb2b8b56e19722b"}]},"branch":"refs/heads/master"},"04ad1f89549687527d186f8c92ef82ef6d8e7fa9":{"kind":"REWORK","_number":3,"created":"2022-09-07 16:11:35.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/3"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 16:11:29.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/04ad1f89549687527d186f8c92ef82ef6d8e7fa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/04ad1f89549687527d186f8c92ef82ef6d8e7fa9"}]},"branch":"refs/heads/master"},"ea28172bb7882546d2f5b57168b87d8f8aad223a":{"kind":"REWORK","_number":4,"created":"2022-09-07 16:34:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/4"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 16:33:57.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/ea28172bb7882546d2f5b57168b87d8f8aad223a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/ea28172bb7882546d2f5b57168b87d8f8aad223a"}]},"branch":"refs/heads/master"},"e7c549e2ff55b42f8098506be90815b3de9d29c1":{"kind":"REWORK","_number":5,"created":"2022-09-07 16:55:38.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/5"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 16:55:31.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/e7c549e2ff55b42f8098506be90815b3de9d29c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/e7c549e2ff55b42f8098506be90815b3de9d29c1"}]},"branch":"refs/heads/master"},"9493e5405b2b0d4dc189ef6f9348758a6c741ade":{"kind":"REWORK","_number":6,"created":"2022-09-07 16:58:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/6"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 16:58:48.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/9493e5405b2b0d4dc189ef6f9348758a6c741ade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/9493e5405b2b0d4dc189ef6f9348758a6c741ade"}]},"branch":"refs/heads/master"},"b4ea37fcc007175d9b550b070d2bd5c93252abd8":{"kind":"REWORK","_number":7,"created":"2022-09-07 17:17:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/7"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 17:17:14.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/b4ea37fcc007175d9b550b070d2bd5c93252abd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/b4ea37fcc007175d9b550b070d2bd5c93252abd8"}]},"branch":"refs/heads/master"},"28cca825dbc0bd86261b8b8132d963833f6b6dce":{"kind":"REWORK","_number":8,"created":"2022-09-07 18:25:52.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/8"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 18:25:11.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nAt the same time, remove some existing hacks around selinux,\nsystemd, PyYAML and libvirt.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/28cca825dbc0bd86261b8b8132d963833f6b6dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/28cca825dbc0bd86261b8b8132d963833f6b6dce"}]},"branch":"refs/heads/master"},"91541073ae446eac75fa4c78c08b1c31ee742fc8":{"kind":"REWORK","_number":9,"created":"2022-09-07 20:00:50.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/9"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 20:00:44.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nAt the same time, remove some existing hacks around selinux,\nsystemd, PyYAML and libvirt.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/91541073ae446eac75fa4c78c08b1c31ee742fc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/91541073ae446eac75fa4c78c08b1c31ee742fc8"}]},"branch":"refs/heads/master"},"6c59da3daf416fa4211858ef3a5154d0894fa85f":{"kind":"REWORK","_number":10,"created":"2022-09-09 15:56:28.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/856296/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/96/856296/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/96/856296/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/96/856296/10"}}},"commit":{"parents":[{"commit":"c41b3edd6775375d96ff43ae801af9b65ed874c0","subject":"Merge \"Add support for EL9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c41b3edd6775375d96ff43ae801af9b65ed874c0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-07 15:47:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-09 15:56:21.000000000","tz":0},"subject":"Use venv instead of virtualenv","message":"Use venv instead of virtualenv\n\nvirtualenv is not a part of the core python installation, but\nvenv is. Let\u0027s use it to be slimmer and avoid issues with\nvirtualenv installation.\n\nAt the same time, remove some existing hacks around selinux,\nsystemd, PyYAML and libvirt.\n\nChange-Id: Icd60aca084323cede944ee261d2cf5941b320e4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/6c59da3daf416fa4211858ef3a5154d0894fa85f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/6c59da3daf416fa4211858ef3a5154d0894fa85f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[]}
