)]}'
{"id":"openstack%2Ftenks~968145","triplet_id":"openstack%2Ftenks~master~I7d14bdd522e69073f69f2c0def2b6b316916e7c3","project":"openstack/tenks","branch":"master","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-23 18:22:29.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-23 16:38:38.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I7d14bdd522e69073f69f2c0def2b6b316916e7c3","subject":"Drop explicit dependency on os-client-config","status":"NEW","created":"2025-11-23 16:25:56.000000000","updated":"2025-11-23 18:22:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a958a74d8f27cca4a95fe88b30da8b95ea328c7a","_number":968145,"virtual_id_number":968145,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-23 16:38:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-23 16:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-23 18:22:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"31c609d5d012f60593ad422b21311d0ac4cb687e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-23 16:25:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e879a5914c42cb6cb90ca072bf9afbf165128153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-23 16:38:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90ca2af5c1da481ab5f4a0297bd6aa1d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a72ff5f004f748eab6bc71f15cbe94e7 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd99f77ea0c4ad5a54d45dca1d4e50b : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36cda2d629ab40b883e0454852847022 : SUCCESS in 2m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a1532c87c84a928130c689efd08d42 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d7c4e5265d1460dab75dd9e3518ec57 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32ff2db1a9cc412fa428a2171e1452ca : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3aceb7e4aafb480d9ac7aa624ec2a729 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d650513733b4cb5b7118479e64606ec : SUCCESS in 3m 35s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/539bd9f21148478680b3c5a4837b8174 : SUCCESS in 3m 31s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/ffcc5c7733fe44bbb36d7a958222eb95 : FAILURE in 7m 57s\n- tenks-deploy-teardown-ovs-bios-debian-bullseye https://zuul.opendev.org/t/openstack/build/4ef3a326158f4db6997d155d39fad6ec : FAILURE in 4m 23s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/26e46e9322e14ccda325313e551b08ef : FAILURE in 6m 19s\n- tenks-deploy-teardown-ovs-bios-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df9db5c6e83f40c3a024e3b106db6964 : SUCCESS in 6m 41s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/ae74e74843094075ace4adba87b1279f : FAILURE in 7m 16s\n- tenks-deploy-teardown-linuxbridge-uefi-debian-bullseye https://zuul.opendev.org/t/openstack/build/4a4fed5279b14c58a46e03c1c99e4c56 : FAILURE in 4m 22s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/3231af9669624f4baab7dcd66f2544f2 : FAILURE in 5m 47s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65ffe3b7c5714e85811d386d1db3fb1c : SUCCESS in 7m 33s\n- tenks-tox-codespell https://zuul.opendev.org/t/openstack/build/cc83f761450e483a8b1d99374417dbec : SUCCESS in 2m 09s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4029ecae8562df0c8068a0f2ea5b8226bc7bf71","revisions":{"e4029ecae8562df0c8068a0f2ea5b8226bc7bf71":{"kind":"REWORK","_number":1,"created":"2025-11-23 16:25:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/968145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/45/968145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/45/968145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/45/968145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/45/968145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/45/968145/1"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-23 16:24:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-23 16:25:39.000000000","tz":540},"subject":"Drop explicit dependency on os-client-config","message":"Drop explicit dependency on os-client-config\n\nIts functionality was merged into openstacksdk in 0.10.1 release[1],\nso we don\u0027t have to install it explicitly.\n\nThe library is now formally deprecated[2].\n\n[1] 65293358a0d3cd7055987b18e16b9be5e8a0261b\n[2] d1e2fe7d2e07767f1a66ed0c057c20d7a95271f4\n\nChange-Id: I7d14bdd522e69073f69f2c0def2b6b316916e7c3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/e4029ecae8562df0c8068a0f2ea5b8226bc7bf71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/e4029ecae8562df0c8068a0f2ea5b8226bc7bf71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
