)]}'
{"id":"openstack%2Ftenks~968200","triplet_id":"openstack%2Ftenks~master~I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed","project":"openstack/tenks","branch":"master","topic":"fix-ci","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-25 15:41:01.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-12-17 16:29:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed","subject":"Modernize and fix CI","status":"MERGED","created":"2025-11-24 14:13:47.000000000","updated":"2025-12-17 16:31:31.000000000","submitted":"2025-12-17 16:29:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"968200-fix-ci","meta_rev_id":"0075d309014fb049ad48a3efe3f107bbd207d347","_number":968200,"virtual_id_number":968200,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-17 16:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2025-12-11 14:33:33.000000000","_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-12-15 00:31:10.000000000","_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-11-25 15:41:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2025-12-17 16:11:14.000000000","_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-24 14:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 15:24:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-11 14:33:33.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-12-15 00:31:10.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2025-12-16 11:45:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"c1ba644813075d35cc22969c8f1bdd1c2409153d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 14:13:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba5a5fc4ec74e576f5ec4308208f352fce80c6f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 14:18:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e1465469f3f94c63c669e08396110989c1059c32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:18:49.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/tenks on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream-8GB\" was not found.\n\nThe problem appears in the the \"tenks-deploy-teardown-ovs-bios-centos10\" job stanza:\n\n  job:\n      name: tenks-deploy-teardown-ovs-bios-centos10\n      parent: tenks-deploy-teardown-ovs-bios-base\n      nodeset: centos-10-stream-8GB\n\n  in \"openstack/tenks/zuul.d/jobs.yaml@master\", line 16","accounts_in_message":[],"_revision_number":2},{"id":"c128832a568bce1fa81d32d13525a6074831f23b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 14:19:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a926fb8f4e5ba9300f0fdaee6057f44b642f7000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:20:17.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/tenks on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream-8GB\" was not found.\n\nThe problem appears in the the \"tenks-deploy-teardown-ovs-bios-centos10\" job stanza:\n\n  job:\n      name: tenks-deploy-teardown-ovs-bios-centos10\n      parent: tenks-deploy-teardown-ovs-bios-base\n      nodeset: centos-10-stream-8GB\n\n  in \"openstack/tenks/zuul.d/jobs.yaml@master\", line 16","accounts_in_message":[],"_revision_number":3},{"id":"b880d9da075773820d25412692c13401e7bbfef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 14:21:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f3a7a36c03e4eeff62c030ba6ccf5db9454ae832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:22:45.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/tenks on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream-8GB\" was not found.\n\nThe problem appears in the the \"tenks-deploy-teardown-ovs-bios-centos10\" job stanza:\n\n  job:\n      name: tenks-deploy-teardown-ovs-bios-centos10\n      parent: tenks-deploy-teardown-ovs-bios-base\n      nodeset: centos-10-stream-8GB\n\n  in \"openstack/tenks/zuul.d/jobs.yaml@master\", line 16","accounts_in_message":[],"_revision_number":4},{"id":"3ebf7b195a507d2b228f3fe4531498b100eceb0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 14:23:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"88ed9f506e63cd4ba087e316ce9e3a5e555295f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:24:11.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/tenks on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream-8GB\" was not found.\n\nThe problem appears in the the \"tenks-deploy-teardown-ovs-bios-centos10\" job stanza:\n\n  job:\n      name: tenks-deploy-teardown-ovs-bios-centos10\n      parent: tenks-deploy-teardown-ovs-bios-base\n      nodeset: centos-10-stream-8GB\n\n  in \"openstack/tenks/zuul.d/jobs.yaml@master\", line 16","accounts_in_message":[],"_revision_number":5},{"id":"8ce02b96922dd014e2e4ee87c1a834536106db12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 14:47:01.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"75b5c13a4752cdb8008cf5bca74db56a93d9aec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:47:18.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/tenks on branch master.  The\nproblem was:\n\n  The nodeset \"centos-10-stream-8GB\" was not found.\n\nThe problem appears in the the \"tenks-deploy-teardown-ovs-bios-centos10\" job stanza:\n\n  job:\n      name: tenks-deploy-teardown-ovs-bios-centos10\n      parent: tenks-deploy-teardown-ovs-bios-base\n      nodeset: centos-10-stream-8GB\n\n  in \"openstack/tenks/zuul.d/jobs.yaml@master\", line 22","accounts_in_message":[],"_revision_number":6},{"id":"e294e497714632ab0b7b77b42fa3f0fe7c2875f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 15:08:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"72ea8b59da2ff7fe1f9982c92f88391598bbd640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 15:29:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/312a2b2c40c6414489142f9d43a9f613\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1249195188634520b0b7ca312e40085c : FAILURE in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fd41acae91f4b7b93a96e18082a7eac : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dbf3bbf37ff45df9c1b881a9a337567 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69e123c58ec54f329af846cf59de024e : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2eeb327d9254bdb9b46f5a3f8d0a7f7 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34e66a672c0341d3b3f18864f13ef048 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f8fa4dc457487bb717612b43ea0339 : SUCCESS in 3m 52s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/87c8e83a5bde41fe8a297108b4bb6278 : SUCCESS in 6m 37s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/f40cbeab7f804f0a9d310505744a30ac : SUCCESS in 9m 06s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/d4c9bf6670974462a3459c8da1098798 : SUCCESS in 4m 57s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/6dc7bcb9e40f4547911906d1763e0e66 : SUCCESS in 7m 49s\n- tenks-deploy-teardown-ovs-bios-centos10 https://zuul.opendev.org/t/openstack/build/d025d392f0634f06aed30bb31a263aed : FAILURE in 3m 20s\n- tenks-deploy-teardown-linuxbridge-uefi-centos10 https://zuul.opendev.org/t/openstack/build/f7df638c9bc74746af42eafc6519cc32 : SUCCESS in 7m 18s\n- tenks-deploy-teardown-ovs-bios-debian-bookworm https://zuul.opendev.org/t/openstack/build/24d215d074a14442badbdb6e82f8b43d : SUCCESS in 7m 23s\n- tenks-deploy-teardown-linuxbridge-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/b629a9f628ac4802afe6e8e667c2a3fc : SUCCESS in 5m 43s\n- tenks-deploy-teardown-ovs-bios-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a16464902158478c9b3ae7c22665bef2 : SUCCESS in 3m 57s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e59b659c22f049c1a60f8a813006cab6 : SUCCESS in 8m 49s\n- tenks-tox-codespell https://zuul.opendev.org/t/openstack/build/274c8f2c9c8d40589df2b657772a5b47 : SUCCESS in 2m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"930e5a76c66a4be23bbb7e09dfb498298594ac0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 15:46:42.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"f78e60e756e4f8d84121df132ee80baeda59ee35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 16:04:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4c075ecfbcd49faa736e8f37e56307e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b08ae9e1f45b4505b9edb6d8c6fab80e : FAILURE in 1m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe0701ec09b24fc697a61cc9fa58216d : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3615c4e97fbb43c4a0efac6808fdb275 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d347591d0e8242de87417c6d2e52b216 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f456df669c9041a181c66ccb0f959b79 : SUCCESS in 3m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d9ab3fb551b4193b98eba540f824afc : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3d985c5ea824f8992027378b58e582b : SUCCESS in 4m 06s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/dca937f031b34f52972154603f80077b : SUCCESS in 2m 17s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/1b5c8b27da2a4f0eb87028ef1098dd59 : RETRY_LIMIT in 3m 24s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/3f8e1281bf764f91bdcdb4d63be7a864 : SUCCESS in 7m 36s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/938204e68329422bada2597d05c88575 : SUCCESS in 10m 13s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/6056bd891d5e40c9984a2af255c52eab : SUCCESS in 8m 17s\n- tenks-deploy-teardown-ovs-bios-centos10 https://zuul.opendev.org/t/openstack/build/5236e600c1a7404a82e9ba267612fbc2 : FAILURE in 3m 06s\n- tenks-deploy-teardown-linuxbridge-uefi-centos10 https://zuul.opendev.org/t/openstack/build/7ea16a0174a243f2a177d7e240e41f17 : SUCCESS in 9m 09s\n- tenks-deploy-teardown-ovs-bios-debian-bookworm https://zuul.opendev.org/t/openstack/build/0d2e41a5430c451eb86716ad03652454 : SUCCESS in 6m 41s\n- tenks-deploy-teardown-linuxbridge-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/d75224fff91741ee9ddeb5e14e48b89c : SUCCESS in 6m 37s\n- tenks-deploy-teardown-ovs-bios-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c64e82e13a2c4ea3bbce309c3d4b5c3c : SUCCESS in 5m 24s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5311ef51d6d84ccfa95c1499a16d8ce8 : SUCCESS in 7m 16s\n- tenks-tox-codespell https://zuul.opendev.org/t/openstack/build/8a3457bf85564351ab42f8287d57a988 : SUCCESS in 2m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0706da946f4085970efca9f04be46073d075170d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 17:19:02.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"0931406697b1dd165b456344d9d33b2efbb80f14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-24 17:31:08.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"fca9cbb27ecf0259b28531fa2946c1823342902a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:52:33.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a31cd29dd5d41e5b73604e0e653e490\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5d640a618f44cefa478e1e0844f8180 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a555842ed23845d0a0cc8860aa3920c5 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/607860be54214904be7d55455e81aaba : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0e9eb5eef7140fabeecdc2a24d19c79 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70d5420ccbde4e9483ffc86b83aeff77 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4571a6a2d54ac8a6641bebab488ca2 : SUCCESS in 6m 40s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/c11544f182544cde8c70b869269c8537 : SUCCESS in 2m 17s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/8e642f39cbbe45ec981b8a4a902ec63a : SUCCESS in 10m 53s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/e5cbb8a80b0a4df1b07ad1ac71e5d1cf : SUCCESS in 9m 29s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/855126d6dbfc4c8b8517941b8c6a2a64 : SUCCESS in 10m 06s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/f010574805134688bb60af19340e4a64 : SUCCESS in 5m 27s\n- tenks-deploy-teardown-ovs-bios-centos10 https://zuul.opendev.org/t/openstack/build/76688d9d68c0431babfb1add69f53ea6 : FAILURE in 4m 01s\n- tenks-deploy-teardown-linuxbridge-uefi-centos10 https://zuul.opendev.org/t/openstack/build/1ff2819c9e094bd6b13072519eb642cf : SUCCESS in 10m 14s\n- tenks-deploy-teardown-ovs-bios-debian-bookworm https://zuul.opendev.org/t/openstack/build/0ef8aa2802cc480f9d679c75b9e389d2 : SUCCESS in 5m 55s\n- tenks-deploy-teardown-linuxbridge-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/cbc55ea723694ce2b48faa76ae16549d : SUCCESS in 6m 04s\n- tenks-deploy-teardown-ovs-bios-ubuntu-noble https://zuul.opendev.org/t/openstack/build/502c131280a5466caabf6cf2b69ca8b2 : SUCCESS in 7m 31s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e20f43444134d518ef830fb9d7da595 : SUCCESS in 3m 48s\n- tenks-tox-codespell https://zuul.opendev.org/t/openstack/build/e702ae710bbe4800a7a3c6417c5610b9 : SUCCESS in 2m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"70914991e2d733aff5f4dad4c03cc49fcae4ae9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-11-25 08:53:13.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"609ca9fffbc7a2f0662b192ff5170e3e3c8b9d4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 09:11:54.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf58bf690aff43b39492b7d8a934dbb9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f56fef709a7a4c93b2333702b513e3ea : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f846d9c9caae4d93ae62a4b4c9564a27 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42b2f437ebd24789b837b878b376bfe6 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a0b3ce664334fcf9eba9be103faff31 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24ccb852d5c84508b7c20ca88d36c73a : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/110c13df271e4878a604f3f690ad426b : SUCCESS in 6m 29s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/de92804aec5b4be690e882b2e474a6f3 : SUCCESS in 1m 57s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/a4fc3e4123f548f6a0fbbabbe89f53dd : SUCCESS in 9m 56s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/19a3d6833de44f3e8207c7bc17226cf9 : SUCCESS in 9m 13s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/d36dab821c094aca8a5dcd91246aebab : SUCCESS in 9m 35s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/83d3e9f2df6048788472f1bd05841eb4 : SUCCESS in 5m 50s\n- tenks-deploy-teardown-ovs-bios-centos10 https://zuul.opendev.org/t/openstack/build/382f4c471d0a47c3898040b52c99f3f2 : SUCCESS in 13m 03s\n- tenks-deploy-teardown-linuxbridge-uefi-centos10 https://zuul.opendev.org/t/openstack/build/a63f7739fc0e4ea79afda2cd02a5b2a3 : SUCCESS in 8m 44s\n- tenks-deploy-teardown-ovs-bios-debian-bookworm https://zuul.opendev.org/t/openstack/build/61cfe77119824678bf62a1ea5026baa5 : SUCCESS in 6m 46s\n- tenks-deploy-teardown-linuxbridge-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/2ba6fd37d67b4beead65e5ae1675ccde : SUCCESS in 3m 24s\n- tenks-deploy-teardown-ovs-bios-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bdc17c5dc10496da09483c026da2080 : SUCCESS in 5m 21s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d8a00a39b3e47be811f84852d865756 : SUCCESS in 5m 43s\n- tenks-tox-codespell https://zuul.opendev.org/t/openstack/build/4caae337e70442c18c4ce3ecf2944698 : SUCCESS in 2m 12s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5df5ba181d337e8073f1423add833d24352e0d71","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-25 15:41:01.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"45b8afe96f6ddf85fb24dc2ac117bfd151b88511","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-12-11 14:33:33.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"9cedb8b00a8d89a8f1b9057038a17ab5a991707d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-12-15 00:31:10.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"637fee2de8f964d9e966626308f89786ca5ce58f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-16 11:45:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"35f4b5d3cb19c702c9f9af8eba17e6b406dd3cbf","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-12-17 16:11:14.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"a7816aa07aede81269831c5f53a4d11c86cb6724","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 16:11:49.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"486b96268e1ea1e750d5af80ba5d382aeb42bf53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 16:29:23.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56c2dc832f1244608a2681007ff61327\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27a4bee4b8574d8ba110cc8d8ccecd40 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/044888e1b663435998d74fe56ef08925 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9868d607fce24149bbc6d1964955e6a6 : SUCCESS in 3m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fadc0f898411474f94b1d631022c4db6 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c8f72ae092b4fe597ea6d7f5d95bbee : SUCCESS in 5m 38s\n- tenks-tox-ansible-lint https://zuul.opendev.org/t/openstack/build/830590b0df964957b6f071a840cc3a72 : SUCCESS in 4m 07s\n- tenks-deploy-teardown-ovs-bios-centos9s https://zuul.opendev.org/t/openstack/build/99b89e9b1ead4f2cbf7c82ea999e7d1f : SUCCESS in 11m 38s\n- tenks-deploy-teardown-ovs-bios-rocky-9 https://zuul.opendev.org/t/openstack/build/4a08337b459949e69023d4d8a23ad7bd : SUCCESS in 11m 50s\n- tenks-deploy-teardown-linuxbridge-uefi-centos9s https://zuul.opendev.org/t/openstack/build/2274b6cf0bba4d29b134a84e8f53f8af : SUCCESS in 9m 33s\n- tenks-deploy-teardown-linuxbridge-uefi-rocky-9 https://zuul.opendev.org/t/openstack/build/32d0e2998c474c27a530ad3efd7d9589 : SUCCESS in 9m 16s\n- tenks-deploy-teardown-ovs-bios-centos10 https://zuul.opendev.org/t/openstack/build/a15ff84c9f444bf682a59340a305e904 : SUCCESS in 9m 56s\n- tenks-deploy-teardown-linuxbridge-uefi-centos10 https://zuul.opendev.org/t/openstack/build/5d51652b52ad470aa168d00cfa17e815 : SUCCESS in 9m 20s\n- tenks-deploy-teardown-ovs-bios-debian-bookworm https://zuul.opendev.org/t/openstack/build/f0789f24e7694b9bb545bf87f49d3efb : SUCCESS in 7m 02s\n- tenks-deploy-teardown-linuxbridge-uefi-debian-bookworm https://zuul.opendev.org/t/openstack/build/45035cf20c7f4c968a00377cf3a7f553 : SUCCESS in 6m 44s\n- tenks-deploy-teardown-ovs-bios-ubuntu-noble https://zuul.opendev.org/t/openstack/build/489a162a600642a4a25b359131442b55 : SUCCESS in 7m 45s\n- tenks-deploy-teardown-linuxbridge-uefi-ubuntu-noble https://zuul.opendev.org/t/openstack/build/724c4febd4014e3c9343c5a2b7d6f12e : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":11},{"id":"b023b73739755eab51e9d00cb70efbbd54045caa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 16:29:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"0075d309014fb049ad48a3efe3f107bbd207d347","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 16:31:31.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82f14310c50645238d28c47afc34f78f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f51d9aabd9a41e08d97ce31d940dd81 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"c25757f16d8e1c3438261a268a21ff6a897174df","revisions":{"1af48e209f255cdc5ab2afb3873165691ea70d6a":{"kind":"REWORK","_number":1,"created":"2025-11-24 14:13:47.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/1"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:13:43.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/1af48e209f255cdc5ab2afb3873165691ea70d6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/1af48e209f255cdc5ab2afb3873165691ea70d6a"}]},"branch":"refs/heads/master"},"c32a1f0e5c60905981732cecfd1b22a4e288bd65":{"kind":"REWORK","_number":2,"created":"2025-11-24 14:18:11.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/2"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:17:48.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c32a1f0e5c60905981732cecfd1b22a4e288bd65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c32a1f0e5c60905981732cecfd1b22a4e288bd65"}]},"branch":"refs/heads/master"},"17573c7d90e91cc494b43df74aa39b26791da463":{"kind":"REWORK","_number":3,"created":"2025-11-24 14:19:56.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/3"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:19:36.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\nRemove ancient jammy and bullseye jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/17573c7d90e91cc494b43df74aa39b26791da463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/17573c7d90e91cc494b43df74aa39b26791da463"}]},"branch":"refs/heads/master"},"a1c33f03725843d02ed767261cd4c936a3571bfa":{"kind":"REWORK","_number":4,"created":"2025-11-24 14:21:57.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/4"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:21:44.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\nRemove ancient jammy and bullseye jobs\nAdd bookworm and noble jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/a1c33f03725843d02ed767261cd4c936a3571bfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/a1c33f03725843d02ed767261cd4c936a3571bfa"}]},"branch":"refs/heads/master"},"51fc91d4cd7475a3caa94fdea8bcd6485853e378":{"kind":"REWORK","_number":5,"created":"2025-11-24 14:23:32.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/5"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:23:28.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\nRemove ancient jammy and bullseye jobs\nAdd bookworm and noble jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/51fc91d4cd7475a3caa94fdea8bcd6485853e378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/51fc91d4cd7475a3caa94fdea8bcd6485853e378"}]},"branch":"refs/heads/master"},"8a8eb4f279f8ae7312a437c7c240e2afefbe2567":{"kind":"REWORK","_number":6,"created":"2025-11-24 14:47:01.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/6"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:46:56.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\nRemove ancient jammy and bullseye jobs\nAdd bookworm and noble jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/8a8eb4f279f8ae7312a437c7c240e2afefbe2567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/8a8eb4f279f8ae7312a437c7c240e2afefbe2567"}]},"branch":"refs/heads/master"},"4a841fff70b1ebbd7a28e5eb2e24a5d3d42123d6":{"kind":"REWORK","_number":7,"created":"2025-11-24 15:08:38.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/7"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 15:08:33.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\nRemove ancient jammy and bullseye jobs\nAdd bookworm and noble jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/4a841fff70b1ebbd7a28e5eb2e24a5d3d42123d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/4a841fff70b1ebbd7a28e5eb2e24a5d3d42123d6"}]},"branch":"refs/heads/master"},"f9830a83cdb17786c06e9bdd5e4347a37b785880":{"kind":"REWORK","_number":8,"created":"2025-11-24 15:46:42.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/8"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 15:46:38.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\nRemove ancient jammy and bullseye jobs\nAdd bookworm and noble jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/f9830a83cdb17786c06e9bdd5e4347a37b785880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/f9830a83cdb17786c06e9bdd5e4347a37b785880"}]},"branch":"refs/heads/master"},"bbd401e2e65db37fc4be56eab7f0095f18dae6c2":{"kind":"REWORK","_number":9,"created":"2025-11-24 17:19:02.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/9"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 17:18:58.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\nRemove ancient jammy and bullseye jobs\nAdd bookworm and noble jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/bbd401e2e65db37fc4be56eab7f0095f18dae6c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/bbd401e2e65db37fc4be56eab7f0095f18dae6c2"}]},"branch":"refs/heads/master"},"7f04d107c4439f76f02dec9eae32352ca1c8ace4":{"kind":"REWORK","_number":10,"created":"2025-11-24 17:31:08.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/10"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 17:31:04.000000000","tz":60},"subject":"[WIP] Modernize and fix CI","message":"[WIP] Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\nRemove ancient jammy and bullseye jobs\nAdd bookworm and noble jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/7f04d107c4439f76f02dec9eae32352ca1c8ace4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/7f04d107c4439f76f02dec9eae32352ca1c8ace4"}]},"branch":"refs/heads/master"},"c25757f16d8e1c3438261a268a21ff6a897174df":{"kind":"REWORK","_number":11,"created":"2025-11-25 08:53:13.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/00/968200/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tenks","ref":"refs/changes/00/968200/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tenks refs/changes/00/968200/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tenks refs/changes/00/968200/11"}}},"commit":{"parents":[{"commit":"96bdd656238fe31e7a172b1a3e9359fb69b657ab","subject":"Keep support for Python 3.9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/96bdd656238fe31e7a172b1a3e9359fb69b657ab"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-24 14:06:24.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2025-11-25 08:52:57.000000000","tz":60},"subject":"Modernize and fix CI","message":"Modernize and fix CI\n\nUse current openstack python versions to run unit tests, but keep\ntesting Python 3.9 using stable/2025.1\nPin stable/2025.1 for CS9 and Rocky 9 jobs\nAdd CS10 jobs to test current master\nRemove ancient jammy and bullseye jobs\nAdd bookworm and noble jobs\n\nChange-Id: I3a3022f0ffd3d7ea1ef03f05fa047d54afac99ed\nSigned-off-by: Riccardo Pittau \u003celfosardo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c25757f16d8e1c3438261a268a21ff6a897174df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tenks/commit/c25757f16d8e1c3438261a268a21ff6a897174df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
