)]}'
{"id":"openstack%2Ftooz~1005098","triplet_id":"openstack%2Ftooz~master~I54115f413749b7b743f8c7ac74c376d65d1ab62a","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-15 22:28:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I54115f413749b7b743f8c7ac74c376d65d1ab62a","subject":"redis: Deprecate retry_on_timeout","status":"MERGED","created":"2026-09-10 17:30:15.000000000","updated":"2026-09-15 22:31:43.000000000","submitted":"2026-09-15 22:28:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005098","meta_rev_id":"7789a70f11c6aa762ab95ca248905f17f1b83656","_number":1005098,"virtual_id_number":1005098,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 22:28:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 20:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-15 21:10:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 19:17:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 20:03:12.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"dccb687245924365b0b0460e35520ed3d00aa8bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 17:30:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e4ffc05303769a597ac6e48800130c3c9535343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 19:17:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b8d48514fca498eb83f3748436c441d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ab936a2863aa4deab48e67ba8bc273c2 : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bb340b0e9dc46cf97b72118481a5b28 : SUCCESS in 3m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1ffa992affd48beb1619503666e19fc : SUCCESS in 1h 41m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4688dd38b6c54c568e1a95ec851c994f : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b104dbe88ac44160ae68fad0f451abb8 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c5fa22809304a759688cc9e6f8c0ca7 : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d91eb41e12f4598acf2e2b88366cde8 : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/74b1e26675f440f6ba65c378824c469b : SUCCESS in 5m 13s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/e17d94a44f4b4c499122fa1015185896 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245c45282b4d4cbf8dfb17d96dee7c13 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8db14cf1ddb0403ca3c078813f4e495b : SUCCESS in 4m 27s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/6dc1138a7588460ea9e90f71e6abcd37 : SUCCESS in 3m 45s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/ff429aff13cb4dac8cca71430ff815fb : SUCCESS in 5m 51s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/a217a1a528d74139b55b3f75994d5e2d : SUCCESS in 8m 13s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/06b8ec2931194a89ac214da408d7bac5 : SUCCESS in 4m 55s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/94e8ed85c7ac4aca86ec2295c73b474e : SUCCESS in 7m 23s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/4a8d66435d3940eb9f72ff845670d4dc : SUCCESS in 4m 23s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/7b40ae66cdc74c198351adb5b623e3da : SUCCESS in 8m 40s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/9a88d928483f44a89456c375c64cb869 : SUCCESS in 8m 48s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/e83ff4544c134521a8042c2be6689d79 : SUCCESS in 5m 46s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/2333bc3d84fe4e4eb8848309afaf0d9a : SUCCESS in 5m 32s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/5597073fc75d4aae9ebeb362f6766495 : SUCCESS in 7m 36s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/d914490132e44ff4b98000ef36340525 : SUCCESS in 4m 57s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/c715e157e837478bae52ed53c971db7e : SUCCESS in 9m 44s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/4d9dd2d665a84174a2746f7a10f0999b : SUCCESS in 8m 10s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/3e358dec80144f77ba2b06822ca8cc65 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":1},{"id":"566242f3657c5aae48c3ece7ccbafbd461bb2382","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 20:03:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d3887ad04c6f7a3463b2bd7e3d97bc0e86bd514","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 21:10:00.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43b73c9513e6d1e96b9f70f83e4db07f3860a3e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 21:10:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"59596c17924bbdfbbed06090fdf8929c328c5cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:28:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdd398fdaf004b439d17c4a5ca5c7246\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e70c3d6d7068490b8d86df692b15304b : SUCCESS in 3m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb49671c295f477ca6a51196cb0ab562 : SUCCESS in 3m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d78fce4c88e0426ba718dc1dc28b2492 : SUCCESS in 1h 16m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be8d24affc043b2a02b7e7f63492f10 : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a793769e185f49b3a949421cc5464323 : SUCCESS in 7m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f643427646434c079969dd41abee04e0 : SUCCESS in 6m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f34b8e5582a4332b96e6d5f48a36217 : SUCCESS in 8m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057708526d7e484ab7a283407478c32e : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbc591e1a6fc437c92e76c37454a3c7a : SUCCESS in 3m 09s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/eb290b1d934e4edd9b3b3af2c7e5dae5 : SUCCESS in 5m 12s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/c49da90f172b40b6bdd13239392f7906 : SUCCESS in 5m 25s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/1810841de7cb47bf962c5b6862719f67 : SUCCESS in 10m 00s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/d45ff80e3c5c4a6fa088275030772858 : SUCCESS in 5m 18s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/cb53755dedea4a5db9e2ae2ad217c635 : SUCCESS in 8m 24s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/51b68b9272864d1b8b9d7eba7c4c6c59 : SUCCESS in 5m 59s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/aad299ec2d3b47e9a747a19da09456bd : SUCCESS in 5m 53s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/07cdbc74dd134106814c6a73bf1280c5 : SUCCESS in 5m 40s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/ac6b1d51cce74fa08ee3e094680b256c : SUCCESS in 8m 28s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/56a429a12ccf4611889114084e2e68fc : SUCCESS in 5m 45s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/ba8240e805ef49a0a9fefd248c1ec981 : SUCCESS in 15m 29s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/eac8f60273014a1b996f1bb6ce9fb35b : SUCCESS in 4m 41s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/874df0ec64694988971afef37b8d39cf : SUCCESS in 7m 31s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/f2c9a755e810422c86c06e58f574e14d : SUCCESS in 8m 06s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/10a3fc47f272411699280ea725552349 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"84a09eaa64a0b5ce36af197b1e1b8d48ad416695","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:28:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7789a70f11c6aa762ab95ca248905f17f1b83656","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:31:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7cd9a71523d40c89060f45451ad9b71\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4e6fe4b2eca4b19897f3c2f1b5b5042 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0c8582448014186b6052aefd52926bc : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05a011d84b823cd9b47e1206a571e1c71ceadec5","revisions":{"05a011d84b823cd9b47e1206a571e1c71ceadec5":{"kind":"REWORK","_number":1,"created":"2026-09-10 17:30:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/1005098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/1005098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/1005098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/1005098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/1005098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/1005098/1"}}},"commit":{"parents":[{"commit":"1e9fd9798bad3372b2d262503603ff7af6877a64","subject":"typing: Use type for parsed_url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1e9fd9798bad3372b2d262503603ff7af6877a64"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 17:28:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 17:30:07.000000000","tz":540},"subject":"redis: Deprecate retry_on_timeout","message":"redis: Deprecate retry_on_timeout\n\nThis option in redis-py was deprecated in redis-py 6.0.0 and now redis\nclients retry operations on timeout by default[1].\n\n[1] https://github.com/redis/redis-py/commit/f32a1ca4f6\n\nNote that warning message is not implemented in the driver because\nthe same warning is already implemented in redis-py.\n\nChange-Id: I54115f413749b7b743f8c7ac74c376d65d1ab62a\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/05a011d84b823cd9b47e1206a571e1c71ceadec5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/05a011d84b823cd9b47e1206a571e1c71ceadec5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
