)]}'
{"id":"openstack%2Ftooz~466098","triplet_id":"openstack%2Ftooz~master~I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa","project":"openstack/tooz","branch":"master","topic":"466098","hashtags":[],"change_id":"I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa","subject":"Separate etcd3gw driver that uses the etcd3 grpc gateway","status":"MERGED","created":"2017-05-18 20:24:27.000000000","updated":"2017-05-22 21:37:20.000000000","submitted":"2017-05-22 21:37:20.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a3f60f25decbb37758261dd726da94f5f44b33c","_number":466098,"virtual_id_number":466098,"owner":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-05-22 21:37:20.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":2,"date":"2017-05-22 14:00:07.000000000","_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":2,"date":"2017-05-22 19:16:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2017-05-22 14:10:23.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":1,"date":"2017-05-22 19:16:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-05-19 13:54:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-05-22 12:17:14.000000000","updated_by":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"reviewer":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"state":"REVIEWER"},{"updated":"2017-05-22 12:26:29.000000000","updated_by":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"reviewer":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"state":"REVIEWER"},{"updated":"2017-05-22 14:10:23.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2017-05-22 19:16:04.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2017-05-22 21:37:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e3b44ea83688c0c14214fb4d3d9247ae55237470","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-18 20:24:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50ad5eef470f2bf112871ea9fce45940244da366","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-18 20:38:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8c6aa85a4f8536fb6a05e811f9c4f0a0a3082dae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-18 22:08:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tooz-docs-ubuntu-xenial http://docs-draft.openstack.org/98/466098/2/check/gate-tooz-docs-ubuntu-xenial/16d0145//doc/build/html/ : SUCCESS in 1m 18s\n- gate-tooz-pep8-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-pep8-ubuntu-xenial/8623cc1/ : FAILURE in 1m 59s\n- gate-tooz-python27-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-python27-ubuntu-xenial/38775a1/ : SUCCESS in 1m 36s\n- gate-tooz-python35 http://logs.openstack.org/98/466098/2/check/gate-tooz-python35/92ab7a7/ : SUCCESS in 1m 32s\n- gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial/609ec7a/ : SUCCESS in 1h 28m 57s\n- gate-tooz-releasenotes http://docs-draft.openstack.org/98/466098/2/check/gate-tooz-releasenotes/38c6ac9//releasenotes/build/html/ : SUCCESS in 1m 12s\n- gate-tooz-tox-py27-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py27-etcd-ubuntu-xenial/60c08de/ : SUCCESS in 3m 05s\n- gate-tooz-tox-py35-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py35-etcd-ubuntu-xenial/0722644/ : SUCCESS in 3m 48s\n- gate-tooz-tox-py27-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py27-etcd3-ubuntu-xenial/aab720e/ : SUCCESS in 2m 30s\n- gate-tooz-tox-py35-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py35-etcd3-ubuntu-xenial/d211823/ : SUCCESS in 1m 27s\n- gate-tooz-tox-py27-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py27-zookeeper-ubuntu-xenial/d788c33/ : SUCCESS in 2m 15s\n- gate-tooz-tox-py35-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py35-zookeeper-ubuntu-xenial/0a4112d/ : SUCCESS in 1m 34s\n- gate-tooz-tox-py27-redis-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py27-redis-ubuntu-xenial/d77957c/ : SUCCESS in 1m 56s\n- gate-tooz-tox-py35-redis-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py35-redis-ubuntu-xenial/c25a6a3/ : SUCCESS in 1m 35s\n- gate-tooz-tox-py27-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py27-sentinel-ubuntu-xenial/fb626da/ : SUCCESS in 2m 19s\n- gate-tooz-tox-py35-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py35-sentinel-ubuntu-xenial/1e90151/ : SUCCESS in 2m 19s\n- gate-tooz-tox-py27-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py27-memcached-ubuntu-xenial/c048003/ : SUCCESS in 1m 54s\n- gate-tooz-tox-py35-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py35-memcached-ubuntu-xenial/198e0cf/ : SUCCESS in 1m 59s\n- gate-tooz-tox-py27-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py27-postgresql-ubuntu-xenial/69458a6/ : SUCCESS in 2m 15s\n- gate-tooz-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py35-postgresql-ubuntu-xenial/2b32533/ : SUCCESS in 2m 14s\n- gate-tooz-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py27-mysql-ubuntu-xenial/00f5249/ : SUCCESS in 1m 39s\n- gate-tooz-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py35-mysql-ubuntu-xenial/cf5b2a2/ : SUCCESS in 2m 04s\n- gate-tooz-tox-py27-consul-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py27-consul-ubuntu-xenial/bb2d5a8/ : SUCCESS in 2m 10s\n- gate-tooz-tox-py35-consul-ubuntu-xenial http://logs.openstack.org/98/466098/2/check/gate-tooz-tox-py35-consul-ubuntu-xenial/e427c71/ : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2},{"id":"d9a34e6088ae1fd5dd793af6b15e772936653248","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-19 17:21:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4bf9c177d4e3074d1291720969011e52fba71d7b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-19 18:18:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tooz-docs-ubuntu-xenial http://docs-draft.openstack.org/98/466098/3/check/gate-tooz-docs-ubuntu-xenial/e3441e3//doc/build/html/ : SUCCESS in 1m 07s\n- gate-tooz-pep8-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-pep8-ubuntu-xenial/fda755e/ : SUCCESS in 2m 23s\n- gate-tooz-python27-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-python27-ubuntu-xenial/0bbc863/ : SUCCESS in 1m 42s\n- gate-tooz-python35 http://logs.openstack.org/98/466098/3/check/gate-tooz-python35/05a861c/ : SUCCESS in 2m 02s\n- gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial/5e86a0d/ : FAILURE in 56m 02s\n- gate-tooz-releasenotes http://docs-draft.openstack.org/98/466098/3/check/gate-tooz-releasenotes/67dafa1//releasenotes/build/html/ : SUCCESS in 1m 28s\n- gate-tooz-tox-py27-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py27-etcd-ubuntu-xenial/6547ff6/ : SUCCESS in 1m 58s\n- gate-tooz-tox-py35-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py35-etcd-ubuntu-xenial/a5880a0/ : SUCCESS in 1m 22s\n- gate-tooz-tox-py27-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py27-etcd3-ubuntu-xenial/73aa2f0/ : SUCCESS in 1m 57s\n- gate-tooz-tox-py35-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py35-etcd3-ubuntu-xenial/f0679ba/ : SUCCESS in 2m 03s\n- gate-tooz-tox-py27-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py27-zookeeper-ubuntu-xenial/2446d79/ : SUCCESS in 1m 33s\n- gate-tooz-tox-py35-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py35-zookeeper-ubuntu-xenial/8fbc86c/ : SUCCESS in 2m 14s\n- gate-tooz-tox-py27-redis-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py27-redis-ubuntu-xenial/db0b863/ : SUCCESS in 2m 53s\n- gate-tooz-tox-py35-redis-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py35-redis-ubuntu-xenial/010a700/ : SUCCESS in 2m 02s\n- gate-tooz-tox-py27-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py27-sentinel-ubuntu-xenial/bd4b5d4/ : SUCCESS in 2m 01s\n- gate-tooz-tox-py35-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py35-sentinel-ubuntu-xenial/773ae65/ : SUCCESS in 1m 30s\n- gate-tooz-tox-py27-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py27-memcached-ubuntu-xenial/b6fb0fb/ : SUCCESS in 1m 30s\n- gate-tooz-tox-py35-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py35-memcached-ubuntu-xenial/d7bbfb6/ : SUCCESS in 1m 21s\n- gate-tooz-tox-py27-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py27-postgresql-ubuntu-xenial/f9285d5/ : SUCCESS in 2m 24s\n- gate-tooz-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py35-postgresql-ubuntu-xenial/ed5b6da/ : SUCCESS in 1m 24s\n- gate-tooz-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py27-mysql-ubuntu-xenial/21db983/ : SUCCESS in 3m 24s\n- gate-tooz-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py35-mysql-ubuntu-xenial/517cdb6/ : SUCCESS in 2m 19s\n- gate-tooz-tox-py27-consul-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py27-consul-ubuntu-xenial/c126784/ : SUCCESS in 2m 40s\n- gate-tooz-tox-py35-consul-ubuntu-xenial http://logs.openstack.org/98/466098/3/check/gate-tooz-tox-py35-consul-ubuntu-xenial/057a802/ : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":3},{"id":"4fb5e4e0f5e5ff5ee26a0c8770f66b698f8d79e7","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-19 18:53:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"12afdcb1a31e6a5a0f69278ad3043148a75ccf4c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-19 19:00:35.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"08dfa6cf214a30d23359fa35679691578af78db2","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-19 19:00:45.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3559dbb4753885af4d1d40f991c8b12b3b9115d1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-19 19:17:24.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"a35245af3d9fa53790d49740e7a6a9b4816bef9f","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-19 19:17:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"70fec4544cfce9ced05ec4887223b8b1d764f3cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-19 20:38:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tooz-docs-ubuntu-xenial http://docs-draft.openstack.org/98/466098/4/check/gate-tooz-docs-ubuntu-xenial/536c87b//doc/build/html/ : SUCCESS in 1m 44s\n- gate-tooz-pep8-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-pep8-ubuntu-xenial/205fe90/ : SUCCESS in 2m 45s\n- gate-tooz-python27-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-python27-ubuntu-xenial/a7dab95/ : SUCCESS in 2m 30s\n- gate-tooz-python35 http://logs.openstack.org/98/466098/4/check/gate-tooz-python35/9fa0422/ : SUCCESS in 1m 35s\n- gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial/a33ff98/ : SUCCESS in 1h 19m 30s\n- gate-tooz-releasenotes http://docs-draft.openstack.org/98/466098/4/check/gate-tooz-releasenotes/a4549bc//releasenotes/build/html/ : SUCCESS in 1m 10s\n- gate-tooz-tox-py27-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-etcd-ubuntu-xenial/f9cbdea/ : SUCCESS in 3m 06s\n- gate-tooz-tox-py35-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-etcd-ubuntu-xenial/a1190b5/ : SUCCESS in 4m 16s\n- gate-tooz-tox-py27-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-etcd3-ubuntu-xenial/957e3dd/ : SUCCESS in 2m 01s\n- gate-tooz-tox-py35-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-etcd3-ubuntu-xenial/59e33b0/ : SUCCESS in 1m 29s\n- gate-tooz-tox-py27-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-zookeeper-ubuntu-xenial/18d8d21/ : SUCCESS in 2m 10s\n- gate-tooz-tox-py35-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-zookeeper-ubuntu-xenial/4b6a139/ : SUCCESS in 2m 08s\n- gate-tooz-tox-py27-redis-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-redis-ubuntu-xenial/de80203/ : SUCCESS in 1m 56s\n- gate-tooz-tox-py35-redis-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-redis-ubuntu-xenial/3eeef2c/ : SUCCESS in 3m 59s\n- gate-tooz-tox-py27-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-sentinel-ubuntu-xenial/747da13/ : SUCCESS in 1m 25s\n- gate-tooz-tox-py35-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-sentinel-ubuntu-xenial/e4c47c8/ : SUCCESS in 1m 32s\n- gate-tooz-tox-py27-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-memcached-ubuntu-xenial/7bebd44/ : SUCCESS in 1m 23s\n- gate-tooz-tox-py35-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-memcached-ubuntu-xenial/67c8bfd/ : SUCCESS in 2m 10s\n- gate-tooz-tox-py27-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-postgresql-ubuntu-xenial/6999819/ : SUCCESS in 3m 42s\n- gate-tooz-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-postgresql-ubuntu-xenial/f4f7576/ : SUCCESS in 2m 52s\n- gate-tooz-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-mysql-ubuntu-xenial/4219dd6/ : SUCCESS in 1m 17s\n- gate-tooz-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-mysql-ubuntu-xenial/ff477d6/ : SUCCESS in 2m 18s\n- gate-tooz-tox-py27-consul-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-consul-ubuntu-xenial/a839799/ : SUCCESS in 1m 23s\n- gate-tooz-tox-py35-consul-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-consul-ubuntu-xenial/ede49ed/ : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":4},{"id":"ed027d061e4e60f749fd55f7f8d9425d2adae1e7","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-19 21:05:41.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8d4fc6877293c5326f3254b889e016b3f252c920","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-19 22:30:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tooz-docs-ubuntu-xenial http://docs-draft.openstack.org/98/466098/4/check/gate-tooz-docs-ubuntu-xenial/07b3183//doc/build/html/ : SUCCESS in 2m 52s\n- gate-tooz-pep8-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-pep8-ubuntu-xenial/0e5245a/ : SUCCESS in 3m 28s\n- gate-tooz-python27-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-python27-ubuntu-xenial/7507ec8/ : SUCCESS in 3m 15s\n- gate-tooz-python35 http://logs.openstack.org/98/466098/4/check/gate-tooz-python35/de7cf25/ : FAILURE in 3m 18s\n- gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial/770d981/ : SUCCESS in 1h 16m 45s\n- gate-tooz-releasenotes http://docs-draft.openstack.org/98/466098/4/check/gate-tooz-releasenotes/31d785b//releasenotes/build/html/ : SUCCESS in 1m 29s\n- gate-tooz-tox-py27-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-etcd-ubuntu-xenial/2a522f5/ : SUCCESS in 2m 14s\n- gate-tooz-tox-py35-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-etcd-ubuntu-xenial/c1cdcd2/ : SUCCESS in 2m 39s\n- gate-tooz-tox-py27-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-etcd3-ubuntu-xenial/c3bf8aa/ : SUCCESS in 2m 39s\n- gate-tooz-tox-py35-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-etcd3-ubuntu-xenial/d118214/ : SUCCESS in 2m 42s\n- gate-tooz-tox-py27-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-zookeeper-ubuntu-xenial/f4a8afa/ : SUCCESS in 1m 53s\n- gate-tooz-tox-py35-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-zookeeper-ubuntu-xenial/3fcba4f/ : SUCCESS in 1m 54s\n- gate-tooz-tox-py27-redis-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-redis-ubuntu-xenial/6c43ec1/ : SUCCESS in 2m 14s\n- gate-tooz-tox-py35-redis-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-redis-ubuntu-xenial/ff2a2da/ : SUCCESS in 1m 41s\n- gate-tooz-tox-py27-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-sentinel-ubuntu-xenial/b55b573/ : SUCCESS in 1m 39s\n- gate-tooz-tox-py35-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-sentinel-ubuntu-xenial/6a0c10c/ : SUCCESS in 2m 07s\n- gate-tooz-tox-py27-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-memcached-ubuntu-xenial/f1914dc/ : SUCCESS in 2m 12s\n- gate-tooz-tox-py35-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-memcached-ubuntu-xenial/1ee0296/ : SUCCESS in 1m 40s\n- gate-tooz-tox-py27-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-postgresql-ubuntu-xenial/affaeae/ : SUCCESS in 2m 46s\n- gate-tooz-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-postgresql-ubuntu-xenial/92f12d9/ : SUCCESS in 3m 01s\n- gate-tooz-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-mysql-ubuntu-xenial/6035ddb/ : SUCCESS in 1m 34s\n- gate-tooz-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-mysql-ubuntu-xenial/404c189/ : SUCCESS in 2m 02s\n- gate-tooz-tox-py27-consul-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py27-consul-ubuntu-xenial/8b600b9/ : SUCCESS in 1m 24s\n- gate-tooz-tox-py35-consul-ubuntu-xenial http://logs.openstack.org/98/466098/4/check/gate-tooz-tox-py35-consul-ubuntu-xenial/9f70c6d/ : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":4},{"id":"77a41787c05b831ce348df2c37cdacd76059bcf2","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-19 23:10:32.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e1d38f466e1e579fa43f887ede2e958223ca7ad3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-20 00:26:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tooz-docs-ubuntu-xenial http://docs-draft.openstack.org/98/466098/5/check/gate-tooz-docs-ubuntu-xenial/5df31a7//doc/build/html/ : SUCCESS in 1m 19s\n- gate-tooz-pep8-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-pep8-ubuntu-xenial/607343b/ : SUCCESS in 2m 40s\n- gate-tooz-python27-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-python27-ubuntu-xenial/8853823/ : SUCCESS in 2m 35s\n- gate-tooz-python35 http://logs.openstack.org/98/466098/5/check/gate-tooz-python35/5ab4b2c/ : SUCCESS in 2m 35s\n- gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial/a04ceef/ : SUCCESS in 1h 14m 46s\n- gate-tooz-releasenotes http://docs-draft.openstack.org/98/466098/5/check/gate-tooz-releasenotes/107a8f6//releasenotes/build/html/ : SUCCESS in 2m 17s\n- gate-tooz-tox-py27-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-etcd-ubuntu-xenial/37806e5/ : SUCCESS in 1m 53s\n- gate-tooz-tox-py35-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-etcd-ubuntu-xenial/5526c40/ : SUCCESS in 1m 51s\n- gate-tooz-tox-py27-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-etcd3-ubuntu-xenial/d3101ba/ : FAILURE in 2m 43s\n- gate-tooz-tox-py35-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-etcd3-ubuntu-xenial/93f2403/ : SUCCESS in 2m 15s\n- gate-tooz-tox-py27-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-zookeeper-ubuntu-xenial/77563bf/ : SUCCESS in 3m 23s\n- gate-tooz-tox-py35-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-zookeeper-ubuntu-xenial/a5ed0fe/ : SUCCESS in 3m 06s\n- gate-tooz-tox-py27-redis-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-redis-ubuntu-xenial/4adc0f5/ : SUCCESS in 2m 36s\n- gate-tooz-tox-py35-redis-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-redis-ubuntu-xenial/bc715c0/ : SUCCESS in 2m 12s\n- gate-tooz-tox-py27-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-sentinel-ubuntu-xenial/bf4e6e9/ : SUCCESS in 1m 30s\n- gate-tooz-tox-py35-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-sentinel-ubuntu-xenial/67d41df/ : SUCCESS in 2m 37s\n- gate-tooz-tox-py27-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-memcached-ubuntu-xenial/07a8595/ : SUCCESS in 2m 34s\n- gate-tooz-tox-py35-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-memcached-ubuntu-xenial/c2fb6b4/ : SUCCESS in 2m 04s\n- gate-tooz-tox-py27-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-postgresql-ubuntu-xenial/efc39f0/ : SUCCESS in 1m 30s\n- gate-tooz-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-postgresql-ubuntu-xenial/d4afb85/ : SUCCESS in 2m 21s\n- gate-tooz-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-mysql-ubuntu-xenial/453ded4/ : SUCCESS in 2m 11s\n- gate-tooz-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-mysql-ubuntu-xenial/8926d31/ : SUCCESS in 2m 49s\n- gate-tooz-tox-py27-consul-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-consul-ubuntu-xenial/e66c0df/ : SUCCESS in 2m 31s\n- gate-tooz-tox-py35-consul-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-consul-ubuntu-xenial/3d4ec38/ : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":5},{"id":"95e670cdbfb72731bd0f8433d9174af50c515b2e","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-20 00:47:02.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"45ff762b3129d08986483a6a51a05787d3788f3f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-20 01:59:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tooz-docs-ubuntu-xenial http://docs-draft.openstack.org/98/466098/5/check/gate-tooz-docs-ubuntu-xenial/f05d8f3//doc/build/html/ : SUCCESS in 1m 41s\n- gate-tooz-pep8-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-pep8-ubuntu-xenial/58ba352/ : SUCCESS in 1m 53s\n- gate-tooz-python27-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-python27-ubuntu-xenial/7002fb0/ : SUCCESS in 2m 06s\n- gate-tooz-python35 http://logs.openstack.org/98/466098/5/check/gate-tooz-python35/e5c1347/ : SUCCESS in 2m 12s\n- gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial/be0e604/ : SUCCESS in 1h 11m 41s\n- gate-tooz-releasenotes http://docs-draft.openstack.org/98/466098/5/check/gate-tooz-releasenotes/f403971//releasenotes/build/html/ : SUCCESS in 54s\n- gate-tooz-tox-py27-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-etcd-ubuntu-xenial/81f6795/ : SUCCESS in 1m 23s\n- gate-tooz-tox-py35-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-etcd-ubuntu-xenial/dc055b8/ : SUCCESS in 1m 27s\n- gate-tooz-tox-py27-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-etcd3-ubuntu-xenial/4111f76/ : SUCCESS in 1m 55s\n- gate-tooz-tox-py35-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-etcd3-ubuntu-xenial/5608a34/ : SUCCESS in 1m 23s\n- gate-tooz-tox-py27-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-zookeeper-ubuntu-xenial/2d89da0/ : SUCCESS in 1m 26s\n- gate-tooz-tox-py35-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-zookeeper-ubuntu-xenial/23915fa/ : SUCCESS in 1m 29s\n- gate-tooz-tox-py27-redis-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-redis-ubuntu-xenial/69f6128/ : SUCCESS in 1m 20s\n- gate-tooz-tox-py35-redis-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-redis-ubuntu-xenial/b016bb2/ : SUCCESS in 1m 22s\n- gate-tooz-tox-py27-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-sentinel-ubuntu-xenial/712cdf6/ : SUCCESS in 1m 22s\n- gate-tooz-tox-py35-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-sentinel-ubuntu-xenial/a5ef099/ : SUCCESS in 4m 09s\n- gate-tooz-tox-py27-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-memcached-ubuntu-xenial/075c40c/ : SUCCESS in 2m 10s\n- gate-tooz-tox-py35-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-memcached-ubuntu-xenial/892874e/ : SUCCESS in 2m 09s\n- gate-tooz-tox-py27-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-postgresql-ubuntu-xenial/5ae5a26/ : SUCCESS in 2m 14s\n- gate-tooz-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-postgresql-ubuntu-xenial/a99fc2a/ : SUCCESS in 1m 40s\n- gate-tooz-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-mysql-ubuntu-xenial/b14f67e/ : SUCCESS in 2m 04s\n- gate-tooz-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-mysql-ubuntu-xenial/9610dc4/ : SUCCESS in 1m 59s\n- gate-tooz-tox-py27-consul-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py27-consul-ubuntu-xenial/3bdba52/ : SUCCESS in 1m 56s\n- gate-tooz-tox-py35-consul-ubuntu-xenial http://logs.openstack.org/98/466098/5/check/gate-tooz-tox-py35-consul-ubuntu-xenial/ae85d20/ : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":5},{"id":"3426ea28f7e78d8a1c33d7021b340401b2cf6a61","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-05-20 16:16:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e853f34528cc24aa29f267f1e696adae5f3fe88b","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-20 21:21:22.000000000","message":"Patch Set 6: Published edit on patch set 5","accounts_in_message":[],"_revision_number":6},{"id":"1d370dba850498ab971ac2f21fa23804af7a5542","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-20 21:21:53.000000000","message":"Patch Set 7: Published edit on patch set 6","accounts_in_message":[],"_revision_number":7},{"id":"7c307c1cf3bd90bf1c25c1814e4aa3d35604a5db","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-20 22:39:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tooz-docs-ubuntu-xenial http://docs-draft.openstack.org/98/466098/7/check/gate-tooz-docs-ubuntu-xenial/a9e55f4//doc/build/html/ : SUCCESS in 1m 31s\n- gate-tooz-pep8-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-pep8-ubuntu-xenial/3f57c4d/ : SUCCESS in 1m 49s\n- gate-tooz-python27-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-python27-ubuntu-xenial/227e175/ : SUCCESS in 1m 30s\n- gate-tooz-python35 http://logs.openstack.org/98/466098/7/check/gate-tooz-python35/81aded1/ : SUCCESS in 2m 17s\n- gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial/d6796a6/ : SUCCESS in 1h 17m 07s\n- gate-tooz-releasenotes http://docs-draft.openstack.org/98/466098/7/check/gate-tooz-releasenotes/44edff3//releasenotes/build/html/ : SUCCESS in 57s\n- gate-tooz-tox-py27-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py27-etcd-ubuntu-xenial/2f6f002/ : SUCCESS in 1m 21s\n- gate-tooz-tox-py35-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py35-etcd-ubuntu-xenial/7fd180b/ : SUCCESS in 1m 19s\n- gate-tooz-tox-py27-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py27-etcd3-ubuntu-xenial/64df53a/ : SUCCESS in 1m 22s\n- gate-tooz-tox-py35-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py35-etcd3-ubuntu-xenial/cc0f052/ : SUCCESS in 1m 20s\n- gate-tooz-tox-py27-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py27-zookeeper-ubuntu-xenial/6488bf2/ : SUCCESS in 1m 45s\n- gate-tooz-tox-py35-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py35-zookeeper-ubuntu-xenial/4585e90/ : SUCCESS in 1m 18s\n- gate-tooz-tox-py27-redis-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py27-redis-ubuntu-xenial/5c28ab5/ : SUCCESS in 1m 20s\n- gate-tooz-tox-py35-redis-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py35-redis-ubuntu-xenial/c6d65da/ : SUCCESS in 1m 19s\n- gate-tooz-tox-py27-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py27-sentinel-ubuntu-xenial/9388db7/ : SUCCESS in 1m 54s\n- gate-tooz-tox-py35-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py35-sentinel-ubuntu-xenial/6b17b97/ : SUCCESS in 2m 02s\n- gate-tooz-tox-py27-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py27-memcached-ubuntu-xenial/5cfc552/ : SUCCESS in 2m 43s\n- gate-tooz-tox-py35-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py35-memcached-ubuntu-xenial/e138b67/ : SUCCESS in 4m 41s\n- gate-tooz-tox-py27-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py27-postgresql-ubuntu-xenial/cbeb3fc/ : SUCCESS in 1m 19s\n- gate-tooz-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py35-postgresql-ubuntu-xenial/e877823/ : SUCCESS in 1m 20s\n- gate-tooz-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py27-mysql-ubuntu-xenial/d45fa4c/ : SUCCESS in 2m 06s\n- gate-tooz-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py35-mysql-ubuntu-xenial/30b95d6/ : SUCCESS in 2m 35s\n- gate-tooz-tox-py27-consul-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py27-consul-ubuntu-xenial/0d43f6d/ : SUCCESS in 1m 48s\n- gate-tooz-tox-py35-consul-ubuntu-xenial http://logs.openstack.org/98/466098/7/check/gate-tooz-tox-py35-consul-ubuntu-xenial/3035b51/ : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":7},{"id":"f0f293c11782858d66cebff2dac3b6b1c464e3d4","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-05-22 12:15:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d34c1280fbcd18603a776663057fbbd6ed115409","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-05-22 12:16:14.000000000","message":"Patch Set 8: Code-Review+2\n\nsimplified encoding to be done once only and avoiding puttin utf8 everywhere :D\n\ntested locally and it works","accounts_in_message":[],"_revision_number":8},{"id":"cd18f77aeb6e9f49c3f8f99d12f222e1f3f130a3","author":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"date":"2017-05-22 12:16:21.000000000","message":"Patch Set 7:\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ae4c797d7adb8019b6060beb7fac42f02978244b","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-05-22 12:16:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7bf162a5b0190a327084d0df754f798f2c0169c9","author":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"date":"2017-05-22 12:17:14.000000000","message":"Patch Set 8: Code-Review-1\n\nSome comment on the previous patch if you don\u0027t mind.","accounts_in_message":[],"_revision_number":8},{"id":"c7732c85879bc2f9cda4849c23b43aab6d3cae92","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-05-22 12:25:28.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"241fa2fd7c348dd6460875f5100baea881feedf6","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-05-22 12:26:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4f6613acec7277594da286237b873eb8e6f8140c","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-05-22 12:26:29.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"0ab8aa605f1a768380b939be2fceafc723765eff","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-05-22 12:27:20.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"9b4a53856a542b6591c11a1b9dbcbdbd01b6b275","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2017-05-22 12:33:54.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"09c784c1620fdb03cdc58a88e3df9d1b45919c5b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-22 13:57:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tooz-docs-ubuntu-xenial http://docs-draft.openstack.org/98/466098/11/check/gate-tooz-docs-ubuntu-xenial/72e08fe//doc/build/html/ : SUCCESS in 1m 19s\n- gate-tooz-pep8-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-pep8-ubuntu-xenial/b5233bc/ : SUCCESS in 1m 57s\n- gate-tooz-python27-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-python27-ubuntu-xenial/416724f/ : SUCCESS in 3m 53s\n- gate-tooz-python35 http://logs.openstack.org/98/466098/11/check/gate-tooz-python35/65b2012/ : SUCCESS in 2m 28s\n- gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial/e5da691/ : SUCCESS in 1h 21m 09s\n- gate-tooz-releasenotes http://docs-draft.openstack.org/98/466098/11/check/gate-tooz-releasenotes/ce4f89b//releasenotes/build/html/ : SUCCESS in 2m 24s\n- gate-tooz-tox-py27-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py27-etcd-ubuntu-xenial/1407221/ : SUCCESS in 2m 08s\n- gate-tooz-tox-py35-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py35-etcd-ubuntu-xenial/7d66fcb/ : SUCCESS in 2m 05s\n- gate-tooz-tox-py27-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py27-etcd3-ubuntu-xenial/57987f4/ : SUCCESS in 1m 36s\n- gate-tooz-tox-py35-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py35-etcd3-ubuntu-xenial/246fc59/ : SUCCESS in 1m 33s\n- gate-tooz-tox-py27-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py27-zookeeper-ubuntu-xenial/69d7ecf/ : SUCCESS in 2m 04s\n- gate-tooz-tox-py35-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py35-zookeeper-ubuntu-xenial/a58ba52/ : SUCCESS in 2m 15s\n- gate-tooz-tox-py27-redis-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py27-redis-ubuntu-xenial/9b9dc29/ : SUCCESS in 1m 30s\n- gate-tooz-tox-py35-redis-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py35-redis-ubuntu-xenial/40b2ed8/ : SUCCESS in 1m 28s\n- gate-tooz-tox-py27-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py27-sentinel-ubuntu-xenial/efbb587/ : SUCCESS in 2m 08s\n- gate-tooz-tox-py35-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py35-sentinel-ubuntu-xenial/0afd3ad/ : SUCCESS in 1m 45s\n- gate-tooz-tox-py27-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py27-memcached-ubuntu-xenial/967efc0/ : SUCCESS in 1m 32s\n- gate-tooz-tox-py35-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py35-memcached-ubuntu-xenial/45df929/ : SUCCESS in 2m 37s\n- gate-tooz-tox-py27-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py27-postgresql-ubuntu-xenial/7600902/ : SUCCESS in 2m 43s\n- gate-tooz-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py35-postgresql-ubuntu-xenial/1edf159/ : SUCCESS in 2m 17s\n- gate-tooz-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py27-mysql-ubuntu-xenial/3227408/ : SUCCESS in 1m 30s\n- gate-tooz-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py35-mysql-ubuntu-xenial/cbeb57e/ : SUCCESS in 1m 59s\n- gate-tooz-tox-py27-consul-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py27-consul-ubuntu-xenial/4419fcc/ : SUCCESS in 2m 14s\n- gate-tooz-tox-py35-consul-ubuntu-xenial http://logs.openstack.org/98/466098/11/check/gate-tooz-tox-py35-consul-ubuntu-xenial/83df664/ : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":11},{"id":"e95329b023d55243038c0ef513707f72e3ca35bc","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2017-05-22 14:00:07.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"5e40be39414e7a8a03049ef1e1ce50b8a74ee6a7","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2017-05-22 19:16:04.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nThis appears to do the same thing as the other etcd3 driver, using the different connection mechanism.","accounts_in_message":[],"_revision_number":11},{"id":"4aa8cac9c552f0e4050d73d600c6a9b697b9ff91","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-22 20:16:33.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":11},{"id":"05374a1b175b15c04f4388662e148a6b8fdfbb99","date":"2017-05-22 21:37:20.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":11},{"id":"2fcd55eb89c1439ec5f09619c4fd5f1579bea4fa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-05-22 21:37:20.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tooz-docs-ubuntu-xenial http://docs-draft.openstack.org/98/466098/11/gate/gate-tooz-docs-ubuntu-xenial/acaa991//doc/build/html/ : SUCCESS in 1m 13s\n- gate-tooz-pep8-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-pep8-ubuntu-xenial/08520e5/ : SUCCESS in 2m 27s\n- gate-tooz-python27-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-python27-ubuntu-xenial/10ce6d7/ : SUCCESS in 1m 48s\n- gate-tooz-python35 http://logs.openstack.org/98/466098/11/gate/gate-tooz-python35/e792ac6/ : SUCCESS in 2m 27s\n- gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tempest-dsvm-neutron-src-tooz-ubuntu-xenial/e333e3d/ : SUCCESS in 1h 15m 53s\n- gate-tooz-releasenotes http://docs-draft.openstack.org/98/466098/11/gate/gate-tooz-releasenotes/596ca28//releasenotes/build/html/ : SUCCESS in 1m 40s\n- gate-tooz-tox-py27-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py27-etcd-ubuntu-xenial/1c50854/ : SUCCESS in 2m 34s\n- gate-tooz-tox-py35-etcd-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py35-etcd-ubuntu-xenial/9302e63/ : SUCCESS in 4m 26s\n- gate-tooz-tox-py27-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py27-etcd3-ubuntu-xenial/bc0baa4/ : SUCCESS in 1m 36s\n- gate-tooz-tox-py35-etcd3-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py35-etcd3-ubuntu-xenial/4b5d3b4/ : SUCCESS in 2m 28s\n- gate-tooz-tox-py27-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py27-zookeeper-ubuntu-xenial/11ad941/ : SUCCESS in 3m 51s\n- gate-tooz-tox-py35-zookeeper-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py35-zookeeper-ubuntu-xenial/976c2ed/ : SUCCESS in 1m 42s\n- gate-tooz-tox-py27-redis-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py27-redis-ubuntu-xenial/04283fb/ : SUCCESS in 3m 08s\n- gate-tooz-tox-py35-redis-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py35-redis-ubuntu-xenial/121bcb5/ : SUCCESS in 3m 27s\n- gate-tooz-tox-py27-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py27-sentinel-ubuntu-xenial/ae9870d/ : SUCCESS in 3m 19s\n- gate-tooz-tox-py35-sentinel-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py35-sentinel-ubuntu-xenial/b706dc2/ : SUCCESS in 2m 48s\n- gate-tooz-tox-py27-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py27-memcached-ubuntu-xenial/2c0b9e1/ : SUCCESS in 4m 42s\n- gate-tooz-tox-py35-memcached-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py35-memcached-ubuntu-xenial/ae8ecd4/ : SUCCESS in 3m 21s\n- gate-tooz-tox-py27-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py27-postgresql-ubuntu-xenial/d57290f/ : SUCCESS in 4m 37s\n- gate-tooz-tox-py35-postgresql-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py35-postgresql-ubuntu-xenial/3a88964/ : SUCCESS in 2m 54s\n- gate-tooz-tox-py27-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py27-mysql-ubuntu-xenial/15a25d2/ : SUCCESS in 2m 52s\n- gate-tooz-tox-py35-mysql-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py35-mysql-ubuntu-xenial/1e56cf0/ : SUCCESS in 2m 55s\n- gate-tooz-tox-py27-consul-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py27-consul-ubuntu-xenial/a70e9f8/ : SUCCESS in 2m 28s\n- gate-tooz-tox-py35-consul-ubuntu-xenial http://logs.openstack.org/98/466098/11/gate/gate-tooz-tox-py35-consul-ubuntu-xenial/13dcea8/ : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"6ec73a0bed7382790819151a0a84db2a8a464438","revisions":{"bfe16c995a98a537f25cdb77d8a687adf9ccd6e4":{"kind":"REWORK","_number":1,"created":"2017-05-18 20:24:27.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/98/466098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/1"}}},"commit":{"parents":[{"commit":"9eb820bfde896cef92ca79b2c6e1434023d0cb6c","subject":"etcd3: fix test run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9eb820bfde896cef92ca79b2c6e1434023d0cb6c"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"subject":"[WIP] Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"[WIP] Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bfe16c995a98a537f25cdb77d8a687adf9ccd6e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bfe16c995a98a537f25cdb77d8a687adf9ccd6e4"}]},"branch":"refs/heads/master"},"8019910e98370bfddc4780271b79f75b3c9beda2":{"kind":"REWORK","_number":2,"created":"2017-05-18 20:38:44.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/98/466098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/2"}}},"commit":{"parents":[{"commit":"9eb820bfde896cef92ca79b2c6e1434023d0cb6c","subject":"etcd3: fix test run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9eb820bfde896cef92ca79b2c6e1434023d0cb6c"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:39:13.000000000","tz":-240},"subject":"[WIP] Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"[WIP] Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8019910e98370bfddc4780271b79f75b3c9beda2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8019910e98370bfddc4780271b79f75b3c9beda2"}]},"branch":"refs/heads/master"},"e41253bc8b2f9b55892667eeb1f6c7a778fb91b4":{"kind":"REWORK","_number":3,"created":"2017-05-19 17:21:01.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/98/466098/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/3"}}},"commit":{"parents":[{"commit":"9eb820bfde896cef92ca79b2c6e1434023d0cb6c","subject":"etcd3: fix test run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9eb820bfde896cef92ca79b2c6e1434023d0cb6c"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-19 17:21:19.000000000","tz":-240},"subject":"[WIP] Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"[WIP] Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nDepends-On: Ia187e1a86413edf25b909b6bb57e84fb4930a696\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e41253bc8b2f9b55892667eeb1f6c7a778fb91b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e41253bc8b2f9b55892667eeb1f6c7a778fb91b4"}]},"branch":"refs/heads/master"},"5fbe12b90f8ea37eca16ac466ae25188b1036a47":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2017-05-19 19:00:45.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/98/466098/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/4"}}},"commit":{"parents":[{"commit":"1a57bbcc24021a6c237cad6a26f5035f0fc8fa92","subject":"etcd: fix acquire(blocking\u003dTrue) on request exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a57bbcc24021a6c237cad6a26f5035f0fc8fa92"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Davanum Srinivas (dims)","email":"davanum@gmail.com","date":"2017-05-19 19:00:45.000000000","tz":0},"subject":"[WIP] Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"[WIP] Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nDepends-On: Ia187e1a86413edf25b909b6bb57e84fb4930a696\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5fbe12b90f8ea37eca16ac466ae25188b1036a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5fbe12b90f8ea37eca16ac466ae25188b1036a47"}]},"branch":"refs/heads/master"},"fd9a0e88bca84870e38c88b47db288dda3b1f51c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2017-05-19 23:10:32.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/98/466098/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/5"}}},"commit":{"parents":[{"commit":"1a57bbcc24021a6c237cad6a26f5035f0fc8fa92","subject":"etcd: fix acquire(blocking\u003dTrue) on request exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a57bbcc24021a6c237cad6a26f5035f0fc8fa92"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-19 23:06:51.000000000","tz":-240},"subject":"Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nUsing grpc to directly access etcd3 does not work well with\neventlet based services. So we need to use the grpc HTTP\ngateway API (/v3alpha).\nhttps://coreos.com/etcd/docs/latest/dev-guide/api_grpc_gateway.html\n\nFor this we use the etcd3gw package:\nhttps://pypi.python.org/pypi/etcd3gw\n\nThe structure of the code is very similar to the one we\nalready have that uses the etcd3 package.\n\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fd9a0e88bca84870e38c88b47db288dda3b1f51c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fd9a0e88bca84870e38c88b47db288dda3b1f51c"}]},"branch":"refs/heads/master"},"97ef0dd01ef223083e9e74a5fb88ecd8327b140b":{"kind":"REWORK","_number":6,"created":"2017-05-20 21:21:22.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/98/466098/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/6"}}},"commit":{"parents":[{"commit":"1a57bbcc24021a6c237cad6a26f5035f0fc8fa92","subject":"etcd: fix acquire(blocking\u003dTrue) on request exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a57bbcc24021a6c237cad6a26f5035f0fc8fa92"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Davanum Srinivas (dims)","email":"davanum@gmail.com","date":"2017-05-20 21:21:16.000000000","tz":0},"subject":"Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nUsing grpc to directly access etcd3 does not work well with\neventlet based services. So we need to use the grpc HTTP\ngateway API (/v3alpha).\nhttps://coreos.com/etcd/docs/latest/dev-guide/api_grpc_gateway.html\n\nFor this we use the etcd3gw package:\nhttps://pypi.python.org/pypi/etcd3gw\n\nThe structure of the code is very similar to the one we\nalready have that uses the etcd3 package.\n\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/97ef0dd01ef223083e9e74a5fb88ecd8327b140b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/97ef0dd01ef223083e9e74a5fb88ecd8327b140b"}]},"branch":"refs/heads/master"},"bba051e5094b4a5d0d9f1235995fd432aade0cce":{"kind":"REWORK","_number":7,"created":"2017-05-20 21:21:53.000000000","uploader":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"ref":"refs/changes/98/466098/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/7"}}},"commit":{"parents":[{"commit":"1a57bbcc24021a6c237cad6a26f5035f0fc8fa92","subject":"etcd: fix acquire(blocking\u003dTrue) on request exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a57bbcc24021a6c237cad6a26f5035f0fc8fa92"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Davanum Srinivas (dims)","email":"davanum@gmail.com","date":"2017-05-20 21:21:50.000000000","tz":0},"subject":"Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nUsing grpc to directly access etcd3 does not work well with\neventlet based services. So we need to use the grpc HTTP\ngateway API (/v3alpha).\nhttps://coreos.com/etcd/docs/latest/dev-guide/api_grpc_gateway.html\n\nFor this we use the etcd3gw package:\nhttps://pypi.python.org/pypi/etcd3gw\n\nThe structure of the code is very similar to the one we\nalready have that uses the etcd3 package.\n\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bba051e5094b4a5d0d9f1235995fd432aade0cce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bba051e5094b4a5d0d9f1235995fd432aade0cce"}]},"branch":"refs/heads/master"},"66a5a3577c3edf13ec91e07a87cd7911e1e84a51":{"kind":"REWORK","_number":8,"created":"2017-05-22 12:15:44.000000000","uploader":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"ref":"refs/changes/98/466098/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/8"}}},"commit":{"parents":[{"commit":"1a57bbcc24021a6c237cad6a26f5035f0fc8fa92","subject":"etcd: fix acquire(blocking\u003dTrue) on request exception","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a57bbcc24021a6c237cad6a26f5035f0fc8fa92"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2017-05-22 12:14:45.000000000","tz":120},"subject":"Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nUsing grpc to directly access etcd3 does not work well with\neventlet based services. So we need to use the grpc HTTP\ngateway API (/v3alpha).\nhttps://coreos.com/etcd/docs/latest/dev-guide/api_grpc_gateway.html\n\nFor this we use the etcd3gw package:\nhttps://pypi.python.org/pypi/etcd3gw\n\nThe structure of the code is very similar to the one we\nalready have that uses the etcd3 package.\n\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/66a5a3577c3edf13ec91e07a87cd7911e1e84a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/66a5a3577c3edf13ec91e07a87cd7911e1e84a51"}]},"branch":"refs/heads/master"},"b07e0f4cf7e03b4211522103980c30b3e9476925":{"kind":"REWORK","_number":9,"created":"2017-05-22 12:26:10.000000000","uploader":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"ref":"refs/changes/98/466098/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/9"}}},"commit":{"parents":[{"commit":"ae1e42274b9d0b4f517c7cfae81cdf9a181a4101","subject":"etcd: fix blocking argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ae1e42274b9d0b4f517c7cfae81cdf9a181a4101"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2017-05-22 12:25:40.000000000","tz":120},"subject":"Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nUsing grpc to directly access etcd3 does not work well with\neventlet based services. So we need to use the grpc HTTP\ngateway API (/v3alpha).\nhttps://coreos.com/etcd/docs/latest/dev-guide/api_grpc_gateway.html\n\nFor this we use the etcd3gw package:\nhttps://pypi.python.org/pypi/etcd3gw\n\nThe structure of the code is very similar to the one we\nalready have that uses the etcd3 package.\n\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b07e0f4cf7e03b4211522103980c30b3e9476925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b07e0f4cf7e03b4211522103980c30b3e9476925"}]},"branch":"refs/heads/master"},"dc7a02cddec4621373b4157f8186fc68dc105930":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2017-05-22 12:27:20.000000000","uploader":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"ref":"refs/changes/98/466098/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/10"}}},"commit":{"parents":[{"commit":"24101cf2643e6e8226a99988989ed37371a830f1","subject":"etcd3: use discard() rather than remove()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/24101cf2643e6e8226a99988989ed37371a830f1"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2017-05-22 12:27:09.000000000","tz":120},"subject":"Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nUsing grpc to directly access etcd3 does not work well with\neventlet based services. So we need to use the grpc HTTP\ngateway API (/v3alpha).\nhttps://coreos.com/etcd/docs/latest/dev-guide/api_grpc_gateway.html\n\nFor this we use the etcd3gw package:\nhttps://pypi.python.org/pypi/etcd3gw\n\nThe structure of the code is very similar to the one we\nalready have that uses the etcd3 package.\n\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dc7a02cddec4621373b4157f8186fc68dc105930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dc7a02cddec4621373b4157f8186fc68dc105930"}]},"branch":"refs/heads/master"},"6ec73a0bed7382790819151a0a84db2a8a464438":{"kind":"REWORK","_number":11,"created":"2017-05-22 12:33:54.000000000","uploader":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"ref":"refs/changes/98/466098/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/98/466098/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/98/466098/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/98/466098/11"}}},"commit":{"parents":[{"commit":"24101cf2643e6e8226a99988989ed37371a830f1","subject":"etcd3: use discard() rather than remove()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/24101cf2643e6e8226a99988989ed37371a830f1"}]}],"author":{"name":"Davanum Srinivas","email":"davanum@gmail.com","date":"2017-05-18 20:24:54.000000000","tz":-240},"committer":{"name":"Julien Danjou","email":"julien@danjou.info","date":"2017-05-22 12:33:44.000000000","tz":120},"subject":"Separate etcd3gw driver that uses the etcd3 grpc gateway","message":"Separate etcd3gw driver that uses the etcd3 grpc gateway\n\nUsing grpc to directly access etcd3 does not work well with\neventlet based services. So we need to use the grpc HTTP\ngateway API (/v3alpha).\nhttps://coreos.com/etcd/docs/latest/dev-guide/api_grpc_gateway.html\n\nFor this we use the etcd3gw package:\nhttps://pypi.python.org/pypi/etcd3gw\n\nThe structure of the code is very similar to the one we\nalready have that uses the etcd3 package.\n\nChange-Id: I97bd7ffb05a7e40cb08c9b9d62cc45236ad292aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6ec73a0bed7382790819151a0a84db2a8a464438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6ec73a0bed7382790819151a0a84db2a8a464438"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
