)]}'
{"id":"openstack%2Ftooz~636038","triplet_id":"openstack%2Ftooz~master~Ie23d88f1a05983d7465f18a329cabc23d789a628","project":"openstack/tooz","branch":"master","topic":"bug/1815330","hashtags":[],"change_id":"Ie23d88f1a05983d7465f18a329cabc23d789a628","subject":"RedisLock release() should not check if the lock has been acquired","status":"MERGED","created":"2019-02-10 03:27:48.000000000","updated":"2019-12-23 16:52:56.000000000","submitted":"2019-12-23 16:51:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"636038-1577119899883-71032318","meta_rev_id":"53983569ba10dc4b3e1287fd1ab74a358b6e78ca","_number":636038,"virtual_id_number":636038,"owner":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-23 16:51:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-12-19 09:12:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":2,"date":"2019-12-19 09:12:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":1,"date":"2019-12-19 09:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-11 08:33:49.000000000","updated_by":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"reviewer":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"state":"REVIEWER"},{"updated":"2019-02-12 16:06:18.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-12-19 09:12:38.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-23 16:51:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0fa38c1d0951490f57b84a6640890cb71607128d","author":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"date":"2019-02-10 03:27:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43c8f4b8534771a04af7f3e4f553e42b71bdbae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-10 05:10:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/38/636038/1/check/tempest-full/e1be182/ : SUCCESS in 1h 40m 39s\n- tempest-full-py3 http://logs.openstack.org/38/636038/1/check/tempest-full-py3/0d82626/ : SUCCESS in 1h 34m 29s\n- openstack-tox-cover http://logs.openstack.org/38/636038/1/check/openstack-tox-cover/40353fd/cover/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/38/636038/1/check/openstack-tox-pep8/642e1d4/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/38/636038/1/check/openstack-tox-py27/9035f62/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/38/636038/1/check/openstack-tox-py35/8525473/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/38/636038/1/check/openstack-tox-py36/9e62ebe/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/38/636038/1/check/openstack-tox-docs/b0268f9/html/ : SUCCESS in 4m 32s\n- tooz-tox-py27-etcd http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-etcd/9b0e25e/ : SUCCESS in 4m 50s\n- tooz-tox-py35-etcd http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-etcd/ba7c7fb/ : SUCCESS in 4m 55s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-etcd3/2b3789b/ : TIMED_OUT in 30m 54s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-etcd3/beb2d48/ : SUCCESS in 4m 46s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-etcd3gw/6ca9150/ : SUCCESS in 4m 18s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-etcd3gw/b45f2ba/ : SUCCESS in 3m 23s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-zookeeper/7812934/ : SUCCESS in 4m 06s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-zookeeper/543de86/ : SUCCESS in 5m 17s\n- tooz-tox-py27-redis http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-redis/b7097fc/ : SUCCESS in 4m 04s\n- tooz-tox-py35-redis http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-redis/d8d5a13/ : SUCCESS in 4m 45s\n- tooz-tox-py27-sentinel http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-sentinel/193258b/ : SUCCESS in 5m 16s\n- tooz-tox-py35-sentinel http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-sentinel/e3366a9/ : SUCCESS in 4m 12s\n- tooz-tox-py27-memcached http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-memcached/bb3b92f/ : SUCCESS in 4m 22s\n- tooz-tox-py35-memcached http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-memcached/57ed26c/ : SUCCESS in 4m 13s\n- tooz-tox-py27-postgresql http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-postgresql/e497219/ : SUCCESS in 3m 36s\n- tooz-tox-py35-postgresql http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-postgresql/5f73689/ : SUCCESS in 3m 16s\n- tooz-tox-py27-mysql http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-mysql/91843b6/ : SUCCESS in 3m 30s\n- tooz-tox-py35-mysql http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-mysql/748c66b/ : SUCCESS in 4m 23s\n- tooz-tox-py27-consul http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-consul/38b67a0/ : SUCCESS in 4m 07s\n- tooz-tox-py35-consul http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-consul/41e035f/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"53ab336f8e9e6a672f873bc558902b0300d0eace","author":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"date":"2019-02-10 13:52:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"204bb29af43c79db94ade8b3913c6a2ec2e17899","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-10 15:30:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/38/636038/1/check/tempest-full/e04ef6f/ : SUCCESS in 1h 36m 45s\n- tempest-full-py3 http://logs.openstack.org/38/636038/1/check/tempest-full-py3/940e9a6/ : SUCCESS in 1h 25m 34s\n- openstack-tox-cover http://logs.openstack.org/38/636038/1/check/openstack-tox-cover/8960f70/cover/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/38/636038/1/check/openstack-tox-pep8/2c22fd1/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/38/636038/1/check/openstack-tox-py27/83f856f/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/38/636038/1/check/openstack-tox-py35/fefb7d1/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/38/636038/1/check/openstack-tox-py36/b6a1eda/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/38/636038/1/check/openstack-tox-docs/2b6c5f5/html/ : SUCCESS in 3m 53s\n- tooz-tox-py27-etcd http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-etcd/9e0ffd4/ : SUCCESS in 4m 24s\n- tooz-tox-py35-etcd http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-etcd/ad21a36/ : SUCCESS in 4m 21s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-etcd3/a5fde87/ : SUCCESS in 4m 06s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-etcd3/1f6566e/ : SUCCESS in 4m 13s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-etcd3gw/5282634/ : SUCCESS in 4m 00s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-etcd3gw/fbafa2e/ : SUCCESS in 4m 04s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-zookeeper/81b377c/ : SUCCESS in 3m 50s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-zookeeper/5f693ce/ : SUCCESS in 4m 14s\n- tooz-tox-py27-redis http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-redis/c55daa9/ : SUCCESS in 3m 41s\n- tooz-tox-py35-redis http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-redis/e88c852/ : SUCCESS in 3m 48s\n- tooz-tox-py27-sentinel http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-sentinel/edb4699/ : SUCCESS in 3m 58s\n- tooz-tox-py35-sentinel http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-sentinel/78b65ab/ : SUCCESS in 3m 53s\n- tooz-tox-py27-memcached http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-memcached/47e0071/ : SUCCESS in 3m 43s\n- tooz-tox-py35-memcached http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-memcached/89ce529/ : SUCCESS in 4m 14s\n- tooz-tox-py27-postgresql http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-postgresql/e5bdddb/ : SUCCESS in 3m 38s\n- tooz-tox-py35-postgresql http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-postgresql/1777c5f/ : SUCCESS in 3m 47s\n- tooz-tox-py27-mysql http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-mysql/48f41c1/ : SUCCESS in 4m 04s\n- tooz-tox-py35-mysql http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-mysql/77b31ca/ : SUCCESS in 3m 48s\n- tooz-tox-py27-consul http://logs.openstack.org/38/636038/1/check/tooz-tox-py27-consul/7ec9d08/ : SUCCESS in 3m 40s\n- tooz-tox-py35-consul http://logs.openstack.org/38/636038/1/check/tooz-tox-py35-consul/6d93dc0/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"767e4d716070fb04f1fd6aaa42d24fd2866113c6","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2019-02-11 08:33:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82bbecf71a63022eeb4faaaa3e30638fa7dfbcdb","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2019-02-11 08:33:56.000000000","message":"Patch Set 1:\n\nHaving new tests would be better but ok","accounts_in_message":[],"_revision_number":1},{"id":"8b4930a480af4a9e7ae4415a5e2743aed20a3588","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-11 08:34:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a8a97eaf76bc8e6f6ae51c9c4f2f23fc70eb91e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-11 10:41:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/38/636038/1/gate/tempest-full/bb1a039/ : TIMED_OUT in 2h 05m 47s\n- tempest-full-py3 http://logs.openstack.org/38/636038/1/gate/tempest-full-py3/71e60af/ : SUCCESS in 1h 47m 59s\n- openstack-tox-pep8 http://logs.openstack.org/38/636038/1/gate/openstack-tox-pep8/e84dbbf/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/38/636038/1/gate/openstack-tox-py27/7f20b0b/ : SUCCESS in 6m 08s\n- openstack-tox-py35 http://logs.openstack.org/38/636038/1/gate/openstack-tox-py35/ee57be8/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/38/636038/1/gate/openstack-tox-py36/9eace85/ : SUCCESS in 6m 23s\n- openstack-tox-docs http://logs.openstack.org/38/636038/1/gate/openstack-tox-docs/00ed66b/html/ : SUCCESS in 6m 49s\n- tooz-tox-py27-etcd http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-etcd/ac2b0f8/ : SUCCESS in 7m 25s\n- tooz-tox-py35-etcd http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-etcd/43bd557/ : SUCCESS in 7m 25s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-etcd3/496d86f/ : SUCCESS in 7m 30s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-etcd3/5f3f2d4/ : SUCCESS in 6m 53s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-etcd3gw/b5cead1/ : SUCCESS in 6m 13s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-etcd3gw/8af6bfa/ : SUCCESS in 6m 26s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-zookeeper/9848399/ : SUCCESS in 6m 49s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-zookeeper/50503a6/ : SUCCESS in 6m 45s\n- tooz-tox-py27-redis http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-redis/0578445/ : SUCCESS in 6m 47s\n- tooz-tox-py35-redis http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-redis/8d0d6e3/ : SUCCESS in 6m 08s\n- tooz-tox-py27-sentinel http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-sentinel/79366ee/ : SUCCESS in 6m 18s\n- tooz-tox-py35-sentinel http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-sentinel/7b91d64/ : SUCCESS in 6m 44s\n- tooz-tox-py27-memcached http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-memcached/6ad649d/ : SUCCESS in 5m 44s\n- tooz-tox-py35-memcached http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-memcached/0481097/ : SUCCESS in 6m 16s\n- tooz-tox-py27-postgresql http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-postgresql/74d2ea5/ : SUCCESS in 6m 21s\n- tooz-tox-py35-postgresql http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-postgresql/18625d0/ : SUCCESS in 6m 28s\n- tooz-tox-py27-mysql http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-mysql/df3d0aa/ : SUCCESS in 6m 05s\n- tooz-tox-py35-mysql http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-mysql/1f1028f/ : SUCCESS in 5m 48s\n- tooz-tox-py27-consul http://logs.openstack.org/38/636038/1/gate/tooz-tox-py27-consul/b922dab/ : SUCCESS in 6m 21s\n- tooz-tox-py35-consul http://logs.openstack.org/38/636038/1/gate/tooz-tox-py35-consul/98944a0/ : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"70bacbf45bb49b2f2b44006aac3c0f4de2c93803","author":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"date":"2019-02-11 13:08:41.000000000","message":"Patch Set 1: Workflow-1\n\nI\u0027ll add a test for this case","accounts_in_message":[],"_revision_number":1},{"id":"0d6e7e6e435670c242dc32be074b2f2e44971613","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-02-12 16:06:18.000000000","message":"Patch Set 1:\n\nIn addition to a test to verify that _stop works correctly now, it would be good to verify that releasing an unacquired lock doesn\u0027t cause problems since that was the case this was guarding against before.","accounts_in_message":[],"_revision_number":1},{"id":"b3d13ce3bc6bb8201afc5500d546a5513448f1dc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-19 09:12:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bdeead46d45c14552a4b745296db6a65f484d2e9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-19 09:12:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1213458f95e88037f60fe1c674037c6178e72fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 10:50:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/98811a807f5447e59ef373255cec15f3 : SUCCESS in 1h 36m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8089694581ae4c4bbc14d6e2436a1a95 : SUCCESS in 1h 20m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b9567fa43364dd8b47aef62c4306a4d : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6701d6634847449ab4b312ff0c190247 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/759e7733bf674f51b187f7b86c393686 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e39e39386f7c4e9d9e4586130dff00a3 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03d5b91b7c8c42548bb7900d4d3bf1ea : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb190a1ed7964b0f98f252f56be17326 : SUCCESS in 5m 15s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/cb5e468cdddb45f7a3e1e58a1c4730f8 : SUCCESS in 7m 50s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/0257e332bffc4e148de69210e8863632 : SUCCESS in 6m 21s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/4811faed816d48d6a352d643ed4f9929 : SUCCESS in 7m 29s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/fc74551e869d47a0a4299510a04c77b7 : TIMED_OUT in 31m 14s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/ed7ccefc652c4a96a1969537f3ecf5fa : SUCCESS in 5m 57s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/72e81e15de7e479b976c4ab43f9943b4 : SUCCESS in 6m 59s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/cdbf666759b1454fae1aa5cb0a5707a1 : SUCCESS in 7m 32s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/4d3cd5e1be5a4476b559b40af07ba8e4 : SUCCESS in 4m 51s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/157eaa6586a0401ab37664787ce2ab99 : SUCCESS in 7m 34s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/47a7addf39dd495f8a9e6948650c76ef : SUCCESS in 5m 31s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/3c9b79777c9b4ba0889bf5b32a4c79db : SUCCESS in 6m 00s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/6cb4501d62004298913e4edacec84e04 : SUCCESS in 6m 06s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/f78aac0df0d44114a7c831bd8c3d92d8 : SUCCESS in 4m 01s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/15778afa3ac746c6ad1fe5fb0abd146c : SUCCESS in 7m 24s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/14c7886edbbc4716bb0cd19da685348a : SUCCESS in 7m 03s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/3df8610662894f2781f06fd2e60d5665 : SUCCESS in 4m 24s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/31fd1db7fd3243f09673647a91259dda : SUCCESS in 6m 07s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/4bd1b6508c894f58b2754dfed2978848 : SUCCESS in 4m 02s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/652bb600cc8f4bd7a3f1ba9fa959e2c0 : SUCCESS in 6m 32s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/dbc027a8f7f74010aa3e558f8228bb14 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f9baa8cd48597971e0d414d5209fe92c6f0ab9c3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:49:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"db7c5ae3c3bc9d8cd6aa9f9b3f70631caab9c54f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 14:50:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/897cc38886154407b5f9c7ed28067d34 : SUCCESS in 1h 57m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cba6b64f99b745f48d768b487926eb79 : SUCCESS in 1h 25m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebde01ca03e84005a32005e5430955c8 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf9b2ac79a948f380c632c3fafcbc6a : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92f4bf9a9d1b4be5953128d184601d68 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce4120c7edd54d6d8f0c263dd1fb83dd : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe594edbf2cd4db29aca73309b28e73a : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb2cf357d38f4ee7a70404fd5f48091e : SUCCESS in 4m 17s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/e156119226b54c3f914ca2652a9239b6 : SUCCESS in 7m 33s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/0149366d88104915b70162d12b44ff21 : SUCCESS in 4m 15s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/c2989249e6e9440ab56e724470490ba6 : SUCCESS in 7m 34s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/e49b273779e849289bed0b9dc1c2aeb9 : SUCCESS in 4m 34s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/b4aed70b183444dbb30c9e63a756eef6 : SUCCESS in 3m 46s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/5555175c50a5464a9bf80e6eb879712a : SUCCESS in 4m 24s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/cbabf7303f1e447aa99bd354c7963fb0 : SUCCESS in 4m 27s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/0940c1a82a8b4be7a671e17a3f1a918c : SUCCESS in 7m 34s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/69d2436b1feb45f3b71e0b3c676eeff7 : SUCCESS in 6m 59s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/c47b3b2cc6404a28b017cffeb91351c3 : SUCCESS in 7m 40s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/9007010c24c54760b20b0602ea73a6d0 : SUCCESS in 3m 59s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b1199a4d7f1e4e6fb1323524a85f2bec : SUCCESS in 7m 22s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/8bd0187168dc42e2a805b3fd2c4da3d6 : SUCCESS in 4m 17s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/31c871ffb635493f838be7eb40ca9d40 : SUCCESS in 6m 52s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/84e5fc9a5d8e416886766f79b9d08f50 : SUCCESS in 4m 00s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/13a4b8a8a7184c19b339f2b792d4fcc1 : SUCCESS in 6m 39s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/d46001b8c5104ff38ad12f8c8afccb1a : SUCCESS in 4m 00s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/e21d197207d94352b3c8db2d6ad0f039 : SUCCESS in 6m 23s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/4e03c9347415446d8be7fa44ad6b656e : SUCCESS in 6m 09s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/28e2644b4ff248f487f7340f6a8f4e46 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"b7a7fef178607fa53985f4d9598a2bc9470a89d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 14:50:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5a5b06079cbe6ca7ee42aab5dcb5f8acafaa67c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 16:51:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b4f69f3917d2dca8fcaca0142441236e10804eb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 16:51:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9a57a8d5e07d46a997fc6827349c5eb8 : SUCCESS in 1h 55m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34cea3914eea489ea06e77f5c9dacc82 : SUCCESS in 1h 25m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13848aee7c954f368a51ac327c0a52b4 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc5cd0b96ce34170a03fdbd8593038e7 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f54bc459bf640a28eed072f3355bfa1 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14bca288f7b64d28b9e3cf797fd4549f : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12b4cf1a60d54209921bf794ea935656 : SUCCESS in 4m 38s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/22bb80bbec5d4e49805634284dec2557 : SUCCESS in 4m 26s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/087237b2a54d45e4892c4fa716b94926 : SUCCESS in 5m 48s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/b05b7994e32d42178153da3419c32b33 : SUCCESS in 4m 28s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/ca8376d9234a4cef8796efc456342324 : SUCCESS in 4m 29s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/5976065587084f09b353fe80f15af4a0 : SUCCESS in 7m 22s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/c85c0bbff66241298a32bf8db83724d9 : SUCCESS in 4m 20s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/40e15c6d05e2483a830a04b0cbb9b3f4 : SUCCESS in 7m 38s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/a8811159ddea43c8a5fca9a03def67e8 : SUCCESS in 4m 24s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/e21947b7025f4eb89240a6869e18ba15 : SUCCESS in 6m 51s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/3a5a7757ca2b4e9eaf03ff45b37f1e00 : SUCCESS in 4m 34s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/bce46898d6454874be15141109800082 : SUCCESS in 6m 59s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/6f546c510f914ae88263c6067ea5b2a5 : SUCCESS in 4m 16s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/09315963565b41d0b2764b5f74d1e161 : SUCCESS in 6m 43s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/f24b135972ae46f98780bad35354a4f5 : SUCCESS in 4m 42s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/2f798ecea03e497090a88d4bdfdc4bae : SUCCESS in 6m 27s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/b15f4a5ee882497793eea5e0b27cc8b6 : SUCCESS in 4m 45s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/63a269d8181d4c09a5cc2955cb5f0cc6 : SUCCESS in 6m 49s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/60e4dfee4378451b8cb7169b4f9338aa : SUCCESS in 4m 29s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/4a386a3d3aae417e997db5aa538bb55c : SUCCESS in 7m 21s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/f87f6ccb7fed4b46a6c288deb8fc3647 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"f0076d5a24407ca769d75a6e43ddf17dbdbba1c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-23 16:52:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e855bf10f43a4014982841099adbe5b6 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"842c7404b2d01c5dd718c14bb55f0cd47471f0e1","revisions":{"c8d85e545d853cd962e348db37a289ce153bfc73":{"kind":"REWORK","_number":1,"created":"2019-02-10 03:27:48.000000000","uploader":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"ref":"refs/changes/38/636038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/38/636038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/38/636038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/38/636038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/38/636038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/38/636038/1"}}},"commit":{"parents":[{"commit":"17b9e9d42c7d79e1a6e4fd60312ed39108eebab6","subject":"Fixed UnicodeEncodeError for Python2 unicode objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/17b9e9d42c7d79e1a6e4fd60312ed39108eebab6"}]}],"author":{"name":"Bob.Haddleton","email":"bob.haddleton@nokia.com","date":"2019-02-10 03:24:54.000000000","tz":-360},"committer":{"name":"Bob.Haddleton","email":"bob.haddleton@nokia.com","date":"2019-02-10 03:24:54.000000000","tz":-360},"subject":"RedisLock release() should not check if the lock has been acquired","message":"RedisLock release() should not check if the lock has been acquired\n\nThe RedisDriver coordinator _stop() method removes the lock from the\n_acquired_locks set before calling the release() method on the lock,\nso the release() method should not check to see if the lock is in the\nset, since it never will be.\n\nThis was preventing the coordinator\u0027s _stop() method from releasing\nthe locks, so they had to timeout before another group member could\ntake over as leader.\n\nChange-Id: Ie23d88f1a05983d7465f18a329cabc23d789a628\nCloses-Bug: 1815330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c8d85e545d853cd962e348db37a289ce153bfc73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c8d85e545d853cd962e348db37a289ce153bfc73"}]},"branch":"refs/heads/master"},"842c7404b2d01c5dd718c14bb55f0cd47471f0e1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-19 09:12:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/38/636038/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/38/636038/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/38/636038/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/38/636038/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/38/636038/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/38/636038/2"}}},"commit":{"parents":[{"commit":"4c85e958a53e66d995c0fb150d3f26531c19db08","subject":"Merge \"Fix membership lease issue on the etcd3gw driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4c85e958a53e66d995c0fb150d3f26531c19db08"}]}],"author":{"name":"Bob.Haddleton","email":"bob.haddleton@nokia.com","date":"2019-02-10 03:24:54.000000000","tz":-360},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-12-19 09:12:28.000000000","tz":0},"subject":"RedisLock release() should not check if the lock has been acquired","message":"RedisLock release() should not check if the lock has been acquired\n\nThe RedisDriver coordinator _stop() method removes the lock from the\n_acquired_locks set before calling the release() method on the lock,\nso the release() method should not check to see if the lock is in the\nset, since it never will be.\n\nThis was preventing the coordinator\u0027s _stop() method from releasing\nthe locks, so they had to timeout before another group member could\ntake over as leader.\n\nChange-Id: Ie23d88f1a05983d7465f18a329cabc23d789a628\nCloses-Bug: 1815330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/842c7404b2d01c5dd718c14bb55f0cd47471f0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/842c7404b2d01c5dd718c14bb55f0cd47471f0e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
