)]}'
{"id":"openstack%2Ftooz~640695","triplet_id":"openstack%2Ftooz~master~I0a0215f56b37d9a4286097b7f748f68d59afcb35","project":"openstack/tooz","branch":"master","topic":"640695","hashtags":[],"change_id":"I0a0215f56b37d9a4286097b7f748f68d59afcb35","subject":"Move grpcio from requirements.txt to extras","status":"MERGED","created":"2019-03-04 09:17:39.000000000","updated":"2019-07-08 16:56:07.000000000","submitted":"2019-07-08 16:56:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"640695-1562604967737-d1a6a7d1","meta_rev_id":"bbaeda7e6b078504cee78ea634e4ab2e923ac89b","_number":640695,"virtual_id_number":640695,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-08 16:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-07-08 15:01:22.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-07-04 09:21:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-07-08 15:01:22.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 09:21:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-08 15:01:22.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-07-08 16:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d48e4edb77bf52834007bd4aeaa19de65c246ef3","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-03-04 09:17:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7d771a9ad612dc87c740c6421697bb8609b68bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-04 11:22:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/95/640695/1/check/requirements-check/d549a94/ : SUCCESS in 4m 49s\n- tempest-full http://logs.openstack.org/95/640695/1/check/tempest-full/b3aceec/ : SUCCESS in 1h 40m 50s\n- tempest-full-py3 http://logs.openstack.org/95/640695/1/check/tempest-full-py3/7591c36/ : SUCCESS in 1h 42m 34s\n- openstack-tox-cover http://logs.openstack.org/95/640695/1/check/openstack-tox-cover/85363b7/cover/ : SUCCESS in 5m 14s\n- openstack-tox-pep8 http://logs.openstack.org/95/640695/1/check/openstack-tox-pep8/90056a5/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/95/640695/1/check/openstack-tox-py27/9e3c779/ : SUCCESS in 5m 30s\n- openstack-tox-py35 http://logs.openstack.org/95/640695/1/check/openstack-tox-py35/09b33cf/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/95/640695/1/check/openstack-tox-py36/95769d2/ : SUCCESS in 5m 11s\n- openstack-tox-docs http://logs.openstack.org/95/640695/1/check/openstack-tox-docs/0499a15/html/ : SUCCESS in 4m 43s\n- tooz-tox-py27-etcd http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-etcd/a012ba1/ : SUCCESS in 4m 55s\n- tooz-tox-py35-etcd http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-etcd/c05e8a7/ : SUCCESS in 5m 44s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-etcd3/c6749cb/ : SUCCESS in 5m 17s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-etcd3/c75f66c/ : SUCCESS in 4m 51s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-etcd3gw/3fde19c/ : SUCCESS in 4m 43s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-etcd3gw/38b1751/ : SUCCESS in 4m 55s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-zookeeper/b85e14c/ : SUCCESS in 5m 39s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-zookeeper/a96d6aa/ : SUCCESS in 5m 19s\n- tooz-tox-py27-redis http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-redis/13df620/ : SUCCESS in 5m 30s\n- tooz-tox-py35-redis http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-redis/c7e38b9/ : SUCCESS in 5m 10s\n- tooz-tox-py27-sentinel http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-sentinel/3d6f6b6/ : SUCCESS in 5m 32s\n- tooz-tox-py35-sentinel http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-sentinel/8658708/ : SUCCESS in 5m 32s\n- tooz-tox-py27-memcached http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-memcached/0223ccc/ : SUCCESS in 5m 20s\n- tooz-tox-py35-memcached http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-memcached/c002b55/ : SUCCESS in 5m 47s\n- tooz-tox-py27-postgresql http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-postgresql/1ec404d/ : SUCCESS in 5m 26s\n- tooz-tox-py35-postgresql http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-postgresql/a52eab4/ : SUCCESS in 5m 53s\n- tooz-tox-py27-mysql http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-mysql/3081fb8/ : SUCCESS in 5m 07s\n- tooz-tox-py35-mysql http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-mysql/ebf1e63/ : SUCCESS in 5m 12s\n- tooz-tox-py27-consul http://logs.openstack.org/95/640695/1/check/tooz-tox-py27-consul/f642fc1/ : SUCCESS in 5m 14s\n- tooz-tox-py35-consul http://logs.openstack.org/95/640695/1/check/tooz-tox-py35-consul/9e1817d/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"87416b6b7eeb91fb6644cca0e64bc13ada5ee3e0","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-03-05 14:55:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55d778995d785beac81c08885338b5bb9057356d","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-03-05 14:57:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f67b70893d86c5d2e22a85a3e07602647c2a526","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-03-05 16:31:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e83d0ab1edbb36bc899defd740ae3f04db49d989","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-04-12 11:41:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46a0035f68892e7495ff60fb59e22de677545663","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-04-12 11:41:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a7ae4c4b495c4701de90496b8c8b7aef1584045","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 13:32:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/95/640695/2/check/test-release-openstack/dded463/ : SUCCESS in 2m 53s\n- requirements-check http://logs.openstack.org/95/640695/2/check/requirements-check/bdb9ca0/ : SUCCESS in 3m 11s\n- tempest-full http://logs.openstack.org/95/640695/2/check/tempest-full/85bd8af/ : SUCCESS in 1h 49m 32s\n- tempest-full-py3 http://logs.openstack.org/95/640695/2/check/tempest-full-py3/3c90265/ : SUCCESS in 1h 35m 22s\n- openstack-tox-cover http://logs.openstack.org/95/640695/2/check/openstack-tox-cover/3fe5a5e/cover/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/95/640695/2/check/openstack-tox-pep8/800d6a5/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/95/640695/2/check/openstack-tox-py27/63862a4/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/95/640695/2/check/openstack-tox-py35/6fcadb5/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/95/640695/2/check/openstack-tox-py36/a3b44be/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/95/640695/2/check/openstack-tox-py37/984fcc0/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/95/640695/2/check/openstack-tox-docs/db61be6/html/ : SUCCESS in 4m 27s\n- tooz-tox-py27-etcd http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-etcd/dd878f7/ : SUCCESS in 4m 31s\n- tooz-tox-py35-etcd http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-etcd/4ae6eac/ : FAILURE in 3m 09s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-etcd3/82bbf0b/ : SUCCESS in 4m 10s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-etcd3/b4c8335/ : FAILURE in 3m 28s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-etcd3gw/d2e5735/ : SUCCESS in 4m 12s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-etcd3gw/00179e3/ : FAILURE in 3m 08s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-zookeeper/b2b8988/ : SUCCESS in 4m 48s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-zookeeper/83281a8/ : FAILURE in 3m 33s\n- tooz-tox-py27-redis http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-redis/79cd9fb/ : FAILURE in 3m 57s\n- tooz-tox-py35-redis http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-redis/6287147/ : FAILURE in 3m 14s\n- tooz-tox-py27-sentinel http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-sentinel/c59b7b1/ : FAILURE in 3m 33s\n- tooz-tox-py35-sentinel http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-sentinel/4410c18/ : FAILURE in 3m 12s\n- tooz-tox-py27-memcached http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-memcached/b6f764d/ : SUCCESS in 4m 30s\n- tooz-tox-py35-memcached http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-memcached/2ba6e98/ : FAILURE in 3m 40s\n- tooz-tox-py27-postgresql http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-postgresql/58b7b3f/ : SUCCESS in 4m 07s\n- tooz-tox-py35-postgresql http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-postgresql/e30ccdf/ : FAILURE in 3m 27s\n- tooz-tox-py27-mysql http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-mysql/86ef5f7/ : SUCCESS in 4m 02s\n- tooz-tox-py35-mysql http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-mysql/598d850/ : FAILURE in 3m 11s\n- tooz-tox-py27-consul http://logs.openstack.org/95/640695/2/check/tooz-tox-py27-consul/661e70a/ : SUCCESS in 4m 15s\n- tooz-tox-py35-consul http://logs.openstack.org/95/640695/2/check/tooz-tox-py35-consul/97cda72/ : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"5416e52e5e7e9e98c32f6e918aefb11d0344a503","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-05-14 04:51:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1595530d96ee8891968c9fb128507177d0f345c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 06:46:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/95/640695/3/check/test-release-openstack/24a13aa/ : SUCCESS in 3m 39s\n- requirements-check http://logs.openstack.org/95/640695/3/check/requirements-check/3630163/ : FAILURE in 3m 30s\n- tempest-full http://logs.openstack.org/95/640695/3/check/tempest-full/44022f1/ : SUCCESS in 1h 53m 57s\n- tempest-full-py3 http://logs.openstack.org/95/640695/3/check/tempest-full-py3/ac78642/ : SUCCESS in 1h 28m 45s\n- openstack-tox-cover http://logs.openstack.org/95/640695/3/check/openstack-tox-cover/d33d117/cover/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/95/640695/3/check/openstack-tox-pep8/cea66f7/ : SUCCESS in 4m 00s\n- openstack-tox-py27 http://logs.openstack.org/95/640695/3/check/openstack-tox-py27/31b88d5/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/95/640695/3/check/openstack-tox-py35/9fb9412/ : SUCCESS in 3m 49s\n- openstack-tox-py36 http://logs.openstack.org/95/640695/3/check/openstack-tox-py36/613ddfb/ : SUCCESS in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/95/640695/3/check/openstack-tox-py37/01db9f4/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/95/640695/3/check/openstack-tox-docs/43ed9f6/html/ : SUCCESS in 5m 07s\n- tooz-tox-py27-etcd http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-etcd/dfac665/ : SUCCESS in 5m 58s\n- tooz-tox-py35-etcd http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-etcd/49d1725/ : FAILURE in 3m 34s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-etcd3/fd3dd57/ : SUCCESS in 3m 47s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-etcd3/8adf093/ : FAILURE in 2m 42s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-etcd3gw/5418c96/ : SUCCESS in 4m 27s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-etcd3gw/f1fbd15/ : FAILURE in 2m 52s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-zookeeper/1214486/ : SUCCESS in 5m 21s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-zookeeper/90f1008/ : FAILURE in 3m 33s\n- tooz-tox-py27-redis http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-redis/e745691/ : FAILURE in 3m 22s\n- tooz-tox-py35-redis http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-redis/dc3a50d/ : FAILURE in 3m 37s\n- tooz-tox-py27-sentinel http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-sentinel/1701cf5/ : FAILURE in 3m 49s\n- tooz-tox-py35-sentinel http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-sentinel/151ffa8/ : FAILURE in 2m 44s\n- tooz-tox-py27-memcached http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-memcached/bcafca9/ : SUCCESS in 5m 28s\n- tooz-tox-py35-memcached http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-memcached/ef2177e/ : FAILURE in 3m 28s\n- tooz-tox-py27-postgresql http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-postgresql/960795a/ : SUCCESS in 3m 34s\n- tooz-tox-py35-postgresql http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-postgresql/751946a/ : FAILURE in 2m 49s\n- tooz-tox-py27-mysql http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-mysql/3034f9f/ : SUCCESS in 4m 13s\n- tooz-tox-py35-mysql http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-mysql/e641fca/ : FAILURE in 4m 41s\n- tooz-tox-py27-consul http://logs.openstack.org/95/640695/3/check/tooz-tox-py27-consul/8985a0e/ : SUCCESS in 3m 25s\n- tooz-tox-py35-consul http://logs.openstack.org/95/640695/3/check/tooz-tox-py35-consul/800d3ac/ : FAILURE in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"176145a177137696923f527c3a4e740d41e36518","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-21 08:57:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"cc6fdf95bedf46fcf8261847ee6e12db22d1265c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-21 08:57:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3e53cdbb68b29a036f7b893e50a5d45937ee3d8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 10:44:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/95/640695/4/check/test-release-openstack/9ebcd1d/ : SUCCESS in 2m 58s\n- requirements-check http://logs.openstack.org/95/640695/4/check/requirements-check/70617e1/ : FAILURE in 2m 57s\n- tempest-full http://logs.openstack.org/95/640695/4/check/tempest-full/bfc2a65/ : SUCCESS in 1h 34m 46s\n- tempest-full-py3 http://logs.openstack.org/95/640695/4/check/tempest-full-py3/bcb1726/ : SUCCESS in 1h 30m 22s\n- openstack-tox-cover http://logs.openstack.org/95/640695/4/check/openstack-tox-cover/f723e1b/cover/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/95/640695/4/check/openstack-tox-pep8/00f9e5e/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/95/640695/4/check/openstack-tox-py27/5cef0e8/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/95/640695/4/check/openstack-tox-py36/0766afe/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/95/640695/4/check/openstack-tox-py37/e885655/ : SUCCESS in 4m 11s\n- openstack-tox-docs http://logs.openstack.org/95/640695/4/check/openstack-tox-docs/d0ccaa8/html/ : SUCCESS in 3m 33s\n- tooz-tox-py27-etcd http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-etcd/98adbb0/ : SUCCESS in 3m 58s\n- tooz-tox-py36-etcd http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-etcd/4c734a0/ : SUCCESS in 4m 13s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-etcd3/6bb6c35/ : SUCCESS in 3m 58s\n- tooz-tox-py36-etcd3 http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-etcd3/2b8f6b7/ : SUCCESS in 4m 27s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-etcd3gw/665768b/ : SUCCESS in 4m 36s\n- tooz-tox-py36-etcd3gw http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-etcd3gw/23f9eba/ : SUCCESS in 4m 09s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-zookeeper/4be3da7/ : SUCCESS in 4m 24s\n- tooz-tox-py36-zookeeper http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-zookeeper/9e29591/ : SUCCESS in 4m 11s\n- tooz-tox-py27-redis http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-redis/7c15518/ : SUCCESS in 4m 40s\n- tooz-tox-py36-redis http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-redis/282181b/ : SUCCESS in 4m 00s\n- tooz-tox-py27-sentinel http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-sentinel/ec1e9bd/ : SUCCESS in 4m 16s\n- tooz-tox-py36-sentinel http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-sentinel/1b613e5/ : SUCCESS in 4m 20s\n- tooz-tox-py27-memcached http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-memcached/7fdf358/ : SUCCESS in 4m 27s\n- tooz-tox-py36-memcached http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-memcached/ca583d9/ : SUCCESS in 4m 01s\n- tooz-tox-py27-postgresql http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-postgresql/f5a1d11/ : SUCCESS in 3m 30s\n- tooz-tox-py36-postgresql http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-postgresql/8791dff/ : SUCCESS in 3m 49s\n- tooz-tox-py27-mysql http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-mysql/f4a9790/ : SUCCESS in 5m 24s\n- tooz-tox-py36-mysql http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-mysql/aa3fd6b/ : SUCCESS in 4m 24s\n- tooz-tox-py27-consul http://logs.openstack.org/95/640695/4/check/tooz-tox-py27-consul/358bd22/ : SUCCESS in 5m 02s\n- tooz-tox-py36-consul http://logs.openstack.org/95/640695/4/check/tooz-tox-py36-consul/c9abdf0/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":4},{"id":"10f7a872e46235260550915262850eab32b9c645","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-04 06:14:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d15fe01da6e9d3fcd3126f012e403a286fad7960","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 06:15:08.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"d94d1da6aed5c83fdf4e836c93db211f1699291b","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-07-04 06:24:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8756b2b6554cff250a07a475bf9ee12b30bb3e19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 08:14:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/95/640695/5/check/test-release-openstack/448c9f3/ : SUCCESS in 4m 18s\n- requirements-check http://logs.openstack.org/95/640695/5/check/requirements-check/a062e23/ : SUCCESS in 4m 15s\n- tempest-full http://logs.openstack.org/95/640695/5/check/tempest-full/0f6ebce/ : SUCCESS in 1h 44m 38s\n- tempest-full-py3 http://logs.openstack.org/95/640695/5/check/tempest-full-py3/2bec0b0/ : SUCCESS in 1h 25m 13s\n- openstack-tox-cover http://logs.openstack.org/95/640695/5/check/openstack-tox-cover/1e9cfed/cover/ : SUCCESS in 8m 21s\n- openstack-tox-pep8 http://logs.openstack.org/95/640695/5/check/openstack-tox-pep8/1c60e51/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/95/640695/5/check/openstack-tox-py27/861dcc0/ : SUCCESS in 6m 30s\n- openstack-tox-py36 http://logs.openstack.org/95/640695/5/check/openstack-tox-py36/df5ccf7/ : SUCCESS in 5m 24s\n- openstack-tox-py37 http://logs.openstack.org/95/640695/5/check/openstack-tox-py37/1403888/ : SUCCESS in 8m 43s\n- openstack-tox-docs http://logs.openstack.org/95/640695/5/check/openstack-tox-docs/f9712de/html/ : SUCCESS in 7m 20s\n- tooz-tox-py27-etcd http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-etcd/ddd3087/ : SUCCESS in 7m 46s\n- tooz-tox-py36-etcd http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-etcd/34f9998/ : SUCCESS in 6m 27s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-etcd3/7de4f60/ : SUCCESS in 3m 56s\n- tooz-tox-py36-etcd3 http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-etcd3/618279f/ : SUCCESS in 3m 53s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-etcd3gw/cac1b26/ : SUCCESS in 3m 50s\n- tooz-tox-py36-etcd3gw http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-etcd3gw/a846f04/ : SUCCESS in 6m 14s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-zookeeper/6ede2f3/ : SUCCESS in 8m 29s\n- tooz-tox-py36-zookeeper http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-zookeeper/9490f5f/ : SUCCESS in 4m 30s\n- tooz-tox-py27-redis http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-redis/04769b2/ : SUCCESS in 3m 42s\n- tooz-tox-py36-redis http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-redis/3a70851/ : SUCCESS in 6m 12s\n- tooz-tox-py27-sentinel http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-sentinel/92df42d/ : SUCCESS in 5m 41s\n- tooz-tox-py36-sentinel http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-sentinel/4fe3873/ : SUCCESS in 4m 17s\n- tooz-tox-py27-memcached http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-memcached/c7a7fe4/ : SUCCESS in 4m 01s\n- tooz-tox-py36-memcached http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-memcached/3d93c22/ : SUCCESS in 4m 22s\n- tooz-tox-py27-postgresql http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-postgresql/55d5b2b/ : SUCCESS in 5m 45s\n- tooz-tox-py36-postgresql http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-postgresql/765bed3/ : SUCCESS in 4m 39s\n- tooz-tox-py27-mysql http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-mysql/aa309dc/ : SUCCESS in 4m 02s\n- tooz-tox-py36-mysql http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-mysql/bf3d600/ : SUCCESS in 7m 10s\n- tooz-tox-py27-consul http://logs.openstack.org/95/640695/5/check/tooz-tox-py27-consul/e62986b/ : SUCCESS in 7m 32s\n- tooz-tox-py36-consul http://logs.openstack.org/95/640695/5/check/tooz-tox-py36-consul/16dc977/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":5},{"id":"34c2f68b79637c84cddfe6bafa3b46f32c5e2955","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-04 09:21:13.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0a3fd068a8c017326be1f2b6bac36976121dd1c6","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-07-08 15:01:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bd9e3213dd2e8d8641b7698c4618e5979e1fb1c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 15:01:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"51bb9d5f79c2d31c8c0477f1d7a9b603d3777f73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:56:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ef81ed83203098a277ae073ae3cbb3555298ba97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:56:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/95/640695/5/gate/test-release-openstack/8bc72db/ : SUCCESS in 2m 56s\n- requirements-check http://logs.openstack.org/95/640695/5/gate/requirements-check/776381a/ : SUCCESS in 3m 16s\n- tempest-full http://logs.openstack.org/95/640695/5/gate/tempest-full/4d45aec/ : SUCCESS in 1h 46m 37s\n- tempest-full-py3 http://logs.openstack.org/95/640695/5/gate/tempest-full-py3/19c49d6/ : SUCCESS in 1h 43m 14s\n- openstack-tox-pep8 http://logs.openstack.org/95/640695/5/gate/openstack-tox-pep8/64b6d0d/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/95/640695/5/gate/openstack-tox-py27/6ae8c6a/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/95/640695/5/gate/openstack-tox-py36/437cc6a/ : SUCCESS in 4m 07s\n- openstack-tox-py37 http://logs.openstack.org/95/640695/5/gate/openstack-tox-py37/7bdbec2/ : SUCCESS in 6m 24s\n- openstack-tox-docs http://logs.openstack.org/95/640695/5/gate/openstack-tox-docs/c13fb32/html/ : SUCCESS in 5m 28s\n- tooz-tox-py27-etcd http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-etcd/c252816/ : SUCCESS in 5m 37s\n- tooz-tox-py36-etcd http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-etcd/bc3c2d8/ : SUCCESS in 8m 06s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-etcd3/f952247/ : SUCCESS in 7m 39s\n- tooz-tox-py36-etcd3 http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-etcd3/8b4cdca/ : SUCCESS in 3m 58s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-etcd3gw/d70ffda/ : SUCCESS in 3m 37s\n- tooz-tox-py36-etcd3gw http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-etcd3gw/ec5461e/ : SUCCESS in 6m 00s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-zookeeper/015ecb7/ : SUCCESS in 6m 00s\n- tooz-tox-py36-zookeeper http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-zookeeper/5f6c2ca/ : SUCCESS in 6m 20s\n- tooz-tox-py27-redis http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-redis/010aaef/ : SUCCESS in 6m 08s\n- tooz-tox-py36-redis http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-redis/dd2241f/ : SUCCESS in 7m 35s\n- tooz-tox-py27-sentinel http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-sentinel/10e332c/ : SUCCESS in 4m 05s\n- tooz-tox-py36-sentinel http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-sentinel/7cad90e/ : SUCCESS in 3m 58s\n- tooz-tox-py27-memcached http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-memcached/3b5d805/ : SUCCESS in 7m 06s\n- tooz-tox-py36-memcached http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-memcached/f7af522/ : SUCCESS in 6m 26s\n- tooz-tox-py27-postgresql http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-postgresql/29c7fb6/ : SUCCESS in 3m 55s\n- tooz-tox-py36-postgresql http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-postgresql/ccbe069/ : SUCCESS in 3m 58s\n- tooz-tox-py27-mysql http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-mysql/826c1ad/ : SUCCESS in 6m 19s\n- tooz-tox-py36-mysql http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-mysql/115175a/ : SUCCESS in 3m 50s\n- tooz-tox-py27-consul http://logs.openstack.org/95/640695/5/gate/tooz-tox-py27-consul/928d6fe/ : SUCCESS in 3m 56s\n- tooz-tox-py36-consul http://logs.openstack.org/95/640695/5/gate/tooz-tox-py36-consul/5d29590/ : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"535121d6133af5d8cf108cb0594511d426fa38ae","revisions":{"9c2807281be01c28643471ebe9f9ef3d40b4c2c5":{"kind":"REWORK","_number":1,"created":"2019-03-04 09:17:39.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/95/640695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/95/640695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/95/640695/1"}}},"commit":{"parents":[{"commit":"d55449258a0e73c9d38cba5a02120c6a8d5ea4d9","subject":"Merge \"More explicitly document driver connection strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d55449258a0e73c9d38cba5a02120c6a8d5ea4d9"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-03-04 09:17:07.000000000","tz":60},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-03-04 09:17:37.000000000","tz":60},"subject":"Remove grpcio from requirements.txt","message":"Remove grpcio from requirements.txt\n\nIt is not used anywhere.\n\nChange-Id: I0a0215f56b37d9a4286097b7f748f68d59afcb35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9c2807281be01c28643471ebe9f9ef3d40b4c2c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9c2807281be01c28643471ebe9f9ef3d40b4c2c5"}]},"branch":"refs/heads/master"},"44f7aeece69041beba94399a312b601a02a31546":{"kind":"REWORK","_number":2,"created":"2019-04-12 11:41:31.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/95/640695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/95/640695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/95/640695/2"}}},"commit":{"parents":[{"commit":"d55449258a0e73c9d38cba5a02120c6a8d5ea4d9","subject":"Merge \"More explicitly document driver connection strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d55449258a0e73c9d38cba5a02120c6a8d5ea4d9"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-03-04 09:17:07.000000000","tz":60},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-04-12 11:40:56.000000000","tz":120},"subject":"Move grpcio from requirements.txt to extras","message":"Move grpcio from requirements.txt to extras\n\nIt is only needed for the etcd3 driver.\n\nChange-Id: I0a0215f56b37d9a4286097b7f748f68d59afcb35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/44f7aeece69041beba94399a312b601a02a31546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/44f7aeece69041beba94399a312b601a02a31546"}]},"branch":"refs/heads/master"},"595e50c3163138a217de74145590ad621b31941d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-14 04:51:34.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/95/640695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/95/640695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/95/640695/3"}}},"commit":{"parents":[{"commit":"c02f9dd6c65cc34f9007af82773349efe13507c0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c02f9dd6c65cc34f9007af82773349efe13507c0"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-03-04 09:17:07.000000000","tz":60},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-05-14 04:51:34.000000000","tz":0},"subject":"Move grpcio from requirements.txt to extras","message":"Move grpcio from requirements.txt to extras\n\nIt is only needed for the etcd3 driver.\n\nChange-Id: I0a0215f56b37d9a4286097b7f748f68d59afcb35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/595e50c3163138a217de74145590ad621b31941d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/595e50c3163138a217de74145590ad621b31941d"}]},"branch":"refs/heads/master"},"08905935ef0f98be0320a0f1189eed80fdc46084":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-21 08:57:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/640695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/95/640695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/95/640695/4"}}},"commit":{"parents":[{"commit":"8207427fffc72547387369ce7ba2696a86bf7626","subject":"Unblock tooz gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8207427fffc72547387369ce7ba2696a86bf7626"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-03-04 09:17:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-21 08:57:17.000000000","tz":0},"subject":"Move grpcio from requirements.txt to extras","message":"Move grpcio from requirements.txt to extras\n\nIt is only needed for the etcd3 driver.\n\nChange-Id: I0a0215f56b37d9a4286097b7f748f68d59afcb35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/08905935ef0f98be0320a0f1189eed80fdc46084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/08905935ef0f98be0320a0f1189eed80fdc46084"}]},"branch":"refs/heads/master"},"535121d6133af5d8cf108cb0594511d426fa38ae":{"kind":"REWORK","_number":5,"created":"2019-07-04 06:24:06.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/95/640695/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/95/640695/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/640695/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/95/640695/5"}}},"commit":{"parents":[{"commit":"ef5277d75c367ddf71fcddff3e7b47459914e0dc","subject":"Move test deps to test-requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ef5277d75c367ddf71fcddff3e7b47459914e0dc"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-03-04 09:17:07.000000000","tz":60},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-07-04 06:23:54.000000000","tz":120},"subject":"Move grpcio from requirements.txt to extras","message":"Move grpcio from requirements.txt to extras\n\nIt is only needed for the etcd3 driver.\n\nChange-Id: I0a0215f56b37d9a4286097b7f748f68d59afcb35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/535121d6133af5d8cf108cb0594511d426fa38ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/535121d6133af5d8cf108cb0594511d426fa38ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
