)]}'
{"id":"openstack%2Ftooz~644228","triplet_id":"openstack%2Ftooz~stable%2Fstein~I699c698754f498cacff4998f33ea0738682b86e7","project":"openstack/tooz","branch":"stable/stein","topic":"create-stein","hashtags":[],"change_id":"I699c698754f498cacff4998f33ea0738682b86e7","subject":"Update .gitreview for stable/stein","status":"MERGED","created":"2019-03-18 14:57:56.000000000","updated":"2020-05-09 14:56:33.000000000","submitted":"2020-05-09 14:55:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"644228-1589036127834-4f1731f4","meta_rev_id":"a65e3d108e97729a40fec252e83f49acae2953a8","_number":644228,"virtual_id_number":644228,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-05-09 12:15:29.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-03-31 18:37:23.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-09 14:55:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-05-24 14:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2019-05-23 15:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2019-05-23 15:42:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 15:42:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-05-24 14:09:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-31 18:37:23.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-05-09 12:15:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-09 14:55:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7bd76a5f922585d92484556a3fe7e942933ecf4c","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-03-18 14:57:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21f2d86411251ecd8852631f813f0cd8e7558370","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 15:49:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/28/644228/1/check/openstack-tox-cover/dccb1ad/cover/ : SUCCESS in 5m 36s\n- openstack-tox-pep8 http://logs.openstack.org/28/644228/1/check/openstack-tox-pep8/4b8716c/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/28/644228/1/check/openstack-tox-py27/85de60c/ : SUCCESS in 6m 48s\n- openstack-tox-py35 http://logs.openstack.org/28/644228/1/check/openstack-tox-py35/154d0fa/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/28/644228/1/check/openstack-tox-py36/ed4dd03/ : SUCCESS in 5m 46s\n- openstack-tox-docs http://logs.openstack.org/28/644228/1/check/openstack-tox-docs/4592ab5/html/ : SUCCESS in 4m 45s\n- tooz-tox-py27-etcd http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-etcd/341eb9d/ : SUCCESS in 5m 34s\n- tooz-tox-py35-etcd http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-etcd/f053a4a/ : FAILURE in 4m 33s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-etcd3/a6b7359/ : SUCCESS in 5m 59s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-etcd3/3eb4f78/ : FAILURE in 5m 19s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-etcd3gw/d99ff7b/ : SUCCESS in 5m 02s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-etcd3gw/ea8dbbe/ : FAILURE in 4m 41s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-zookeeper/f947a15/ : SUCCESS in 6m 00s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-zookeeper/1961f4a/ : FAILURE in 4m 37s\n- tooz-tox-py27-redis http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-redis/432a5ac/ : FAILURE in 6m 11s\n- tooz-tox-py35-redis http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-redis/bad9b1e/ : FAILURE in 5m 47s\n- tooz-tox-py27-sentinel http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-sentinel/24f3048/ : FAILURE in 4m 33s\n- tooz-tox-py35-sentinel http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-sentinel/44e28e5/ : FAILURE in 3m 51s\n- tooz-tox-py27-memcached http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-memcached/ed09e1c/ : SUCCESS in 6m 02s\n- tooz-tox-py35-memcached http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-memcached/b15752f/ : FAILURE in 4m 46s\n- tooz-tox-py27-postgresql http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-postgresql/514abb3/ : SUCCESS in 5m 39s\n- tooz-tox-py35-postgresql http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-postgresql/793c826/ : FAILURE in 4m 21s\n- tooz-tox-py27-mysql http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-mysql/41207ad/ : SUCCESS in 4m 30s\n- tooz-tox-py35-mysql http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-mysql/76ee9a9/ : FAILURE in 4m 52s\n- tooz-tox-py27-consul http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-consul/844b3d8/ : SUCCESS in 6m 05s\n- tooz-tox-py35-consul http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-consul/2c738fd/ : FAILURE in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3f845bccabf48572e31077f346fdcc4e303e3722","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-03-20 13:58:22.000000000","message":"Patch Set 1:\n\nLooks like issues with running on bionic that need to be addressed first.","accounts_in_message":[],"_revision_number":1},{"id":"8f3726675170fb15ea9c2c72c7542683abe06f90","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-22 16:42:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26b6778775179d9f67384d208720386ed94dc031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 17:08:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/28/644228/1/check/tempest-full/ebba8cc/ : FAILURE in 19m 13s\n- tempest-full-py3 http://logs.openstack.org/28/644228/1/check/tempest-full-py3/2cbafa5/ : FAILURE in 17m 44s\n- openstack-tox-cover http://logs.openstack.org/28/644228/1/check/openstack-tox-cover/65f18cb/cover/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/28/644228/1/check/openstack-tox-pep8/2b5c3b7/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/28/644228/1/check/openstack-tox-py27/9412f49/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/28/644228/1/check/openstack-tox-py35/bdd0074/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/28/644228/1/check/openstack-tox-py36/ca8dfbe/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/28/644228/1/check/openstack-tox-docs/b22f94a/html/ : SUCCESS in 5m 50s\n- tooz-tox-py27-etcd http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-etcd/5ead69b/ : SUCCESS in 4m 13s\n- tooz-tox-py35-etcd http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-etcd/663d6a4/ : FAILURE in 2m 37s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-etcd3/2ac9b15/ : SUCCESS in 4m 16s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-etcd3/e7898e1/ : FAILURE in 3m 06s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-etcd3gw/5e864d2/ : SUCCESS in 5m 40s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-etcd3gw/01351e7/ : FAILURE in 4m 30s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-zookeeper/b79d4d9/ : SUCCESS in 3m 56s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-zookeeper/64a477b/ : FAILURE in 2m 38s\n- tooz-tox-py27-redis http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-redis/c3df2fd/ : FAILURE in 3m 20s\n- tooz-tox-py35-redis http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-redis/cef7983/ : FAILURE in 4m 53s\n- tooz-tox-py27-sentinel http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-sentinel/721046c/ : FAILURE in 3m 32s\n- tooz-tox-py35-sentinel http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-sentinel/55aaf27/ : FAILURE in 3m 28s\n- tooz-tox-py27-memcached http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-memcached/56bc1ee/ : SUCCESS in 3m 51s\n- tooz-tox-py35-memcached http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-memcached/603781b/ : FAILURE in 2m 53s\n- tooz-tox-py27-postgresql http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-postgresql/2208ee6/ : SUCCESS in 3m 51s\n- tooz-tox-py35-postgresql http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-postgresql/e612467/ : FAILURE in 4m 39s\n- tooz-tox-py27-mysql http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-mysql/080aaae/ : SUCCESS in 3m 49s\n- tooz-tox-py35-mysql http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-mysql/42126bf/ : FAILURE in 2m 47s\n- tooz-tox-py27-consul http://logs.openstack.org/28/644228/1/check/tooz-tox-py27-consul/6591bc6/ : SUCCESS in 3m 51s\n- tooz-tox-py35-consul http://logs.openstack.org/28/644228/1/check/tooz-tox-py35-consul/ba3413a/ : FAILURE in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8f1282abab6ee63c9a0b27f2facd9ff4effca879","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 15:42:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"66c640ffc4ba416967b87d55f4f2d4d01ddc9660","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 15:42:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nFast approving since this is needed","accounts_in_message":[],"_revision_number":2},{"id":"d21805baf02a9d170ddbba7b82541942ab2da65a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 17:26:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/28/644228/2/check/tempest-full/d874c6c/ : SUCCESS in 1h 38m 49s\n- tempest-full-py3 http://logs.openstack.org/28/644228/2/check/tempest-full-py3/4220344/ : FAILURE in 1h 16m 46s\n- openstack-tox-cover http://logs.openstack.org/28/644228/2/check/openstack-tox-cover/fe4b082/cover/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 http://logs.openstack.org/28/644228/2/check/openstack-tox-pep8/0d30c5f/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/28/644228/2/check/openstack-tox-py27/e03eb89/ : SUCCESS in 4m 08s\n- openstack-tox-py35 http://logs.openstack.org/28/644228/2/check/openstack-tox-py35/90cad19/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/28/644228/2/check/openstack-tox-py36/f519935/ : SUCCESS in 4m 13s\n- openstack-tox-docs http://logs.openstack.org/28/644228/2/check/openstack-tox-docs/9f902c3/html/ : SUCCESS in 3m 24s\n- tooz-tox-py27-etcd http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-etcd/b835166/ : SUCCESS in 4m 11s\n- tooz-tox-py35-etcd http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-etcd/24dd233/ : FAILURE in 2m 33s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-etcd3/d02d332/ : SUCCESS in 3m 52s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-etcd3/b359621/ : FAILURE in 3m 10s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-etcd3gw/f9b505c/ : SUCCESS in 3m 42s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-etcd3gw/6290158/ : FAILURE in 3m 07s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-zookeeper/48b195c/ : SUCCESS in 3m 55s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-zookeeper/4656bf5/ : FAILURE in 4m 55s\n- tooz-tox-py27-redis http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-redis/8715434/ : FAILURE in 4m 39s\n- tooz-tox-py35-redis http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-redis/efa545a/ : FAILURE in 3m 19s\n- tooz-tox-py27-sentinel http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-sentinel/1f98eda/ : FAILURE in 3m 14s\n- tooz-tox-py35-sentinel http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-sentinel/17fc7aa/ : FAILURE in 3m 21s\n- tooz-tox-py27-memcached http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-memcached/22af21a/ : SUCCESS in 3m 49s\n- tooz-tox-py35-memcached http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-memcached/6b7b89f/ : FAILURE in 3m 22s\n- tooz-tox-py27-postgresql http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-postgresql/f30d2e2/ : SUCCESS in 4m 14s\n- tooz-tox-py35-postgresql http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-postgresql/e34bdd9/ : FAILURE in 2m 56s\n- tooz-tox-py27-mysql http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-mysql/184ade6/ : SUCCESS in 4m 15s\n- tooz-tox-py35-mysql http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-mysql/70b16c0/ : FAILURE in 3m 25s\n- tooz-tox-py27-consul http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-consul/7a912ed/ : SUCCESS in 5m 30s\n- tooz-tox-py35-consul http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-consul/5e9bb1e/ : FAILURE in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2408a027e85d937324a5d1a292380eb877805ba7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-24 14:09:48.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9e3536fa425a13fa72b6bd7af15f6b6a8fa8f79c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 16:17:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/28/644228/2/check/tempest-full/87142f2/ : SUCCESS in 1h 51m 35s\n- tempest-full-py3 http://logs.openstack.org/28/644228/2/check/tempest-full-py3/13d5ca1/ : SUCCESS in 1h 40m 49s\n- openstack-tox-cover http://logs.openstack.org/28/644228/2/check/openstack-tox-cover/7342d15/cover/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/28/644228/2/check/openstack-tox-pep8/816816c/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/28/644228/2/check/openstack-tox-py27/32d9935/ : SUCCESS in 5m 21s\n- openstack-tox-py35 http://logs.openstack.org/28/644228/2/check/openstack-tox-py35/2631e2f/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/28/644228/2/check/openstack-tox-py36/e716aeb/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/28/644228/2/check/openstack-tox-docs/3a97289/html/ : SUCCESS in 3m 59s\n- tooz-tox-py27-etcd http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-etcd/3a2d830/ : SUCCESS in 4m 05s\n- tooz-tox-py35-etcd http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-etcd/db63df9/ : FAILURE in 3m 27s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-etcd3/a224e5c/ : SUCCESS in 4m 05s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-etcd3/57ace38/ : FAILURE in 5m 31s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-etcd3gw/15711db/ : SUCCESS in 4m 02s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-etcd3gw/c2417bc/ : FAILURE in 3m 21s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-zookeeper/8c9e4bd/ : SUCCESS in 4m 04s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-zookeeper/37d7a8a/ : FAILURE in 3m 19s\n- tooz-tox-py27-redis http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-redis/b0a5c11/ : FAILURE in 4m 46s\n- tooz-tox-py35-redis http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-redis/57e8f9b/ : FAILURE in 4m 05s\n- tooz-tox-py27-sentinel http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-sentinel/e1fe30e/ : FAILURE in 3m 27s\n- tooz-tox-py35-sentinel http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-sentinel/3fdfc9b/ : FAILURE in 2m 52s\n- tooz-tox-py27-memcached http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-memcached/0f77801/ : SUCCESS in 4m 13s\n- tooz-tox-py35-memcached http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-memcached/f69bffb/ : FAILURE in 2m 46s\n- tooz-tox-py27-postgresql http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-postgresql/e30c500/ : SUCCESS in 3m 56s\n- tooz-tox-py35-postgresql http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-postgresql/04796fd/ : FAILURE in 2m 46s\n- tooz-tox-py27-mysql http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-mysql/0b37292/ : SUCCESS in 5m 02s\n- tooz-tox-py35-mysql http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-mysql/4d30d85/ : FAILURE in 3m 33s\n- tooz-tox-py27-consul http://logs.openstack.org/28/644228/2/check/tooz-tox-py27-consul/092afb1/ : SUCCESS in 4m 50s\n- tooz-tox-py35-consul http://logs.openstack.org/28/644228/2/check/tooz-tox-py35-consul/45e2307/ : FAILURE in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"9b8357c3df153c85263f935de5b3b3be632b6832","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-10-02 18:19:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d80975b648b2f3aadc3ba04d46f4a35fd9b622b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 19:59:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ed022798cb1a4a17916503f70f3c7b63 : SUCCESS in 1h 37m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5db3e27780264adc975846794943499a : SUCCESS in 1h 23m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a29c2b24c10b421eb3ae8c5ddd966044 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d3c5b31faac480cb7eb09484e627830 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6944a2e0dda49c9afe3e1ebdc10d891 : SUCCESS in 4m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cc414a9d1cfa4475ac386ad975bd4225 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06727466ced54a6b87bd03808fd811c5 : FAILURE in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41d6488b2f0b47d4a74e1f12ede95650 : SUCCESS in 5m 07s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/419bcc6d8a8d4122b7ebfb964dcb07eb : SUCCESS in 4m 07s\n- tooz-tox-py35-etcd https://zuul.opendev.org/t/openstack/build/eff396d633f640e2a9746d017f16e7ea : FAILURE in 3m 08s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/f61a297bcf3f4079b8fa58715d000ff1 : SUCCESS in 5m 53s\n- tooz-tox-py35-etcd3 https://zuul.opendev.org/t/openstack/build/ea183ba88bca4ba1ad1990b21779507c : FAILURE in 3m 02s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/6ccb23be455049b0b6ca549db4346ece : SUCCESS in 5m 39s\n- tooz-tox-py35-etcd3gw https://zuul.opendev.org/t/openstack/build/61b66b977e824d57848951315fd4a619 : FAILURE in 5m 02s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/8fe7ba08920449ac80fa6065e1cad87e : SUCCESS in 5m 16s\n- tooz-tox-py35-zookeeper https://zuul.opendev.org/t/openstack/build/2965ff7bfc3f4d089447627b136d7089 : FAILURE in 4m 04s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/bc6d6397b13741ffa98d41013c5aac81 : FAILURE in 3m 39s\n- tooz-tox-py35-redis https://zuul.opendev.org/t/openstack/build/6f4c4ca17e444ba3ae17dffed9acecfb : FAILURE in 3m 04s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/a498a040e0e44c34afc329ff4ec9a414 : FAILURE in 3m 40s\n- tooz-tox-py35-sentinel https://zuul.opendev.org/t/openstack/build/9527c302874b40a6a005101017bde82a : FAILURE in 3m 07s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/175def41d8e34057853bcefd3f18901a : SUCCESS in 5m 04s\n- tooz-tox-py35-memcached https://zuul.opendev.org/t/openstack/build/137446b71db9464888336761b355c35c : FAILURE in 4m 32s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/bcdf4779c3ed402389cf658ec697ff5d : SUCCESS in 4m 55s\n- tooz-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/e80449ea953740ae9df9b123ae2ccaa5 : FAILURE in 2m 46s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/e3148ca0f14c407ba3d2d36949740edd : SUCCESS in 4m 00s\n- tooz-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/2067e8cb475a433daf5a8a4eac7652f3 : FAILURE in 3m 41s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/ca7319bc1b194b8990b74052630a9a3e : SUCCESS in 4m 02s\n- tooz-tox-py35-consul https://zuul.opendev.org/t/openstack/build/9d00358bff1d477fb0f83efc7e4c7c0b : FAILURE in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"99f391603bd4799626fb63023d6b9867604f3fd0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-03 10:11:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8d597e0681603cbae6afba74def26f4b290c8df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 11:51:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ed970c9cbdd94223b8d294c6e5d61f69 : SUCCESS in 1h 37m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0abd595cc3234e8aaa01e664b0e4e502 : SUCCESS in 1h 27m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9fa2673565146eeb3bbe0eff5932674 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a628bc0816de422fb39782914122fe3f : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f7ada6552724bbd9ce8a9e7d2e4991f : SUCCESS in 4m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cf41820007994b2b95dd2c2365115877 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/379cf538229649a18f573f0b5bfa7195 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beff451d05554ae78ba4f6255798a830 : SUCCESS in 4m 26s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/cc5be96861e4449599af7e0655a1703a : SUCCESS in 4m 40s\n- tooz-tox-py35-etcd https://zuul.opendev.org/t/openstack/build/ca504c19ae9943a09a9c0d26b15b7c6b : FAILURE in 2m 55s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/45bf150cefec4b13ba3bc76d2bac27c2 : SUCCESS in 5m 26s\n- tooz-tox-py35-etcd3 https://zuul.opendev.org/t/openstack/build/4e81248a76c74467a3f4daacb48e8e78 : FAILURE in 3m 41s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/8d53760cca774fed88c85fc6a3bc1bfc : SUCCESS in 4m 08s\n- tooz-tox-py35-etcd3gw https://zuul.opendev.org/t/openstack/build/d5868527ec1e466cb3e9ff9f88763cb1 : FAILURE in 3m 33s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/a101601f041a4f1791a4d28d5811f0b6 : SUCCESS in 4m 02s\n- tooz-tox-py35-zookeeper https://zuul.opendev.org/t/openstack/build/6094b94e075844759a9fdadac2342408 : FAILURE in 2m 56s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/796c44f2705b431fa2bcfc5f8beeaad0 : FAILURE in 3m 24s\n- tooz-tox-py35-redis https://zuul.opendev.org/t/openstack/build/70757219f9ed48ce893a4580e8f58a97 : FAILURE in 5m 04s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/ed889156d1f94d9ab8d1e020fafa068b : FAILURE in 4m 17s\n- tooz-tox-py35-sentinel https://zuul.opendev.org/t/openstack/build/b791e72fe3cc471788e5271868780fb7 : FAILURE in 2m 48s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/bd4abd49d4834735b45fb8b117446af5 : SUCCESS in 4m 21s\n- tooz-tox-py35-memcached https://zuul.opendev.org/t/openstack/build/4be4f81a72b6436980e98b6788f32e6b : FAILURE in 3m 28s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/09ff8edf77d842e1b9e8bbf3bcdaf9eb : SUCCESS in 4m 41s\n- tooz-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/478c297aae6044bb95e6ef120c1a714c : FAILURE in 3m 02s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/3a462c71df1d48e88a577877fee5b681 : SUCCESS in 6m 35s\n- tooz-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/1f027bf4c0d542238544695ef847da6c : FAILURE in 3m 04s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/ad3613e3893e45a1ad09f73452e4a78b : SUCCESS in 3m 48s\n- tooz-tox-py35-consul https://zuul.opendev.org/t/openstack/build/e15cb131308d4bfb86f984c0a143ba0b : FAILURE in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e799ea712111e621a5c903f92dba55c46693a6d6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-31 18:37:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dec0fe7dbec70e6398a693ed4f9e9da2901b1f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 20:16:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/fd980bd451ba4fabada0b7c9133327c7 : SUCCESS in 1h 33m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f03acff0d57b43a0bd5a04a4e2d315f3 : SUCCESS in 1h 19m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1661415a74454733a87b48e323df70d8 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09d5ecf4c75401fbe7b9732810bb412 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a8e9b6262e04f599d64d1a1f9da2480 : FAILURE in 4m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d804083fc503483aadf90d35d0f46135 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44ded92dddf64310b4ca7b3de0ffdae2 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43d172e1d0084fb38132a71b0b5ed9c3 : SUCCESS in 5m 42s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/a3abb869c80c4a02b6f096d21f272c1f : FAILURE in 5m 10s\n- tooz-tox-py35-etcd https://zuul.opendev.org/t/openstack/build/78e8d9a9ccf641bf9efff07de5ea9062 : FAILURE in 3m 15s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/92910583aba34b3e90fca2d28df1920f : FAILURE in 7m 13s\n- tooz-tox-py35-etcd3 https://zuul.opendev.org/t/openstack/build/5676bb235d524589bd3695bdc7e780fe : FAILURE in 6m 10s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/29c29da7e64f47a7a49a95339610197d : FAILURE in 7m 52s\n- tooz-tox-py35-etcd3gw https://zuul.opendev.org/t/openstack/build/12cd821e491a4bb39b64514846ff7da9 : FAILURE in 3m 50s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/d93882722f2f4c1db40922dfb960807e : FAILURE in 4m 26s\n- tooz-tox-py35-zookeeper https://zuul.opendev.org/t/openstack/build/7cc5fb8eab7244f0bb2bc414360cab8a : FAILURE in 4m 07s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/84b4c2fb104e442cb22947f726278fb0 : FAILURE in 5m 54s\n- tooz-tox-py35-redis https://zuul.opendev.org/t/openstack/build/efc1692ebffc437894c2a7517696d201 : FAILURE in 3m 39s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/ed39d2a247144c03b781f78b9ab0869d : FAILURE in 5m 50s\n- tooz-tox-py35-sentinel https://zuul.opendev.org/t/openstack/build/dfee15601dae4efdaabe91964e486238 : FAILURE in 5m 03s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/e382e76c2c124737b55a50a154db6e39 : FAILURE in 5m 15s\n- tooz-tox-py35-memcached https://zuul.opendev.org/t/openstack/build/0363cb7088ee4905a8a762c1b3ed7ab8 : FAILURE in 3m 26s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/6e8792e7171d4d54883fb3f30dcf059e : FAILURE in 5m 14s\n- tooz-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/14a2d6b4e92b4e5080ca3dde3f9263ea : FAILURE in 6m 12s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/03f775fc0e9b4783bec04026b53734e6 : FAILURE in 6m 32s\n- tooz-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/e3239e301287413a81d26018a7073e65 : FAILURE in 5m 34s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/58b5c280db8f42868a738a65079c966a : SUCCESS in 5m 33s\n- tooz-tox-py35-consul https://zuul.opendev.org/t/openstack/build/da102d554e6f423d8a0dfa8a3345d457 : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e362fa624f5ff6b335558cedcfd1a6f5b871ebc2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-09 12:15:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a42efb6260b6972d159322f794af32bade0f66ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 13:36:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/96fe2aee3ef0471b8ce2f4d179583127 : SUCCESS in 1h 20m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/091f7bccf94f466e95bb8aa29541354a : SUCCESS in 1h 14m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/831624c1584a4bddb15388c6f938cc91 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d497c25a4796481080d5b6b941a34563 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01cecbc892e54f3495dd3845ec75fb2f : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3a23e3d4913417294bdf96ba58807c9 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d29e6db7c5f4641b37a89e02fc70dc1 : SUCCESS in 9m 29s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/693c1661703e4775b1f780958d454658 : SUCCESS in 4m 46s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/d19e83494a4746eb9afa784e71d8522f : SUCCESS in 5m 20s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/2d2f4f9c1b2b4039bb33afaf247a7da7 : SUCCESS in 4m 34s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/0b768f69688b47a19c1d1d959deb6c36 : SUCCESS in 4m 35s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/df75f2ac19da4feea4faf027bdc702b6 : SUCCESS in 5m 08s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/7c7ef58351aa46e08b7b38e12ff68e7a : SUCCESS in 4m 35s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/8ae4176e674e47b1aff70593db938f5a : SUCCESS in 5m 47s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/af5781c9e744418eacf5442edf224955 : SUCCESS in 4m 36s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/39f6c6c15f664f14b6460ff555d8bab9 : SUCCESS in 5m 00s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/9be76feefdca443794574b286d4b0342 : SUCCESS in 4m 52s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/352fbd00490649728053db095604da4d : SUCCESS in 5m 08s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/ec2ca17251af40e692ce3f4c23075830 : SUCCESS in 5m 16s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/0832b0f63b37441088ace03baff4de4e : SUCCESS in 4m 35s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/5447f89348324cd6b902ed72d169f0b2 : SUCCESS in 5m 20s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/5b3ada04a5fb4e7bab08a02f44b0b342 : SUCCESS in 5m 15s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/dcabc51312f04170a717da1a2d2c20c9 : SUCCESS in 5m 02s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/4be53963554047a5acc18bc8d9d7741c : SUCCESS in 4m 59s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/854bc685fdee48e6be4f7c2124e6d91d : SUCCESS in 5m 27s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/d4481043f75a4f5380be365ec798e750 : SUCCESS in 4m 27s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/d8abffe8f5d644b8b702c9cde0f98a23 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"18dfddd608d6abad3f5dbcdba28892ec8c2f3cd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 13:36:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"351d556ed74cc39d6cf6e98e9b92f5c4865e2c04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 14:55:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba61c9d3f7b74413af415e48102b4260 : SUCCESS in 1h 15m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ff8d931dc624bbbb0add0d3afba97f6 : SUCCESS in 1h 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20cfd43d1c684090ab19671c1d40f3d0 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ef7d7bf72f44970a40d1b7751cb4b9b : SUCCESS in 9m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/579d575b87454bb18e9f980853ea2346 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77f3e7434a94ee8bd2e1e2a417dc58d : SUCCESS in 5m 34s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/11c1c31081de4e5c9ddf0b17f6671f81 : SUCCESS in 6m 20s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/027b24af3a7e449a8f4b02b99b720d43 : SUCCESS in 6m 21s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/fa2b8fdc94a54ac28d52314472e02452 : SUCCESS in 5m 10s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/58b944567b9c4c79aed79ca3a73603ab : SUCCESS in 6m 23s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/3bcebda7a5e447c1be4f25c2ee362eb8 : SUCCESS in 5m 27s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/b626d0c6654941ff959d6745b7cbb182 : SUCCESS in 6m 37s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/1612cc1adc104b21911a60189b22e30c : SUCCESS in 6m 36s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/60e22a1f29564a568c56a39fc5acda03 : SUCCESS in 6m 46s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/bfba7c7d9c4d4d08ab4cc00e0cc0fa0c : SUCCESS in 6m 33s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/0204e900285b4409b95c90db6c5dee17 : SUCCESS in 5m 22s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/57b85b98f7a549baaad9771d849b7dde : SUCCESS in 4m 45s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/5c4a8a8661e64983a89a56cbdec4353c : SUCCESS in 4m 56s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/2f965220207240beac6209bfe2657aba : SUCCESS in 4m 15s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/bc6abcb01f4d4fbdb49ddedee31bfb40 : SUCCESS in 4m 46s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/e120bc3522954db6aa7859d400b8205b : SUCCESS in 4m 08s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/2a262c477b97466293cd3803eab69cf7 : SUCCESS in 4m 51s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/ba101b955cc547148fdf477b5b18ba9a : SUCCESS in 4m 20s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/5a7c7812b1294cd1a0db8c85b05b8c03 : SUCCESS in 6m 55s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/0302640cb35c4254b4e3f433cf4dd846 : SUCCESS in 4m 24s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/d12b0fc170594b3eb9e972c08e6486dc : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d165809336d781fdd62128fb07908c6895a27cbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 14:55:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c609c203a45a540b5d2cd54a0ea7ccfc71817125","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 14:56:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcdab62a8ce244ef8152cd7c126a3290 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1551b9006a7d1b64b46c5f3e4490f6ea30b6fbb4","revisions":{"c2ee1d22b6b6789ed839e82e69d2268465ceaa60":{"kind":"REWORK","_number":1,"created":"2019-03-18 14:57:56.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/28/644228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/28/644228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/28/644228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/28/644228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/28/644228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/28/644228/1"}}},"commit":{"parents":[{"commit":"d55449258a0e73c9d38cba5a02120c6a8d5ea4d9","subject":"Merge \"More explicitly document driver connection strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d55449258a0e73c9d38cba5a02120c6a8d5ea4d9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-03-18 14:57:54.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-03-18 14:57:54.000000000","tz":0},"subject":"Update .gitreview for stable/stein","message":"Update .gitreview for stable/stein\n\nChange-Id: I699c698754f498cacff4998f33ea0738682b86e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c2ee1d22b6b6789ed839e82e69d2268465ceaa60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c2ee1d22b6b6789ed839e82e69d2268465ceaa60"}]},"branch":"refs/heads/stable/stein"},"1551b9006a7d1b64b46c5f3e4490f6ea30b6fbb4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-23 15:42:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/644228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/28/644228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/28/644228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/28/644228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/28/644228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/28/644228/2"}}},"commit":{"parents":[{"commit":"6204beeb035b6107545fbac1677c4b29ced8156c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6204beeb035b6107545fbac1677c4b29ced8156c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-03-18 14:57:54.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2019-05-23 15:42:32.000000000","tz":0},"subject":"Update .gitreview for stable/stein","message":"Update .gitreview for stable/stein\n\nChange-Id: I699c698754f498cacff4998f33ea0738682b86e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1551b9006a7d1b64b46c5f3e4490f6ea30b6fbb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1551b9006a7d1b64b46c5f3e4490f6ea30b6fbb4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
