)]}'
{"id":"openstack%2Ftooz~648557","triplet_id":"openstack%2Ftooz~master~I821e2cd1f835ff1b46f334480ea8b53d6dd9205c","project":"openstack/tooz","branch":"master","topic":"bug/1752225","hashtags":[],"change_id":"I821e2cd1f835ff1b46f334480ea8b53d6dd9205c","subject":"Adds heartbeating to the consul driver. Also fixes a bug with the log warning level for the tenacity library.","status":"ABANDONED","created":"2019-03-28 23:29:15.000000000","updated":"2020-02-13 16:07:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"46882dbb38c39fcd34218f7d431baf14b0506b3c","_number":648557,"virtual_id_number":648557,"owner":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-30 03:47:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-13 14:52:04.000000000","updated_by":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"reviewer":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"state":"REVIEWER"}],"messages":[{"id":"86c04b19bc94eb27b1878cb846caa330401e15a6","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2019-03-28 23:29:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e3f458cdac5fc51ae3c200be610750e6dfb5de8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 01:26:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/57/648557/1/check/tempest-full/d42d0f6/ : SUCCESS in 1h 54m 34s\n- tempest-full-py3 http://logs.openstack.org/57/648557/1/check/tempest-full-py3/5ee14c5/ : SUCCESS in 1h 30m 56s\n- openstack-tox-cover http://logs.openstack.org/57/648557/1/check/openstack-tox-cover/2823337/cover/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/57/648557/1/check/openstack-tox-pep8/15b9e20/ : FAILURE in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/57/648557/1/check/openstack-tox-py27/23703d2/ : SUCCESS in 5m 51s\n- openstack-tox-py35 http://logs.openstack.org/57/648557/1/check/openstack-tox-py35/1186756/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/57/648557/1/check/openstack-tox-py36/197230c/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/57/648557/1/check/openstack-tox-py37/7476be9/ : SUCCESS in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/57/648557/1/check/openstack-tox-docs/a014f09/html/ : SUCCESS in 3m 41s\n- tooz-tox-py27-etcd http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-etcd/c233494/ : SUCCESS in 4m 17s\n- tooz-tox-py35-etcd http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-etcd/571d25f/ : FAILURE in 2m 56s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-etcd3/47ef1b8/ : SUCCESS in 3m 49s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-etcd3/2b81795/ : FAILURE in 3m 10s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-etcd3gw/ff36095/ : SUCCESS in 3m 28s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-etcd3gw/cbce29a/ : FAILURE in 4m 17s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-zookeeper/9e5b1c6/ : SUCCESS in 3m 45s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-zookeeper/aa241b7/ : FAILURE in 2m 51s\n- tooz-tox-py27-redis http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-redis/71bf710/ : FAILURE in 4m 12s\n- tooz-tox-py35-redis http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-redis/e1c8ea0/ : FAILURE in 4m 03s\n- tooz-tox-py27-sentinel http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-sentinel/231a388/ : FAILURE in 3m 22s\n- tooz-tox-py35-sentinel http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-sentinel/68d8a56/ : FAILURE in 2m 47s\n- tooz-tox-py27-memcached http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-memcached/e832e27/ : SUCCESS in 4m 41s\n- tooz-tox-py35-memcached http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-memcached/6cf68d8/ : FAILURE in 2m 40s\n- tooz-tox-py27-postgresql http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-postgresql/7b690b8/ : SUCCESS in 3m 45s\n- tooz-tox-py35-postgresql http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-postgresql/f399c24/ : FAILURE in 3m 49s\n- tooz-tox-py27-mysql http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-mysql/cc4c086/ : SUCCESS in 3m 40s\n- tooz-tox-py35-mysql http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-mysql/b1f2024/ : FAILURE in 2m 45s\n- tooz-tox-py27-consul http://logs.openstack.org/57/648557/1/check/tooz-tox-py27-consul/f626652/ : SUCCESS in 3m 42s\n- tooz-tox-py35-consul http://logs.openstack.org/57/648557/1/check/tooz-tox-py35-consul/48930cc/ : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"297c70b353e437cfb83e969b3c28f29c201bf788","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2019-03-30 01:46:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8399ada1bfaacc4c751e3b9b3faf0f799e7df749","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-30 03:47:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full http://logs.openstack.org/57/648557/2/check/tempest-full/927a78a/ : SUCCESS in 1h 45m 55s\n- tempest-full-py3 http://logs.openstack.org/57/648557/2/check/tempest-full-py3/c7c13a9/ : SUCCESS in 1h 44m 21s\n- openstack-tox-cover http://logs.openstack.org/57/648557/2/check/openstack-tox-cover/0dfc160/cover/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/57/648557/2/check/openstack-tox-pep8/978cdbe/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/57/648557/2/check/openstack-tox-py27/2c1cb85/ : SUCCESS in 4m 12s\n- openstack-tox-py35 http://logs.openstack.org/57/648557/2/check/openstack-tox-py35/9860381/ : SUCCESS in 6m 04s\n- openstack-tox-py36 http://logs.openstack.org/57/648557/2/check/openstack-tox-py36/9f3e18f/ : SUCCESS in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/57/648557/2/check/openstack-tox-py37/98d90d7/ : SUCCESS in 6m 24s\n- openstack-tox-docs http://logs.openstack.org/57/648557/2/check/openstack-tox-docs/a7ebf8a/html/ : SUCCESS in 3m 38s\n- tooz-tox-py27-etcd http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-etcd/64f9f23/ : SUCCESS in 6m 05s\n- tooz-tox-py35-etcd http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-etcd/084690e/ : FAILURE in 3m 16s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-etcd3/96accc6/ : SUCCESS in 5m 17s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-etcd3/8006c3b/ : FAILURE in 3m 14s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-etcd3gw/637b233/ : SUCCESS in 5m 27s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-etcd3gw/5e75275/ : FAILURE in 3m 23s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-zookeeper/f8e98fc/ : SUCCESS in 5m 35s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-zookeeper/25e8ac1/ : FAILURE in 3m 22s\n- tooz-tox-py27-redis http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-redis/eb84c3b/ : FAILURE in 4m 48s\n- tooz-tox-py35-redis http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-redis/861a2a6/ : FAILURE in 3m 21s\n- tooz-tox-py27-sentinel http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-sentinel/bd2059c/ : FAILURE in 4m 49s\n- tooz-tox-py35-sentinel http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-sentinel/5433fb5/ : FAILURE in 3m 33s\n- tooz-tox-py27-memcached http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-memcached/488defe/ : SUCCESS in 5m 04s\n- tooz-tox-py35-memcached http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-memcached/5a1a100/ : FAILURE in 3m 04s\n- tooz-tox-py27-postgresql http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-postgresql/31100eb/ : SUCCESS in 5m 12s\n- tooz-tox-py35-postgresql http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-postgresql/65ec82d/ : FAILURE in 3m 33s\n- tooz-tox-py27-mysql http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-mysql/d268df9/ : SUCCESS in 5m 20s\n- tooz-tox-py35-mysql http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-mysql/1fadbe4/ : FAILURE in 3m 33s\n- tooz-tox-py27-consul http://logs.openstack.org/57/648557/2/check/tooz-tox-py27-consul/f262627/ : SUCCESS in 5m 23s\n- tooz-tox-py35-consul http://logs.openstack.org/57/648557/2/check/tooz-tox-py35-consul/93fd1bd/ : FAILURE in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f1be7fd5a19e10dabbadfa949c6b076611cf631a","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-02-13 14:52:04.000000000","message":"Patch Set 2: Code-Review+1\n\nIs something holding this up at the moment? I also need this fix.","accounts_in_message":[],"_revision_number":2},{"id":"6f323aa7652a86d4f57cd49efc82f876d4064fe6","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-13 15:49:24.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c8c33a4f38e39f08428dc3fb199fce5d7d182573","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-13 16:03:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a7e9016b5c96fb87290e5778fcac6b24ccec2bc5","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-13 16:07:54.000000000","message":"Abandoned\n\nMessed up rebase","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c5b6db1b55f184705892a74ea5bb3b8323b9e81b","revisions":{"2e1d2ccf4648bd763b286a0482734e61cbe087a0":{"kind":"REWORK","_number":1,"created":"2019-03-28 23:29:15.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/57/648557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/57/648557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/57/648557/1"}}},"commit":{"parents":[{"commit":"de3a5f818b261e53f19feebcc0bfb64c7395c2aa","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/de3a5f818b261e53f19feebcc0bfb64c7395c2aa"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2019-03-28 23:26:01.000000000","tz":-240},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2019-03-28 23:29:15.000000000","tz":-240},"subject":"Adds heartbeating to the consul driver. Also fixes a bug with the log warning level for the tenacity library.","message":"Adds heartbeating to the consul driver. Also fixes a bug with the log warning level for the tenacity library.\n\nChange-Id: I821e2cd1f835ff1b46f334480ea8b53d6dd9205c\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e1d2ccf4648bd763b286a0482734e61cbe087a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e1d2ccf4648bd763b286a0482734e61cbe087a0"}]},"branch":"refs/heads/master"},"35940ec5d328a800b4957e7f49ccb9655b096a87":{"kind":"REWORK","_number":2,"created":"2019-03-30 01:46:53.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/57/648557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/57/648557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/57/648557/2"}}},"commit":{"parents":[{"commit":"de3a5f818b261e53f19feebcc0bfb64c7395c2aa","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/de3a5f818b261e53f19feebcc0bfb64c7395c2aa"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2019-03-28 23:26:01.000000000","tz":-240},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2019-03-30 01:46:46.000000000","tz":-240},"subject":"Adds heartbeating to the consul driver. Also fixes a bug with the log warning level for the tenacity library.","message":"Adds heartbeating to the consul driver. Also fixes a bug with the log warning level for the tenacity library.\n\nChange-Id: I821e2cd1f835ff1b46f334480ea8b53d6dd9205c\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/35940ec5d328a800b4957e7f49ccb9655b096a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/35940ec5d328a800b4957e7f49ccb9655b096a87"}]},"branch":"refs/heads/master"},"9a76be49d5e9b1e73dbdcaf328e647ca7a5176a8":{"kind":"NO_CHANGE","_number":3,"created":"2020-02-13 15:49:24.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/57/648557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/57/648557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/57/648557/3"}}},"commit":{"parents":[{"commit":"de3a5f818b261e53f19feebcc0bfb64c7395c2aa","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/de3a5f818b261e53f19feebcc0bfb64c7395c2aa"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2019-03-28 23:26:01.000000000","tz":-240},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 15:49:01.000000000","tz":-300},"subject":"Adds heartbeating to the consul driver. Also fixes a bug with the log warning level for the tenacity library.","message":"Adds heartbeating to the consul driver. Also fixes a bug with the log warning level for the tenacity library.\n\nChange-Id: I821e2cd1f835ff1b46f334480ea8b53d6dd9205c\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9a76be49d5e9b1e73dbdcaf328e647ca7a5176a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9a76be49d5e9b1e73dbdcaf328e647ca7a5176a8"}]},"branch":"refs/heads/master"},"c5b6db1b55f184705892a74ea5bb3b8323b9e81b":{"kind":"REWORK","_number":4,"created":"2020-02-13 16:03:03.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/57/648557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/57/648557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/57/648557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/57/648557/4"}}},"commit":{"parents":[{"commit":"de3a5f818b261e53f19feebcc0bfb64c7395c2aa","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/de3a5f818b261e53f19feebcc0bfb64c7395c2aa"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2019-03-28 23:26:01.000000000","tz":-240},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:01:48.000000000","tz":-300},"subject":"Adds heartbeating to the consul driver. Also fixes a bug with the log warning level for the tenacity library.","message":"Adds heartbeating to the consul driver. Also fixes a bug with the log warning level for the tenacity library.\n\nChange-Id: I821e2cd1f835ff1b46f334480ea8b53d6dd9205c\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c5b6db1b55f184705892a74ea5bb3b8323b9e81b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c5b6db1b55f184705892a74ea5bb3b8323b9e81b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
