)]}'
{"id":"openstack%2Ftooz~657454","triplet_id":"openstack%2Ftooz~master~I94c109144712c01197ad37da0a59ae45c16204bd","project":"openstack/tooz","branch":"master","topic":"bump-grpcio","hashtags":[],"change_id":"I94c109144712c01197ad37da0a59ae45c16204bd","subject":"Uncap grpcio","status":"ABANDONED","created":"2019-05-06 21:06:35.000000000","updated":"2019-05-28 16:45:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e35278407b9799730c71163982d25c84d8023ab2","_number":657454,"virtual_id_number":657454,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-2,"date":"2019-05-20 10:01:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2019-05-11 12:16:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2019-05-20 07:45:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":-1,"date":"2019-05-23 15:58:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"all":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2019-05-20 07:45:52.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-11 12:16:01.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2019-05-20 07:45:52.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-05-20 10:01:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-23 15:58:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"9054c312d0949d2dc6160e2b200147d3b0ce446d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-06 21:06:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c96fa5282f36dd4d75b817ef892d0535a713b2b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 22:25:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/54/657454/1/check/requirements-check/6150ec2/ : SUCCESS in 3m 02s\n- tempest-full http://logs.openstack.org/54/657454/1/check/tempest-full/bf82fae/ : FAILURE in 41m 20s\n- tempest-full-py3 http://logs.openstack.org/54/657454/1/check/tempest-full-py3/fb8073c/ : SUCCESS in 1h 18m 03s\n- openstack-tox-cover http://logs.openstack.org/54/657454/1/check/openstack-tox-cover/5f6d6cd/cover/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/54/657454/1/check/openstack-tox-pep8/d58749f/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/54/657454/1/check/openstack-tox-py27/3fddc13/ : SUCCESS in 4m 26s\n- openstack-tox-py35 http://logs.openstack.org/54/657454/1/check/openstack-tox-py35/7880645/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/54/657454/1/check/openstack-tox-py36/6b458ef/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/54/657454/1/check/openstack-tox-py37/3db44c7/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/54/657454/1/check/openstack-tox-docs/ba6a32c/html/ : SUCCESS in 4m 04s\n- tooz-tox-py27-etcd http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-etcd/c9dd1f8/ : SUCCESS in 3m 57s\n- tooz-tox-py35-etcd http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-etcd/dc68b1e/ : FAILURE in 3m 14s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-etcd3/de6c459/ : SUCCESS in 4m 16s\n- tooz-tox-py35-etcd3 http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-etcd3/e5f4ce5/ : FAILURE in 3m 26s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-etcd3gw/84b057c/ : SUCCESS in 3m 44s\n- tooz-tox-py35-etcd3gw http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-etcd3gw/f0dfeef/ : FAILURE in 2m 49s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-zookeeper/9a2cb13/ : SUCCESS in 4m 04s\n- tooz-tox-py35-zookeeper http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-zookeeper/62900c8/ : FAILURE in 3m 00s\n- tooz-tox-py27-redis http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-redis/8d7b0f4/ : FAILURE in 3m 11s\n- tooz-tox-py35-redis http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-redis/a78f68f/ : FAILURE in 3m 29s\n- tooz-tox-py27-sentinel http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-sentinel/ad55a66/ : FAILURE in 3m 48s\n- tooz-tox-py35-sentinel http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-sentinel/17748e9/ : FAILURE in 4m 31s\n- tooz-tox-py27-memcached http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-memcached/21d1b01/ : SUCCESS in 4m 29s\n- tooz-tox-py35-memcached http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-memcached/e876932/ : FAILURE in 2m 37s\n- tooz-tox-py27-postgresql http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-postgresql/595cd1d/ : SUCCESS in 4m 19s\n- tooz-tox-py35-postgresql http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-postgresql/efb794a/ : FAILURE in 2m 50s\n- tooz-tox-py27-mysql http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-mysql/7365e90/ : SUCCESS in 3m 42s\n- tooz-tox-py35-mysql http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-mysql/e6dbd54/ : FAILURE in 2m 56s\n- tooz-tox-py27-consul http://logs.openstack.org/54/657454/1/check/tooz-tox-py27-consul/3386c14/ : SUCCESS in 4m 45s\n- tooz-tox-py35-consul http://logs.openstack.org/54/657454/1/check/tooz-tox-py35-consul/b0d497a/ : FAILURE in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6777f5b03262455db6d40764fc4bc0ca438229c8","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-09 20:21:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"054c68bbc4123a75b964c44322e7aaa266711813","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 22:15:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/54/657454/2/check/requirements-check/0cf6325/ : SUCCESS in 3m 44s\n- tempest-full http://logs.openstack.org/54/657454/2/check/tempest-full/0fe7768/ : SUCCESS in 1h 48m 56s\n- tempest-full-py3 http://logs.openstack.org/54/657454/2/check/tempest-full-py3/6560853/ : SUCCESS in 1h 34m 16s\n- openstack-tox-cover http://logs.openstack.org/54/657454/2/check/openstack-tox-cover/22de0ee/cover/ : SUCCESS in 4m 04s\n- openstack-tox-pep8 http://logs.openstack.org/54/657454/2/check/openstack-tox-pep8/47cd2b7/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/54/657454/2/check/openstack-tox-py27/a15086a/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/54/657454/2/check/openstack-tox-py36/f2eb18d/ : SUCCESS in 5m 05s\n- openstack-tox-py37 http://logs.openstack.org/54/657454/2/check/openstack-tox-py37/4ec4628/ : SUCCESS in 4m 38s\n- openstack-tox-docs http://logs.openstack.org/54/657454/2/check/openstack-tox-docs/d0c02b0/html/ : SUCCESS in 3m 39s\n- tooz-tox-py27-etcd http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-etcd/8572f2b/ : SUCCESS in 3m 49s\n- tooz-tox-py36-etcd http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-etcd/34bac83/ : SUCCESS in 3m 51s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-etcd3/e3e6eac/ : SUCCESS in 3m 53s\n- tooz-tox-py36-etcd3 http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-etcd3/f2c9ff6/ : SUCCESS in 3m 47s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-etcd3gw/ede86ab/ : SUCCESS in 3m 27s\n- tooz-tox-py36-etcd3gw http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-etcd3gw/4de9c33/ : SUCCESS in 3m 28s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-zookeeper/570619f/ : SUCCESS in 3m 41s\n- tooz-tox-py36-zookeeper http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-zookeeper/4428a6f/ : SUCCESS in 3m 58s\n- tooz-tox-py27-redis http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-redis/97b2e91/ : FAILURE in 3m 21s\n- tooz-tox-py36-redis http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-redis/60cf240/ : FAILURE in 3m 21s\n- tooz-tox-py27-sentinel http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-sentinel/c718f3e/ : FAILURE in 3m 05s\n- tooz-tox-py36-sentinel http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-sentinel/2b6357e/ : FAILURE in 4m 20s\n- tooz-tox-py27-memcached http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-memcached/9a1aa3c/ : SUCCESS in 3m 50s\n- tooz-tox-py36-memcached http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-memcached/1c2671d/ : SUCCESS in 4m 06s\n- tooz-tox-py27-postgresql http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-postgresql/504ab9a/ : SUCCESS in 3m 36s\n- tooz-tox-py36-postgresql http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-postgresql/b79831b/ : SUCCESS in 3m 55s\n- tooz-tox-py27-mysql http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-mysql/9b0d348/ : SUCCESS in 3m 53s\n- tooz-tox-py36-mysql http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-mysql/d0c93d0/ : SUCCESS in 4m 13s\n- tooz-tox-py27-consul http://logs.openstack.org/54/657454/2/check/tooz-tox-py27-consul/5d03a31/ : SUCCESS in 3m 49s\n- tooz-tox-py36-consul http://logs.openstack.org/54/657454/2/check/tooz-tox-py36-consul/4285dc8/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1f763a19dece19b90242887d747fe5b3d56fa14e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-10 20:04:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d761d83efee373c1f24e3d3eae84f6aaebb4a7a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 22:04:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/54/657454/3/check/requirements-check/fb78fd1/ : SUCCESS in 2m 59s\n- tempest-full http://logs.openstack.org/54/657454/3/check/tempest-full/1442c9a/ : SUCCESS in 1h 52m 54s\n- tempest-full-py3 http://logs.openstack.org/54/657454/3/check/tempest-full-py3/c9e8391/ : SUCCESS in 1h 39m 04s\n- openstack-tox-cover http://logs.openstack.org/54/657454/3/check/openstack-tox-cover/1ff4383/cover/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/54/657454/3/check/openstack-tox-pep8/3b98ab9/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/54/657454/3/check/openstack-tox-py27/442e400/ : SUCCESS in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/54/657454/3/check/openstack-tox-py36/91bf7ff/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/54/657454/3/check/openstack-tox-py37/0e2dcbe/ : SUCCESS in 4m 26s\n- openstack-tox-docs http://logs.openstack.org/54/657454/3/check/openstack-tox-docs/ef5523b/html/ : SUCCESS in 3m 43s\n- tooz-tox-py27-etcd http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-etcd/1478fac/ : SUCCESS in 3m 47s\n- tooz-tox-py36-etcd http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-etcd/39e3533/ : SUCCESS in 4m 17s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-etcd3/3652121/ : SUCCESS in 4m 05s\n- tooz-tox-py36-etcd3 http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-etcd3/63b9f7a/ : SUCCESS in 4m 12s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-etcd3gw/1c79beb/ : SUCCESS in 3m 47s\n- tooz-tox-py36-etcd3gw http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-etcd3gw/63eec40/ : SUCCESS in 4m 02s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-zookeeper/8e68897/ : SUCCESS in 3m 52s\n- tooz-tox-py36-zookeeper http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-zookeeper/644cff9/ : SUCCESS in 4m 34s\n- tooz-tox-py27-redis http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-redis/337c2e0/ : SUCCESS in 3m 57s\n- tooz-tox-py36-redis http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-redis/c7ca731/ : SUCCESS in 4m 10s\n- tooz-tox-py27-sentinel http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-sentinel/b51da3e/ : SUCCESS in 4m 14s\n- tooz-tox-py36-sentinel http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-sentinel/bd4843a/ : SUCCESS in 4m 07s\n- tooz-tox-py27-memcached http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-memcached/9f98d54/ : SUCCESS in 4m 11s\n- tooz-tox-py36-memcached http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-memcached/c5273b5/ : SUCCESS in 4m 02s\n- tooz-tox-py27-postgresql http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-postgresql/a487158/ : SUCCESS in 3m 45s\n- tooz-tox-py36-postgresql http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-postgresql/35ae974/ : SUCCESS in 4m 00s\n- tooz-tox-py27-mysql http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-mysql/b2e1eba/ : SUCCESS in 3m 56s\n- tooz-tox-py36-mysql http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-mysql/0845c84/ : SUCCESS in 4m 04s\n- tooz-tox-py27-consul http://logs.openstack.org/54/657454/3/check/tooz-tox-py27-consul/a72419a/ : SUCCESS in 3m 45s\n- tooz-tox-py36-consul http://logs.openstack.org/54/657454/3/check/tooz-tox-py36-consul/ba56212/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"f00af6534136304679d75bec0540c956c8ed9c22","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-05-11 12:16:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a08b4f23d14ad9495e6125b0f12b80f091fdca1e","author":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"date":"2019-05-20 07:45:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"558b4fbef3ebe57d1532eb6ae425cb4bb16f17f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:46:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9a29b98a5f6affdd59390ea9f1cb2903a26dd824","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:01:10.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/54/657454/3/gate/requirements-check/54832e2/ : FAILURE in 4m 13s\n- tempest-full http://logs.openstack.org/54/657454/3/gate/tempest-full/24428d6/ : TIMED_OUT in 2h 14m 17s\n- tempest-full-py3 http://logs.openstack.org/54/657454/3/gate/tempest-full-py3/44f66de/ : SUCCESS in 1h 46m 09s\n- openstack-tox-pep8 http://logs.openstack.org/54/657454/3/gate/openstack-tox-pep8/84db312/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/54/657454/3/gate/openstack-tox-py27/b06c72e/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/54/657454/3/gate/openstack-tox-py36/0a81a3a/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/54/657454/3/gate/openstack-tox-py37/a0fa7cc/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/54/657454/3/gate/openstack-tox-docs/a486161/html/ : SUCCESS in 5m 52s\n- tooz-tox-py27-etcd http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-etcd/713c2de/ : SUCCESS in 4m 39s\n- tooz-tox-py36-etcd http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-etcd/3523a7f/ : SUCCESS in 4m 31s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-etcd3/5e7918a/ : SUCCESS in 4m 55s\n- tooz-tox-py36-etcd3 http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-etcd3/0c3f7ae/ : SUCCESS in 4m 13s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-etcd3gw/11161fc/ : SUCCESS in 4m 28s\n- tooz-tox-py36-etcd3gw http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-etcd3gw/998a074/ : SUCCESS in 4m 23s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-zookeeper/db95549/ : SUCCESS in 4m 29s\n- tooz-tox-py36-zookeeper http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-zookeeper/cd9f18e/ : SUCCESS in 4m 32s\n- tooz-tox-py27-redis http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-redis/3f94e1d/ : SUCCESS in 5m 34s\n- tooz-tox-py36-redis http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-redis/0360cff/ : SUCCESS in 4m 31s\n- tooz-tox-py27-sentinel http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-sentinel/3b19e59/ : SUCCESS in 4m 16s\n- tooz-tox-py36-sentinel http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-sentinel/9fb99a0/ : SUCCESS in 4m 39s\n- tooz-tox-py27-memcached http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-memcached/fb957f2/ : SUCCESS in 4m 19s\n- tooz-tox-py36-memcached http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-memcached/4db12d8/ : SUCCESS in 4m 45s\n- tooz-tox-py27-postgresql http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-postgresql/feac4bc/ : SUCCESS in 4m 40s\n- tooz-tox-py36-postgresql http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-postgresql/44a01d0/ : SUCCESS in 4m 09s\n- tooz-tox-py27-mysql http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-mysql/e28cf62/ : SUCCESS in 4m 38s\n- tooz-tox-py36-mysql http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-mysql/9274d9c/ : SUCCESS in 4m 26s\n- tooz-tox-py27-consul http://logs.openstack.org/54/657454/3/gate/tooz-tox-py27-consul/fddf780/ : SUCCESS in 4m 52s\n- tooz-tox-py36-consul http://logs.openstack.org/54/657454/3/gate/tooz-tox-py36-consul/e9bce9a/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":3},{"id":"166a412f875c8f5072131d8652f698f70fd65b48","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-05-23 15:58:48.000000000","message":"Patch Set 3: Code-Review-1\n\nLooks like this needs a Sphinx fix too like https://review.opendev.org/#/c/659590/. We should choose one and abandon the other","accounts_in_message":[],"_revision_number":3},{"id":"5170e4b73df6c9977b429f68862ac90dfd73847c","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-05-28 16:45:35.000000000","message":"Abandoned\n\nIntegrated with https://review.opendev.org/#/c/659590/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fdb1d7fb76cd21aad57f362005a8cd92f9cd6090","revisions":{"10cee44f766475f9cfc8d299df7802d0d73e6362":{"kind":"REWORK","_number":1,"created":"2019-05-06 21:06:35.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/54/657454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/54/657454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/54/657454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/54/657454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/54/657454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/54/657454/1"}}},"commit":{"parents":[{"commit":"c02f9dd6c65cc34f9007af82773349efe13507c0","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c02f9dd6c65cc34f9007af82773349efe13507c0"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-06 21:05:16.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-06 21:06:17.000000000","tz":0},"subject":"Uncap grpcio","message":"Uncap grpcio\n\nBump the minimum version to 1.18.0, which fixes the problem that\nprompted the cap in the first place.\n\nChange-Id: I94c109144712c01197ad37da0a59ae45c16204bd\nDepends-On: https://review.opendev.org/625010\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/10cee44f766475f9cfc8d299df7802d0d73e6362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/10cee44f766475f9cfc8d299df7802d0d73e6362"}]},"branch":"refs/heads/master"},"b958736081b5f09e5534e83ed05b6d65eb6db818":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-09 20:21:21.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/54/657454/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/54/657454/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/54/657454/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/54/657454/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/54/657454/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/54/657454/2"}}},"commit":{"parents":[{"commit":"ad1e28fd8c5cc99f1b3f3f7a97ee3c13744d13c1","subject":"Move testing to py36","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ad1e28fd8c5cc99f1b3f3f7a97ee3c13744d13c1"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-06 21:05:16.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-09 20:21:08.000000000","tz":0},"subject":"Uncap grpcio","message":"Uncap grpcio\n\nBump the minimum version to 1.18.0, which fixes the problem that\nprompted the cap in the first place.\n\nChange-Id: I94c109144712c01197ad37da0a59ae45c16204bd\nDepends-On: https://review.opendev.org/625010\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b958736081b5f09e5534e83ed05b6d65eb6db818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b958736081b5f09e5534e83ed05b6d65eb6db818"}]},"branch":"refs/heads/master"},"fdb1d7fb76cd21aad57f362005a8cd92f9cd6090":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-10 20:04:36.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/54/657454/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/54/657454/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/54/657454/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/54/657454/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/54/657454/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/54/657454/3"}}},"commit":{"parents":[{"commit":"8207427fffc72547387369ce7ba2696a86bf7626","subject":"Unblock tooz gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8207427fffc72547387369ce7ba2696a86bf7626"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-06 21:05:16.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-10 20:03:49.000000000","tz":0},"subject":"Uncap grpcio","message":"Uncap grpcio\n\nBump the minimum version to 1.18.0, which fixes the problem that\nprompted the cap in the first place.\n\nChange-Id: I94c109144712c01197ad37da0a59ae45c16204bd\nDepends-On: https://review.opendev.org/625010\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fdb1d7fb76cd21aad57f362005a8cd92f9cd6090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fdb1d7fb76cd21aad57f362005a8cd92f9cd6090"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
