)]}'
{"id":"openstack%2Ftooz~669472","triplet_id":"openstack%2Ftooz~master~Ie594497012dd65372c0014890f1fc989714a94b1","project":"openstack/tooz","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ie594497012dd65372c0014890f1fc989714a94b1","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 19:20:19.000000000","updated":"2019-08-12 14:57:01.000000000","submitted":"2019-08-12 14:57:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669472-1565621821915-888803a0","meta_rev_id":"f408e16e9ac4fd60101127d5b819621b1fe5292f","_number":669472,"virtual_id_number":669472,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2019-08-12 14:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-09 20:45:38.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-08-08 19:39:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-07-16 20:30:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2019-08-12 12:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":1,"date":"2019-07-25 06:27:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2019-08-12 12:57:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 20:30:07.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-07-25 06:27:19.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-08-08 19:39:24.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-08-12 12:57:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-08-12 14:57:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51ee132cc8f8a1134d319a1e89d6eb613fb37c04","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 19:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e24c9be76e626ed48a863dd3862a2e54d725cb25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:14:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full http://logs.openstack.org/72/669472/1/check/tempest-full/8608088/ : SUCCESS in 1h 51m 10s\n- tempest-full-py3 http://logs.openstack.org/72/669472/1/check/tempest-full-py3/9d5eb1e/ : SUCCESS in 1h 40m 30s\n- openstack-tox-cover http://logs.openstack.org/72/669472/1/check/openstack-tox-cover/3ba5ad1/cover/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/72/669472/1/check/openstack-tox-pep8/b5e87d4/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/72/669472/1/check/openstack-tox-py27/c01b81d/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/72/669472/1/check/openstack-tox-py36/ca75c7c/ : SUCCESS in 7m 49s\n- openstack-tox-py37 http://logs.openstack.org/72/669472/1/check/openstack-tox-py37/8efbc7e/ : SUCCESS in 4m 22s\n- openstack-tox-docs http://logs.openstack.org/72/669472/1/check/openstack-tox-docs/4bafd10/html/ : SUCCESS in 4m 12s\n- build-openstack-releasenotes http://logs.openstack.org/72/669472/1/check/build-openstack-releasenotes/19bb5ac/html/ : SUCCESS in 5m 23s\n- tooz-tox-py27-etcd http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-etcd/c23bf0a/ : SUCCESS in 3m 46s\n- tooz-tox-py36-etcd http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-etcd/b18c160/ : SUCCESS in 5m 33s\n- tooz-tox-py27-etcd3 http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-etcd3/c99384a/ : SUCCESS in 4m 10s\n- tooz-tox-py36-etcd3 http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-etcd3/cc14531/ : SUCCESS in 6m 41s\n- tooz-tox-py27-etcd3gw http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-etcd3gw/b07a5ea/ : SUCCESS in 3m 45s\n- tooz-tox-py36-etcd3gw http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-etcd3gw/3844708/ : SUCCESS in 3m 37s\n- tooz-tox-py27-zookeeper http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-zookeeper/c394bd8/ : SUCCESS in 3m 51s\n- tooz-tox-py36-zookeeper http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-zookeeper/8afcbe9/ : SUCCESS in 3m 55s\n- tooz-tox-py27-redis http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-redis/315c6d1/ : SUCCESS in 3m 50s\n- tooz-tox-py36-redis http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-redis/dfe9d61/ : SUCCESS in 5m 25s\n- tooz-tox-py27-sentinel http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-sentinel/6c28fb2/ : SUCCESS in 4m 31s\n- tooz-tox-py36-sentinel http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-sentinel/b4f090e/ : SUCCESS in 4m 55s\n- tooz-tox-py27-memcached http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-memcached/214c42b/ : SUCCESS in 4m 42s\n- tooz-tox-py36-memcached http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-memcached/ac4e744/ : SUCCESS in 7m 39s\n- tooz-tox-py27-postgresql http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-postgresql/73bb341/ : SUCCESS in 11m 04s\n- tooz-tox-py36-postgresql http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-postgresql/01ee30b/ : SUCCESS in 4m 21s\n- tooz-tox-py27-mysql http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-mysql/bb73a50/ : SUCCESS in 4m 08s\n- tooz-tox-py36-mysql http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-mysql/b6b4370/ : SUCCESS in 3m 41s\n- tooz-tox-py27-consul http://logs.openstack.org/72/669472/1/check/tooz-tox-py27-consul/a8e7d18/ : SUCCESS in 3m 43s\n- tooz-tox-py36-consul http://logs.openstack.org/72/669472/1/check/tooz-tox-py36-consul/2c40968/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"bf351265a3e6d9d4f27483f5ccd32502e0efb704","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-07-16 20:30:07.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46558909b6377b8cabb39e8582f0f6fa84ea3e36","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-25 06:27:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c0313725eff6987b1bef2fdb2a4650c8054776da","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-08-05 14:14:11.000000000","message":"Patch Set 1: Code-Review+1\n\nHello Corey,\n\nHave you seen the Zane\u0027s question?\n\nI\u0027ve compared this patch and your changes to the others patches of this serie and I also think we need to switch the `tooz-tox-py36-*` to python 3.7.\n\nThoughts?","accounts_in_message":[],"_revision_number":1},{"id":"180502aaec307f71c70065b4fa92f13e53ead479","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-08 19:39:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nAgree that we should add py37 jobs for the drivers, but since this is about switching to the new standard unit test template I\u0027m okay with it going in as-is and doing the other changes as a followup.","accounts_in_message":[],"_revision_number":1},{"id":"c43ef898454cc6a063d7374578c84b439c803bef","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-08-08 19:47:53.000000000","message":"Patch Set 1:\n\nAdded the driver jobs in https://review.opendev.org/#/c/675454","accounts_in_message":[],"_revision_number":1},{"id":"df163572fb719760ace53b89ae1aa0e6d7240b66","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-08-09 20:45:38.000000000","message":"Patch Set 1:\n\nThanks Ben. I had the drivers on my todo list but hadn\u0027t gotten to it. Appreciate that.","accounts_in_message":[],"_revision_number":1},{"id":"425a409f66d67ad863d323260bfe0687426e4af7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-08-12 12:57:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c99db3c95f8fe694ca06544b6c0b5733542aece","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 12:57:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1d6a421ab35f93d929b966b3cba37b7cebb42ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:57:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://logs.opendev.org/72/669472/1/gate/tempest-full/75743d5/ : SUCCESS in 1h 50m 04s\n- tempest-full-py3 https://logs.opendev.org/72/669472/1/gate/tempest-full-py3/a32d216/ : SUCCESS in 1h 33m 16s\n- openstack-tox-pep8 https://logs.opendev.org/72/669472/1/gate/openstack-tox-pep8/cdac82b/ : SUCCESS in 3m 52s\n- openstack-tox-py27 https://logs.opendev.org/72/669472/1/gate/openstack-tox-py27/6472571/ : SUCCESS in 4m 01s\n- openstack-tox-py36 https://logs.opendev.org/72/669472/1/gate/openstack-tox-py36/9314fe2/ : SUCCESS in 4m 35s\n- openstack-tox-py37 https://logs.opendev.org/72/669472/1/gate/openstack-tox-py37/94d3bdb/ : SUCCESS in 5m 16s\n- openstack-tox-docs https://logs.opendev.org/72/669472/1/gate/openstack-tox-docs/75380ea/html/ : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://logs.opendev.org/72/669472/1/gate/build-openstack-releasenotes/1bcd9e3/html/ : SUCCESS in 3m 33s\n- tooz-tox-py27-etcd https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-etcd/0bd3f93/ : SUCCESS in 3m 47s\n- tooz-tox-py36-etcd https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-etcd/e2a1665/ : SUCCESS in 3m 57s\n- tooz-tox-py27-etcd3 https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-etcd3/2d68728/ : SUCCESS in 3m 35s\n- tooz-tox-py36-etcd3 https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-etcd3/8f828da/ : SUCCESS in 4m 21s\n- tooz-tox-py27-etcd3gw https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-etcd3gw/8c6f267/ : SUCCESS in 5m 06s\n- tooz-tox-py36-etcd3gw https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-etcd3gw/9fbaa12/ : SUCCESS in 4m 31s\n- tooz-tox-py27-zookeeper https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-zookeeper/d97fb59/ : SUCCESS in 5m 16s\n- tooz-tox-py36-zookeeper https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-zookeeper/7149386/ : SUCCESS in 5m 14s\n- tooz-tox-py27-redis https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-redis/c087f33/ : SUCCESS in 4m 05s\n- tooz-tox-py36-redis https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-redis/d4ffdf1/ : SUCCESS in 5m 48s\n- tooz-tox-py27-sentinel https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-sentinel/c0085e0/ : SUCCESS in 5m 13s\n- tooz-tox-py36-sentinel https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-sentinel/63c50a3/ : SUCCESS in 5m 51s\n- tooz-tox-py27-memcached https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-memcached/43bcdd9/ : SUCCESS in 4m 03s\n- tooz-tox-py36-memcached https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-memcached/a700e2b/ : SUCCESS in 5m 29s\n- tooz-tox-py27-postgresql https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-postgresql/e5b1af8/ : SUCCESS in 4m 18s\n- tooz-tox-py36-postgresql https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-postgresql/b7658fa/ : SUCCESS in 4m 40s\n- tooz-tox-py27-mysql https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-mysql/c766860/ : SUCCESS in 4m 40s\n- tooz-tox-py36-mysql https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-mysql/6b3b894/ : SUCCESS in 4m 12s\n- tooz-tox-py27-consul https://logs.opendev.org/72/669472/1/gate/tooz-tox-py27-consul/e8cf874/ : SUCCESS in 3m 45s\n- tooz-tox-py36-consul https://logs.opendev.org/72/669472/1/gate/tooz-tox-py36-consul/d4fe869/ : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c76b3c2cd6e319e6427959e32930dcdbec76c8b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 14:57:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f7a0e290e4a7c02d6998fe3a83421510ab157fb","revisions":{"5f7a0e290e4a7c02d6998fe3a83421510ab157fb":{"kind":"REWORK","_number":1,"created":"2019-07-05 19:20:19.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/72/669472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/72/669472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/72/669472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/72/669472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/72/669472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/72/669472/1"}}},"commit":{"parents":[{"commit":"ef5277d75c367ddf71fcddff3e7b47459914e0dc","subject":"Move test deps to test-requirements.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ef5277d75c367ddf71fcddff3e7b47459914e0dc"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:57:01.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:20:19.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ie594497012dd65372c0014890f1fc989714a94b1\nStory: #2005924\nTask: #34234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5f7a0e290e4a7c02d6998fe3a83421510ab157fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5f7a0e290e4a7c02d6998fe3a83421510ab157fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
