)]}'
{"id":"openstack%2Ftooz~680286","triplet_id":"openstack%2Ftooz~master~I8572b0df2615f8f6e8f666be988b49e63a549679","project":"openstack/tooz","branch":"master","hashtags":[],"change_id":"I8572b0df2615f8f6e8f666be988b49e63a549679","subject":"Add shared arg in metaclass Lock","status":"MERGED","created":"2019-09-05 03:17:22.000000000","updated":"2019-10-10 15:21:09.000000000","submitted":"2019-10-10 15:19:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680286-1570720782393-94a06386","meta_rev_id":"de14cc9b8824d52d30ffa27a1abeeba506ba59bb","_number":680286,"virtual_id_number":680286,"owner":{"_account_id":25837,"name":"yenai","email":"yenai_yewu@cmss.chinamobile.com","username":"yenai2008"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 15:19:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":2,"date":"2019-09-23 12:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-10-10 13:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"date":"2019-09-11 07:40:57.000000000","_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-10-10 13:29:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 07:40:57.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2019-09-23 12:15:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-10 13:29:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-10-10 15:19:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc649525b0d089ba9900a4bc0b2e47f62e33e2ec","author":{"_account_id":25837,"name":"yenai","email":"yenai_yewu@cmss.chinamobile.com","username":"yenai2008"},"date":"2019-09-05 03:17:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfc4a11b9d888aa2bca2a6e91d543044ab747a06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 05:01:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/290913f42fdb41d19476d12b23e8f750 : SUCCESS in 1h 37m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8135c7812e5c457cb3bca70bd4aac0e7 : SUCCESS in 1h 33m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16415cd814b7406babd60ddf330d06a3 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29477abd88224b93aa97960d792ded64 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df6878f31ed944848d9cb07424f23646 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a24c69e506a24c36a6bb9706885c0b66 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fee74d416df48d88658404b4df8ca76 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/955dfc981fec4dc48b46edb7eee8b0ee : SUCCESS in 4m 56s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/c7de7567b688401bb1be3685f939299b : SUCCESS in 6m 16s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/4259109e20bb4f139bcd7c4074b82868 : SUCCESS in 5m 14s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/b23d66deb5454833b68db5d95d78d348 : SUCCESS in 4m 20s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/0cbdf76f4a5d441c9b0f41ffe4e969a0 : SUCCESS in 4m 59s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/26ba943192174d368bc2ea432db325af : SUCCESS in 4m 34s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/50260cc3e3ba4c0aa416c73c8496bfce : SUCCESS in 4m 37s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/3db9c0edbc7d4eeb8bf20726d8d73d5c : SUCCESS in 4m 34s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/3a74df0fd87b4d25ab35a79df31c5a22 : SUCCESS in 4m 58s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/22d454aedd314aae9b2a8f68b3f5ecb0 : SUCCESS in 5m 19s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/b6f7befdbb674840871e2f41aaf1af77 : SUCCESS in 4m 34s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/570f5f4d74704d30997a3e14602d2cad : SUCCESS in 5m 11s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/df2bb31363d84d40a75292772d549ffe : SUCCESS in 6m 05s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/21ea15353e7145d3bac5ea31388c9048 : SUCCESS in 6m 02s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/0347b14883324d29bcccd649f4c7570a : SUCCESS in 5m 37s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/cb3a4f1492e54a8fa37f1edc921078c1 : SUCCESS in 4m 54s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/7618aec5624c43c09a4066f627717339 : SUCCESS in 4m 34s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/16c5083907624cefa5a60797d8aae1e9 : SUCCESS in 5m 31s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/d0b804c1a9fa47e0b9744274b1b5a485 : SUCCESS in 4m 58s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/953aeb0660e94d8bab1a2e0405f47096 : SUCCESS in 4m 25s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/40f80c3bfac445abad58f594ffc9ab87 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c1762cbc5f549f7df1e72aa49ae5ae41bca209f2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-23 12:15:13.000000000","message":"Patch Set 1: Code-Review+2\n\nYup, all the implementations of this have that argument. Suprised abc allows this, tbh","accounts_in_message":[],"_revision_number":1},{"id":"0aa524d1522ce6700186d50809294349a216285b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-10 13:29:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nPython and ABC allow methods who inherited from a given abstract method to override the signature of this one, in other words it\u0027s not forbidden [1] like with principles like LSP [2].\n\n[1] https://docs.python.org/3/library/abc.html#abc.abstractmethod\n[2] https://en.wikipedia.org/wiki/Liskov_substitution_principle","accounts_in_message":[],"_revision_number":1},{"id":"633d9269e0cd109fcdde578121a4e27f886ceff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:29:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c157be8c9b80e30878bced7643dcb4848fdfbf7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:19:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/997b8cf47b9c4e59b83f234232ad4df5 : SUCCESS in 1h 48m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/554377f0d9264e0e8e361e5ece016a16 : SUCCESS in 1h 38m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/957a4ad599814ce2a2d4c35694d1e5ae : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7e77b83b2be40fa8abaa4fe5dbaf092 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68be151051604841bcb1afac26271eb8 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7922f3acd48b49489d04a083d71a1f7c : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88423353c6d7482d93f077585368ee93 : SUCCESS in 4m 07s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/271cb5a95cbc40b0b9d9faa43b0af5b4 : SUCCESS in 4m 12s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/78acd5c0548644708eefc54da73860c5 : SUCCESS in 4m 28s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/b69829492132475ab495cb97f8df0549 : SUCCESS in 6m 35s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/6029cb448b614894b92e46c53af0c097 : SUCCESS in 4m 29s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/6cef2941c948490681a7ae525764806d : SUCCESS in 4m 04s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/5ab147b3c08046feb9a77d27fe0d326d : SUCCESS in 4m 12s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/5c000d6c263544609c5e009fa693bba8 : SUCCESS in 4m 59s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/1f4c510647724d07b0452a9e4c457285 : SUCCESS in 4m 12s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/5f267e7a3ae5491e9586247a8d5ce6c5 : SUCCESS in 3m 42s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/347d8a2726bb4bd2821e2134dd22c473 : SUCCESS in 5m 07s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/3344ebbe3bc04e22b025ba3cea144f25 : SUCCESS in 4m 23s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/e7a60bfa64534118835cd24ccfeefbe6 : SUCCESS in 4m 08s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/033a9e2c07f54c2b9ed2738c922c014a : SUCCESS in 4m 11s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/e118bd8fc6c14f0f9fa326f7e8c789d1 : SUCCESS in 3m 39s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/14b0da0216334062a67c644b4247ed6f : SUCCESS in 3m 52s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/6253320f9e374f93ba32363a922df815 : SUCCESS in 4m 48s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/768df77fdacf480b8df981eeae17cb01 : SUCCESS in 4m 26s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/903f439e386e4451ad1ffd76c40fac81 : SUCCESS in 4m 49s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/7a050ee47c7e4459afd1c02d564d91aa : SUCCESS in 4m 25s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/987d2ec444814cd79594e9f6556bd407 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"94b557f518cb801436002c9606e954ab41ee249c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:19:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"de14cc9b8824d52d30ffa27a1abeeba506ba59bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:21:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8921e1343c04ef9b200431b0f49507b : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76cd3e8e568ffc0762442c51ae6e968f89c88293","revisions":{"76cd3e8e568ffc0762442c51ae6e968f89c88293":{"kind":"REWORK","_number":1,"created":"2019-09-05 03:17:22.000000000","uploader":{"_account_id":25837,"name":"yenai","email":"yenai_yewu@cmss.chinamobile.com","username":"yenai2008"},"ref":"refs/changes/86/680286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/86/680286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/86/680286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/86/680286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/86/680286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/86/680286/1"}}},"commit":{"parents":[{"commit":"5b280a85c548ecc247e4bd491841aef4d8575257","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5b280a85c548ecc247e4bd491841aef4d8575257"}]}],"author":{"name":"yenai","email":"yenai@cmss.chinamobile.com","date":"2019-09-05 03:11:16.000000000","tz":480},"committer":{"name":"yenai","email":"yenai@cmss.chinamobile.com","date":"2019-09-05 03:17:19.000000000","tz":480},"subject":"Add shared arg in metaclass Lock","message":"Add shared arg in metaclass Lock\n\nAdd the missing arg \u0027shared\u0027 in the abstract method \u0027acquire\u0027 of\nthe metaclass \u0027Lock\u0027.\n\nChange-Id: I8572b0df2615f8f6e8f666be988b49e63a549679\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/76cd3e8e568ffc0762442c51ae6e968f89c88293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/76cd3e8e568ffc0762442c51ae6e968f89c88293"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
