)]}'
{"id":"openstack%2Ftooz~683667","triplet_id":"openstack%2Ftooz~master~I42bd5786c06043c21010ef7260184b5d9d4de00f","project":"openstack/tooz","branch":"master","topic":"reno-train","hashtags":[],"change_id":"I42bd5786c06043c21010ef7260184b5d9d4de00f","subject":"Update master for stable/train","status":"MERGED","created":"2019-09-20 17:45:19.000000000","updated":"2019-10-10 13:19:37.000000000","submitted":"2019-10-10 13:17:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683667-1570713459475-70f2cce9","meta_rev_id":"c7942f9681d1b0c3d9b839a21f9f9017bf9d1cb1","_number":683667,"virtual_id_number":683667,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 13:17:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-09-23 12:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-10-10 12:02:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-10-10 12:02:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-23 12:03:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-10 12:02:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-10-10 13:17:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"965337dab383f04e0a6cd2306973a430e5019cdf","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-09-20 17:45:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6093dc3f74f5cbe65943301b72bc8dbc26bcd4e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 18:39:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2f77c15fb9d4786a4a96cd90f6f1db5 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63b447d815534f03b598efa59832b9cc : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46ffa39688654409a1d2dde7e98161b8 : SUCCESS in 4m 14s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/6609d45e6e3c4a7ea805fc94df4514d2 : SUCCESS in 4m 27s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/3ab4b2c341ca4218be5d69e06bac2eb8 : SUCCESS in 3m 47s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/6cbdee4d2db944629724bd8364835231 : SUCCESS in 4m 40s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/b74d5a0afd5d4de8991a056869bc9d76 : SUCCESS in 4m 34s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/0b9b9aca3b6a4359b6e64cf157a29641 : SUCCESS in 3m 56s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/34f1aa8792044ad3a9136fee69863ad0 : SUCCESS in 4m 54s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/5d99a65120c4421c95e12bb165b0f6e5 : SUCCESS in 5m 00s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/d3ec31f02652410abfade9b66fc2f56b : SUCCESS in 4m 50s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/7a646546f72d4d93ad89bbf34d825142 : SUCCESS in 4m 16s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/5193aa38b7034f0d84667d63876adcba : SUCCESS in 7m 01s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/20a61aac36b14345af74f390aa2a4489 : SUCCESS in 3m 56s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/975b530d228f497c92de7b884fabc75d : SUCCESS in 5m 41s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/d138708a68794839987d56c0b3b2bfb6 : SUCCESS in 5m 28s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/6cf93d7b9d0a4614940e6525c4da4307 : SUCCESS in 3m 51s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/a3fff46d282f4cd9a5cd71257005ab8e : SUCCESS in 4m 23s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/76c6460f812d4654b189ea52b4855ab3 : SUCCESS in 5m 21s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/d50b38d3d6a6418597ae7a9587a5d83f : SUCCESS in 5m 34s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/49cac104937b465b96827b3bec73a84a : SUCCESS in 5m 48s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/396fd4051ba449829f26e9691493ebd6 : SUCCESS in 5m 27s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/55fcb9e4cce843c1b2ec4f30816b8795 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8e3e1ef0f8f66204173c4c899be332e5ae1544b2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-23 12:03:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c7ec07956ffcdf1723619c69f942a33b5ca7abb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-10 12:02:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"689b143f854c31d09d77913419b1968a31adb122","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 12:02:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"268a150c4d99b1f23d8eec794a25670d28d7405c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 12:40:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57426f7f73148e291e40e26c1a86326 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820cf68e28304f96a6760a3c0c47b68b : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9683be74cfd84011b5ce1f75d74d5ce0 : SUCCESS in 3m 47s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/73a646676be5428aa9fc4900acbfcc4d : SUCCESS in 4m 29s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/e89ce23a16f84f058ea1272ee0fd0ba1 : SUCCESS in 4m 21s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/fcf4dfcc7f154f83b89578614e3d9294 : SUCCESS in 4m 04s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/88cb32e276d74c44896e6d8e3f7f6252 : TIMED_OUT in 31m 13s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/5a99b0fcd1724ece832d32632034d739 : SUCCESS in 4m 16s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/d99b2646c8d54e97ad9b7be8429f03e4 : SUCCESS in 4m 06s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/06a0f476bb304ba5bc587a6477b41e1d : SUCCESS in 5m 17s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/d6af8cda1eba4a0bb42af748e9666536 : SUCCESS in 4m 35s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/bf6af10d25b64656abbd0b94e9548cdb : SUCCESS in 3m 58s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/3e6af3ca965041ba9ba08d1d84310f65 : SUCCESS in 4m 55s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/98766fe5be674e9f9be73502f06c6a7e : SUCCESS in 6m 56s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/8cb65f97676b48d9aed44438c3691810 : SUCCESS in 5m 10s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/4865159cc7b44fdea1a06416c12c2105 : SUCCESS in 5m 55s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/b62772c336b34c5e9ff11f37e313f0c0 : SUCCESS in 5m 56s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/bf933b8d42234a348dfeb0d2ff43b9ec : SUCCESS in 5m 53s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/0ecf73c4e5574112a999eb9adbc97daf : SUCCESS in 4m 11s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/e234e580604a4634b9dc5ad9c02ab13d : SUCCESS in 3m 47s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/775211a1728e4bea9610a0309bc1b2e3 : SUCCESS in 4m 25s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/475b4495531e4598ba202759d4f274de : SUCCESS in 5m 50s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/70992da422364f6a8857cb5af2fd75a7 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"06a9c67c2d5fee2f76833b117680fa03e5a5510a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-10 12:46:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"976a9f32049185a6b08195589d870d1c75dade9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 12:57:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0ba12e2a8614dcea746d7cf4a0c7f51 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbdcb8a4f5334fa2bcbb73d60cec6993 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeb6f8b957a04278971184dff7141601 : SUCCESS in 6m 41s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/648cd1f72e16466f98b1aaf681f73695 : SUCCESS in 6m 14s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/966040faf7f147e1bcec3c275d0359ce : SUCCESS in 7m 25s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/c7c17092a3a842579a17e3da4acb46d6 : SUCCESS in 5m 19s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/c49b56415aaf4c5db3625d9beedebecb : SUCCESS in 5m 47s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/74527149da974550be181ee45e7938aa : SUCCESS in 4m 51s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/33f58cff414d4ff08217d726c05e54b4 : SUCCESS in 6m 05s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/fba3677b0909461b831844f4237d0738 : SUCCESS in 5m 50s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/069047dc7fd6403e87c3bec0051af9c5 : SUCCESS in 4m 06s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/31b6c21e7b8844d5a4a645289ce0dbd2 : SUCCESS in 4m 11s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/ef284e8c858d4450804cbc66042770af : SUCCESS in 6m 51s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/3c75ba028ef4437d8d57386c82dc7f0b : SUCCESS in 4m 34s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/c6c71fc59b464a5eb3ab38dadbf0751d : SUCCESS in 5m 35s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/e02efdd257494394a40b6090a61175ac : SUCCESS in 5m 51s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/1ced68dec78049fc91e90f1244845f3f : SUCCESS in 4m 46s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/1a5c2cc558da466d9872a9cad5232ca9 : SUCCESS in 6m 35s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/fdc3b15d56ea458486d19e2fde3403a9 : SUCCESS in 5m 34s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/ea582bc7cd1e4bd695d0506064794d23 : SUCCESS in 5m 48s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/f234ebe2d80c4cb6b8c5bfa560b53c72 : SUCCESS in 6m 30s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/6ad4bd8e2a4747488c6227f74e637393 : SUCCESS in 5m 13s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/a44bd4c3a38e4da0b80850902a9901e0 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"65854e4c3e547d4bcf4305960882916997b21059","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 12:58:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ddd63c1902a7736e7a0381b385acda1254d7b13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:17:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7730755ec0b35d4503ace061fc70f2088818e925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:17:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be6ee95b6bda4072b7c303ce0bf7c233 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c833558679405896138129775140d9 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a335a3fff725403092c3bc63a209d149 : SUCCESS in 4m 53s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/df8282ae20614ab0addfb8911f8245a7 : SUCCESS in 3m 41s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/7e7dd420fa4142f78a2ef93a4d355a23 : SUCCESS in 3m 43s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/db0d65d9289d417e9b690a086ff67816 : SUCCESS in 3m 39s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/9d2d9ac51d684288a42939652104d0b1 : SUCCESS in 3m 34s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/d897499f03de41f187ffd0496f3bf80d : SUCCESS in 4m 10s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/8f21c5554895418689778b27bebaa81d : SUCCESS in 3m 39s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/31cf4df970ae44bb8f0912ba8fd33f08 : SUCCESS in 3m 42s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/29110f274ffc47fab3c28e2ecd663be7 : SUCCESS in 6m 48s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/44046e7cb5664e2d8df3226adc6dbe5a : SUCCESS in 4m 18s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/80defeb3f7154b2fa60af5fe4fee964c : SUCCESS in 6m 14s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/769e10bbac7640d09dbcd88632d82b3b : SUCCESS in 13m 21s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/ed243d34ea124ba89eed4f86b420406f : SUCCESS in 5m 50s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/f935745e3a774753bf2426c03f407c9c : SUCCESS in 11m 46s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/75caa39e7b2d479cbc7057515b326d93 : SUCCESS in 4m 02s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/e3f79a94fee74c7d9211b2be37bf0507 : SUCCESS in 11m 40s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/a6983232270e4a63a756fcbf0238d689 : SUCCESS in 12m 07s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/b68be72b63a54a8f9dc6ba0445a20a4b : SUCCESS in 4m 14s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/1773bf17a7534234a55bfc174d869676 : SUCCESS in 5m 55s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/e7e2ea0be1ad4cdf9540b556dd2c7f1f : SUCCESS in 6m 57s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/a228c5e1834c4cdfa5e49c85b70fa4e8 : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1a25e4247c1daa2d3992cb5548cf60d301613c88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:19:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3344bb6303142198fb2d451659f86cf : SUCCESS in 1m 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45f13e33a004485084a3b2f4651737e8 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c430dbaa69e5a0329adfc13c6e2726dddb38405a","revisions":{"c430dbaa69e5a0329adfc13c6e2726dddb38405a":{"kind":"REWORK","_number":1,"created":"2019-09-20 17:45:19.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/67/683667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/67/683667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/67/683667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/67/683667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/67/683667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/67/683667/1"}}},"commit":{"parents":[{"commit":"5b280a85c548ecc247e4bd491841aef4d8575257","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5b280a85c548ecc247e4bd491841aef4d8575257"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-20 17:45:17.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-09-20 17:45:17.000000000","tz":0},"subject":"Update master for stable/train","message":"Update master for stable/train\n\nAdd file to the reno documentation build to show release notes for\nstable/train.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/train.\n\nChange-Id: I42bd5786c06043c21010ef7260184b5d9d4de00f\nSem-Ver: feature\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c430dbaa69e5a0329adfc13c6e2726dddb38405a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c430dbaa69e5a0329adfc13c6e2726dddb38405a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
